)]}'
{"id":"openstack%2Fswift~982769","triplet_id":"openstack%2Fswift~master~Id9e0fb0231097a3d5ffbdfcc6837b06bc14c75d5","project":"openstack/swift","branch":"master","topic":"p-versions-pagination-test","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2026-04-10 23:22:44.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-04-10 23:22:44.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-04-10 22:09:00.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"hashtags":[],"change_id":"Id9e0fb0231097a3d5ffbdfcc6837b06bc14c75d5","subject":"versioning: fix listing gap when version_marker\u003dnull","status":"MERGED","created":"2026-03-31 10:59:52.000000000","updated":"2026-04-13 09:37:03.000000000","submitted":"2026-04-10 23:22:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":46,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"982769-p-versions-pagination-test","meta_rev_id":"c843f6ca214b62042d070c0e5d4adbc333421156","_number":982769,"virtual_id_number":982769,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-10 23:22:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2026-04-10 22:09:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2026-04-10 22:09:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-31 11:38:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-31 12:12:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-06 15:57:34.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2026-04-10 07:43:49.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"10dface1114d8e8c7c43adbeca49bd5f19422c84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-31 10:59:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a1288d9ca05953be874f711af2f49de2cd4713d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 11:38:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/737121bef2484424a76fae0fbf653f99\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/db461677a8aa40bfa4d07b0aba5c5ace : SUCCESS in 33m 42s","accounts_in_message":[],"_revision_number":1},{"id":"663e0db474cc48ada04e4c3f31ed97f3cf7ed3af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 12:12:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/358a3d41bdc749bea642a4c5f1e122b7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c7fa16f5a564d16b0994f2e985c4efe : SUCCESS in 10m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1daae3653bf646cda9ea4f9ebe7d20a3 : SUCCESS in 20m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/04735ab6a14c432c94ea86b38e99d03e : SUCCESS in 22m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a651d4597d2d49139d026296b3b424fe : SUCCESS in 12m 47s (non-voting)\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3eb0635cd04a4db689dafb465154972f : SUCCESS in 15m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/20629d1ebc8d4f209ddf4bb31892f5cf : SUCCESS in 16m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c08be5d84e074733b2e6965fce3f8826 : SUCCESS in 10m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5ef5a5864355436a90e6c9f80c3a76f4 : SUCCESS in 56m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ee0c348ddd5c49c6b8ceb9e625d1ba6b : SUCCESS in 42m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e38c715e2b9e4772bcbbb1f9ee07b17a : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/63e93812b4a2492082b86b50c6b7d6d4 : FAILURE in 48m 55s","accounts_in_message":[],"_revision_number":1},{"id":"3654b3ee1782697f2d036780948b1db4dd6f1e80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-01 15:46:12.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"854c5f1826bc21863d205d9ccc87a97de0da62cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 16:33:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09a3f81e7b7748e6816a4d1a6e648de6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0c7fd366d94a40bbada3cf043067ae1d : SUCCESS in 29m 43s","accounts_in_message":[],"_revision_number":2},{"id":"58c84880b264a0c4fa1545f9ab9934c755fffac2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 16:59:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d92514e3c9654b60bdc73d2060f94e71\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bd560c057d048e8adf1cd0f175ac57d : SUCCESS in 9m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/60b8fc888a0044969367ad3a2b8d544f : SUCCESS in 36m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4432b1dbe1cb4a4e902c9d71d536e314 : SUCCESS in 1h 08m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/59c03f2313864693b77ad211b6d07a65 : SUCCESS in 58m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd643dc8c6a6412897ec5f1d849836d0 : SUCCESS in 1h 02m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3f28bb1522544ab9ae971683a0f524ae : SUCCESS in 11m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/375911b362ea4a6e8d4bbb5784a5a7f9 : SUCCESS in 10m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b4b27bd9082945a9a89c0c99aa306aaf : SUCCESS in 20m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a7ab5e6d263d435399a86f558b30a186 : SUCCESS in 8m 03s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f17eadbc510e4d2681d42f7f92ecaaf4 : SUCCESS in 25m 53s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8c5722b200504d92a4860ea94a0b218f : SUCCESS in 9m 46s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3864ac67a3724f90967e5c6f5f648ceb : SUCCESS in 25m 53s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/e35045cd3e8243a29b16365dbfa725ab : SUCCESS in 9m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6628509ebfc74e4bae8b08661b32afdf : SUCCESS in 14m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9780477e45204079850bd765fd4b00b9 : SUCCESS in 8m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/514d74ef15264a24ad9c1f61b48e7cbb : SUCCESS in 16m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d9c01ec3c97e43f183ca59b40b6bff40 : SUCCESS in 11m 06s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5e281404e1064de5a3718cc24369200d : SUCCESS in 8m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d71f18ac74f14257b6dc1b4d0a184204 : SUCCESS in 5m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c10fbc690b734797a31824ca73d0ab94 : SUCCESS in 48m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e8ef158ae433435383dde946a73afc0a : SUCCESS in 49m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9c62fb21d7d8451cae173ffc223863d1 : SUCCESS in 25m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b7c4626cf944598aff1171ac8c8ea6b : SUCCESS in 30m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8db3e4b6b6904ad58cf6ebaf35c70ee2 : SUCCESS in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/656ba85f81ea4192b0ae4e12c3439fe7 : SUCCESS in 9m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b0f3a2e637c4cf38f183e7850da15c6 : SUCCESS in 47m 55s","accounts_in_message":[],"_revision_number":2},{"id":"441e7d0712e25705b9e3b8ae18f499dff31d3b6e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-04-06 15:57:34.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bf02ab9214c9c0e94200f8bc048a0ea2c5619829","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-04-10 01:07:55.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"40b7321c5d91eb16f5a3aad21309767dfc2e81d8","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-04-10 07:43:49.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ae04d43573597da1cc7613ea06995538390a8de3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-10 10:25:42.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"1d806bec1b25b22620bcac76aadf817faa8677ce","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-10 10:55:47.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"324ff0fe3dd80c502fa06069ea3c4e460d261d30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 11:39:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0334c47ef65142c8ab9bf2d6fec81e74\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40ed833dc553476bb36039b815dff664 : SUCCESS in 9m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/3d6926cbf9e24bd4b1cd152f32557a92 : SUCCESS in 1h 06m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bd8ac2decaa24f78a1c30b6d7e014533 : SUCCESS in 30m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/82835a8a097d420792de293b14b166ee : SUCCESS in 1h 06m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bc566cf7f29743f49536db785a7f6f20 : SUCCESS in 33m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c03103d88f344e8a9b8d5ffe6d3d52f0 : SUCCESS in 11m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/83011622ce524e068f2ea0edc3c9e3d6 : SUCCESS in 20m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fab325d905b74854a11b45042010409a : SUCCESS in 21m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9c95bab9379141d0b3d66d992bef3487 : SUCCESS in 6m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3194923fe2f04deeae119b9798375bc5 : SUCCESS in 20m 15s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/97b202c985674452b2f1cd2185c94fce : SUCCESS in 16m 57s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/829565f36b6d422d90c191a43f7806bf : SUCCESS in 28m 38s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/72a9c93ab19b49a7bef0603db303ef8f : SUCCESS in 20m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/556addcc8b7542659690e76c40b3d993 : SUCCESS in 11m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ba0ced1f312c4c0da4537e137eda7720 : SUCCESS in 15m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ea8a9c2d7f0844bf997061ce3e459f4c : SUCCESS in 9m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/675d614f445c414f80147042a61d7b85 : SUCCESS in 6m 02s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/380b23a1eda542dc8f8fd6d2a4be2c50 : SUCCESS in 15m 03s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6bbc19d259c4431c877957cb7a656834 : SUCCESS in 8m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/615f90aa14074410a900509576311b6a : SUCCESS in 59m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e5a8c9a1192e47949c543a3d66043420 : SUCCESS in 27m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/feb1f6cfb89e4d9abe9a589e6ee1f46b : SUCCESS in 49m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9db7179572d481ba6ce1b9737ad7b4a : SUCCESS in 27m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/104beec62e4441dfbd49004d7ab9ea8e : SUCCESS in 3m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/caa52cca37fc4c978729064270fecece : SUCCESS in 18m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9bd4e7146ebd4d83baa8baa546ae13cb : SUCCESS in 43m 35s","accounts_in_message":[],"_revision_number":3},{"id":"e8e3475062832bd73fd0c41b5b664b3e3461a015","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 11:49:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70f75ca1740c4e9e9cd0b852ac689df2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cebe03e4a70b45be9cc16f1b65428cd6 : SUCCESS in 33m 16s","accounts_in_message":[],"_revision_number":3},{"id":"136616773cde3f5ab7093ee8a1a0de9c84543aa6","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-04-10 22:09:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(19 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e20482bab291b98aa326c5bf708fcb6b93ea9b28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 22:10:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"63e26bd739b7bac0c7d5491700736adef759edd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 23:22:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42fb0a4e7aaf47569146636bf1898100\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac572efdf05947b89e3f838d4a065692 : SUCCESS in 9m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/a02ed96a7e9040bca16e1cbe7e6ee1c0 : SUCCESS in 1h 02m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3d30ca5402cf438e9d64b869c104b20d : SUCCESS in 1h 06m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/857b02d509e04a74a500346dfe5b8329 : SUCCESS in 36m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/926a1eec9727451a8a2ced9eb3d5e505 : SUCCESS in 1h 03m 52s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b4c2aba3974d4b3395a93fc9726d80ff : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5f76fbf2707447a3b4386d473458ecf4 : SUCCESS in 25m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/87a84d58a8d6471483fc74fe39f7322a : SUCCESS in 24m 47s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4cdd10b8e02e4a23bc28c249a8a4296e : SUCCESS in 29m 06s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/fa73a9dd7aa44521a0763ced8bc5a4c2 : SUCCESS in 21m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2ef6c0f2e1f8490b85ca0871b0041b47 : SUCCESS in 8m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/53564be40459435eb04b0696efa1e419 : SUCCESS in 16m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/23220b63e8fd47758ef763796476fc2d : SUCCESS in 14m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bbe0448c60064a16805718e29c7dacb5 : SUCCESS in 6m 21s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/58e96a06a7f54e70b6b3d69bfce09ecd : SUCCESS in 8m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4881ff14b5bb416696d9efde041c1787 : SUCCESS in 1h 01m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6e33f0ca19f645ab86530e6a5902c524 : SUCCESS in 55m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/97f22b81168a40b08fbca65efb08baba : SUCCESS in 40m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1e2e0fc0c79430680ef05d23702e2a9 : SUCCESS in 28m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8f9669e77e3458aaf0d93c0e89ef706 : SUCCESS in 2m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4688c17397c247b891a62a140d48ae41 : SUCCESS in 18m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4500a9608d2b4e8794d2a923a82d1062 : SUCCESS in 33m 49s","accounts_in_message":[],"_revision_number":3},{"id":"584598f2c9fdaf8e8147b830aed899e40f5fe8ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 23:22:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"cc5c1e8cf52d701db9ca077afe3cc8cc8995fd82","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 23:23:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33574b8255d04a49b78459a5496a23e8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12354b8d07c846f5ac4ef0b71d49b8df : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/78e54e5994ff403aabf3a581be159e13 : SUCCESS in 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c843f6ca214b62042d070c0e5d4adbc333421156","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-13 09:37:03.000000000","message":"Patch Set 3:\n\n(12 comments)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"003bffee3abd4bf6cf0e6e389f034ccc8b142ee6","revisions":{"af0631ecdefd1a4a05f8bdbb7d37b8f3790edfd4":{"kind":"REWORK","_number":1,"created":"2026-03-31 10:59:52.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/69/982769/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/69/982769/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/982769/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/982769/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/982769/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/69/982769/1"}}},"commit":{"parents":[{"commit":"70d43b052e6a0806ee33c1d7714091138ae8d967","subject":"Merge \"S3Request: reinstate timestamp property for sigv2 requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70d43b052e6a0806ee33c1d7714091138ae8d967"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-31 09:58:14.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-31 10:59:31.000000000","tz":60},"subject":"tests: add versions pagination functional test","message":"tests: add versions pagination functional test\n\nChange-Id: Id9e0fb0231097a3d5ffbdfcc6837b06bc14c75d5\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af0631ecdefd1a4a05f8bdbb7d37b8f3790edfd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af0631ecdefd1a4a05f8bdbb7d37b8f3790edfd4"}]},"branch":"refs/heads/master"},"52b9de4a395a8dbd481213c64f19c29028bdce61":{"kind":"REWORK","_number":2,"created":"2026-04-01 15:46:12.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/69/982769/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/69/982769/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/982769/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/982769/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/982769/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/69/982769/2"}}},"commit":{"parents":[{"commit":"694d25bb1a87f9426ac7ad1f3817c51f5bf13a34","subject":"Improved offset support in Timestamp.__invert__","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/694d25bb1a87f9426ac7ad1f3817c51f5bf13a34"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-31 09:58:14.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-01 15:43:37.000000000","tz":60},"subject":"versioning: fix listing gap when version_marker\u003dnull","message":"versioning: fix listing gap when version_marker\u003dnull\n\nThe Related-Change ensured that when version_marker\u003dnull a versions listing\nwould not exclude *all* versions of *all* objects. However, when\nversion_marker\u003dnull, a versions listing may still erroneously exclude older\nversions of the *marker object*. This is because \u0027null\u0027 was wrongly\nconsidered to be the final version of an object that would be listed\nwhen in fact it is the first version to be listed since by definition\na null version, if it exists, is the latest version.\n\nThis patch ensures that if version_marker\u003dnull the converted marker\nused when listing versions is less than any version object name in the\nversions container that corresponds to the marker object.\n\nCloses-Bug: #2146885\nRelated-Change: I4b619bc20316eed61ca257a151fcca4f18c9af76\nChange-Id: Id9e0fb0231097a3d5ffbdfcc6837b06bc14c75d5\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52b9de4a395a8dbd481213c64f19c29028bdce61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52b9de4a395a8dbd481213c64f19c29028bdce61"}]},"branch":"refs/heads/master"},"003bffee3abd4bf6cf0e6e389f034ccc8b142ee6":{"kind":"REWORK","_number":3,"created":"2026-04-10 10:25:42.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/69/982769/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/69/982769/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/982769/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/982769/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/982769/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/69/982769/3"}}},"commit":{"parents":[{"commit":"694d25bb1a87f9426ac7ad1f3817c51f5bf13a34","subject":"Improved offset support in Timestamp.__invert__","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/694d25bb1a87f9426ac7ad1f3817c51f5bf13a34"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-31 09:58:14.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-10 10:24:19.000000000","tz":60},"subject":"versioning: fix listing gap when version_marker\u003dnull","message":"versioning: fix listing gap when version_marker\u003dnull\n\nThe Related-Change ensured that when version_marker\u003dnull a versions listing\nwould not exclude *all* versions of *all* objects. However, when\nversion_marker\u003dnull, a versions listing may still erroneously exclude older\nversions of the *marker object*. This is because \u0027null\u0027 was wrongly\nconsidered to be the final version of an object that would be listed\nwhen in fact it is the first version to be listed since by definition\na null version, if it exists, is the latest version.\n\nThis patch ensures that if version_marker\u003dnull the converted marker\nused when listing versions is less than any version object name in the\nversions container that corresponds to the marker object.\n\nCloses-Bug: #2146885\nRelated-Change: I4b619bc20316eed61ca257a151fcca4f18c9af76\nChange-Id: Id9e0fb0231097a3d5ffbdfcc6837b06bc14c75d5\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/003bffee3abd4bf6cf0e6e389f034ccc8b142ee6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/003bffee3abd4bf6cf0e6e389f034ccc8b142ee6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
