)]}'
{"id":"openstack%2Fswift~982911","triplet_id":"openstack%2Fswift~master~Ib60fbade97009d9400db6bdae109bb83f92338af","project":"openstack/swift","branch":"master","topic":"p-timestamp-collision","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-04-02 10:50:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib60fbade97009d9400db6bdae109bb83f92338af","subject":"test: more Timestamp inversion sanity checks","status":"MERGED","created":"2026-03-31 19:15:54.000000000","updated":"2026-04-02 10:51:08.000000000","submitted":"2026-04-02 10:50:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982911-p-timestamp-collision","meta_rev_id":"65db7490be5a1ab34de098ca3ed82c447a03fef8","_number":982911,"virtual_id_number":982911,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-02 10:50:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2026-04-02 09:01:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2026-04-02 09:01:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-31 20:11:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-31 20:21:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-02 09:01:18.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"6b72be81bfe51b537efc34c27e4a55988cfe5892","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-03-31 19:15:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b824439acfdf5bc9f1fe5d00c6bd946980fae88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 20:11:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9829ea40d9c4479dbab52fd60228bf6a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/53c158e2c91546808cef4bc31fe95a9d : SUCCESS in 38m 00s","accounts_in_message":[],"_revision_number":1},{"id":"a964dabc26b186c54817999c884544bc780ce142","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 20:21:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b150e4c636a4ec2a7ba9c183f0b6647\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c80962792d6749349f139380701657c6 : SUCCESS in 5m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7b30acf9c24e411e8c6da19deb5cce7d : SUCCESS in 19m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad3015ac59a349c3aeac96e7e0caaff1 : SUCCESS in 19m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6498b6c2d13f4e5dae1e88601c472ead : SUCCESS in 20m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2ce6ee50603749b6aa28851e60869f55 : SUCCESS in 6m 36s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f30740fe40984dd6b13bf3f272db28e1 : SUCCESS in 11m 22s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5a0f5bc3078d4d83a88cb8e326c4e551 : SUCCESS in 20m 48s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e844cda465c14b748e44e2efb88dd19c : SUCCESS in 28m 31s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/f3ff50138adb4208b602e4f41354f5a1 : SUCCESS in 19m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/00f2ad0dd8d748c1a348a93e242dd2f0 : SUCCESS in 14m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/70039ee8b84b492abc33359e2d43a08e : SUCCESS in 15m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6ac837a7fb4e42f08dbf373ed47517f6 : SUCCESS in 16m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ed524edd25e74773ae9624acdf7fed9d : SUCCESS in 53m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f072163fb1bd4753ae2bac3afb22ba78 : SUCCESS in 59m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cce613d3816e41a5a2669322037833e5 : SUCCESS in 30m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/057c1745ee75417989ed1b7fa5ab4189 : SUCCESS in 3m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f5b38821fc104fea9e096059daa61c74 : SUCCESS in 19m 03s","accounts_in_message":[],"_revision_number":1},{"id":"c14fbad9d2fecb74813654e71fb2a4c972c1fd8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-02 09:00:21.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"45df365b62e796ab1a8b2933811a7a1f22b4dd38","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-02 09:01:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"865f8505771a5631ccb2ee5106d2260d9b9cb3ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 09:34:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e568ed46d554eccbfeecfeb9120801c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/04b48df84be84f61854458ad8061bd27 : SUCCESS in 30m 36s","accounts_in_message":[],"_revision_number":2},{"id":"4a8c0b120405ad50b635ebff0945a76e78f9ce55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 09:54:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4894295af754c2b9b8ab138ed71fb57\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fc6c2ee506a40189779f95a21c34fca : SUCCESS in 5m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/38dd4b10b992457a9a88e80a4350f9bd : SUCCESS in 11m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d430c6d660244b07b198de1e7ee49649 : SUCCESS in 11m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2f2990bfc1d4498aa99ab549cfcd5253 : SUCCESS in 11m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a28d23e6bfb345cc9954e9772fd4a350 : SUCCESS in 4m 18s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/098da92e6ea646acaea8148840646a12 : SUCCESS in 18m 32s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4e1e182272ca45e58a25c91d6cf7d7f5 : SUCCESS in 14m 35s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e322201223fe4d42a843977a33a02cdd : SUCCESS in 12m 20s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/52596303326e482faa18e91e62658891 : SUCCESS in 9m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6188c89feafd453495e3492f9e9fae69 : SUCCESS in 15m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5dbaa4607ac94b3e86f9b1936ceb9f91 : SUCCESS in 14m 56s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a4e5f32fd79d424dba16fcc6876e039c : SUCCESS in 8m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/07e83b7acde345648a0f8e79d469db42 : SUCCESS in 48m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a42028368f164bc984372773f5458b75 : SUCCESS in 21m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86c0a6b1ed374af9becfcc3badc849a8 : SUCCESS in 27m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6e34c1bc66a47ecbb304aa05c5d7026 : SUCCESS in 2m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f36d535e844d44d7a8652e500377f278 : SUCCESS in 18m 29s","accounts_in_message":[],"_revision_number":2},{"id":"61233b83b35401283a26f0359a40456cafc9aa41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 09:54:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"71c509b7a50605fb4bd2dd76a06706cce2b5b77a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 10:50:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3574d30c4a04e13b58dbce6d4547641\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2a66e11ac8043bca6dae56cda797800 : SUCCESS in 8m 15s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/dc6248263aff49eca1a34c8cd4bae87b : SUCCESS in 6m 38s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/91c0ea2bf4ee4c1aa1a7b8ce02fdd6f4 : SUCCESS in 10m 01s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/76900811d8614bf399c4b48dc15f535e : SUCCESS in 8m 37s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/496e951260b54c89b84b0b9a3c9f52ff : SUCCESS in 29m 19s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/f978cba569e14c8f8675f961475262e1 : SUCCESS in 27m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f4c54e9190d14527818f68bee197834e : SUCCESS in 14m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2b524347ce1f45f59930114c9b3eabf2 : SUCCESS in 11m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/519227d1324441adadf2db6f4071d08f : SUCCESS in 13m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4c43fc3c60054eae956bf748f31a6dee : SUCCESS in 6m 38s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/aec1f7346e7e4fa2b67392ec7d761551 : SUCCESS in 7m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/397ffae353d14a5299e5510fe3d99b85 : SUCCESS in 50m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bbfc51d2fed54a2aafb03ead3f9f484b : SUCCESS in 36m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e99f323f098483f871b411d0e868392 : SUCCESS in 10m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcd56a42663f445b907ec67c79cb1151 : SUCCESS in 5m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0dd8c085827444b6b70fbcb05c998b8c : SUCCESS in 9m 48s","accounts_in_message":[],"_revision_number":2},{"id":"cb1582629f2e0e297022ea0a3b7abe2f8b94a396","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 10:50:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"65db7490be5a1ab34de098ca3ed82c447a03fef8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 10:51:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3125fd9e6bc7446c941c7d36e3d92ab9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52ccb4b2960f4064983f6539b1ecb37f : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f0da315fa92a4f80bf66632f1c28c2d1 : SUCCESS in 46s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7ae2871a36c7a066ee346ec83e36bd07237f3f7c","revisions":{"af89c3eaebb3296994a9424c66012ec66a33b238":{"kind":"REWORK","_number":1,"created":"2026-03-31 19:15:54.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/11/982911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/982911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/982911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/982911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/982911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/982911/1"}}},"commit":{"parents":[{"commit":"694d25bb1a87f9426ac7ad1f3817c51f5bf13a34","subject":"Improved offset support in Timestamp.__invert__","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/694d25bb1a87f9426ac7ad1f3817c51f5bf13a34"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-03-31 19:15:16.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-03-31 19:15:49.000000000","tz":-300},"subject":"test: more inversion sanity checks","message":"test: more inversion sanity checks\n\nChange-Id: Ib60fbade97009d9400db6bdae109bb83f92338af\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af89c3eaebb3296994a9424c66012ec66a33b238"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af89c3eaebb3296994a9424c66012ec66a33b238"}]},"branch":"refs/heads/master"},"7ae2871a36c7a066ee346ec83e36bd07237f3f7c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-04-02 09:00:21.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/11/982911/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/982911/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/982911/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/982911/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/982911/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/982911/2"}}},"commit":{"parents":[{"commit":"694d25bb1a87f9426ac7ad1f3817c51f5bf13a34","subject":"Improved offset support in Timestamp.__invert__","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/694d25bb1a87f9426ac7ad1f3817c51f5bf13a34"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-03-31 19:15:16.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-02 09:00:21.000000000","tz":0},"subject":"test: more Timestamp inversion sanity checks","message":"test: more Timestamp inversion sanity checks\n\nRelated-Change: If0b5f11f38405e69ae1b54d73b667abf0abfb89b\nChange-Id: Ib60fbade97009d9400db6bdae109bb83f92338af\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7ae2871a36c7a066ee346ec83e36bd07237f3f7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7ae2871a36c7a066ee346ec83e36bd07237f3f7c"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
