)]}'
{"id":"openstack%2Fswift~983143","triplet_id":"openstack%2Fswift~master~I873c349e965d886d9f5e02f0643a78885d86b5f4","project":"openstack/swift","branch":"master","topic":"just-raise-diskfileerror","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-04-02 00:27:27.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I873c349e965d886d9f5e02f0643a78885d86b5f4","subject":"not MORE logging; BETTER logging","status":"NEW","created":"2026-04-01 23:16:21.000000000","updated":"2026-04-02 17:30:22.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4a4cdc0fdc12fef2353f7abff28c3ad1b1983aa1","_number":983143,"virtual_id_number":983143,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-04-02 17:07:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-01 23:57:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-02 00:27:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c2969015c56e38c4913aa6d81cc9f46127274e31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-04-01 23:16:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a51d6179c5c746e8edab2d3268a6ce014f2c7a8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 23:57:46.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b81d1f0ab364ca0b86b965c54456558\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/43d5914e308f4a4c9508028a98062304 : FAILURE in 38m 12s","accounts_in_message":[],"_revision_number":1},{"id":"554af71d58c524bb125f1406fba1af2a75714458","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 00:27:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/124a63e932cb4e6597a20c762df672d0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c8820a907b049c4a61ef1eae6d2edf2 : SUCCESS in 10m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/b0f6a43ad5c24fbca29ce46a261f10fd : SUCCESS in 1h 03m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b7edc99782be4dbbb4279e58a9911c2b : SUCCESS in 1h 05m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5dfe7dc861d04615b81bb21bc5fbdc1c : SUCCESS in 52m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1eb528e9862641a99ce916e4b0cf0530 : SUCCESS in 33m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/84c3dd5bc5424aee90ded1034f92d68f : SUCCESS in 18m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/53dadd0d78a744ccae1639c335d3e4c6 : SUCCESS in 21m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/55af4b77e50c4d33aef75d6a941e1611 : SUCCESS in 21m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/af6168744b6a458aa23f10cdfb2d175f : FAILURE in 4m 18s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c4ae4fa365d0481aa634ce3037f9d57c : FAILURE in 25m 30s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/acc73af854e64990bfd4da2634982ac7 : FAILURE in 14m 11s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/63f305f62ace4f408d6c817b8f37b500 : FAILURE in 28m 11s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/c80aa02446744c5d8723c112f3b83bbc : FAILURE in 16m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8948ab33a2ff42409db96b0b9cb15ca9 : SUCCESS in 8m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/da3f8feb083c4574ae30930aee3473ff : SUCCESS in 13m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/66a040dafc984d53a491352f85516e59 : SUCCESS in 13m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9e3fe47b7358486cb1e9bf3ad59ee163 : SUCCESS in 6m 14s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8b8bd757f80a4d22b145d81fbea1ee1e : SUCCESS in 23m 57s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/31856eb8eb3849bd889dfbbbad4f9d5c : SUCCESS in 5m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ba82a25a342486bbafad7c2b42f0604 : FAILURE in 1h 00m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f5809ecc9a8146488be2910fdde1a305 : SUCCESS in 22m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eebf0fe309cb4778b6e7980f2284036c : SUCCESS in 22m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/098fabf289e04694a1f34a962d92313a : FAILURE in 28m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eda4f7bdb6f545ba8adb86421520a19d : SUCCESS in 2m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/93abdb94f5934043b71185e7edf076a4 : SUCCESS in 18m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1fb55894367742a08508fdc455bb71fc : SUCCESS in 31m 11s","accounts_in_message":[],"_revision_number":1},{"id":"04f4d0c6e17d2bb8295c8d91907e813cc1c15dfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-04-02 15:45:38.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"13954d7e78b0bc736b5af38caac79b9827404c58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 17:07:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/de33c7f4248f4ef980bf30881f5a5a56\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/466b617a3cfe4a309e86e59bd1864bb1 : SUCCESS in 10m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/18adbca23f2f4e5f90282f530566a84d : SUCCESS in 33m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/927858b3076842a0bc90e72aca333b25 : SUCCESS in 32m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e6558eed06a4466e80dce5d0b8d13f52 : SUCCESS in 34m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f01ee1561f7946a3a5ecefd32247bb32 : SUCCESS in 1h 15m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/83b5832d4d7a4515828d9b36909de70f : SUCCESS in 19m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8799bcae6132476d8d44028f6be06b3a : SUCCESS in 20m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a3ea1f17eff84cfda4a48d7573cbebb4 : SUCCESS in 21m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5a1d9f01b3474fe28ba5969fa8960123 : SUCCESS in 6m 34s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b4cf9c5c1e0c47468cdc247b13357be6 : SUCCESS in 10m 52s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/23ac23ddc6954cb7b6b0f43b41fc8e21 : SUCCESS in 22m 11s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9aff24e4ee414577bc3a18da8152f9a0 : SUCCESS in 24m 45s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/6be90699cc44484e95ab0ccfa63ce6b3 : SUCCESS in 20m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/12f61a13b45f4ed4805d6557f710630e : SUCCESS in 16m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/12f8811b0d114821b1080b22a33781ac : SUCCESS in 8m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3d164e6b1aee4c749de990b58497199d : SUCCESS in 9m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6752696b7f4a44e38500804323bd97da : SUCCESS in 9m 59s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6bba547253fc4eb98923f1c6f31ccd3a : SUCCESS in 19m 18s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e6f5623a01944dad9c24daf10d1d2664 : SUCCESS in 5m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/17a653487e534a6aa45e30aa098bf0be : FAILURE in 56m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/96e311e484a04faaaa1c0dd83ff8b360 : SUCCESS in 57m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a7b1f518a30f4a60b3122ea251049d1c : SUCCESS in 49m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83cf09eb7c6141dc96989368539a00d0 : SUCCESS in 28m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1b972bb5a314e449c28260a240a05e2 : SUCCESS in 2m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0cdbd62c44ba4fb8bcb036bf16f76324 : SUCCESS in 18m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/487c904f93774e889588747ec65f2229 : SUCCESS in 56m 16s","accounts_in_message":[],"_revision_number":2},{"id":"4a4cdc0fdc12fef2353f7abff28c3ad1b1983aa1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 17:30:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1081f55078f743ebb9aac5d069f36abd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/874a1116512d4b79ae309f692607def0 : SUCCESS in 39m 22s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dc8f87f1d1aaa9b33642c23cacbbc57a5b9c4ca0","revisions":{"a3588918dd354d3bcc6b6641de3aea13931e2b3b":{"kind":"REWORK","_number":1,"created":"2026-04-01 23:16:21.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/43/983143/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/983143/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/983143/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/983143/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/983143/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/983143/1"}}},"commit":{"parents":[{"commit":"4e2f36fff3d1fe8c19e730e8ae58617e820b1bab","subject":"raise DiskfileError on \"bad\" useful_bucket","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e2f36fff3d1fe8c19e730e8ae58617e820b1bab"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-04-01 23:10:10.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-04-01 23:16:15.000000000","tz":-300},"subject":"not MORE logging; BETTER logging","message":"not MORE logging; BETTER logging\n\nChange-Id: I873c349e965d886d9f5e02f0643a78885d86b5f4\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a3588918dd354d3bcc6b6641de3aea13931e2b3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a3588918dd354d3bcc6b6641de3aea13931e2b3b"}]},"branch":"refs/heads/master"},"dc8f87f1d1aaa9b33642c23cacbbc57a5b9c4ca0":{"kind":"REWORK","_number":2,"created":"2026-04-02 15:45:38.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/43/983143/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/983143/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/983143/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/983143/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/983143/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/983143/2"}}},"commit":{"parents":[{"commit":"4e2f36fff3d1fe8c19e730e8ae58617e820b1bab","subject":"raise DiskfileError on \"bad\" useful_bucket","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e2f36fff3d1fe8c19e730e8ae58617e820b1bab"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-04-01 23:10:10.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-04-02 15:45:24.000000000","tz":-300},"subject":"not MORE logging; BETTER logging","message":"not MORE logging; BETTER logging\n\nChange-Id: I873c349e965d886d9f5e02f0643a78885d86b5f4\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc8f87f1d1aaa9b33642c23cacbbc57a5b9c4ca0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc8f87f1d1aaa9b33642c23cacbbc57a5b9c4ca0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
