)]}'
{"id":"openstack%2Fswift~983242","triplet_id":"openstack%2Fswift~master~Ie4d6da97cf6b11bf0c5358b6bd2e43d437dd997e","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"38993":{"account":{"_account_id":38993,"name":"Lumir Sliva","display_name":"Lumir Sliva","email":"lumirsliva@gmail.com","username":"lumirsliva","status":"Senior Storage Engineer @ seznam.cz "},"last_update":"2026-04-07 11:30:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie4d6da97cf6b11bf0c5358b6bd2e43d437dd997e","subject":"doc: fix object versioning wording","status":"MERGED","created":"2026-04-02 18:38:20.000000000","updated":"2026-04-07 11:31:48.000000000","submitted":"2026-04-07 11:30:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983242","meta_rev_id":"33f7363ba61583f802abbb1f9954fc0e636e0e29","_number":983242,"virtual_id_number":983242,"owner":{"_account_id":38993,"name":"Lumir Sliva","display_name":"Lumir Sliva","email":"lumirsliva@gmail.com","username":"lumirsliva","status":"Senior Storage Engineer @ seznam.cz "},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-07 11:30:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2026-04-07 11:15:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2026-04-07 11:15:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-02 18:54:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-07 11:15:43.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"1d3acc8c781c1c6debc1c0c92c4db97d8f199a41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38993,"name":"Lumir Sliva","display_name":"Lumir Sliva","email":"lumirsliva@gmail.com","username":"lumirsliva","status":"Senior Storage Engineer @ seznam.cz "},"date":"2026-04-02 18:38:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a6d4a87967306b98f0524dcc6c691fdcef0e002","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 18:54:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/472b71bf594f4ec3859a4b2da02b1575\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57b597fbb4e345faa91e4f021851cc91 : SUCCESS in 10m 48s","accounts_in_message":[],"_revision_number":1},{"id":"89453d6c6ca7df73621157a4dbf94fbc77a32835","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-07 11:15:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"20bb9f3a2d62331bbdf1b4ccaa5114387ddbbc20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 11:15:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d9938de26bbb26d32214d17d27857cf10e719987","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 11:30:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc171a4146dc45d2a5d4a10ad97bebdc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/760e00ce0bf743e980586d8f0464f768 : SUCCESS in 5m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/619e57ef27fa4422a1513d79dda1c3ee : SUCCESS in 3m 37s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/52fae4c76c7d4918a336659dba25a298 : SUCCESS in 9m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d2add9c1a62f535bbd0816c391514a5961138647","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 11:30:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"33f7363ba61583f802abbb1f9954fc0e636e0e29","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 11:31:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68d4682f9e724532ab8a4740fcfeb5bf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7c0eb51a0e340a68d098cabf668157b : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/3f7abb7ed0574126836abcaa98610a65 : FAILURE in 33s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc3c6936256d24c2ef1452fed7c3676bf3063e60","revisions":{"fc3c6936256d24c2ef1452fed7c3676bf3063e60":{"kind":"REWORK","_number":1,"created":"2026-04-02 18:38:20.000000000","uploader":{"_account_id":38993,"name":"Lumir Sliva","display_name":"Lumir Sliva","email":"lumirsliva@gmail.com","username":"lumirsliva","status":"Senior Storage Engineer @ seznam.cz "},"ref":"refs/changes/42/983242/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/983242/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/983242/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/983242/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/983242/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/983242/1"}}},"commit":{"parents":[{"commit":"7ae2871a36c7a066ee346ec83e36bd07237f3f7c","subject":"test: more Timestamp inversion sanity checks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7ae2871a36c7a066ee346ec83e36bd07237f3f7c"}]}],"author":{"name":"Lumir Sliva","email":"lumirsliva@gmail.com","date":"2026-04-02 18:38:12.000000000","tz":120},"committer":{"name":"Lumir Sliva","email":"lumirsliva@gmail.com","date":"2026-04-02 18:38:12.000000000","tz":120},"subject":"doc: fix object versioning wording","message":"doc: fix object versioning wording\n\nThe documentation says ``length`` is the character length of\nthe \"object\", but it is actually the character length of the\n\"object name\". The Python docstring in the versioned_writes\nmiddleware already uses the correct wording.\n\nCloses-Bug: #2095381\nChange-Id: Ie4d6da97cf6b11bf0c5358b6bd2e43d437dd997e\nSigned-off-by: Lumir Sliva \u003clumirsliva@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc3c6936256d24c2ef1452fed7c3676bf3063e60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc3c6936256d24c2ef1452fed7c3676bf3063e60"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
