)]}'
{"id":"openstack%2Fswift~983269","triplet_id":"openstack%2Fswift~master~I5f16db524a39783e998addc9e707c43cda4cfd39","project":"openstack/swift","branch":"master","topic":"sq-ec-fix","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-04-03 18:13:12.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I5f16db524a39783e998addc9e707c43cda4cfd39","subject":"reconstructor: do not write corrupt data","status":"NEW","created":"2026-04-02 23:19:08.000000000","updated":"2026-04-03 20:16:37.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"16b7b8b7183993faf202b025893848035142d9fc","_number":983269,"virtual_id_number":983269,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-04-03 20:16:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-03 00:29:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3f32f54271e778accfbb6a90aaafb3eb99dd3dfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-04-02 23:19:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8dda1fe82538850262b769fb83d1ae52d724a97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 00:29:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ddc3188eae7540dda5c556124a82b68d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16f680a1fb064f93933a7672aeb808d7 : SUCCESS in 10m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/0073d68a8859424593b9aa28e38ac8ea : SUCCESS in 1h 05m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/11bd78e10fca4d16b02f40feacb3d9af : SUCCESS in 32m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/32dbe552337a43c7a725c198ee06b929 : SUCCESS in 39m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/002a945c680040d886a875f0d4bc0ccb : SUCCESS in 43m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/54b7927264594be8bafac37674dd4347 : SUCCESS in 19m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/63f53669e06a4f84bc8020cb78de0196 : SUCCESS in 20m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3ffc30001c864f21b552d65c2a02f0b6 : SUCCESS in 22m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d3be491be92b4c23b3078ce1b179f9e0 : SUCCESS in 6m 32s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/36379da941a44149a0b9b56e2c93f9ee : SUCCESS in 9m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9fe273341b524ff7b85cbf68b04280cc : SUCCESS in 20m 57s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/eef441fe1f974da28ea14e81fa4fdd96 : SUCCESS in 19m 32s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/058a2600d48e4651872b1e21c0f9b304 : SUCCESS in 10m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6a13346711524193b077e650cc856492 : SUCCESS in 15m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9ae4656ed5e6415ebb6b919fa609484c : SUCCESS in 8m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d262acc0437342fb89bc117a6472167b : SUCCESS in 16m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e15d5dc3d87241329a244099fba0696c : SUCCESS in 3m 42s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/278b69d1ceca47a4b3e2cd24a3550caa : SUCCESS in 13m 20s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6705ab98f03d4dad8f253adb84f40beb : SUCCESS in 8m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b2909bd48e14090a6ca1507720287b3 : FAILURE in 35m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b5d8ccc912c64bab8afcef4adb879fb0 : SUCCESS in 31m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e39046a069244557947446a541f02307 : SUCCESS in 52m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d234651389a74d6bbff7ebd4f0651f60 : SUCCESS in 21m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a08652048b5c410a9312dc26b96bd4ba : SUCCESS in 5m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/19bc767ce53949029fb7e27154e90db2 : SUCCESS in 24m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0440975f231f4ac4995ea699fc25daf2 : SUCCESS in 59m 46s","accounts_in_message":[],"_revision_number":1},{"id":"f3a8601be9f14163a6cefb355a611e8beec4a1a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 00:29:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c47bd3c0c97946f280be0dc3242c6bdd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0adf93bc5845465a8886960746d60751 : SUCCESS in 37m 43s","accounts_in_message":[],"_revision_number":1},{"id":"9799416011d670720abc62d10ee29da844c5aa6b","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-04-03 15:38:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4d38c853f2dde6707d0d598b66f667f28f62e6d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-04-03 17:06:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"78e7b413a7e15db0daaa7dc0617d6c2f5cd5b49f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 17:35:55.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e6c62066f8f421f98d2ddbc6bb6202f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0665545aa11f43f5b5a1eead989b87f5 : FAILURE in 28m 23s","accounts_in_message":[],"_revision_number":2},{"id":"84d6e2f3161831812bf4453885f68387a6c83087","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 18:13:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/66b1129493294a89874d5bc33327aa08\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f361f36c3734ee9acea537a1a6c4dd8 : SUCCESS in 8m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/1551169801c4471d897b700c44b2fe04 : SUCCESS in 1h 02m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/636bafec7a014b87a7fedfec1f1f4da3 : SUCCESS in 43m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c1b6d3a60efd43f091b26b376f9f0303 : SUCCESS in 59m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49534163f06d4c6eb815672a72550998 : SUCCESS in 36m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5e7a64afbe7b4455a2d685d182004d87 : SUCCESS in 11m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6f33e8bec0bf4a37a1626b0db2b5a363 : SUCCESS in 11m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ce5fc62c30f1421fa46078b1a2f95df9 : SUCCESS in 20m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/75444117a5aa4ae4959b9f89c742a217 : SUCCESS in 4m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b265488111f74dceb45625b9be22f338 : FAILURE in 10m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/bdf6ef2948e24f77a643f5b573ccb6e9 : FAILURE in 21m 22s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e35d19f632e44d559e3fa9773a0a2a58 : FAILURE in 19m 41s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/6980629631024b9d972e8a8e65f8ed9b : FAILURE in 20m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0a09105fe0784d52be41562b05d68791 : SUCCESS in 8m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d504c5358456491989872786c8bfcd11 : SUCCESS in 15m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/40eccdd26e704558a6eb8fec333954c8 : SUCCESS in 12m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/848fb9487fc54e2ba5b9d41c354191db : SUCCESS in 6m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a2ea455eeb9b4c778f6269c23e1275c7 : SUCCESS in 19m 26s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/300dd99509b94532ac9dc0300b1b1aea : SUCCESS in 7m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6d17c581c374fa7935754dcf43f5843 : SUCCESS in 48m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9d5f8c262fd24a92a6f827c1a0f4bc4c : SUCCESS in 35m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3a4977d6363b4bd68afa7511407c9ace : SUCCESS in 31m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15afa6d69f194351be7b65a430944c89 : FAILURE in 28m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b03abea95ca74a61a91137d8044512be : SUCCESS in 3m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ee28d25c01214fd69ab9aa3cfe744332 : SUCCESS in 16m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eddb2ae38fe04c61bec6f5f48a24e92c : SUCCESS in 59m 51s","accounts_in_message":[],"_revision_number":2},{"id":"7ef28162df439177e0438b60571e5d5333b66624","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-04-03 18:49:18.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"9f63fe35cbb7c48cb59f772bc9c9dc84ea99447c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 19:23:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b31ce5d3aba24ee58baa40ebb2c176eb\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/45436749b4d447ffb11ebce970795aef : SUCCESS in 32m 24s","accounts_in_message":[],"_revision_number":3},{"id":"16b7b8b7183993faf202b025893848035142d9fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 20:16:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10bbefa0b9104a0a9d2e8330bfd46aad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/134d82d517f34b56810e6f8c4f76b36e : SUCCESS in 5m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/5fc82ec53294466ea84e2697507393da : SUCCESS in 1h 08m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/48332b7858c643b6929c597d12506bfb : SUCCESS in 33m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9c39ac99bd114e5ebe2f1ef0f3a7ea09 : SUCCESS in 33m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6072bf9decc944fca3550a439e7ac72b : SUCCESS in 1h 08m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e5bf3a91b8e74af5badd735402259532 : SUCCESS in 16m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0657f7d951294715ad270aa89e53e188 : SUCCESS in 11m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9de4336e69d54867b1c5e6eda338617a : SUCCESS in 17m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9283181fe3694a4f8153bdff49ef7430 : SUCCESS in 6m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d89dc50794ef4c52a818b3d66caccc3b : SUCCESS in 18m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b5ad407a36c54abaa4b896946e7f9fb3 : SUCCESS in 20m 20s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8dbb2c2d07be4e41ada540e0333a4e78 : SUCCESS in 12m 36s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/b60248ba2fb1430392d55c85683c83bc : SUCCESS in 19m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0cab85f4482240ac9a04cd87c53745e1 : SUCCESS in 7m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7776065f6db546d098016013d954d7e7 : SUCCESS in 15m 52s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/159f83cdf2ac41d2bb4be3ef7fd2a45f : SUCCESS in 14m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/152b1253c1a94a36bd3d84013918ce7a : SUCCESS in 7m 18s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/70cc81fcd30c4b3a8a07f2972e5f6795 : SUCCESS in 12m 56s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0cbf18bad6a74576a9b7cbe03660b204 : SUCCESS in 4m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa08e479414a4d4cb0ba8d662d7105c9 : SUCCESS in 35m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f86fce98351c4e8c9fb6fb227b6ecc73 : SUCCESS in 53m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3623ff0fa94c4029ac15a2f42541aae0 : SUCCESS in 1h 17m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fcd6416bb39e4a2899d5fe026bd2b937 : SUCCESS in 27m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f99a6f358072447aab696a783eb903c1 : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7cae99bb88424bedaf3edcf196b0087a : SUCCESS in 9m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe6553d9716c4736a20183cab7d4bdd2 : SUCCESS in 58m 04s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"54df866740a7f268959b75183e814d2cbad0802b","revisions":{"07deb7b4c747222fc2b748505c3fc7cd6c6391ce":{"kind":"REWORK","_number":1,"created":"2026-04-02 23:19:08.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/69/983269/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/69/983269/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/983269/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/983269/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/983269/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/69/983269/1"}}},"commit":{"parents":[{"commit":"e9c92154b4bc75454810ba0c779f000f4d620532","subject":"test: remove dead code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9c92154b4bc75454810ba0c779f000f4d620532"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-04-02 19:00:59.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-04-02 23:18:57.000000000","tz":-300},"subject":"reconstructor: do not write corrupt data","message":"reconstructor: do not write corrupt data\n\nWhen rebuilding a missing EC fragment, the reconstructor fetches\nprimary fragments and gathers a consistent set of responses.\n\nHowever when it writes the resulting rebuilt fragment to it\u0027s peer it\nuses metadata (e.g. X-Timestamp, X-Object-Sysmeta-Ec-Etag, etc) from\nit\u0027s local datafile; it mustn\u0027t do that if the metadata doesn\u0027t match.\n\nCloses-Bug: #2143202\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I5f16db524a39783e998addc9e707c43cda4cfd39\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07deb7b4c747222fc2b748505c3fc7cd6c6391ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07deb7b4c747222fc2b748505c3fc7cd6c6391ce"}]},"branch":"refs/heads/master"},"8ef2518a9f5dac8734be98fa4ea5250b0c22cc97":{"kind":"REWORK","_number":2,"created":"2026-04-03 17:06:15.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/69/983269/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/69/983269/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/983269/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/983269/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/983269/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/69/983269/2"}}},"commit":{"parents":[{"commit":"e9c92154b4bc75454810ba0c779f000f4d620532","subject":"test: remove dead code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9c92154b4bc75454810ba0c779f000f4d620532"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-04-02 19:00:59.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-04-03 17:06:03.000000000","tz":-300},"subject":"reconstructor: do not write corrupt data","message":"reconstructor: do not write corrupt data\n\nWhen rebuilding a missing EC fragment, the reconstructor fetches\nprimary fragments and gathers a consistent set of responses.\n\nHowever when it writes the resulting rebuilt fragment to it\u0027s peer it\nuses metadata (e.g. X-Timestamp, X-Object-Sysmeta-Ec-Etag, etc) from\nit\u0027s local datafile; it mustn\u0027t do that if the metadata doesn\u0027t match.\n\nCloses-Bug: #2143202\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I5f16db524a39783e998addc9e707c43cda4cfd39\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ef2518a9f5dac8734be98fa4ea5250b0c22cc97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ef2518a9f5dac8734be98fa4ea5250b0c22cc97"}]},"branch":"refs/heads/master"},"54df866740a7f268959b75183e814d2cbad0802b":{"kind":"REWORK","_number":3,"created":"2026-04-03 18:49:18.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/69/983269/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/69/983269/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/983269/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/983269/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/983269/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/69/983269/3"}}},"commit":{"parents":[{"commit":"e9c92154b4bc75454810ba0c779f000f4d620532","subject":"test: remove dead code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9c92154b4bc75454810ba0c779f000f4d620532"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-04-02 19:00:59.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-04-03 18:48:39.000000000","tz":-300},"subject":"reconstructor: do not write corrupt data","message":"reconstructor: do not write corrupt data\n\nWhen rebuilding a missing EC fragment, the reconstructor fetches\nprimary fragments and gathers a consistent set of responses.\n\nHowever when it writes the resulting rebuilt fragment to it\u0027s peer it\nuses metadata (e.g. X-Timestamp, X-Object-Sysmeta-Ec-Etag, etc) from\nit\u0027s local datafile; it mustn\u0027t do that if the metadata doesn\u0027t match.\n\nCloses-Bug: #2143202\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I5f16db524a39783e998addc9e707c43cda4cfd39\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/54df866740a7f268959b75183e814d2cbad0802b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/54df866740a7f268959b75183e814d2cbad0802b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
