)]}'
{"id":"openstack%2Fswift~983325","triplet_id":"openstack%2Fswift~master~Ifc61fcca8a740d41d933573c0d1536319416418d","project":"openstack/swift","branch":"master","topic":"green-pile-close","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-04-03 19:51:57.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ifc61fcca8a740d41d933573c0d1536319416418d","subject":"sq? better concurrency","status":"NEW","created":"2026-04-03 18:39:52.000000000","updated":"2026-04-03 19:51:57.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"69a922b59bfce9a504afc21c9a1f064dcc505961","_number":983325,"virtual_id_number":983325,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-04-03 19:51:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-03 19:08:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-03 19:51:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3b08bacdce928034dce743fa1a8c93fb0319053e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-04-03 18:39:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8bccd2ee09ff607d917aec230dbb953aefb6b198","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 19:08:27.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18076a623c2d4d56a4d53f7e16733315\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8770d0df8b6142aea5e9884f504932f4 : FAILURE in 27m 07s","accounts_in_message":[],"_revision_number":1},{"id":"69a922b59bfce9a504afc21c9a1f064dcc505961","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 19:51:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f38f2753269941fcbebadaf260cc9ce7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19cd5c461968426488ee4ab85d294646 : SUCCESS in 7m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/790363773758471b98b14c0aeff41a6e : SUCCESS in 1h 04m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4f50168ec462419f89fd777eb4a5082d : SUCCESS in 1h 06m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c893dff71c1e4fc09f91d8f01c25c540 : SUCCESS in 33m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2bfaf77f91524a65bd7f2e43eedd2aa8 : SUCCESS in 36m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bc9098b5c14c44748dc0f14b94a81a8e : SUCCESS in 10m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/000b481032334829b0f6d0e2d0ac4ef8 : SUCCESS in 21m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d23873506a1c4c5fa83d36d76a1acb92 : SUCCESS in 21m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cdf4cb3150a54ed09cc28cceb5396821 : SUCCESS in 7m 00s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/36b5b812fa8a47b795b9995db5ce5fee : FAILURE in 16m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8930e6e54b05462a8805da06c7ed76bf : FAILURE in 8m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9c4a7f72c72248ef9b5a666ef8107ff5 : FAILURE in 28m 57s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/dbb3073e7200456db5a62a4821475287 : FAILURE in 14m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/71b015247e80427e9d560500dd6cf138 : SUCCESS in 16m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/17c081eb2a0e44b2a02474a29ecf12ee : SUCCESS in 16m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9520e4ad01e5402495ea6ea43f98076b : SUCCESS in 8m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fa31ec64d4264f5d963a233a7c494136 : SUCCESS in 5m 42s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/08e4d5220ac44f3b8b992cc2f89c4a99 : SUCCESS in 19m 40s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/78ee5ac7c3eb45f9b3e7a3f0fc2f73d1 : SUCCESS in 7m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/600efa10d88f4a419ff162c5a859be83 : SUCCESS in 1h 00m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eba2666c86bf4306b57f8e4dc2c2517b : SUCCESS in 48m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/04c1901e6a1b4729b3fe9deaa8c22811 : SUCCESS in 56m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0036d3a4827b457db7a52c0e8f107b43 : FAILURE in 31m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bd014feca1140d6882d93610450ff10 : SUCCESS in 3m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5c48cf452ca241a4a396779c9f07703a : SUCCESS in 21m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/814a829fad8e4b91b13ea8aae041ebbf : SUCCESS in 55m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d62272418e68d3ed44e40d0d9b9487d4015dd20e","revisions":{"d62272418e68d3ed44e40d0d9b9487d4015dd20e":{"kind":"REWORK","_number":1,"created":"2026-04-03 18:39:52.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/25/983325/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/983325/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/983325/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/983325/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/983325/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/983325/1"}}},"commit":{"parents":[{"commit":"0d354667b2d2c1a87b660c4b2271bdd4d3679f50","subject":"utils: add GreenAsyncPile.close","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0d354667b2d2c1a87b660c4b2271bdd4d3679f50"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-04-03 18:10:50.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-04-03 18:17:47.000000000","tz":-300},"subject":"sq? better concurrency","message":"sq? better concurrency\n\nChange-Id: Ifc61fcca8a740d41d933573c0d1536319416418d\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d62272418e68d3ed44e40d0d9b9487d4015dd20e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d62272418e68d3ed44e40d0d9b9487d4015dd20e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
