)]}'
{"id":"openstack%2Fswift~983561","triplet_id":"openstack%2Fswift~master~I5cbda7c1ba5171447635de102678221a30c847e5","project":"openstack/swift","branch":"master","topic":"p-timestamp-v2","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-04-08 16:31:15.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I5cbda7c1ba5171447635de102678221a30c847e5","subject":"tests_multi_get: fix timestamp formats","status":"MERGED","created":"2026-04-07 13:17:09.000000000","updated":"2026-04-08 19:07:08.000000000","submitted":"2026-04-08 19:04:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983561-p-timestamp-v2","meta_rev_id":"f89411c221069519517c7beaba3a8e9b4ae2f316","_number":983561,"virtual_id_number":983561,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-08 19:04:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2026-04-08 16:31:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2026-04-08 16:30:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2026-04-08 16:31:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-07 14:40:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-08 16:12:43.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"ac8469d8d41dd5383c82c6f746328aa0ebe6c94f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-07 13:17:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08f709edf4a519eebdb3647432f56a842817d464","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 14:40:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e438b489eae74fad8fb3eae159122312\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc9aaedbf7234819aa42e3b3123f9171 : SUCCESS in 11m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e053da8a83b74400850e11392dc1ae5c : SUCCESS in 19m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d422caf4b8c24d4f871614eaa2756abc : SUCCESS in 22m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6de8a9b376c4209b3856005d4cb8cca : SUCCESS in 23m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4abedf0ed1e445e79a14646e6a759a01 : SUCCESS in 7m 18s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c1adfcffb1b94f59be61e8979d088f8a : SUCCESS in 17m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e93fa626df064a7e9a210dd196fa4fc4 : SUCCESS in 22m 35s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8d86f5437d7b43ec9437d516b3c043b4 : SUCCESS in 32m 25s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/be7ae5a60a93447ba3c1e74555063af3 : SUCCESS in 22m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d406eeb73aff4e2084fd7218a9c47098 : SUCCESS in 17m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/28d1a7952d834d70abc03da1bbe4e839 : SUCCESS in 15m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/814b25d158024a0f8dc5a35cf8fc7a53 : SUCCESS in 15m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/df281e6ac4e54cf59509cfd62421558d : SUCCESS in 1h 02m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b835e38904f3400d91e7835acee7b79d : SUCCESS in 53m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7eb83ec0edd148568f53ff863b71fd1d : SUCCESS in 29m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0836c89383fa4e1bae4d1c8d1edb55e9 : SUCCESS in 3m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a9f19d1bd77346819760e038b58a196e : SUCCESS in 10m 37s","accounts_in_message":[],"_revision_number":1},{"id":"addb6a390133fec73a03582299d1a5a6c94fc8db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 14:46:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f57f818c7bc49ba993ff7438b43efd5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/375ece074f9a4ec69184e16a267d6887 : SUCCESS in 34m 21s","accounts_in_message":[],"_revision_number":1},{"id":"f6449d9a3919aae95b35ee1879ab29235e2bd110","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-04-08 16:12:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4dfad9beaee76a21e730bd26237df38d3901f306","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-08 16:30:39.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"40cd09e7841c2902ec83af8a0a1a8a7dc46885c3","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-08 16:31:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"64542ac8d73a74a10842d1cdd9e19c989ba19455","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 17:02:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d54d438918eb4c379b153904338e438e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b1c3175801124c1da86b5690d894c785 : SUCCESS in 29m 47s","accounts_in_message":[],"_revision_number":2},{"id":"97619dde99a5258427a19abf4e0bed2f4e44c196","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 17:28:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c862ef470e4a4a27a5e9dc93387a8bd8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e47b9d332a24471a8a5bbfed2c4d14a4 : SUCCESS in 10m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/429cb471217d406b9bf7025989b95c59 : SUCCESS in 22m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/14b675b614934a4c93a5656c3e0a26fc : SUCCESS in 12m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f497ebfc6de14bf4b922ce2293281d59 : SUCCESS in 21m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/278a16231a2a4c9e835e29d35fbf87d8 : SUCCESS in 4m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e1d1ab3c7b0d4a3a9561ff0f6c434eff : SUCCESS in 20m 23s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d75f8bfc67ad4f7696a54b72820864dd : SUCCESS in 23m 55s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/71bb24018f3c48e682e78fc0e9a8f1a1 : SUCCESS in 21m 17s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/5dfb2e43b7dc4700a6efb8d1808cafce : SUCCESS in 9m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/38c2a8469a3c47bca3cea00c518ff032 : SUCCESS in 15m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f42507b5d3564a17965bcf5066467bab : SUCCESS in 14m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5d7baf2250294dbb8b4f01031c99c501 : SUCCESS in 15m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f73e151a3f924fc08674a59ff27f9f40 : SUCCESS in 36m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8ac40a805f434249a6a9cb685e6f4820 : SUCCESS in 52m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ccf49576ade44ba3a2f5c3b9e6f065d8 : SUCCESS in 10m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ee8ca0dfa3f41bcb3903ad0c8b33b56 : SUCCESS in 3m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a4485a637e0347778f4a6980e9bbc9f3 : SUCCESS in 19m 35s","accounts_in_message":[],"_revision_number":2},{"id":"8914a29faa8f207485c2a15e4647ab823ff623ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 17:28:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c21bf9706f1aaa715a9f3d316b4b179409309ae9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 19:04:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a72eb6532dd4e84af230bbd3c7f4466\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9bb2cd1fe644a729299f94b5dc824dc : SUCCESS in 9m 51s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3bb28223016a458daba2a2da0297235b : SUCCESS in 5m 47s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6b4246d8fae64e41b7b3d3191b62e18b : SUCCESS in 25m 15s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/765865bc9ec34aaa8311aa4d4324a8a0 : SUCCESS in 16m 21s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1fede1e1704e4b8ea9502f90566f1e34 : SUCCESS in 28m 15s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/055dd34bff944469a6e16d652ccc9335 : SUCCESS in 9m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1fb9e9c05b744bb28a1be26cfc24b69f : SUCCESS in 12m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/826873c760f648cb9756abb2a8dbcc10 : SUCCESS in 13m 48s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/29a26d094e6f40bca7e33decfc0c0edb : SUCCESS in 10m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9db472d75b28468aad43753faeb70f61 : SUCCESS in 6m 47s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/45730015a8a842469bb61c8c52631934 : SUCCESS in 8m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9445443a166c47968bc05d1d4f72ecf0 : SUCCESS in 38m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9245d0a4197248019c879f1e17bc93ba : SUCCESS in 38m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dce2b777c854406fbfa1cc22d4ba5b42 : SUCCESS in 11m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8268729dfd044ee2945aa7997f83a80c : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/80435ae852204aeabd331424eb715180 : SUCCESS in 20m 39s","accounts_in_message":[],"_revision_number":2},{"id":"db7c76b9385f3eb8cc858ca547326363569a3978","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 19:04:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f89411c221069519517c7beaba3a8e9b4ae2f316","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 19:07:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b79a26285b44858acf3cf2cb0baf5d2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a45692ed76514364ad5e0b4d2beeb978 : SUCCESS in 41s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/a5947738a9204836b0e2c0beddd3e861 : SUCCESS in 41s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bafeaa6ea9e5d8c3b07de661ae784aa4cb0d5a4a","revisions":{"d2b8945bd99ee0af98fc37fe1d841449acbd957c":{"kind":"REWORK","_number":1,"created":"2026-04-07 13:17:09.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/61/983561/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/983561/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/983561/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/983561/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/983561/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/983561/1"}}},"commit":{"parents":[{"commit":"1e818bdd5f757de1411e685f5d2baf04018d0259","subject":"timestamps: add NormalTimestamp for ShardRanges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e818bdd5f757de1411e685f5d2baf04018d0259"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-07 13:05:58.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-07 13:16:47.000000000","tz":60},"subject":"tests_multi_get: fix timestamp formats","message":"tests_multi_get: fix timestamp formats\n\nUse internal timestamp format when building version-id\u0027s in tests, for\nconsistency with object_versioning middleware.\n\nChange-Id: I5cbda7c1ba5171447635de102678221a30c847e5\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2b8945bd99ee0af98fc37fe1d841449acbd957c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2b8945bd99ee0af98fc37fe1d841449acbd957c"}]},"branch":"refs/heads/master"},"bafeaa6ea9e5d8c3b07de661ae784aa4cb0d5a4a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-04-08 16:30:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/61/983561/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/983561/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/983561/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/983561/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/983561/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/983561/2"}}},"commit":{"parents":[{"commit":"047b37bc291397fed1df218867ed15da36b9164a","subject":"Merge \"doc: fix object versioning wording\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/047b37bc291397fed1df218867ed15da36b9164a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-07 13:05:58.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-08 16:29:58.000000000","tz":60},"subject":"tests_multi_get: fix timestamp formats","message":"tests_multi_get: fix timestamp formats\n\nUse internal timestamp format when building version-id\u0027s in tests, for\nconsistency with object_versioning middleware.\n\nChange-Id: I5cbda7c1ba5171447635de102678221a30c847e5\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bafeaa6ea9e5d8c3b07de661ae784aa4cb0d5a4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bafeaa6ea9e5d8c3b07de661ae784aa4cb0d5a4a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
