)]}'
{"id":"openstack%2Fswift~983825","triplet_id":"openstack%2Fswift~master~Ib7cac00bb1143bdb59517a325218afabcd296adc","project":"openstack/swift","branch":"master","topic":"add_cleaving_state","attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2026-09-08 09:49:26.000000000","reason":"\u003cGERRIT_ACCOUNT_38368\u003e replied on the change","reason_account":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"}}},"removed_from_attention_set":{"38368":{"account":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"last_update":"2026-09-08 09:49:26.000000000","reason":"\u003cGERRIT_ACCOUNT_38368\u003e replied on the change","reason_account":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"}}},"hashtags":[],"change_id":"Ib7cac00bb1143bdb59517a325218afabcd296adc","subject":"Track the db_id of db that found shard ranges","status":"NEW","created":"2026-04-09 08:04:25.000000000","updated":"2026-09-08 09:49:26.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":6,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"d29e11ee20b469857c749ca7dad421fbfd575da0","_number":983825,"virtual_id_number":983825,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-04 07:15:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"all":[{"value":-1,"date":"2026-09-08 09:49:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-09 09:20:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-03 00:02:27.000000000","updated_by":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"reviewer":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"state":"CC"},{"updated":"2026-09-08 09:49:26.000000000","updated_by":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"reviewer":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"state":"REVIEWER"}],"messages":[{"id":"bb2b8feea736abd22e7a94f708b7bde41af90ab9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-04-09 08:04:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a8ed6792b524fe47aebb9252b7a428939959516","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 09:20:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3420ed6182c34597bfd5304bcdf9bf2a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a579fceb88874402a4840b9ed68b5a93 : SUCCESS in 8m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/8b910c9990334ce29d8d8cc446aa41c9 : SUCCESS in 1h 10m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e0a598de770a4332abd424b292f7117a : SUCCESS in 1h 10m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/942df191c7af4028b1ab10b39e939c2b : SUCCESS in 34m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2232bc361e942eabaf609272e962cb4 : SUCCESS in 55m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4c8a1bbcd97a47a88c98125e242bc30c : SUCCESS in 20m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f635ae7defa4dd4887e2b7321dc244b : SUCCESS in 19m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0e294d9fd82241d3a5e9652fb5c28999 : SUCCESS in 12m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/38ef111965ff40e287850e96a8a760a4 : SUCCESS in 8m 57s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/af9ffdd4212a4fd8a7d35af34a3ffac7 : SUCCESS in 26m 58s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/19665ad00b0e4af095b3043224c3538d : SUCCESS in 10m 59s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2396254be73f400b983dfa70dde7ad65 : SUCCESS in 14m 02s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/c8cba29f4e64409ba089bc7176d8b418 : SUCCESS in 11m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f93c6f566e1048b38d5f8b6c9a347eec : SUCCESS in 16m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5e0835bd0ac64006a5d326a79e14efca : SUCCESS in 15m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/619f4373cbc54e8ebd6678c713754d64 : SUCCESS in 16m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9a62c459b8b942ca810c7c9884f96db6 : SUCCESS in 9m 25s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/86ad9ecfbfee488882baf757941ee5a5 : SUCCESS in 20m 51s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/11771266870b40cb95cf1ec89ae9ce33 : SUCCESS in 11m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b1e25ce418a432398a9adff83d5c525 : SUCCESS in 1h 00m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7c92d2663441445e996c13f61826f5a3 : SUCCESS in 53m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/70b2f0bca06d43b0a033f6a606626e1f : SUCCESS in 1h 02m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f2e075d0e8447889c92f11b1bd55f24 : SUCCESS in 31m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc8554c936094730a0d1aac8020f32c2 : SUCCESS in 5m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5668fa17180d452e910d203305fef71d : SUCCESS in 20m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79cd131be8c247838f3cb8a5ea08cf8d : SUCCESS in 1h 05m 19s","accounts_in_message":[],"_revision_number":1},{"id":"8395583939bd296be25961116be829ad929d42c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 09:29:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c2dcfef3c754cef92143f605d5bd561\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bb4accb33c624cde9279b330963ad0fb : SUCCESS in 31m 02s","accounts_in_message":[],"_revision_number":1},{"id":"8c5c022932254dbbf00001a51cde379131f9e780","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-06-16 04:11:00.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"3ffe7f76d64ccc94bd3f353c10cfa9468af84e40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 05:26:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e697d06c387b4031a6ed9bf26d1b1b39\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daccb1e7723a46378c5ba759b8c016a1 : SUCCESS in 8m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ca5873ddf4d4ff69808a18466af4c57 : SUCCESS in 9m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/5f3c58e8375d4873a8911cafcd8f500b : SUCCESS in 33m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c42c8f376c1b46298a5ed043b0e2a8eb : SUCCESS in 38m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/896c9f9188604a5baa453c9badaabae9 : SUCCESS in 55m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/194f184721db4d42b84d8e36933343ad : SUCCESS in 1h 09m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3e5bcfdf16f94eaab5e0bc143f15e216 : SUCCESS in 10m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/61f00ab8a70848feb964c4d1cfbb18aa : SUCCESS in 10m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b4639f83d5534ad3b6f3adef62876af1 : SUCCESS in 12m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bff2da8fa39e45f8955ed8dfd580d963 : SUCCESS in 10m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/eaab01f7b0ba41899187eca50fc41c4e : SUCCESS in 13m 13s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6322288614594a168c44c4540c8fb937 : SUCCESS in 9m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/16b67533336c473a9bc273fb4dc67a36 : SUCCESS in 30m 57s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/44457c5573c04c77a9ae1fb1dbee6611 : SUCCESS in 11m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ef722abcf36244e59646b8938a8a2f8e : SUCCESS in 9m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6105fc862e094b2583b611f2fc1342f2 : SUCCESS in 16m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f689aaeb7bff4184ae01bb11daf2f742 : SUCCESS in 9m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/414671f20fc343fdb7cadc18bd9ce716 : SUCCESS in 8m 06s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5015567c7bc4468ca0c0445547cd9eb5 : SUCCESS in 9m 46s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/339045339f7f4b1b8af11c5115ab1ae4 : SUCCESS in 9m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b5a2a2e0b58412b8b37baa80bf0ab08 : SUCCESS in 41m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b0533afb52c44a3bb67ef82bb9a59ee7 : SUCCESS in 57m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/444e83f887d344579c811b7102bfb228 : SUCCESS in 55m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a254d6d44c054ac9a7ce1497e2b20bdd : SUCCESS in 28m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/597045b48ad84825a82b80c86ea2ee4c : SUCCESS in 3m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/362442d1c5404523941a1bc2bdae24fa : SUCCESS in 16m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1dbdff0c848454b9a8b833e1273a0e5 : SUCCESS in 1h 04m 08s","accounts_in_message":[],"_revision_number":2},{"id":"03895838e9ae45067e77d90b75f0dd315ab75550","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 09:21:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a852953b1b849bcb1a599a139d964a1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/146d8f2bc8ae4551a2a5fae21a1548af : SUCCESS in 34m 57s","accounts_in_message":[],"_revision_number":2},{"id":"4916ce3f5e7c8f518b6a6a01753e0d1817e14a26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-09-02 07:31:47.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"44ca699115556960f9cbb049699f72e3eab90a9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 08:28:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afa2fb3558934f6aa3e9b541b07f5eae\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/de40f3cdb8254db7ad3567764b67a344 : SUCCESS in 54m 12s","accounts_in_message":[],"_revision_number":3},{"id":"f1c17850af7b65483149e622ea72700adbca280b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 08:45:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03bb52f02dae484880b0b950ce74763d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aa987a5abce4cfba9ef4d34f59e03b4 : SUCCESS in 10m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36d91bf5b050453c822d4f161594c4b1 : SUCCESS in 6m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/9ca9163b1fdf42879d18464a7fd26950 : SUCCESS in 31m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/722b2d7fcd43496b9970c9e719b913a4 : SUCCESS in 59m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1f934baef2ce4e68934ca3cc115f7fd3 : SUCCESS in 52m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b769064928b4326aee65f1ad5aa3aa2 : SUCCESS in 1h 04m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/935c1a8c12e949c3a6da51181684f07b : SUCCESS in 16m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1fdbb0f4d974414b8605ff175f6ca7d7 : SUCCESS in 20m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6154705ac1ee418986ade89bc108ff05 : SUCCESS in 21m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2b43b44954204fc6bcd6b7c9e6092be2 : SUCCESS in 6m 32s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/01a2d14c75c04442b6072689d4fa7463 : SUCCESS in 26m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5bf8b8b3e2d746f28210e5d39ef503d4 : SUCCESS in 18m 17s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/46060bc7289a44f29e5fd4b944ee3d63 : SUCCESS in 32m 30s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/dc90236ce74a4970bf8c84b70515ecd1 : SUCCESS in 20m 50s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/23f73fae3d874019a14873b8f53495be : SUCCESS in 11m 00s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a8f613c27ce24b3a9895c7940e4b56e0 : SUCCESS in 16m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dae706fbd6424a3cbd9147cd49d61a80 : SUCCESS in 17m 29s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/d01c29d334cd41358a9d1492ad29827e : SUCCESS in 17m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e7b4b06b453f4f4d9f5dad31314eb986 : SUCCESS in 13m 15s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dac05a1da8be4f99a0850be69e7970f0 : FAILURE in 27m 00s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/62019ca3db2e4c76a8ff9577655895f5 : SUCCESS in 5m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8bf46ded925247dfa2770cab6a76ac63 : SUCCESS in 54m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/94dae38c41a442c7afeae509a548fff5 : SUCCESS in 39m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9ddd3d6a201747c2943630ddc9b8d343 : SUCCESS in 32m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2bc00a1261244bd99a208dff1f23f562 : SUCCESS in 28m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1296db0d242414ab108a545f860336b : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1f7080c81ab1470fa4f2d3fbf65303cb : SUCCESS in 21m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3c74fedfa8747088a28a3e2df913e81 : SUCCESS in 58m 10s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"02ecffd55b1b70b27fc7c039eafc18353416a810","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2026-09-03 00:02:27.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8111994bf8292b48336a717c81c7a5430616e3d4","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-09-04 03:36:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9249941bf276b7676318b16767c6bb4bcb44527d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-09-04 05:52:44.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"dcc03f085d870bf835131828f3aa82ad9c2e0129","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 07:04:10.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/310e0e8aaca34d088272213364d64454\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b1ecffed2dd141ce9a045e886f03a97d : POST_FAILURE in 1h 09m 19s","accounts_in_message":[],"_revision_number":4},{"id":"aad93be7bb677ca2dada64a4b1d750766c8d593b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 07:15:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78a8c9bb95ab4820be464bdae39217f9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0c28def82df4386a6539a175459ea14 : SUCCESS in 10m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6661ac3f829043a19b506600a9e64069 : SUCCESS in 5m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/9124241dd5134aefa06b865c111c46e3 : SUCCESS in 1h 06m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d60b9168182041bd8e97b49cc8658067 : SUCCESS in 1h 15m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ed6fb2b72b784f0ab9c36223fbe704e8 : SUCCESS in 55m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b1742f424f74dc4a023edf26420196d : SUCCESS in 40m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f28b9e3cd8a43698f2f874a6e597b62 : SUCCESS in 20m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ea720c8b47ed498bb56e7c65b8276d77 : SUCCESS in 23m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c97d24013acd464fb265bff964592108 : SUCCESS in 13m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f5efbc5da54d4c8ca5216a5cea584b30 : SUCCESS in 7m 06s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6a7d8b4a186048e5ac4fdacf054e1373 : SUCCESS in 22m 01s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9bb01d31c6da407dbfff78fb9655c229 : SUCCESS in 18m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9f2750d2a6b94c148852eb0efa0602a5 : SUCCESS in 15m 08s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/f8ae5d4ad9084eabaf2d64462b0559be : SUCCESS in 21m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0541134f5ddc4577a24af141ebcea3b3 : SUCCESS in 16m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1726d9df97f248ec9834eec3a9475111 : SUCCESS in 18m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0791105bb735432389e909df34eae6ed : SUCCESS in 17m 01s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/f0556ab444a240fdb2ee690df3b65f4d : SUCCESS in 15m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1ad58b5a99b443c482e65bea79cac6b3 : SUCCESS in 7m 44s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c52210602a3248e99be1aa1669c3586e : FAILURE in 32m 34s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7b42141c0f1e465480f9c52f88bfbb1a : SUCCESS in 9m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a0594c1449a41b0a166a189de702d2c : SUCCESS in 1h 04m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/da3cb411678e4b4b963a5cbf011be797 : SUCCESS in 59m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/90de94b03141455ca9ca1f3f9d37a441 : SUCCESS in 51m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8d893a15242488eb22685b62ad15429 : SUCCESS in 33m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c055c157c944be49f6216c8b6aee6f4 : SUCCESS in 4m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e889270143f5492386f25da9541dc68f : SUCCESS in 21m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a6ec1bcebe646f2bdbe2aae0732e7c5 : SUCCESS in 56m 47s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4},{"id":"d29e11ee20b469857c749ca7dad421fbfd575da0","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2026-09-08 09:49:26.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"78dab0b3d26b4ccbb41db344be4e5394c901aa1d","revisions":{"15fac281009f7028145b3b91e09b7cb510723a3a":{"kind":"REWORK","_number":1,"created":"2026-04-09 08:04:25.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/25/983825/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/983825/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/983825/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/983825/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/983825/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/983825/1"}}},"commit":{"parents":[{"commit":"e0b5aa3a26c53cfbf9969bda9b8f9541f9252bb3","subject":"Merge \"tests_multi_get: fix timestamp formats\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e0b5aa3a26c53cfbf9969bda9b8f9541f9252bb3"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-04-09 04:05:02.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-04-09 04:16:20.000000000","tz":600},"subject":"Track the db_id of db that found shard ranges","message":"Track the db_id of db that found shard ranges\n\nWhen shards ranges are found searching a container db, at least at some\npoint in time this DB picked ranges that made sense. All replicas may\nnot agree, so by marking the one that did the scanning we know that if\nsharding takes along time and there has been a massive amount of deletes\nwe can check the authoritive DB and know things have definitely changed\nsince scanning happened.\n\nThis patch adds an extra piece of sharding sysmata called\nX-Container-Sysmeta-Shard-Authority which stores the db_id of the db\nthat scanned for ranges.\n\nBy the db_id, because the db_id is unique to a DB and so we can use it\nto track it to a particular DB. And now that db_id\u0027s also have the\ndevice appended to the ID we can also check the sys-meta on any replica\nand track the authoritive source.\n\nChange-Id: Ib7cac00bb1143bdb59517a325218afabcd296adc\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/15fac281009f7028145b3b91e09b7cb510723a3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/15fac281009f7028145b3b91e09b7cb510723a3a"}]},"branch":"refs/heads/master"},"f1ee772ea1e6af52a8fe9077a179792fd890f2ea":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-06-16 04:11:00.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/25/983825/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/983825/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/983825/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/983825/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/983825/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/983825/2"}}},"commit":{"parents":[{"commit":"47dbcf97230f6b1f8c7f3de9e7d28c940102a596","subject":"Fix: drop the page cache for the bytes a ranged GET read, not the file head","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/47dbcf97230f6b1f8c7f3de9e7d28c940102a596"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-04-09 04:05:02.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-06-16 04:05:49.000000000","tz":600},"subject":"Track the db_id of db that found shard ranges","message":"Track the db_id of db that found shard ranges\n\nWhen shards ranges are found searching a container db, at least at some\npoint in time this DB picked ranges that made sense. All replicas may\nnot agree, so by marking the one that did the scanning we know that if\nsharding takes along time and there has been a massive amount of deletes\nwe can check the authoritive DB and know things have definitely changed\nsince scanning happened.\n\nThis patch adds an extra piece of sharding sysmata called\nX-Container-Sysmeta-Shard-Authority which stores the db_id of the db\nthat scanned for ranges.\n\nBy the db_id, because the db_id is unique to a DB and so we can use it\nto track it to a particular DB. And now that db_id\u0027s also have the\ndevice appended to the ID we can also check the sys-meta on any replica\nand track the authoritive source.\n\nChange-Id: Ib7cac00bb1143bdb59517a325218afabcd296adc\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1ee772ea1e6af52a8fe9077a179792fd890f2ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1ee772ea1e6af52a8fe9077a179792fd890f2ea"}]},"branch":"refs/heads/master"},"eb63fac6fe545e615dc20e30750405de2d89d6a9":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2026-09-02 07:31:47.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/25/983825/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/983825/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/983825/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/983825/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/983825/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/983825/3"}}},"commit":{"parents":[{"commit":"4ff57666f016d4054a46dc2dcaa3ca2ad7ed4766","subject":"s3api: add normalize_checksum_algorithm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ff57666f016d4054a46dc2dcaa3ca2ad7ed4766"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-04-09 04:05:02.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-09-02 05:49:45.000000000","tz":600},"subject":"Track the db_id of db that found shard ranges","message":"Track the db_id of db that found shard ranges\n\nWhen shards ranges are found searching a container db, at least at some\npoint in time this DB picked ranges that made sense. All replicas may\nnot agree, so by marking the one that did the scanning we know that if\nsharding takes along time and there has been a massive amount of deletes\nwe can check the authoritive DB and know things have definitely changed\nsince scanning happened.\n\nThis patch adds an extra piece of sharding sysmata called\nX-Container-Sysmeta-Shard-Authority which stores the db_id of the db\nthat scanned for ranges.\n\nWhy the db_id? because the db_id is unique to a DB and so we can use it\nto track it to a particular DB. And now that db_id\u0027s also have the\ndevice appended to the ID we can also check the sys-meta on any replica\nand track the authoritive source.\n\nChange-Id: Ib7cac00bb1143bdb59517a325218afabcd296adc\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb63fac6fe545e615dc20e30750405de2d89d6a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb63fac6fe545e615dc20e30750405de2d89d6a9"}]},"branch":"refs/heads/master"},"78dab0b3d26b4ccbb41db344be4e5394c901aa1d":{"kind":"REWORK","_number":4,"created":"2026-09-04 05:52:44.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/25/983825/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/983825/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/983825/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/983825/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/983825/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/983825/4"}}},"commit":{"parents":[{"commit":"4ff57666f016d4054a46dc2dcaa3ca2ad7ed4766","subject":"s3api: add normalize_checksum_algorithm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ff57666f016d4054a46dc2dcaa3ca2ad7ed4766"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-04-09 04:05:02.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-09-04 04:32:36.000000000","tz":600},"subject":"Track the db_id of db that found shard ranges","message":"Track the db_id of db that found shard ranges\n\nWhen shards ranges are found searching a container db, at least at some\npoint in time this DB picked ranges that made sense. All replicas may\nnot agree, so by marking the one that did the scanning we know that if\nsharding takes along time and there has been a massive amount of deletes\nwe can check the authoritive DB and know things have definitely changed\nsince scanning happened.\n\nThis patch adds an extra piece of sharding sysmata called\nX-Container-Sysmeta-Shard-Authority which stores the db_id of the db\nthat scanned for ranges.\n\nWhy the db_id? because the db_id is unique to a DB and so we can use it\nto track it to a particular DB. And now that db_id\u0027s also have the\ndevice appended to the ID we can also check the sys-meta on any replica\nand track the authoritive source.\n\nChange-Id: Ib7cac00bb1143bdb59517a325218afabcd296adc\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78dab0b3d26b4ccbb41db344be4e5394c901aa1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78dab0b3d26b4ccbb41db344be4e5394c901aa1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
