)]}'
{"id":"openstack%2Fswift~983927","triplet_id":"openstack%2Fswift~stable%2F2026.1~If0b5f11f38405e69ae1b54d73b667abf0abfb89b","project":"openstack/swift","branch":"stable/2026.1","topic":"offset-inversion","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-05-27 19:01:54.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"If0b5f11f38405e69ae1b54d73b667abf0abfb89b","subject":"Improved offset support in Timestamp.__invert__","status":"MERGED","created":"2026-04-09 19:15:57.000000000","updated":"2026-05-27 20:41:16.000000000","submitted":"2026-05-27 20:41:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983927-offset-inversion","meta_rev_id":"fcdcd41461f78c7392fde74f51fb41920fff6772","_number":983927,"virtual_id_number":983927,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-27 20:41:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2026-04-13 10:02:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-27 19:01:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-27 19:01:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-09 20:13:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-09 20:32:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-13 10:02:50.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"5646af52a334a3402ebd51c740b52ba1b6378123","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-09 19:15:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b684e5ae137cdf080cb0827aa19c1fc139a74c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 20:13:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40cd1db77e884e54a45a5f0ba8a64efb\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/66f73ff80e61437db4eefa98f55039f1 : SUCCESS in 34m 31s","accounts_in_message":[],"_revision_number":1},{"id":"d5528f4735014df7ddf7b3c5924d425226cc08cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 20:32:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcfe870188244e90bbc5b3eee169798b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce6909e0a4444866b2429db43a41bb68 : SUCCESS in 11m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/ef9db5a64cb6489cb67a88a57dd23b4d : SUCCESS in 1h 11m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/27637a108f8a4a818da65fbedd50b786 : SUCCESS in 30m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1136520ddfb94ab9b93352258e898a7b : SUCCESS in 57m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f212c1969fbc41778d8553eb849d735c : SUCCESS in 1h 08m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bc7054d353df4f83b39a105f43d4bd1e : SUCCESS in 22m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/221bfeebba54458182e321a6db8f8c81 : SUCCESS in 21m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/778e74e846c24eecbe7b5ddd540afa2c : SUCCESS in 10m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f1e4e7f6566a49bcaf1ee0c08a0bd194 : SUCCESS in 9m 13s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1a834d60a12c44f88844fab5bf074ab3 : SUCCESS in 24m 43s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/13ff43424fcc40c1af18eb43323f103f : SUCCESS in 10m 06s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f145f7ff09a84177be58fc250eb5a499 : SUCCESS in 29m 09s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/d96252cda3d24769b24105c776c58be2 : SUCCESS in 10m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f4b760aa9a9b46c7ac75c0dbe800e2ac : SUCCESS in 15m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fac90b87c5454394bf95f4b2665e8ea8 : SUCCESS in 15m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4aaa3a623d9543c8b8660ab24633fb25 : SUCCESS in 14m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7070fdeb61084fff9ba6912d4e83bb49 : SUCCESS in 6m 59s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fe5fd2643df04235b5679489df7e0706 : SUCCESS in 26m 05s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b1cad09a6bc149a2b4ac289369ab017f : SUCCESS in 8m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/37f00ce0d71c42d6bd5ee7d1e1f3bc69 : SUCCESS in 1h 01m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0ff6b3846c8b40518f573b82b64d9771 : SUCCESS in 51m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f30ee481039d4eb998639eca40003e0d : SUCCESS in 54m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8506b9e6b73c47c6907b482615d0ee97 : SUCCESS in 31m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60cc02242d8943e6924cdf4a949d1fff : SUCCESS in 4m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/64dec48de1f844fe958e2f417ff2de41 : SUCCESS in 19m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0df35e84a5874a548606ea29a2244ad7 : SUCCESS in 1h 09m 50s","accounts_in_message":[],"_revision_number":1},{"id":"17ac99190515677cf590fa15768e552743d40ab2","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-13 10:02:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5863224517d51c128fb4f6368b712ef4b2599e63","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-13 10:03:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6bc5a5f37951046546890931b6355a97f2101573","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-05-27 19:01:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"646ebf0137f17df62ff8bc88f038ced8b4d8d31f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 19:03:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2960d8ee1ece53873525e3b6505def909145294e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 20:41:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90ef8e1cba6d4588b34f45ee00cbbf9e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c7014a55d494eb2b881a64c22335651 : SUCCESS in 11m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e231e45f779148b9892fb951cd25b84a : SUCCESS in 3m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/584e1a237732465a87533fff9740371b : SUCCESS in 1h 03m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3b0316139ecc4ca7959f0154ef0d280c : SUCCESS in 1h 07m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/234533125c8c4e7085cdc8af586168d3 : SUCCESS in 57m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b669027ae384c578c9d709182321d95 : SUCCESS in 1h 02m 42s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/cc95db2ea16148b9a11d4a442129c057 : SUCCESS in 6m 17s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/afbd76dcfea0498ca2ab9380ddcd352f : SUCCESS in 25m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a257e4fe657f43988dc33de03844a1ec : SUCCESS in 22m 15s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4bd13295b3fa4336b5de3ff8671f4f2b : SUCCESS in 29m 49s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/639082ed762b4fc19ab0795ca808fe73 : SUCCESS in 21m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/55686b50f06040a49d3fa1f36f9f8d2c : SUCCESS in 15m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0bd22a5b3bd74222842a3fedb935a795 : SUCCESS in 11m 53s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/42469b59f6a84432b148c5da46888e88 : SUCCESS in 16m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9f4c50b28ef3401d96ffe48ef350bbe1 : SUCCESS in 3m 33s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/04867c1f466e4055bc7602e26b203330 : SUCCESS in 8m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7bca20385c1a4fa4b168a537bbbea439 : SUCCESS in 57m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f6a459982a8d469b9efb9f6434ef5409 : SUCCESS in 51m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/93af845e36824a11aabd6cfc87522563 : SUCCESS in 22m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bcfbe98b6a20408082023576e234a329 : SUCCESS in 24m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21a390a4584d4fa3967305ed4e109ca0 : SUCCESS in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bba05d01bd2b421e9efaceef4f3ba92b : SUCCESS in 18m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69cf246b8b9044fb9e98e74dd73afacd : SUCCESS in 39m 53s","accounts_in_message":[],"_revision_number":1},{"id":"fcdcd41461f78c7392fde74f51fb41920fff6772","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 20:41:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8c41239fdbcd57164f92328356de07f9b0991c42","revisions":{"8c41239fdbcd57164f92328356de07f9b0991c42":{"kind":"REWORK","_number":1,"created":"2026-04-09 19:15:57.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/27/983927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/983927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/983927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/983927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/983927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/983927/1"}}},"commit":{"parents":[{"commit":"ac7a0abceafe9668ed796ee47c5cda022046cc7c","subject":"timestamps: support offsets with Timestamp.__invert__","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac7a0abceafe9668ed796ee47c5cda022046cc7c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-26 00:24:16.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-09 19:15:22.000000000","tz":-420},"subject":"Improved offset support in Timestamp.__invert__","message":"Improved offset support in Timestamp.__invert__\n\nThe Related-Change added support for inverting a Timestamp with\nnon-zero offset. A disadvantage of that implementation is that\ninverted Timestamps do not always sort in the reverse order of the\noriginal Timestamps. Specifically, a Timestamp with zero offset sorts\n*before* a Timestamp with the same float part but non-zero offset, but\nthe inverse of a Timestamp with zero offset also sorts *before* the\ninverse of a Timestamp with the same float part but non-zero offset.\n\nThis patch improves the inversion to ensure that the inverse of a\nTimestamp with zero offset sorts *after* the inverse of a Timestamp\nwith the same float part but non-zero offset. This is achieved by\nredefining the maximum Timestamp.internal value as\n\"9999999999.99999_0000000000000000\" rather than\n\"9999999999.99999_ffffffffffffffff\".\n\nThe only known use of Timestamp inversion is by object versioning\nmiddleware, where timestamps are not yet expected to ever have\nnon-zero offset. This patch, like the Related-Change, is in\nanticipation of *future* inverted timestamps having non-zero hex\nparts. There is therefore no known impact from modifyng the\nTimestamp.__invert__ implementation at this time.\n\nThis patch preserves the property that the inverse of a Timestamp with\nzero offset also has zero offset.\n\nRelated-Change: I4898c8250fe85dbbf66d0ab3d24b0ae9802820f5\nChange-Id: If0b5f11f38405e69ae1b54d73b667abf0abfb89b\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n(cherry picked from commit 694d25bb1a87f9426ac7ad1f3817c51f5bf13a34)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8c41239fdbcd57164f92328356de07f9b0991c42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8c41239fdbcd57164f92328356de07f9b0991c42"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
