)]}'
{"id":"openstack%2Fswift~984503","triplet_id":"openstack%2Fswift~master~Iaa8f1931aad51df9d023a692c60562a62a2b0019","project":"openstack/swift","branch":"master","topic":"add_cleaving_state","attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2026-04-14 08:28:53.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Iaa8f1931aad51df9d023a692c60562a62a2b0019","subject":"auto-sharding: enable auto-repair","status":"NEW","created":"2026-04-14 07:15:42.000000000","updated":"2026-09-04 07:01:42.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"607604565587f74eccee04b8c070b9f8a308d6c4","_number":984503,"virtual_id_number":984503,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-04 07:01:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-14 08:03:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-14 08:28:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d21233dd1da9b4635b488fdf6936219865e7ca1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-04-14 07:15:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01eb69f9397ffc1f4e845389e3259824a0876b18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 08:03:03.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3ccb71da2b940748c0ec13f560ea541\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/509c3833a8ae4cff837f0f724a6291c4 : FAILURE in 34m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7cdacd6c32434ceb0be0d78bf4e1c48d9cf0b130","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 08:28:53.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e81bdc34e49542c1b1d0ed39d24c1cbe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b00f095b15fe4bd1babbea5703329939 : SUCCESS in 10m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/1920c0ac97bf40c397d0239834cc5972 : SUCCESS in 1h 06m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9249d43d55ae4b069d5187a890874e91 : SUCCESS in 36m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/42d56228ceea4e08bc607752ef07e4b8 : SUCCESS in 58m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/77c8a81e83044ea6a241f6306cc879a4 : SUCCESS in 50m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ade81860bbe9472ab4703beb08bbba83 : SUCCESS in 11m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/afbf785a521841d39fc70974ca863c97 : SUCCESS in 20m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f8fa1e5ee2e1467db6c96bbe9822f4e4 : SUCCESS in 22m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2bde48c441ba4ebda68170b595438ec0 : SUCCESS in 7m 06s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b87204cbab10434fa730f7983c6f8b97 : FAILURE in 14m 01s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ddbc0df6361744dfa48b6a5c5475b4e4 : FAILURE in 22m 19s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a88c6094d9934393be45cb1de6864f21 : FAILURE in 30m 03s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/b9a39919c9ca46aeb0c96898797e925d : FAILURE in 10m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5c5d90cef7a9426cba1fa69dd1b5d801 : SUCCESS in 14m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8b610bde051143f6b34e859c0b91f1c7 : SUCCESS in 9m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/21601df81be147ceb11f3c5deac1529d : SUCCESS in 9m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4206a23492504bbf82fc59972449d1e6 : SUCCESS in 9m 07s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eae06deb4bc64e428e912358b09d4a7f : SUCCESS in 15m 50s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d8f834394f154ba9b4b98006c1ab5c44 : SUCCESS in 9m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d02cdb43bbe4014bbd679bbdd330739 : SUCCESS in 36m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5d448c1cb1204b88a1c763f545e1dabd : SUCCESS in 53m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/262b50c1fa7e478db7433967063b8d60 : SUCCESS in 37m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1916528a0aa64a3ca8bd1e9c21af0b41 : FAILURE in 26m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1fd3f89b34b4b6081d1b7adf82e8b5c : FAILURE in 3m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8c88283d9ee646b08b40a5bb1fb98d2d : SUCCESS in 9m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ff09442fbc8493fa2e9cc22a5a70b2f : SUCCESS in 29m 42s\n\nWarning:\n  Comments left for invalid file test/unit/container/test_sharder.py","accounts_in_message":[],"_revision_number":1},{"id":"9de5ac3aa5fdc3a5f0d569f3d280451c06e8ec1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-06-16 04:11:00.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"8caa07af099fc2c464efeef358c80dc7ebfcbf3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 05:36:29.000000000","message":"Patch Set 2: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab0820662f364557be7b752af222a723\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be027df1068b4adcbefb9714f4a69583 : SUCCESS in 10m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc49dca628984b6399d6e629d6e7fcfa : SUCCESS in 7m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/6de758f45bf643e280d695f9aa83b6b2 : SUCCESS in 1h 17m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ab6eb74b40d145f6a2c6c75765645c7c : SUCCESS in 31m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/62e00374e9024641ac9005cb31e041ac : SUCCESS in 48m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2706a1ce38684d02a762de2a4a8f1c20 : SUCCESS in 1h 09m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cd04f54d3f6e4aceb53ca8173287dfc8 : SUCCESS in 21m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a4a6efa599c48c08d12090eb13defca : SUCCESS in 20m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/169541f4d88042eaa63008871e306d7c : SUCCESS in 20m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8c828265ed52456989176da9ec74af97 : SUCCESS in 6m 23s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2a461ce539a64122b8a94955d35b8ee7 : FAILURE in 27m 21s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/756f4e9bb0344809a5b497d63a5c3246 : FAILURE in 9m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/176bd555a4944f8cb079d67c04ecaa7b : FAILURE in 12m 12s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/64f31cdf9f1e4a7c8bbd119843efa356 : FAILURE in 17m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/423be165b2ba4bee8e445433b592dbc1 : SUCCESS in 18m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/673b4b94ce7b40c2b4686a07d3153b58 : SUCCESS in 15m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/79a15914852947f99b239d2bf4b31713 : SUCCESS in 16m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/239c73fef243499ca892315c7c4eae74 : SUCCESS in 6m 34s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1355de8ab24c4a5a9029424df4ac6b93 : SUCCESS in 19m 06s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e7d6587a210b46f792d83ff4a3d1fa31 : SUCCESS in 8m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1074eb7a6514091982328db3ae34f4e : FAILURE in 1h 04m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f4ab003baa484a8f8999f928ef87b1a9 : SUCCESS in 50m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f8ceae8be6ee47798e87ecac18e4d053 : SUCCESS in 55m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e29a2abfc2b34151aab12e6ce30dd166 : FAILURE in 29m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/762b9d2bdc614f75b99c023463d51cc8 : FAILURE in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/264b6fbb3797420eba7909586eadccb0 : SUCCESS in 20m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dff8805e29594d3594ecf1ebcdab3980 : SUCCESS in 58m 41s\n\nWarning:\n  Comments left for invalid file test/unit/container/test_sharder.py","accounts_in_message":[],"_revision_number":2},{"id":"385669590434e1d4970f39c8a5b58f8956e6747d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 10:53:14.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8107342ff654f0bbdccfef85e88ecb4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d39b35c858dc4cf1ba600bae34bf8eef : FAILURE in 35m 52s","accounts_in_message":[],"_revision_number":2},{"id":"636f405494ca812cb8f68024e92c0034a611c6a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-09-02 07:31:47.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"c1196547dd54b978fb3f14cd19c9992d69ff6b62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 08:12:01.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b78c70aab97c49dcacff4596863a342b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0fe0803f25744d708507997929ad3dd5 : FAILURE in 38m 16s","accounts_in_message":[],"_revision_number":3},{"id":"ac059a45323acd7e98e341b383c3c375174c167d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 08:51:17.000000000","message":"Patch Set 3: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ade319be39294ba6ad79cc263d9333ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/943294ca4aa94e039666d2c969b18220 : SUCCESS in 9m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed269ae4af5c49afbe81f6cebde96e38 : SUCCESS in 10m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/114f24c32c9b4f49a2c29ce7ec48d02f : SUCCESS in 1h 13m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/75c04e52bdc443e093d7886d50482505 : SUCCESS in 1h 09m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0a0b35b38e4547b7ae6b14891aac64d9 : SUCCESS in 54m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d5df8d5c18c14cc5a9cf327fe8a778da : SUCCESS in 1h 02m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a175c1043c0a42d49e64811ce372606b : SUCCESS in 20m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e89c8b8830774dc9a407b6b8d08fe5ce : SUCCESS in 20m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9c9bffc9caa742eb9de69fc856a3e4f7 : SUCCESS in 19m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/384d193f8818484e81b167d7cce60296 : SUCCESS in 7m 27s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/288af34cc6fb4dea8c7fc9712f62ad16 : FAILURE in 27m 46s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/738060e0433449a1a01ee1a02cf8fb66 : FAILURE in 9m 57s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/04df1759bf824fa0b3c76dab40b7d91c : FAILURE in 36m 31s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/0c3dc5919c29440281f39ec781747e5c : FAILURE in 20m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4e3950617fb247d58bef7852a782bc0c : SUCCESS in 16m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6d1d4378a77b45a48b741b538a0efd9c : SUCCESS in 16m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e41fd3d56bdb4d44927148fd1f67168a : SUCCESS in 16m 43s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/ada191ddffda4ea5bd69ee63da2d3d7d : SUCCESS in 21m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a86a96e2692643fe9f0c2c9d2a5eba9d : SUCCESS in 14m 40s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0ac401c4ce9f4639b4f63fe51d72921a : FAILURE in 30m 28s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/02022bdf00ba4d49a6a5540093c62a40 : SUCCESS in 11m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1ed897a5eb5442edaf90d728e342e891 : FAILURE in 44m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0efd1cbed8e74cb5b39b07084bbfd8b4 : SUCCESS in 57m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/285b4378278e48d0a5965624c1ae90fd : SUCCESS in 23m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cfdafc852ecc48e4801fd64199c184b2 : FAILURE in 12m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df98869c94d84d168bd1da4fa644e9ef : FAILURE in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3520131b93a9453e9f0ff08d018962b3 : SUCCESS in 19m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61a145fb3a9e4979ab55ea4a32d861e8 : SUCCESS in 51m 55s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"ff9d970a43241f6d0d1b4fa1cdf27548a88704ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-09-04 05:52:44.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"9dacef004c966f89613910d529fc0b1a3f33d42e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 06:25:49.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/680d9ca4e8b24fa9ad49df69635a6b50\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e0fdf50a1e32453ba9e9e1155a884871 : FAILURE in 31m 03s","accounts_in_message":[],"_revision_number":4},{"id":"607604565587f74eccee04b8c070b9f8a308d6c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 07:01:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b97cb5a238c4ae38e8903bea9735242\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ddd56da522c4fb8a69299eeca7ea62c : SUCCESS in 11m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73861d61018b4b10ae1e17a7d2aaa05e : SUCCESS in 4m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/dbf71ec5d4b446f3bac288eb8fbce8bb : SUCCESS in 59m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9ba8f138794b4ef6859ccc3df9c59096 : SUCCESS in 40m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6b6671001405492a8a75fcd61d1510dc : SUCCESS in 36m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d7edf1fbc56448a9135d381540ad0c3 : SUCCESS in 42m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e6a9981a12514a039412ffaf2a1c8cb2 : SUCCESS in 17m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b3e01ea0e23444fd8b1bb12d6bc1a512 : SUCCESS in 20m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d45dbea160b34befab9ed33da624d7f3 : SUCCESS in 22m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/335f36880e6e46c5856199c5a82f2a3b : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/34b00f7136e747ceb12c98236a142d1d : FAILURE in 11m 11s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7838934db2c24cfb883a470cf01cf992 : FAILURE in 23m 16s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f1559e6d08ab49e5ab976546cb8305e8 : FAILURE in 36m 48s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/cce92d0f7de744a68c5779011b634559 : FAILURE in 22m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0f2eb7fc09d2455b854f1b10eff5ba5e : SUCCESS in 15m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/821d437cf88a4537bf2b61fb985dc78d : SUCCESS in 16m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/665e7f0d0df14882af8e0ca0c94cf3d8 : SUCCESS in 15m 47s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/fe56aa64aa7f4609ae6dab5abb316e39 : SUCCESS in 18m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8e762492e83147f7831460b4b63a0a30 : SUCCESS in 3m 49s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d38db702eaa741b180fabc42ff9a2d49 : FAILURE in 31m 38s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/97ca7087fc304d8ea7df941eec3ef8d0 : SUCCESS in 8m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/56bc785940a94e208dee0495d852aaf5 : SUCCESS in 59m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/68aefadd24f8461fba5490a686357f1d : SUCCESS in 50m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a6ff07eda5094dfbb9f4ed9a4c46446f : SUCCESS in 57m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a74ffe44501d4ae09289f3e93c246679 : FAILURE in 27m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca693398606a4baea01fbb28c908de76 : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bd61f57c8a7a40a8a31730f5dcd25e83 : SUCCESS in 27m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a7b296f97c840748d25cee7b413d400 : SUCCESS in 1h 02m 50s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"07644d98e347e914fc18b9f0c60330298b7f8dea","revisions":{"260a71e7c3459a15d22594b3d358e0ffe4d9c056":{"kind":"REWORK","_number":1,"created":"2026-04-14 07:15:42.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/03/984503/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/984503/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/984503/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/984503/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/984503/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/984503/1"}}},"commit":{"parents":[{"commit":"11f0e5da790857e45b7d9a9c514d7a1a95358d60","subject":"shrinking: Add a CLEAVING state","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11f0e5da790857e45b7d9a9c514d7a1a95358d60"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-04-14 05:23:33.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-04-14 05:23:33.000000000","tz":600},"subject":"auto-sharding: enable auto-repair","message":"auto-sharding: enable auto-repair\n\nWe have repair code that we use in s-m-s-r to repair gaps and overlaps.\nDuring a root container audit we also check for gaps and overlaps and if\nwe find we log saying a repair with s-m-s-r is needed.\n\nThis patch, when auto-sharding in enabled, will call the code that\nrepairs in these cases, as to fix them automatically.\n\nIn order to do so, some refactoring of s-m-s-r ranges code was needed to\nhave a shared space for both code paths to access the repair code.\n\nChange-Id: Iaa8f1931aad51df9d023a692c60562a62a2b0019\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/260a71e7c3459a15d22594b3d358e0ffe4d9c056"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/260a71e7c3459a15d22594b3d358e0ffe4d9c056"}]},"branch":"refs/heads/master"},"935268d1009e55dd1d355cad8b77a672795e7bb7":{"kind":"REWORK","_number":2,"created":"2026-06-16 04:11:00.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/03/984503/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/984503/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/984503/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/984503/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/984503/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/984503/2"}}},"commit":{"parents":[{"commit":"b2e3768af4d84aa0b6e4b8d82b306a0a6c689913","subject":"shrinking: Add a CLEAVING state","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b2e3768af4d84aa0b6e4b8d82b306a0a6c689913"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-04-14 05:23:33.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-06-16 04:09:06.000000000","tz":600},"subject":"auto-sharding: enable auto-repair","message":"auto-sharding: enable auto-repair\n\nWe have repair code that we use in s-m-s-r to repair gaps and overlaps.\nDuring a root container audit we also check for gaps and overlaps and if\nwe find we log saying a repair with s-m-s-r is needed.\n\nThis patch, when auto-sharding in enabled, will call the code that\nrepairs in these cases, as to fix them automatically.\n\nIn order to do so, some refactoring of s-m-s-r ranges code was needed to\nhave a shared space for both code paths to access the repair code.\n\nChange-Id: Iaa8f1931aad51df9d023a692c60562a62a2b0019\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/935268d1009e55dd1d355cad8b77a672795e7bb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/935268d1009e55dd1d355cad8b77a672795e7bb7"}]},"branch":"refs/heads/master"},"0dd5804406bfc03d41715a9c0c289500129c32e0":{"kind":"REWORK","_number":3,"created":"2026-09-02 07:31:47.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/03/984503/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/984503/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/984503/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/984503/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/984503/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/984503/3"}}},"commit":{"parents":[{"commit":"d7befbf3ebce8217128068c183ceaa8a6bd89eaf","subject":"shrinking: Add a CLEAVING state","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7befbf3ebce8217128068c183ceaa8a6bd89eaf"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-04-14 05:23:33.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-09-02 05:49:45.000000000","tz":600},"subject":"auto-sharding: enable auto-repair","message":"auto-sharding: enable auto-repair\n\nWe have repair code that we use in s-m-s-r to repair gaps and overlaps.\nDuring a root container audit we also check for gaps and overlaps and if\nwe find we log saying a repair with s-m-s-r is needed.\n\nThis patch, when auto-sharding in enabled, will call the code that\nrepairs in these cases, as to fix them automatically.\n\nIn order to do so, some refactoring of s-m-s-r ranges code was needed to\nhave a shared space for both code paths to access the repair code.\n\nChange-Id: Iaa8f1931aad51df9d023a692c60562a62a2b0019\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dd5804406bfc03d41715a9c0c289500129c32e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dd5804406bfc03d41715a9c0c289500129c32e0"}]},"branch":"refs/heads/master"},"07644d98e347e914fc18b9f0c60330298b7f8dea":{"kind":"REWORK","_number":4,"created":"2026-09-04 05:52:44.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/03/984503/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/984503/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/984503/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/984503/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/984503/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/984503/4"}}},"commit":{"parents":[{"commit":"cc99fbc714106400dae8f14bdca538e0c1b24541","subject":"shrinking: Add a CLEAVING state","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc99fbc714106400dae8f14bdca538e0c1b24541"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-04-14 05:23:33.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-09-04 04:35:20.000000000","tz":600},"subject":"auto-sharding: enable auto-repair","message":"auto-sharding: enable auto-repair\n\nWe have repair code that we use in s-m-s-r to repair gaps and overlaps.\nDuring a root container audit we also check for gaps and overlaps and if\nwe find we log saying a repair with s-m-s-r is needed.\n\nThis patch, when auto-sharding in enabled, will call the code that\nrepairs in these cases, as to fix them automatically.\n\nIn order to do so, some refactoring of s-m-s-r ranges code was needed to\nhave a shared space for both code paths to access the repair code.\n\nChange-Id: Iaa8f1931aad51df9d023a692c60562a62a2b0019\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07644d98e347e914fc18b9f0c60330298b7f8dea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07644d98e347e914fc18b9f0c60330298b7f8dea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
