)]}'
{"id":"openstack%2Fswift~985092","triplet_id":"openstack%2Fswift~master~Ie2d33b96687bc319b86f375e339d9a99f3892ed5","project":"openstack/swift","branch":"master","topic":"p-timestamp-float-offset","attention_set":{},"removed_from_attention_set":{"38496":{"account":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"last_update":"2026-05-12 15:41:35.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-05-12 15:41:35.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-05-12 12:29:52.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-05-12 15:41:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie2d33b96687bc319b86f375e339d9a99f3892ed5","subject":"timestamps: check offset is an integer during construction","status":"MERGED","created":"2026-04-17 11:15:09.000000000","updated":"2026-05-12 15:43:57.000000000","submitted":"2026-05-12 15:41:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":39,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"985092-p-timestamp-float-offset","meta_rev_id":"86e24b7c26f6d1f2f9709b5e3d18c963fa508ce7","_number":985092,"virtual_id_number":985092,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-12 15:41:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":2,"date":"2026-05-12 12:29:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":1,"date":"2026-05-12 12:29:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-17 12:27:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-27 14:04:23.000000000","updated_by":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"reviewer":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"state":"REVIEWER"},{"updated":"2026-04-28 22:21:40.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2026-05-11 21:05:52.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2026-05-12 12:29:52.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"3eedde9c693a40b7060037d5f4a053327d404093","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-17 11:15:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d98ec36ae8537108978662b450595463032e4361","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 12:27:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/553db137c5044a3ba842580b8148aef8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bb9467977c54a278a56f865872c3a51 : SUCCESS in 8m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/186b7d78517341bebe8fa2c8a9277271 : SUCCESS in 39m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fa809778048941c09c82a79552f9989b : SUCCESS in 34m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cd0ec6d371444bd9b1b9eb1c38d6ef27 : SUCCESS in 57m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4ce4b19213440249c81f9736824f5e3 : SUCCESS in 36m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2fe3a43f033b457482f5accce1cacb63 : SUCCESS in 12m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bf6c6648a8c747e09d3f649912353520 : SUCCESS in 18m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/386f71957c7249b4adb40e948df20bc0 : SUCCESS in 20m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2bf734c3b2004a80946b9a86f77664fd : SUCCESS in 6m 50s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/91773543f1b24435a0acf5120593f11c : SUCCESS in 24m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/54bf592b77924f3a9eb8ed5fad5b450c : SUCCESS in 12m 02s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e49b3734b6ab4dc0afb8ceb2f1e750a5 : SUCCESS in 27m 52s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/b4e490ce53454a43907eda0d3f4ba5c2 : SUCCESS in 9m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9b5fec360703475a8f827c43e952b773 : SUCCESS in 14m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/850b6cab17974450a410c60826f08444 : SUCCESS in 8m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b0e534755c724472b3977db1ec25deef : SUCCESS in 14m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4a59394e656141eaaf91e4f727eebc56 : SUCCESS in 5m 47s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ab1a7ade25524aaa925d6cbdb8cc467a : SUCCESS in 17m 30s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/da6e2be960724170acb2101feafdeeef : SUCCESS in 5m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b83d8884dae64b6fb95d15db667862bf : SUCCESS in 42m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/85daf5a3a82f43c3a264376d8f48256f : SUCCESS in 49m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1e11fb3f6eca4bc883eec2cf56a9823c : SUCCESS in 48m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3056d46d9ee4aa28800b6ff49bfaf1f : SUCCESS in 27m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25f6f6ba798940caa4b1769bae9581b7 : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/30892ffc6f1e49109514346b6e90c2ca : SUCCESS in 17m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f3e631f1cc64fd09481b293ad27d994 : SUCCESS in 1h 02m 29s","accounts_in_message":[],"_revision_number":1},{"id":"db2307c09cd66d77ecb4c5c5763399d93d7a4f9c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 12:36:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ef1a592f1934a27a92e6e3de45c473b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c183b7c095a74155ac2a295a2a5d62e1 : SUCCESS in 27m 11s","accounts_in_message":[],"_revision_number":1},{"id":"700eeb54f305c59ec5b5a9d535970d37c8544e1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-22 09:48:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"3a71ea0b9a94f9540389cf9d1d433ea37422e931","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 10:22:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b5f761ed955449094e3c4ddb41c7277\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1a6439c9bb614a86ab582fe81740443b : SUCCESS in 30m 07s","accounts_in_message":[],"_revision_number":2},{"id":"4c66b47fa99f3c6e6e51102e63ae6f5b1f10189f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 11:10:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6530af736544080bddbea4a72024b81\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/484531b078144f50853c84791495b983 : SUCCESS in 9m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/841e6b05076241aa9e558b70c8953462 : SUCCESS in 1h 03m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b7e9bc6290ce4dccad23585fd4d547ff : SUCCESS in 1h 16m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/574398d37c3a4cda963661fa0a42662c : SUCCESS in 59m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49c93c3c0bd94fd786b24b916d8d67b8 : SUCCESS in 1h 06m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9f3b3b1f2adb4808bd9d9cc600088a4c : SUCCESS in 19m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/afd8a6926d894672a6f13b67dc335496 : SUCCESS in 21m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0f1c4fbe3a0540a4bf5b7322ce31e9ae : SUCCESS in 21m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a1caf1549839424099bbf2cb56448f47 : SUCCESS in 6m 50s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a96ed21547914747a1c4b5f062a60c0e : SUCCESS in 17m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/988216a221b549e9b6562ebb1dcbd8cc : SUCCESS in 22m 35s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f0bc44d328f94b9a9b5dd5e6481182a0 : SUCCESS in 31m 45s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/3ec3365188df4048934799caea881b07 : SUCCESS in 21m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/60a74129b5414d599e1fcf46333a51ea : SUCCESS in 15m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7a9ded9cc45d4cda94b833ee80090d5f : SUCCESS in 13m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3d3a243eee1f472f835cab780aaf92a1 : SUCCESS in 16m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2aff8cd1d8c84574b959f19cd8035b6f : SUCCESS in 7m 01s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8595f0bbc86e4200859c0414b3354319 : SUCCESS in 20m 14s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/27cab4a0587e4a94a7cca44a1c67d9d2 : SUCCESS in 8m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6b429ba4a4694921aa720c241f19d297 : SUCCESS in 57m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ff6dbb0eb9de487da703518422651efd : SUCCESS in 51m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/17d7213ac26747949def42cca91ebe78 : SUCCESS in 25m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3a588246ee6435db7a4859f02b9267b : SUCCESS in 25m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e01c7f37d16b4c16ab349926c6a3ca5f : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3e44a924094f441dac3f03aaceb5c0ef : SUCCESS in 18m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83cd00b36aba49b9815d7cbafae24021 : SUCCESS in 48m 04s","accounts_in_message":[],"_revision_number":2},{"id":"49cac11b31f8bc97aa2cd019eca9e1517f669503","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-27 09:42:46.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"69db36aea073b8a132b6f895b1093469725bc38d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 10:17:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa002dd67ac64d549258dcdd065d63df\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4a8bdbd399074b82a7ce66dfb9d147c5 : SUCCESS in 32m 09s","accounts_in_message":[],"_revision_number":3},{"id":"a9115fadac8dcf1d58d39ba109079519ed4c9c81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 10:55:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/770e53f244b94cffa65d997cda4a434e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8a9305352f84ee5a435764410ed96f9 : SUCCESS in 10m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/44c5a34e9ba04dc2a868bbaaf8577071 : SUCCESS in 1h 00m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eb789092c2804cbdb2b87f7d36bf199b : SUCCESS in 1h 02m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ded713de9ebe4198955e062609f6add2 : SUCCESS in 58m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a083a74934074180b1081b416b64ad25 : SUCCESS in 1h 06m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3838445240dd4cefa2e182993b1eafcb : SUCCESS in 19m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b203904a1d214221826b6dec0d3417ea : SUCCESS in 20m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/79c85ad867424aa99c286ab7148be1ae : SUCCESS in 19m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b4d76367c1a24477818e055b72e90fd6 : SUCCESS in 8m 34s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9848f20e15a044468e741ef8f440d5af : SUCCESS in 26m 15s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1adb439d6f584195a5493584759b5a84 : SUCCESS in 21m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ab4283fa422f45719eb391e40d8cf8fb : SUCCESS in 30m 11s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/79a184d8e46648e5ae63f1f08bd3745a : SUCCESS in 21m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7143437d753a4c2db404ae02f8dbe051 : SUCCESS in 15m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6b828754e2564033a748536b17121c8e : SUCCESS in 16m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/80fddafbaa3048fc8158eb56ad5ba0fc : SUCCESS in 14m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9f8c24b1c1af4408913ec6f788f89216 : SUCCESS in 6m 44s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/85a330185ef749d7bb6c0b43dcf42347 : SUCCESS in 16m 38s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e22672cc47b24ec4ac5eef37ee6b1fbf : SUCCESS in 7m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6609a548bba346f4bcd2759e17763881 : SUCCESS in 55m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3cad90620e444ea683f9f09bd795572d : SUCCESS in 51m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d3d793f381864051ab2aade7a392d9a1 : SUCCESS in 51m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f124f83e5a40466e9fc9ca7e39a09442 : SUCCESS in 30m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5e56093fd9744c099e0cb0a11311281 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1061d8c37c0a45949d09d64bb4ddffa9 : SUCCESS in 19m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15b15771efea4b648cc8a523e7a6726b : SUCCESS in 45m 23s","accounts_in_message":[],"_revision_number":3},{"id":"90631d93fcdc368a2747c5ce873d1fd2d141018c","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-04-27 14:04:23.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"765f2e80aa69115bc4ca7df930a389ad5800aec7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-28 22:21:40.000000000","message":"Patch Set 3: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e71542abf888a77e3f876f716e2136c5220aba66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-05-06 14:00:19.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"545711f4d71d5cdeb18851d979057f0f9d0ac6ae","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-05-06 14:06:05.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b2ee92d6b885f7fc0befd339020ba1b47cbd6758","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 15:03:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ef83520214b41a686bc16835e0c2da3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d9d45fc3d3f44bb6afb34414b2aca935 : SUCCESS in 51m 27s","accounts_in_message":[],"_revision_number":4},{"id":"141e484a9436b729839574629679fe30bff3272d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 15:26:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63d6dedbd276472382e1ec0d0ca7c38b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4fdfdec9ca14bb5aa1017dd5109acfd : SUCCESS in 10m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/a68e8c4887c14e87a9047aa9c6c00338 : SUCCESS in 1h 00m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/219cf8ac1f934ffb81ed5652a04535ea : SUCCESS in 1h 10m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bfc459a600074009a2ae5a7b52f79f31 : SUCCESS in 30m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd5b8ce1b6464b83980bee9ad8f7b06e : SUCCESS in 34m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/916bf249730f483bbe00f5fdc9939fc0 : SUCCESS in 19m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b749e11d0e0b48b49ba8b890dd161b80 : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0a937e9f09be429f88d882971a47299b : SUCCESS in 18m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0d35154cc7cb4e7fb7da34c23237aa49 : SUCCESS in 6m 40s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ab8f5939b59c400896c44f9c11e336c8 : SUCCESS in 19m 18s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9f0197b2716a45c69823237c038b1d4b : SUCCESS in 14m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e75a9fa40a1d45b09af93c4cfdef2c47 : SUCCESS in 12m 14s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/1f4724d5b2594c69a2af762fcaa9852b : SUCCESS in 21m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fb1cba0f6b834ede940d4f68382aeff9 : SUCCESS in 9m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f62fdf63a1e949a5bdc093313e81fa9c : SUCCESS in 8m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dcd5cae387334ea4ae007e435c1fcb70 : SUCCESS in 14m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8d7fc9f782ba4599a04f23b65b5df741 : SUCCESS in 7m 11s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7b2f335db58346eda51a64a707692b37 : SUCCESS in 16m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/cd42996582d94f558bb58b7f709c141c : SUCCESS in 5m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/79e916916c48407195635642d24be6bd : SUCCESS in 1h 17m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/10630903d39b423f9338ad2668ac55bb : SUCCESS in 22m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/162f875e71a5466392c379147a0ec1b8 : SUCCESS in 52m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71f451244201421598a79ac0c68bc6b2 : SUCCESS in 29m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00b57c7bb6024b3a8b4117f7e46eac7f : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cef005f2e3d847978be8f01967892c81 : SUCCESS in 18m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/075fb164f2804d6bb634e1e2f01b45b5 : SUCCESS in 1h 04m 34s","accounts_in_message":[],"_revision_number":4},{"id":"5aa7e00a4b6f730c88dde7eb775d6c3af5f39309","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-05-07 11:05:00.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"25703e6fd3a7fc4017bc6724232db67be0f6261f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-05-11 21:05:52.000000000","message":"Patch Set 4:\n\n(10 comments)","accounts_in_message":[],"_revision_number":4},{"id":"bd7e7f71f367d6d95f39c00b969afec9ac31948b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-05-12 10:15:21.000000000","message":"Patch Set 4:\n\n(13 comments)","accounts_in_message":[],"_revision_number":4},{"id":"bae360b24608e684b655bfd0a6cbefff3452904f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-05-12 10:15:42.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"0d50bf8ee8e89e2e1811c85366d06c004d7dd814","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 11:05:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/888965e3a8094291b56e07cee3875cac\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2bd3cebbe0164ef7a3541a891fee3084 : SUCCESS in 33m 27s","accounts_in_message":[],"_revision_number":5},{"id":"9d205374caccca16443c5ef836ec4dbc9d5e8caa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 11:38:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65fedbe9925c439ba2476593ab38e042\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4926987ae242498aa4685b24499f368b : SUCCESS in 10m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/ec20f38912b14267a587d7160c60530e : SUCCESS in 1h 09m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4df7e3dfe8a9488e9d3a42f0757b44fe : SUCCESS in 1h 06m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/44149af333d04dc4a1ea414659277226 : SUCCESS in 52m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e95fa03337574613aa3c18c55175d3ee : SUCCESS in 1h 01m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fb9356e3ac2447dcab2205c714fe5a5b : SUCCESS in 19m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ef2a256a9a942bea9e672a69aa38429 : SUCCESS in 20m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/26a6ec9458f34148a92826b3ffbec52e : SUCCESS in 21m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f2bb19a30ddd42bea8587b0ecad01107 : SUCCESS in 6m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f54c8cd0053d4e369cbd7d0fb14c84dd : SUCCESS in 26m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/196e27eec89b4533a1a8614ed29724da : SUCCESS in 24m 26s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1402a6a614af46419d221c1288b28bcc : SUCCESS in 29m 38s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/399d884a6ef7475da9981d83ed9e8178 : SUCCESS in 23m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b57ca2a316684396869e6844b6e0c527 : SUCCESS in 14m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8cf6fc1651664129925f475b400b1472 : SUCCESS in 15m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/bdb8453502b84ce9b0db61c39e791225 : SUCCESS in 15m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6993f57b97aa4b58b00b3a48e13bda10 : SUCCESS in 7m 14s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/09fc845edb334baabbd1916a9364b346 : SUCCESS in 18m 18s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c42f55ca26df4057bd15e2d85d5643c4 : SUCCESS in 9m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/df53edf3ab1149298b9a8537b85b9cdc : SUCCESS in 57m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9aae6f4f87fd45c5985ff992459c50c2 : SUCCESS in 54m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2c3c10816553470c970bcc72acee679a : SUCCESS in 50m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2ab6835258c4027b910468ddbf244e4 : SUCCESS in 30m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45487da56775435ea49e187983c703cb : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bf36481b6bd34a7691742e673c0cc0df : SUCCESS in 19m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b7a9506c17924d60a650d00b5778c952 : SUCCESS in 1h 03m 05s","accounts_in_message":[],"_revision_number":5},{"id":"9d2a931210e0862b0b2232f2db6d01ee53698aab","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-05-12 12:29:52.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"c4f80ed3c634908e22f04881bb60f38a6963e28b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 12:30:23.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"305434562c001faf6503607d89d6a0ee38c903dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 15:41:35.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19e05e9ebca5426383f7145c8c91f2b7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa93de52ced94ed49a8744f933260fa2 : SUCCESS in 10m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/ce1da82466904a39be635e3a40a78dfd : SUCCESS in 1h 04m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0c982f1d2265420d801966312654bf32 : SUCCESS in 34m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ba07701d8c2549c190dc3fe81d4edeba : SUCCESS in 1h 02m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2d9fd5d115e44eab9b4357c35d8ae5f4 : SUCCESS in 40m 40s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/7c3ac2ee9148411cae0e822150521321 : SUCCESS in 6m 27s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8cfef1c9a1f64ca49acafd355a134132 : SUCCESS in 11m 15s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f42251576dc14655b8ef6edb19ced795 : SUCCESS in 21m 29s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f8166dd53e65404fb8919c0810ab1de2 : SUCCESS in 37m 14s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/ae33fd3e1b5944609691621659e7383b : SUCCESS in 21m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/50463271997c4675bfd5d7f69436aac0 : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/01a029da2a7e44af926aed11f65ff571 : SUCCESS in 9m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f11b2d63a23445b6a5a7627ec92b7068 : SUCCESS in 15m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4ea439bb715c48249ee8f4f77f326904 : SUCCESS in 8m 40s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/07f90d07e035425ead6f281d7a23f263 : SUCCESS in 10m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee7e9b0eac6e4df2a0bba5763fa9f36c : SUCCESS in 44m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2ab90483c8444573beb8b49c3a970516 : SUCCESS in 56m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e8a13968ab384879b090873a4b78321b : SUCCESS in 53m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45574c03e5f647c29b51218f6775b640 : SUCCESS in 22m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32b274babaf6459b877bc38398d43b0b : SUCCESS in 4m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b750b4ce0a054a97b59134f3e0dd91d8 : SUCCESS in 15m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a93e23184aa74fdd8c52c60bea0b67d9 : SUCCESS in 31m 28s","accounts_in_message":[],"_revision_number":5},{"id":"51362495c2db6cd407e8e2a9854166c72e7ef0dd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 15:41:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"86e24b7c26f6d1f2f9709b5e3d18c963fa508ce7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 15:43:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d48e23fd5cfe445fabd4916340ba4b03\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f81d5960fe44b6187ab2d3234bc1b37 : SUCCESS in 58s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/0732a0f3f258425e9b59c458816019d0 : SUCCESS in 53s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"33967853276f0987a58e5b3aca450173adadc064","revisions":{"3c809dc60dff122eb99d51b9e3d37ff770d33810":{"kind":"REWORK","_number":1,"created":"2026-04-17 11:15:09.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/92/985092/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/985092/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/985092/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/985092/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/985092/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/985092/1"}}},"commit":{"parents":[{"commit":"498350c898ca99d3b4635999145431798d0ae541","subject":"relinker: fix typo in help","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/498350c898ca99d3b4635999145431798d0ae541"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-17 11:10:34.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-17 11:10:34.000000000","tz":60},"subject":"timestamps: check offset is an integer during construction","message":"timestamps: check offset is an integer during construction\n\nPreviously, a Timestamp could be constructed with a floating point\noffset, but Timestamp.internal would subsequently raise a\nTypeError. This patch modifies Timestamp to check that offset is an\ninteger value during construction and otherwise raise a ValueError.\n\nChange-Id: Ie2d33b96687bc319b86f375e339d9a99f3892ed5\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c809dc60dff122eb99d51b9e3d37ff770d33810"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c809dc60dff122eb99d51b9e3d37ff770d33810"}]},"branch":"refs/heads/master"},"d73c7263c8f7e0bf3d4df1f64f002e3bff8e9ffc":{"kind":"REWORK","_number":2,"created":"2026-04-22 09:48:21.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/92/985092/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/985092/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/985092/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/985092/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/985092/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/985092/2"}}},"commit":{"parents":[{"commit":"a4cfcd869642fbc7e6fda68bffabfeb5cd1098b1","subject":"Merge \"timestamps: add NormalTimestamp for ShardRanges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a4cfcd869642fbc7e6fda68bffabfeb5cd1098b1"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-17 11:10:34.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-22 09:47:24.000000000","tz":60},"subject":"timestamps: check offset is an integer during construction","message":"timestamps: check offset is an integer during construction\n\nPreviously, a Timestamp could be constructed with a floating point\noffset, but Timestamp.internal would subsequently raise a\nTypeError. This patch modifies Timestamp to check that offset is an\ninteger value during construction and otherwise raise a ValueError.\n\nThis patch also adds test coverage for Timestamp.increment_offset.\n\nChange-Id: Ie2d33b96687bc319b86f375e339d9a99f3892ed5\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d73c7263c8f7e0bf3d4df1f64f002e3bff8e9ffc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d73c7263c8f7e0bf3d4df1f64f002e3bff8e9ffc"}]},"branch":"refs/heads/master"},"21a5cb4d12190cd2607f0ba1c8e07bc675fe421e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-04-27 09:42:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/92/985092/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/985092/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/985092/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/985092/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/985092/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/985092/3"}}},"commit":{"parents":[{"commit":"18e2528e5077e173b1f950674a27857a0e519706","subject":"Merge \"versioning: add func test for PUT?version-id\u003d 404\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/18e2528e5077e173b1f950674a27857a0e519706"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-17 11:10:34.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-27 09:42:13.000000000","tz":60},"subject":"timestamps: check offset is an integer during construction","message":"timestamps: check offset is an integer during construction\n\nPreviously, a Timestamp could be constructed with a floating point\noffset, but Timestamp.internal would subsequently raise a\nTypeError. This patch modifies Timestamp to check that offset is an\ninteger value during construction and otherwise raise a ValueError.\n\nThis patch also adds test coverage for Timestamp.increment_offset.\n\nChange-Id: Ie2d33b96687bc319b86f375e339d9a99f3892ed5\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21a5cb4d12190cd2607f0ba1c8e07bc675fe421e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21a5cb4d12190cd2607f0ba1c8e07bc675fe421e"}]},"branch":"refs/heads/master"},"864f81bdf0a39aa4e75eb2a61bfd2bcd25664556":{"kind":"REWORK","_number":4,"created":"2026-05-06 14:00:19.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/92/985092/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/985092/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/985092/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/985092/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/985092/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/985092/4"}}},"commit":{"parents":[{"commit":"18e2528e5077e173b1f950674a27857a0e519706","subject":"Merge \"versioning: add func test for PUT?version-id\u003d 404\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/18e2528e5077e173b1f950674a27857a0e519706"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-17 11:10:34.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-05-06 13:58:51.000000000","tz":60},"subject":"timestamps: check offset is an integer during construction","message":"timestamps: check offset is an integer during construction\n\nPreviously, a Timestamp could be constructed with a floating point\noffset, but Timestamp.internal would subsequently raise a\nTypeError. This patch modifies Timestamp to check that offset is an\ninteger value during construction and otherwise raise a ValueError.\n\nNote: non-int types with integer values (e.g. boolean) were previously\nallowed and are therefore still supported.\n\nDrive-By: always return the current offset from\nTimestamp.increment_offset() and add test coverage for that method.\n\nChange-Id: Ie2d33b96687bc319b86f375e339d9a99f3892ed5\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/864f81bdf0a39aa4e75eb2a61bfd2bcd25664556"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/864f81bdf0a39aa4e75eb2a61bfd2bcd25664556"}]},"branch":"refs/heads/master"},"33967853276f0987a58e5b3aca450173adadc064":{"kind":"REWORK","_number":5,"created":"2026-05-12 10:15:42.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/92/985092/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/985092/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/985092/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/985092/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/985092/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/985092/5"}}},"commit":{"parents":[{"commit":"bb378869ba1adc3db90eb8b20c97636cf2213938","subject":"Merge \"tests: more rebalance_missing_suppression_count scenario\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb378869ba1adc3db90eb8b20c97636cf2213938"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-17 11:10:34.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-05-12 09:58:03.000000000","tz":60},"subject":"timestamps: check offset is an integer during construction","message":"timestamps: check offset is an integer during construction\n\nPreviously, a Timestamp could be constructed with a floating point\noffset, but Timestamp.internal would subsequently raise a\nTypeError. This patch modifies Timestamp to check that offset is an\nint value *during construction* and otherwise raise a TypeError.\n\nNote: a boolean value was previously allowed and is therefore still\nsupported.\n\nDrive-By: always return the current offset from\nTimestamp.increment_offset() and add test coverage for that method.\n\nChange-Id: Ie2d33b96687bc319b86f375e339d9a99f3892ed5\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/33967853276f0987a58e5b3aca450173adadc064"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/33967853276f0987a58e5b3aca450173adadc064"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
