)]}'
{"id":"openstack%2Fswift~98518","triplet_id":"openstack%2Fswift~master~Ie6d6dd16c3b5a7e059de689629aeba1bda5953b8","project":"openstack/swift","branch":"master","topic":"bug/1326237","hashtags":[],"change_id":"Ie6d6dd16c3b5a7e059de689629aeba1bda5953b8","subject":"Fix typo in the man page of proxy-server.conf","status":"ABANDONED","created":"2014-06-06 21:01:21.000000000","updated":"2014-06-10 22:21:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d65609987df37874dfd2c3b72cc87cf3a3b29378","_number":98518,"virtual_id_number":98518,"owner":{"_account_id":7479,"name":"paul luse","email":"paul.e.luse@intel.com","username":"peluse"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2014-06-10 03:58:55.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2014-06-06 21:58:18.000000000","_account_id":7479,"name":"paul luse","email":"paul.e.luse@intel.com","username":"peluse"},{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"recommended":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2014-06-10 02:43:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7479,"name":"paul luse","email":"paul.e.luse@intel.com","username":"peluse"},{"value":-2,"date":"2014-06-10 20:14:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":7479,"name":"paul luse","email":"paul.e.luse@intel.com","username":"peluse"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7479,"name":"paul luse","email":"paul.e.luse@intel.com","username":"peluse"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-06-10 02:43:34.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2014-06-10 03:58:55.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2014-06-10 20:14:43.000000000","updated_by":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"reviewer":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"state":"REVIEWER"}],"messages":[{"id":"ae0a6e5a7de7588535a7d49cab2a6058b6643595","author":{"_account_id":7479,"name":"paul luse","email":"paul.e.luse@intel.com","username":"peluse"},"date":"2014-06-06 21:01:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"187ea359d57949a30b78d42cbf70dfb524641d16","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-06 21:04:29.000000000","message":"Patch Set 1:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"3f37e69f57c24e7e9c7e0d6e92fcc177b9e65edd","author":{"_account_id":7479,"name":"paul luse","email":"paul.e.luse@intel.com","username":"peluse"},"date":"2014-06-06 21:24:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"de8f795b1bcb0d84fdf6cc349b114b5b78e791b9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-06 21:24:53.000000000","message":"Patch Set 2:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"ba04832b2a83736140ddff7a9385a415232c5039","author":{"_account_id":7479,"name":"paul luse","email":"paul.e.luse@intel.com","username":"peluse"},"date":"2014-06-06 21:57:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fa7ba5d9c5161b396c4203cbf95c17dc01f2c02e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-06 21:57:29.000000000","message":"Patch Set 3:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"c59df9c84077b53e767133bae608a1cc481536af","author":{"_account_id":7479,"name":"paul luse","email":"paul.e.luse@intel.com","username":"peluse"},"date":"2014-06-06 21:58:18.000000000","message":"Patch Set 3:\n\nyeah, trying to set a record for how many times I can get the bug # wrong in the commit message :)","accounts_in_message":[],"_revision_number":3},{"id":"7df28257bad1fb03704b3cd938644fbcf73ee008","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-06 23:08:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- gate-swift-pep8 http://logs.openstack.org/18/98518/3/check/gate-swift-pep8/b5a68b9 : SUCCESS in 1m 16s\n- gate-swift-docs http://docs-draft.openstack.org/18/98518/3/check/gate-swift-docs/44530da/doc/build/html/ : SUCCESS in 2m 05s\n- gate-swift-python26 http://logs.openstack.org/18/98518/3/check/gate-swift-python26/0897927 : SUCCESS in 4m 35s\n- gate-swift-python27 http://logs.openstack.org/18/98518/3/check/gate-swift-python27/5d5a6a4 : SUCCESS in 2m 54s\n- check-tempest-dsvm-full http://logs.openstack.org/18/98518/3/check/check-tempest-dsvm-full/393c053 : SUCCESS in 58m 17s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/18/98518/3/check/check-tempest-dsvm-postgres-full/86cca81 : SUCCESS in 50m 17s\n- check-tempest-dsvm-neutron http://logs.openstack.org/18/98518/3/check/check-tempest-dsvm-neutron/32c658a : SUCCESS in 1h 04m 47s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/18/98518/3/check/check-tempest-dsvm-neutron-heat-slow/6b4b9da : SUCCESS in 22m 15s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/18/98518/3/check/check-grenade-dsvm/c4d2d57 : SUCCESS in 53m 27s\n- gate-swift-tox-func http://logs.openstack.org/18/98518/3/check/gate-swift-tox-func/d2ec885 : SUCCESS in 3m 34s (non-voting)\n- check-swift-dsvm-functional http://logs.openstack.org/18/98518/3/check/check-swift-dsvm-functional/80df8e7 : SUCCESS in 13m 45s\n- check-devstack-dsvm-cells http://logs.openstack.org/18/98518/3/check/check-devstack-dsvm-cells/94f1ed7 : SUCCESS in 15m 31s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/18/98518/3/check/gate-tempest-dsvm-large-ops/615bc25 : SUCCESS in 24m 13s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/18/98518/3/check/gate-tempest-dsvm-neutron-large-ops/3d2e525 : SUCCESS in 27m 55s","accounts_in_message":[],"_revision_number":3},{"id":"cfc258935a0f6030b7be1a09e1d75bd5ca481d91","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2014-06-10 02:43:34.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"abe2ba0d893565fc331970182503fd28128a96f8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-10 02:43:35.000000000","message":"Patch Set 3: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"5d0c43a52a2a1dd3a6da826680e87cc02b4bf0d7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-10 03:58:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- gate-swift-pep8 http://logs.openstack.org/18/98518/3/check/gate-swift-pep8/32c0adb : SUCCESS in 2m 02s\n- gate-swift-docs http://docs-draft.openstack.org/18/98518/3/check/gate-swift-docs/7af10f0/doc/build/html/ : SUCCESS in 1m 16s\n- gate-swift-python26 http://logs.openstack.org/18/98518/3/check/gate-swift-python26/afb624f : SUCCESS in 3m 44s\n- gate-swift-python27 http://logs.openstack.org/18/98518/3/check/gate-swift-python27/f99cbc8 : SUCCESS in 3m 16s\n- check-tempest-dsvm-full http://logs.openstack.org/18/98518/3/check/check-tempest-dsvm-full/e66b258 : SUCCESS in 43m 40s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/18/98518/3/check/check-tempest-dsvm-postgres-full/7f88775 : SUCCESS in 55m 32s\n- check-tempest-dsvm-neutron http://logs.openstack.org/18/98518/3/check/check-tempest-dsvm-neutron/a18d045 : SUCCESS in 1h 02m 17s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/18/98518/3/check/check-tempest-dsvm-neutron-heat-slow/748e8e9 : SUCCESS in 17m 27s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/18/98518/3/check/check-grenade-dsvm/149efc4 : SUCCESS in 44m 11s\n- gate-swift-tox-func http://logs.openstack.org/18/98518/3/check/gate-swift-tox-func/7606fa5 : SUCCESS in 3m 33s (non-voting)\n- check-swift-dsvm-functional http://logs.openstack.org/18/98518/3/check/check-swift-dsvm-functional/128947e : SUCCESS in 14m 33s\n- check-devstack-dsvm-cells http://logs.openstack.org/18/98518/3/check/check-devstack-dsvm-cells/dd7e4a6 : SUCCESS in 12m 39s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/18/98518/3/check/gate-tempest-dsvm-large-ops/71e2b14 : SUCCESS in 32m 56s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/18/98518/3/check/gate-tempest-dsvm-neutron-large-ops/b8ac858 : SUCCESS in 21m 37s","accounts_in_message":[],"_revision_number":3},{"id":"544d1bd0c033363017486aa290c511b31cf9ee8a","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2014-06-10 20:14:43.000000000","message":"Patch Set 3: Code-Review-2\n\nduplicate of https://review.openstack.org/#/c/97695/","accounts_in_message":[],"_revision_number":3},{"id":"7adae836a85f031fb8c336579e97bb90127a37dc","author":{"_account_id":7479,"name":"paul luse","email":"paul.e.luse@intel.com","username":"peluse"},"date":"2014-06-10 22:21:15.000000000","message":"Abandoned\n\nduplicate","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"98d67278532573470672e2db921d3238468dedd7","revisions":{"16f038d74e40bc612cc317cdd473a55ef58f507b":{"kind":"REWORK","_number":1,"created":"2014-06-06 21:01:21.000000000","uploader":{"_account_id":7479,"name":"paul luse","email":"paul.e.luse@intel.com","username":"peluse"},"ref":"refs/changes/18/98518/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/98518/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/98518/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/98518/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/98518/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/98518/1"}}},"commit":{"parents":[{"commit":"c384d76c575e28d045b86528fb5a6c67f0b8c518","subject":"Merge \"Add --quoted option to swift-temp-url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c384d76c575e28d045b86528fb5a6c67f0b8c518"}]}],"author":{"name":"Paul Luse","email":"paul.e.luse@intel.com","date":"2014-06-06 21:00:04.000000000","tz":-420},"committer":{"name":"Paul Luse","email":"paul.e.luse@intel.com","date":"2014-06-06 21:00:04.000000000","tz":-420},"subject":"Fix typo in the man page of proxy-server.conf","message":"Fix typo in the man page of proxy-server.conf\n\n\"client_timeoutt\" ---\u003e  \"client_timeout\"\n\nChange-Id: Ie6d6dd16c3b5a7e059de689629aeba1bda5953b8\nCloses-Bug: #1326237\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16f038d74e40bc612cc317cdd473a55ef58f507b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16f038d74e40bc612cc317cdd473a55ef58f507b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c384d76c575e28d045b86528fb5a6c67f0b8c518","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"2146fdfa3d744372e978bbebca7ce9b887328656":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2014-06-06 21:24:52.000000000","uploader":{"_account_id":7479,"name":"paul luse","email":"paul.e.luse@intel.com","username":"peluse"},"ref":"refs/changes/18/98518/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/98518/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/98518/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/98518/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/98518/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/98518/2"}}},"commit":{"parents":[{"commit":"c384d76c575e28d045b86528fb5a6c67f0b8c518","subject":"Merge \"Add --quoted option to swift-temp-url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c384d76c575e28d045b86528fb5a6c67f0b8c518"}]}],"author":{"name":"Paul Luse","email":"paul.e.luse@intel.com","date":"2014-06-06 21:00:04.000000000","tz":-420},"committer":{"name":"Paul Luse","email":"paul.e.luse@intel.com","date":"2014-06-06 21:24:21.000000000","tz":-420},"subject":"Fix typo in the man page of proxy-server.conf","message":"Fix typo in the man page of proxy-server.conf\n\n\"client_timeoutt\" ---\u003e  \"client_timeout\"\n\nChange-Id: Ie6d6dd16c3b5a7e059de689629aeba1bda5953b8\nCloses-Bug: 1308987\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2146fdfa3d744372e978bbebca7ce9b887328656"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2146fdfa3d744372e978bbebca7ce9b887328656"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c384d76c575e28d045b86528fb5a6c67f0b8c518","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"98d67278532573470672e2db921d3238468dedd7":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2014-06-06 21:57:16.000000000","uploader":{"_account_id":7479,"name":"paul luse","email":"paul.e.luse@intel.com","username":"peluse"},"ref":"refs/changes/18/98518/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/98518/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/98518/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/98518/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/98518/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/98518/3"}}},"commit":{"parents":[{"commit":"c384d76c575e28d045b86528fb5a6c67f0b8c518","subject":"Merge \"Add --quoted option to swift-temp-url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c384d76c575e28d045b86528fb5a6c67f0b8c518"}]}],"author":{"name":"Paul Luse","email":"paul.e.luse@intel.com","date":"2014-06-06 21:00:04.000000000","tz":-420},"committer":{"name":"Paul Luse","email":"paul.e.luse@intel.com","date":"2014-06-06 21:56:48.000000000","tz":-420},"subject":"Fix typo in the man page of proxy-server.conf","message":"Fix typo in the man page of proxy-server.conf\n\n\"client_timeoutt\" ---\u003e  \"client_timeout\"\n\nChange-Id: Ie6d6dd16c3b5a7e059de689629aeba1bda5953b8\nCloses-Bug: 1326237\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/98d67278532573470672e2db921d3238468dedd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/98d67278532573470672e2db921d3238468dedd7"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c384d76c575e28d045b86528fb5a6c67f0b8c518","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
