)]}'
{"id":"openstack%2Fswift~985842","triplet_id":"openstack%2Fswift~master~Id55312d835a1360e20809d79cec5435b5bf5adc8","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-04-28 15:34:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id55312d835a1360e20809d79cec5435b5bf5adc8","subject":"Be more targetted in our eventlet imports","status":"MERGED","created":"2026-04-22 16:15:54.000000000","updated":"2026-04-28 15:36:24.000000000","submitted":"2026-04-28 15:34:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"985842","meta_rev_id":"ead53b627f876b91187640b0978cea805d7e6475","_number":985842,"virtual_id_number":985842,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-28 15:34:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2026-04-28 14:06:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2026-04-28 14:06:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-22 17:27:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-28 14:06:01.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"da491659e207380f3fe7401738f17965db1e6820","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-22 16:15:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71726d89dd921fe377c129b4a3e2a151845bb56b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 17:27:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7570f7e925ba40cf98fc0c1df6d2a187\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d9b1fb8d252456ca21274697e659a58 : SUCCESS in 10m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/1baef0ec65064a37b28d4ee537d6b751 : SUCCESS in 33m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5e47c78865d54e9f8fdc5fd66b0bd6a1 : SUCCESS in 52m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/63c4e4d1a5744a6f81022ab6787658b3 : SUCCESS in 38m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee6479665cb44486bd2aa3fc25bcf5b3 : SUCCESS in 1h 01m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1a5bc3cd92764cbb96927e08f6fd3a71 : SUCCESS in 19m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/08fc688d583e4967a439e37d88c778ff : SUCCESS in 18m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d21b3fd2b86b49cf812035dfd93b9b1b : SUCCESS in 20m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9431032c3bb5483b80de4d87962889cd : SUCCESS in 5m 12s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c7ea21bb27654880bc8c7d573334617d : SUCCESS in 25m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6c69cec97b7a49dbad3665d60f4688a2 : SUCCESS in 24m 04s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b948a64bdbdb4a129913af24ce9707a1 : SUCCESS in 21m 09s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/2b18f1a5b40b427f9d104e7eb3a5b219 : SUCCESS in 17m 50s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1899e933369f4d5c83fc72576143e317 : SUCCESS in 8m 16s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/58b31bea62e0445c87c488a9bd6ef06c : SUCCESS in 8m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/596383320a6b4d168bdbcdbd162719ed : SUCCESS in 14m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/11ef5aee3a054d09b078a8c1f00b34a7 : SUCCESS in 6m 24s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9e94dc6bdad54a0fa41c282e0b53ccdb : SUCCESS in 10m 47s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9f734941374e4a94acd9be371dbfc94d : SUCCESS in 5m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea000f6de45f4a7c8456a86106bc1bcd : SUCCESS in 39m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ac7aa7b7bb97456f8166ed2d4219a8b6 : SUCCESS in 54m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c150b6a7498241ee836a6474faa3483f : SUCCESS in 38m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76092883f91d400c8e3ce687284cd1d4 : SUCCESS in 27m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2e2daf603ff4ac79096e6e006379cc9 : SUCCESS in 2m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6fb87727c2534ff1b646865b3edd0979 : SUCCESS in 11m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a75b03616e4944c28b9ab08731bab091 : SUCCESS in 35m 20s","accounts_in_message":[],"_revision_number":1},{"id":"45eebac579f1ca958c76b7837008fd5a84974963","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 22:12:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54a8e9528e7e4cac979b0aeb9fed3162\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d4a0bef2cfdb4b94a423234d60b8a380 : SUCCESS in 37m 33s","accounts_in_message":[],"_revision_number":1},{"id":"ddaa691200cad2114853841c3d2a7d74da48f9a9","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-28 14:06:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c4777aa0d8608e60c60cb1247a8860102e618e1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 14:14:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c7ce116203f35a290008a67290d101a1dc2b4b11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 15:34:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08042eca6b5441e09e80444f905c04eb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0f1216571b64a828de32e7b2af1577b : SUCCESS in 8m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/e02da2c4b2bb4d049a9aab16d69f1aaa : SUCCESS in 57m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3709a51fcc0c48c0bda438c85fa224db : SUCCESS in 51m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4ae28dfc913547dc88f7a5c416e55913 : SUCCESS in 56m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c2c3ff039ee4c0690554b4b5284ce7e : SUCCESS in 1h 04m 53s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2f88a9ee0fcb42c386e9e396a1e86574 : SUCCESS in 7m 19s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/646471cf691b4cd1ac71bd4173292620 : SUCCESS in 12m 38s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ca8f37c8340d4181bdf77d6f5a50be47 : SUCCESS in 20m 22s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9958a1c6c31f48e4b1c92ac03809727b : SUCCESS in 27m 13s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/2ef8740eb0b64d3688414c0088e81c0d : SUCCESS in 24m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f67d46122f564bcbbe64761a1d902b12 : SUCCESS in 14m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b8f8d54e959542ef8880d9f975e814ad : SUCCESS in 16m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a4647d1e189d4aa499a2d99939d6cbae : SUCCESS in 15m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c3ff18e062a443cd964525d62de4e502 : SUCCESS in 7m 05s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0f90e6dc0e5d45c3a6b334be51e20092 : SUCCESS in 9m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1dca6a0de2644b0bb575da7306b377a1 : SUCCESS in 57m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f37ec88f5048476f8cf5da9b79b4ded8 : SUCCESS in 56m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1914972aa559415ca72854bca55da5c8 : SUCCESS in 54m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fafdc824530048d6a9b92d91cfaf598d : SUCCESS in 30m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7e7f695aa9f4b80936584fd3b875549 : SUCCESS in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4fe1933b4eef4b4cb30a12f4c2d566d2 : SUCCESS in 18m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3462a942582a4b15a0be21fd9e6fe678 : SUCCESS in 56m 34s","accounts_in_message":[],"_revision_number":1},{"id":"b495940d9df45d5c1561d01911b07df840277050","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 15:34:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ead53b627f876b91187640b0978cea805d7e6475","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 15:36:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fabfc7bc6d8246cfa3c4571a4b071fe4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c198bac64f3b4aa3a75fd610740beb71 : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/35da5c6f37084fdebf3d7244e2907324 : SUCCESS in 1m 02s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bf9d86baea1ca158ed57c81285a8b86cb68cf34e","revisions":{"bf9d86baea1ca158ed57c81285a8b86cb68cf34e":{"kind":"REWORK","_number":1,"created":"2026-04-22 16:15:54.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/42/985842/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/985842/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/985842/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/985842/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/985842/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/985842/1"}}},"commit":{"parents":[{"commit":"aaf20ea08e63a37f28823f806796a6cf1f0ac82c","subject":"docs: Import from swift.common.concurrency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aaf20ea08e63a37f28823f806796a6cf1f0ac82c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-22 16:15:43.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-22 16:15:43.000000000","tz":-420},"subject":"Be more targetted in our eventlet imports","message":"Be more targetted in our eventlet imports\n\nChange-Id: Id55312d835a1360e20809d79cec5435b5bf5adc8\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf9d86baea1ca158ed57c81285a8b86cb68cf34e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf9d86baea1ca158ed57c81285a8b86cb68cf34e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
