)]}'
{"id":"openstack%2Fswift~985896","triplet_id":"openstack%2Fswift~master~I2d77be67624f3c70dd055122c0a2cb5c9702d14c","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-05-05 06:59:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2d77be67624f3c70dd055122c0a2cb5c9702d14c","subject":"probe tests: Pass when there are multiple EC policies","status":"MERGED","created":"2026-04-22 21:40:38.000000000","updated":"2026-05-05 07:00:43.000000000","submitted":"2026-05-05 06:59:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"985896","meta_rev_id":"bf95ca7edae0eb9a3a49f3e2098abe008b8a85b9","_number":985896,"virtual_id_number":985896,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-05 06:59:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2026-05-05 05:11:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2026-05-05 05:11:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-22 22:38:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-05 05:11:24.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"1ec2e35db1534cc28ea52ddb3766ac9e990bbbf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-22 21:40:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"348ab56a39cc2a6956772e31c87605c7c0fe3e5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 22:38:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f04c8baadcd450ca8574f610a067613\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb66243925d24a4c8e3121a66d209d75 : SUCCESS in 6m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/197ee3ab67534272bd44c4b3d574dac9 : SUCCESS in 52m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c92afe033c1744fd88250b7a216839a8 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":1},{"id":"23ee491af8aeb3fb52fe585bcf9c471b2c1ec15e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 02:13:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7bd822835c64aa5a47608062bb9686e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/81e6fee475bf43da8434ba78c40fd3c2 : SUCCESS in 34m 19s","accounts_in_message":[],"_revision_number":1},{"id":"c5eb22066524a758b9e692ac2a56b745076a81af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-24 17:46:08.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"e3c810714a33ee8e4e912f084620c56d46ce462e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 18:45:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d525b8cd46b4d098de05b5675ffbdfb\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/20cfb517937747e9bea153e7d711a1d4 : SUCCESS in 39m 00s","accounts_in_message":[],"_revision_number":2},{"id":"4702897db5ea99c0746d1360e0956dc2b433eb92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 18:51:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/561c58d25819400ea92b12fc20239dba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c417afc161c45d6b21988e7af114ac8 : SUCCESS in 11m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5214030eb7b9409baa69f479b7b88cee : SUCCESS in 1h 00m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38aabfe14237432183bbd57f015e2f5a : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":2},{"id":"babab2c0711dd225b9b215878d5eb468d9f93073","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-05-05 05:11:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f5aac0b42090e25a6a2238e4accd2a3b708ce9b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 05:11:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"53b995e758ea45a10a1bd6d720d60fa544500e49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 06:59:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f51c4228a864acdb89b3282aec2dd7d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a51926528794339ba2ed6a0ac65ec3b : SUCCESS in 7m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/50ececed67d34bc2aded5d933a356006 : SUCCESS in 17m 55s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1165b4a12496419789674c08fa10e80c : SUCCESS in 21m 37s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1aee8ca22dd5441fa6b058cb8b74ff36 : SUCCESS in 29m 17s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/629b7f40a04846d58d23baaa9cd6e986 : SUCCESS in 10m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f30322d2fa934d15bdbc0a6ebe9aa816 : SUCCESS in 11m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/14bf176df172410688077562828a5a42 : SUCCESS in 15m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7000483e015c4274a1dac5e6635263d1 : SUCCESS in 13m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9810e0c2b07648b0be0e0b894669b744 : SUCCESS in 4m 01s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1643ba616af8480ba796e60dcbe8c9fa : SUCCESS in 8m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d806f2ec85c0479fb430cf5eb4267790 : SUCCESS in 38m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02a7cb1b9ac44d88b6129a10be32fd14 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":2},{"id":"bd114a3012ed8573bd2e9337cbd5a65e1ca282c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 06:59:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"bf95ca7edae0eb9a3a49f3e2098abe008b8a85b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 07:00:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91dec9a6266a4a92a9abc3bdb42533e9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5c5ee6cffa94d47bf525c1863e1c4c8 : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4707e5bb77084672ab7f93fef6558e6f : FAILURE in 30s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3948d71adc6a25d4671d1a81eb2b25ef127e9bf4","revisions":{"7a81b5ee58f5b91026d5ec49b820763605dbfa90":{"kind":"REWORK","_number":1,"created":"2026-04-22 21:40:38.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/96/985896/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/985896/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/985896/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/985896/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/985896/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/985896/1"}}},"commit":{"parents":[{"commit":"d39b6b98fd989267e73064317d657d64bc3d63c3","subject":"Merge \"relinker: Test cleaning up consecutive hashes in same suffix directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d39b6b98fd989267e73064317d657d64bc3d63c3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-22 20:26:11.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-22 20:26:25.000000000","tz":-420},"subject":"probe tests: Pass when there are multiple EC policies","message":"probe tests: Pass when there are multiple EC policies\n\nPreviously, we were assuming that the \"other policy\" would always be\na replicated policy, with encryption disabled.\n\nRelated-Change: I051316e54fb62c941cd1801dbd8fb104bd046431\nChange-Id: I2d77be67624f3c70dd055122c0a2cb5c9702d14c\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a81b5ee58f5b91026d5ec49b820763605dbfa90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a81b5ee58f5b91026d5ec49b820763605dbfa90"}]},"branch":"refs/heads/master"},"3948d71adc6a25d4671d1a81eb2b25ef127e9bf4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-04-24 17:46:08.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/96/985896/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/985896/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/985896/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/985896/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/985896/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/985896/2"}}},"commit":{"parents":[{"commit":"05c6eb09bb5a6df4ffc9d45f8b96581014940e78","subject":"Merge \"Timestamp: allow delta kwarg to decrement time to zero\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05c6eb09bb5a6df4ffc9d45f8b96581014940e78"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-22 20:26:11.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-24 17:07:18.000000000","tz":-420},"subject":"probe tests: Pass when there are multiple EC policies","message":"probe tests: Pass when there are multiple EC policies\n\nPreviously, we were assuming that the \"other policy\" would always be\na replicated policy, with encryption disabled.\n\nRelated-Change: I051316e54fb62c941cd1801dbd8fb104bd046431\nChange-Id: I2d77be67624f3c70dd055122c0a2cb5c9702d14c\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3948d71adc6a25d4671d1a81eb2b25ef127e9bf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3948d71adc6a25d4671d1a81eb2b25ef127e9bf4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
