)]}'
{"id":"openstack%2Fswift~985972","triplet_id":"openstack%2Fswift~master~I179a8423f9c8152c1fe14945c6cac8b7597ea7c1","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2026-05-07 17:59:14.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-05-07 19:58:21.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-05-07 19:58:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I179a8423f9c8152c1fe14945c6cac8b7597ea7c1","subject":"container: Split DELETE code paths","status":"MERGED","created":"2026-04-23 15:38:01.000000000","updated":"2026-05-07 19:59:30.000000000","submitted":"2026-05-07 19:58:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"985972","meta_rev_id":"7b95fc8231f8769404cf92bbd9f1ac46a3cfeb5d","_number":985972,"virtual_id_number":985972,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-07 19:58:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":2,"date":"2026-05-07 17:59:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":2,"date":"2026-05-07 17:29:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":1,"date":"2026-05-07 17:59:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-23 16:41:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-23 16:44:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-07 01:22:08.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2026-05-07 16:30:45.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2026-05-07 17:29:29.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"aaa97538f003e94fffb6eb85591df1ef37a86753","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-23 15:38:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be35eccab54d8dfe03b1922a4721bad17658e55c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 16:41:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0471039481144569a9ec17ef506436e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e99a1f0a4ef34307a52f2cacdf1cf672 : SUCCESS in 31m 14s","accounts_in_message":[],"_revision_number":1},{"id":"1fdda85145665d88e3e8dca2933bf6d4933bb000","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 16:44:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b448193c12d499581d9cd9a255d680c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cf7b205e59142bfa292f5638f367144 : SUCCESS in 11m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/daba13d8837a4163916c5a0538919642 : SUCCESS in 1h 03m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8b217d8b069a466591c93c8be53ab01a : SUCCESS in 35m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a61d0aaf52294393a2f14b5d6ccd4cb1 : SUCCESS in 37m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a9a1fd66a27f40c0be03902c4f3916bc : SUCCESS in 39m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/708babaf59d843e6a1947555be661324 : SUCCESS in 15m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/29c530f34b6e476e9a206cfd922e5eff : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/637f2dffc3944444b0a9b04cb84124bc : SUCCESS in 18m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4f6c3d95a6454f399497b4b75caf2f5f : SUCCESS in 7m 09s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8a546825d4274dc7a430c7f04778fab3 : SUCCESS in 30m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4416bc342f6340c9978955f142b52b20 : SUCCESS in 20m 25s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/139717519f3e4649a0ac61126584b46f : SUCCESS in 23m 31s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/00ef965da7ed4d559e03b4dcbae2177f : SUCCESS in 20m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c739562ca40d4b0cbeeab6ca44e45eee : SUCCESS in 14m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ee491987d8a44af4b1a346a31ddc1c23 : SUCCESS in 14m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/66f0a169c8e8413b99110b0b3640ec07 : SUCCESS in 9m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/168730bb7e024a4b94f3190da7f693c7 : SUCCESS in 8m 14s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e2f851e2216743b08beb587c9157658a : SUCCESS in 16m 17s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3e436df694f0409697bff3aec053dcff : SUCCESS in 5m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/39a384d3e870468681ee9509becdafb9 : SUCCESS in 57m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9c5b348f3c964d64812b1d92b590f3f6 : SUCCESS in 23m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ff59a85a46ba4fd0ba247adbc19d0392 : SUCCESS in 54m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab45e208a6684f82a7e93488cee0e982 : SUCCESS in 21m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bfd0fa6f6f84754a5cabe01864360f6 : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/876a28c205dd4b06b3ce42727ca57338 : SUCCESS in 11m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e66e13ffbce44c4397847a8b5b71cb46 : SUCCESS in 30m 06s","accounts_in_message":[],"_revision_number":1},{"id":"fb2a2b62ec9157e096ef3407fd0763b1602a493f","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-05-07 01:22:08.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4ab1b7a88707f656de7ff2bff0091dbdd1d7d7af","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-05-07 16:30:45.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4acd52094fc44485523975c297c30342727260e8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-05-07 16:55:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"11ef9b0d6c4da7c71f1770bd131963e1fa86e833","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-05-07 17:15:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"d3a7bfe8d88831d70927524c9a8317a3b8fa9dea","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-05-07 17:29:29.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b37bde4454fe2ede36e39c7b3a15663426445540","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-05-07 17:59:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cf8e3e014b4ccc4ceeb3b9e94d2c882e325b3e82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 18:16:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d03946f94344e1aa0a04042c1ff9115\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5c1af38d719f409591441abb6fdbd85f : SUCCESS in 48m 30s","accounts_in_message":[],"_revision_number":2},{"id":"c4636f912408299651e2018ce91aa34571362e63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 18:23:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f652c60defbd48b39c3213fb2b72d775\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2282ea02a0ce49779a1f620a1036987d : SUCCESS in 4m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/1d95bc1d56a849cea5ff475d55fd77de : SUCCESS in 1h 01m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8b1bb85d30444de18350190dde0a4244 : SUCCESS in 1h 03m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5a190ef4a1c448b9b9bf87a4810fbda1 : SUCCESS in 43m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c485e5cc6d248f7bb75de316f6e7cc2 : SUCCESS in 34m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e66829dd141a4718a4d99ca4363e762f : SUCCESS in 9m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/37ea6e485a724184869a72f10eb6b00f : SUCCESS in 10m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0e5e2357908547dba3427e1aa338b6dd : SUCCESS in 19m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d83ad58fc7b4491a853bc338e0545b2c : RETRY_LIMIT in 3m 43s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6e81747e6a184657852feda6c90abdf4 : SUCCESS in 25m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b3bcf04fa4f54585bf60a2d8575e8b83 : SUCCESS in 13m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a56021a004f54e5990b0057012c821be : SUCCESS in 30m 25s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/ff1ecc12f3f54c7cae460cb201f58bb7 : SUCCESS in 22m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6a347d74e56f42fbb295814ee6ec9711 : SUCCESS in 11m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/dfedd87b27b34ff6a1df7b92f8505f51 : SUCCESS in 16m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/925f5100aa634de581028aaf84d41a7e : SUCCESS in 17m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/aaeac57068754ea7ac45ef2922552eae : SUCCESS in 9m 06s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/59f1e4c198b5439db759ae3d21635aeb : SUCCESS in 8m 50s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/688ffe5b0ae0481c9c652e9a4bdbf451 : SUCCESS in 6m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/966423f5bb9d478da573b8f23cbeda92 : SUCCESS in 1h 02m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fd2e3585047e40758afb17b5def6e9e1 : SUCCESS in 22m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/20fa5d72497b422da23045c51a04ce66 : SUCCESS in 55m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/338efac8ce464b658291777a50f9586f : SUCCESS in 22m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b1dad140e804f3a9325b7348654cc50 : SUCCESS in 5m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7f1ebe98566a461c9033b3db48d2ac26 : SUCCESS in 19m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0a4cbc379c946edbf10cc06b2623f9e : SUCCESS in 1h 02m 57s","accounts_in_message":[],"_revision_number":2},{"id":"054651a0e02e8c5f308684528fa1febf1ad6bcc6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 18:24:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2851b53099c0ff331f77ee92e973df3aaa131b24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 19:58:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/735fe061274d4bddb86e954552cfc3d2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d4acaa42d5c47659dd4399347413d22 : SUCCESS in 8m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/c4444de3372148f5b5b52c5fd79fc40f : SUCCESS in 1h 09m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/50d8b7b808394bdd8e6cc86229f9dd2c : SUCCESS in 1h 30m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/462635d505e64906910ba73fa597e892 : SUCCESS in 1h 08m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6bf137fcd1834bf4b0fbb2f3541e1bae : SUCCESS in 34m 05s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b482be90502341d4826b56dc791a84d4 : SUCCESS in 10m 11s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/52c867e2cb6049d3abfb8511e1e1a6ea : SUCCESS in 27m 28s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1529b880bf9c4e7d847a003f5ef31c76 : SUCCESS in 22m 25s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9a9765644e3846c193804539435c50d8 : SUCCESS in 11m 39s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/f1ac05f61fc740cd8ebf29892c00ec0c : SUCCESS in 21m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3852ab16439c4a4d866339897bd58ceb : SUCCESS in 16m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5882248ce63e4cb38221e0d6d561d5e5 : SUCCESS in 8m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/141a22368f9d4f7b87a556b918275baa : SUCCESS in 16m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ab97e1b0c6724675a9e9c84db30781dc : SUCCESS in 7m 17s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/803b54f6b46f40fe8788046c056e5ef6 : SUCCESS in 11m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a46318e56b3b4e539cc548e417c7cfa9 : SUCCESS in 1h 00m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bf0544da4aad411ea6f2ab6004c21929 : SUCCESS in 53m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e07a86f59aec4a6d95da386fd7b7cd6c : SUCCESS in 54m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2f76dad47514566bb211f9124d922ea : SUCCESS in 30m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f49b472cd04a4c96b67bb0d908d94efb : SUCCESS in 5m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/00469774e5e445839c26341d8fb86cfa : SUCCESS in 21m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6bc05e49c7e8462a87e3ad8594d45bb1 : SUCCESS in 1h 02m 54s","accounts_in_message":[],"_revision_number":2},{"id":"49b5b3b99fa4e8b3a08cbb289cf7a0a49218ac69","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 19:58:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7b95fc8231f8769404cf92bbd9f1ac46a3cfeb5d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 19:59:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79cea52496484aa6aa36780cd601f3a2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51bacf1464c54e3db62780b6092fd5bf : SUCCESS in 56s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/12cc3fef8ddf4b5fa16dfc806126c102 : SUCCESS in 42s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"43965416e7f09fdf8e2c1d62b01e5f74a26ab437","revisions":{"65e47b755b3a2b7399997133ed208fad97cefdca":{"kind":"REWORK","_number":1,"created":"2026-04-23 15:38:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/72/985972/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/985972/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/985972/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/985972/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/985972/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/985972/1"}}},"commit":{"parents":[{"commit":"a4cfcd869642fbc7e6fda68bffabfeb5cd1098b1","subject":"Merge \"timestamps: add NormalTimestamp for ShardRanges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a4cfcd869642fbc7e6fda68bffabfeb5cd1098b1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-23 15:31:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-23 15:31:26.000000000","tz":-420},"subject":"container: Split DELETE code paths","message":"container: Split DELETE code paths\n\nContainer DELETE request behave very differently depending on whether\nwe\u0027re deleting the container itself or an individual object row within\na container. Split those into different functions.\n\nChange-Id: I179a8423f9c8152c1fe14945c6cac8b7597ea7c1\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/65e47b755b3a2b7399997133ed208fad97cefdca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/65e47b755b3a2b7399997133ed208fad97cefdca"}]},"branch":"refs/heads/master"},"43965416e7f09fdf8e2c1d62b01e5f74a26ab437":{"kind":"REWORK","_number":2,"created":"2026-05-07 17:15:02.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/72/985972/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/985972/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/985972/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/985972/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/985972/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/985972/2"}}},"commit":{"parents":[{"commit":"a4cfcd869642fbc7e6fda68bffabfeb5cd1098b1","subject":"Merge \"timestamps: add NormalTimestamp for ShardRanges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a4cfcd869642fbc7e6fda68bffabfeb5cd1098b1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-23 15:31:23.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-05-07 17:10:01.000000000","tz":60},"subject":"container: Split DELETE code paths","message":"container: Split DELETE code paths\n\nContainer DELETE request behave very differently depending on whether\nwe\u0027re deleting the container itself or an individual object row within\na container. Split those into different functions.\n\nThis patch makes one intentional behavioral change. Previously a\nDELETE of a container would check the validity of a\nX-Backend-Storage-Policy-Index header, if present. This was\nunnecessary because the header is not relevant when deleting a\ncontainer, only when deleting an object. With this patch the\nX-Backend-Storage-Policy-Index is ignored when deleting a container.\n\nChange-Id: I179a8423f9c8152c1fe14945c6cac8b7597ea7c1\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/43965416e7f09fdf8e2c1d62b01e5f74a26ab437"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/43965416e7f09fdf8e2c1d62b01e5f74a26ab437"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
