)]}'
{"id":"openstack%2Fswift~986537","triplet_id":"openstack%2Fswift~master~I1ab158b79135f16f844b4f8d748102b71c62aa19","project":"openstack/swift","branch":"master","topic":"whalbawi.relinker-hash-filter","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-05-06 16:38:19.000000000","reason":"\u003cGERRIT_ACCOUNT_38767\u003e replied on the change","reason_account":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-04-29 09:44:06.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"removed_from_attention_set":{"38767":{"account":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"last_update":"2026-05-06 16:38:19.000000000","reason":"\u003cGERRIT_ACCOUNT_38767\u003e replied on the change","reason_account":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"}}},"hashtags":[],"change_id":"I1ab158b79135f16f844b4f8d748102b71c62aa19","subject":"relinker: more test coverage for rate limited locations","status":"NEW","created":"2026-04-28 13:52:35.000000000","updated":"2026-05-06 16:38:19.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"357382c46dec9a07c82b2931b327e82f29ea4659","_number":986537,"virtual_id_number":986537,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-04-29 10:19:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-06 16:38:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-28 14:27:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-28 16:36:26.000000000","updated_by":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"reviewer":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"state":"REVIEWER"},{"updated":"2026-04-28 21:33:54.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"f0919f1144d044ab5e4f17d097b2667bbc7487ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-28 13:52:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e2143c82ab055186fac0a00432d5b1f6cf270d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 14:27:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd5f29b66ce9411a881c4c5c8cde50bb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8eb7a0f78205493b835056bdb04ba032 : SUCCESS in 5m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/796c94f576d741e38ce3da5a41ebc19b : SUCCESS in 17m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4e940244639f4ce190f961e91e09909f : SUCCESS in 20m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1c580c7b444e4e9c9ad72769c312c90c : SUCCESS in 22m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/992538608e7b4a2883fdcc778c040d73 : SUCCESS in 6m 53s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2f82c60e665447a781ae17d7dbe4b910 : SUCCESS in 21m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1ff6c12bf669425e8a3c659994288665 : SUCCESS in 18m 05s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8694164181ea463891cfc39217c025e2 : SUCCESS in 11m 56s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/0a563ede10944d3aa3bb4280c4cec929 : SUCCESS in 9m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8285eead27cb45fba8ecbd3f686718f9 : SUCCESS in 15m 03s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/71a1e075cd0a45c9b7378e051730f1a5 : SUCCESS in 8m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ad252057bfa6494ba16ef01c352f459d : SUCCESS in 16m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/65d5572b536848dda38dd2328ec422e8 : SUCCESS in 22m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0b431ac0eeec4240b7d3c35ab2ed2bf6 : SUCCESS in 23m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57fa697b07e346fbb427534bbb08c84b : SUCCESS in 27m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8475f78deff4421b94f30b8f2b3786a9 : SUCCESS in 4m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2a5029e751af4020bdbf71d39e085252 : SUCCESS in 18m 56s","accounts_in_message":[],"_revision_number":1},{"id":"787db557c2e8a399f6ea41b3c223c0a5eef36c17","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-04-28 16:36:26.000000000","message":"Patch Set 1: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c75406fb08c77e5f6cfd4937c6cc50eb7a74fc7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 17:50:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f59234991999418db790c6985ce05508\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8191a4a902be4b3aa0c9549accfd1028 : SUCCESS in 43m 57s","accounts_in_message":[],"_revision_number":1},{"id":"b0ba17f59cb8eb1e02cd97937497aa7f9dba1534","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-28 21:33:54.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3bb63e057ab938fbb96c03531bdb48fc0467cf63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-29 09:41:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"2fba05bbdf029d722f4788f50c7d5643f18d7c87","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-29 09:44:06.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3d7e583fb740103fa665d45270f33743572b1dc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 10:19:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48995fb739a64d74a2abb3c499fad3fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07f1052fe7c643ab8aa6f60b29066910 : SUCCESS in 4m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e01a273ff3514108a4debaf67b02bad1 : SUCCESS in 20m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cc1a494b937a495a90ba79f7f0bb0494 : SUCCESS in 17m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fdee8332bf734e1dbb80edb9d44e57a4 : SUCCESS in 20m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f4d8ffc2e65943a5b9bcf29931b5ae9b : SUCCESS in 7m 38s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c518d36a770f4d0bb5ad57cb6683e0cf : SUCCESS in 11m 13s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5d5bf6b883fa4ea7816372d548abba22 : SUCCESS in 22m 21s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6379b0fb0f61428ab35f2c4a71832fd5 : SUCCESS in 31m 03s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/8d4f8795f49345ea866912c825ccecd0 : SUCCESS in 19m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4c0b577c6f534850bd69a2c1330de6b9 : SUCCESS in 15m 16s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/46e6150ea10d47e1a5720c7c70f402e1 : SUCCESS in 17m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/622e5a114ac84b1d859cd5a6170336c6 : SUCCESS in 12m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5b13ad3319094c858f7272ac447cbc4d : SUCCESS in 22m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a7fb71377a27415988de1b121fd61b32 : SUCCESS in 21m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/482e713a3cd74a02a856097a15e68125 : SUCCESS in 27m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/193747c3f15c469ea882e5a517321e4c : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2cc652d0084544eba22e518ebe25d7a2 : SUCCESS in 25m 27s","accounts_in_message":[],"_revision_number":2},{"id":"635c26e7df041d1cbbb0c78be17b462be05e5009","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 11:58:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2000eefd50b4e7b810ad7441699b43a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/40ef831fbf95407db7c52847445d39fb : SUCCESS in 1h 15m 44s","accounts_in_message":[],"_revision_number":2},{"id":"357382c46dec9a07c82b2931b327e82f29ea4659","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-05-06 16:38:19.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0dd1252817c78a002bc8425fdfb00e814dc5f688","revisions":{"dd919c3703f95bbe9f8dc19d07bb5429fa5d0b3a":{"kind":"REWORK","_number":1,"created":"2026-04-28 13:52:35.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/37/986537/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/37/986537/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/986537/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/986537/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/986537/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/37/986537/1"}}},"commit":{"parents":[{"commit":"66af9d2ab0387d1e138b3f39aa4573678b05d0ab","subject":"relinker: Improve hash filtering logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66af9d2ab0387d1e138b3f39aa4573678b05d0ab"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-28 13:35:53.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-28 13:52:25.000000000","tz":60},"subject":"relinker: more test coverage for rate limited locations","message":"relinker: more test coverage for rate limited locations\n\nImprove the unit tests to verify that when rate limiting is configured\nthen the processed locations are iterated via the\nRateLimitedIterator. Existing tests only verified that a\nRateLimitedIterator was constructed and that no-op locations were not\npassed to it.\n\nChange-Id: I1ab158b79135f16f844b4f8d748102b71c62aa19\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd919c3703f95bbe9f8dc19d07bb5429fa5d0b3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd919c3703f95bbe9f8dc19d07bb5429fa5d0b3a"}]},"branch":"refs/heads/master"},"0dd1252817c78a002bc8425fdfb00e814dc5f688":{"kind":"REWORK","_number":2,"created":"2026-04-29 09:41:18.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/37/986537/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/37/986537/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/986537/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/986537/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/986537/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/37/986537/2"}}},"commit":{"parents":[{"commit":"66af9d2ab0387d1e138b3f39aa4573678b05d0ab","subject":"relinker: Improve hash filtering logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66af9d2ab0387d1e138b3f39aa4573678b05d0ab"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-28 13:35:53.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-29 09:40:57.000000000","tz":60},"subject":"relinker: more test coverage for rate limited locations","message":"relinker: more test coverage for rate limited locations\n\nImprove the unit tests to verify that when rate limiting is configured\nthen the processed locations are iterated via the\nRateLimitedIterator. Existing tests only verified that a\nRateLimitedIterator was constructed and that no-op locations were not\npassed to it.\n\nChange-Id: I1ab158b79135f16f844b4f8d748102b71c62aa19\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dd1252817c78a002bc8425fdfb00e814dc5f688"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dd1252817c78a002bc8425fdfb00e814dc5f688"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
