)]}'
{"id":"openstack%2Fswift~986543","triplet_id":"openstack%2Fswift~master~Icbe74d8742d169d7506979acdb70bb5df4e2607b","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-04-28 22:52:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icbe74d8742d169d7506979acdb70bb5df4e2607b","subject":"Consolidate swift.common.concurrency import statments","status":"MERGED","created":"2026-04-28 14:23:52.000000000","updated":"2026-04-28 22:53:09.000000000","submitted":"2026-04-28 22:52:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986543","meta_rev_id":"6182ca6145eaba04287cabbcbaa9d32b8fd35735","_number":986543,"virtual_id_number":986543,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-28 22:52:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-28 21:42:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-28 21:42:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-28 15:32:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-28 21:42:32.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"1790db007ee32e1376c7760d562360dcf5b1a7db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-28 14:23:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8b7f291f98871dfa50bea15b6213c81ed969915","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 15:32:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe534dbbc06f486e8c4041eb77a4a4fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1350616a60541b1aff755f8e4490468 : SUCCESS in 12m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/149ce093782c450ba85100032640d749 : SUCCESS in 43m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7bffb91316ff46d3bc2532ec5f6f12ac : SUCCESS in 1h 01m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/40efa03e66814782bf5c15a975053fe2 : SUCCESS in 31m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe1a4a2060144581875e61eb7fa0c197 : SUCCESS in 41m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/051c6c12e58d489182c80ca50bd48308 : SUCCESS in 10m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f8f157a4a4784affb693aa2d203227ae : SUCCESS in 19m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/548a45584f96423ba74f4e5b0ffa0402 : SUCCESS in 19m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5c06c84d05d3499da672a125a1c9e39a : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4ab096a93c3345d0a5925f11c77eb5f9 : SUCCESS in 23m 12s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/fbedcfb3480c4fd9b282a28ada179fd9 : SUCCESS in 20m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6f7aa4383fd84b5391b942aa83849817 : SUCCESS in 24m 48s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/b9e41ccd6b9a45d6b2ee3363f6e14bd5 : SUCCESS in 18m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/38b66d94dcb941928430f6fd58e3a551 : SUCCESS in 8m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9b38cc82b5cc4b81960b5d735a6677a3 : SUCCESS in 15m 47s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7b7b2e6be85049748655e9a9db197c68 : SUCCESS in 15m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cde06188e06b479ba55f9ecb5b2c4fed : SUCCESS in 3m 41s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/da780d56e02042cca669b16f62f9e6b2 : SUCCESS in 21m 44s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/af58175d1c474c9d9750f8fa87b61a96 : SUCCESS in 4m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/31c5144424a24593ba7b437449da6ca3 : SUCCESS in 35m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5d8634d8e7f349f49318116fb77d8cfa : SUCCESS in 51m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f9d552a0f9424a3b80ec83194b6e5ed4 : SUCCESS in 30m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ffa041b164294a77a4143748addaf351 : SUCCESS in 11m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84cbced63701457dbf7905f315347598 : SUCCESS in 4m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ec777b5883de46a6bbdca73bf05a800f : SUCCESS in 25m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48896eca5158412399053f49f2c7b3ee : SUCCESS in 43m 17s","accounts_in_message":[],"_revision_number":1},{"id":"b4a5c3e05a01923cce9b879977bbd5b4c44d5bb0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 18:37:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/596ce63bbe4040bab20b1a8b22cbacb7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/da96104e99b44f1fb66c6d288fb5acf5 : SUCCESS in 40m 39s","accounts_in_message":[],"_revision_number":1},{"id":"d48df42323fbd9a7f31733499a70e0306ade6cba","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-28 21:42:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c7dc089d4c77f322c0754a3f5a2d81e64c081b1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 21:43:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"51abb5d3398799d5a157abc0c1d0925322173460","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 22:52:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c4f360a6e0d45a099922eb477eee523\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb93a44f253948f49c8d3babd4cb5ed7 : SUCCESS in 4m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/45176886736b4522be86b23329931577 : SUCCESS in 31m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/501e1a17e5704d5bb0693313deffa30b : SUCCESS in 1h 01m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3643651005a34a15aca7c68723add3bf : SUCCESS in 56m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00238adb63ee482aa19423d3d3cbcd14 : SUCCESS in 36m 24s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/18181027a7c94415b380914b02ab5009 : SUCCESS in 6m 45s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8df7ca22becd44d39b5df415e8997a3e : SUCCESS in 10m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0a3acf4fe9364106bf3878876605e998 : SUCCESS in 8m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0e30ce0f38a640efbf97046c0149d51d : SUCCESS in 13m 07s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/06516c733d0a4cd7b8a4724cfe2b73f7 : SUCCESS in 20m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/01c0ddf3a6fe426ba2a12569005965e1 : SUCCESS in 15m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/986d5f91b8a3428aa161e8d228bb8b96 : SUCCESS in 15m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b0895817dde54d7da78a7897fe780992 : SUCCESS in 15m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ca5e04dd87a6411e9ac5b25c8ea37a2b : SUCCESS in 5m 40s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e43adc5bfefd4c2187c959bd2758dfdf : SUCCESS in 8m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/50a7daa14b3d47d08c235743d3bfb44f : SUCCESS in 56m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3d416851134948218c3f56c4a241f8a8 : SUCCESS in 52m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ae15b23f075c4f6ba7c856013abe920d : SUCCESS in 21m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d285a06a55644045a0f5448c402a60ee : SUCCESS in 29m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cb88d7ba36d4ec8a2e066104661f893 : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5f6b70f693734be086a69206cac4b8e2 : SUCCESS in 9m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1877cb93ce64780b098265454eb8443 : SUCCESS in 58m 29s","accounts_in_message":[],"_revision_number":1},{"id":"896fbe0906f7b0b3ef81ba22ba8c6dfe703e1ebf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 22:52:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6182ca6145eaba04287cabbcbaa9d32b8fd35735","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 22:53:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49fa440a6b0f49ceb0135bd52e89a76e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7cc5b70a02c4ec28ff545d3fe7ead55 : SUCCESS in 51s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f9f18bce07a441a1a0433acfa43181e8 : SUCCESS in 48s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ae4186bdb64c83455278010c6f965353ea126ff6","revisions":{"ae4186bdb64c83455278010c6f965353ea126ff6":{"kind":"REWORK","_number":1,"created":"2026-04-28 14:23:52.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/986543/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/986543/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/986543/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/986543/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/986543/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/986543/1"}}},"commit":{"parents":[{"commit":"bf9d86baea1ca158ed57c81285a8b86cb68cf34e","subject":"Be more targetted in our eventlet imports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf9d86baea1ca158ed57c81285a8b86cb68cf34e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-28 14:15:55.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-28 14:23:37.000000000","tz":60},"subject":"Consolidate swift.common.concurrency import statments","message":"Consolidate swift.common.concurrency import statments\n\nChange-Id: Icbe74d8742d169d7506979acdb70bb5df4e2607b\nRelated-Change: Id55312d835a1360e20809d79cec5435b5bf5adc8\nRelated-Change: Iaf74abeaaa8dd7a5321eb8da892b5e3f9d39a9e2\nAssisted-by: Claude Sonnet 4.5 \u003cnoreply@anthropic.com\u003e\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae4186bdb64c83455278010c6f965353ea126ff6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae4186bdb64c83455278010c6f965353ea126ff6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
