)]}'
{"id":"openstack%2Fswift~987677","triplet_id":"openstack%2Fswift~master~Iacdbbcad1f75e96e700b9388fae33a74d00504e1","project":"openstack/swift","branch":"master","topic":"p-timestamp-collision-patch","hashtags":[],"change_id":"Iacdbbcad1f75e96e700b9388fae33a74d00504e1","subject":"s3api: fix inventory config modified_time - extra test","status":"ABANDONED","created":"2026-05-07 14:51:46.000000000","updated":"2026-05-19 09:53:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d97188847377be8b401b1dfb12016411b9795471","_number":987677,"virtual_id_number":987677,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-05-07 16:11:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-07 15:29:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-07 16:11:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"159e57f2e1dbf0c9b7b5135ee557d36f4147e090","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-05-07 14:51:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2189ce610ea0fa08e9749c7d046e3731041c1a37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 15:29:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/842a5d09e10b4fdfa97afa8be051dd6f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d70ea8349b1c43ce8c968b5e2a60bc6f : SUCCESS in 34m 31s","accounts_in_message":[],"_revision_number":1},{"id":"87c616464481b7e6ea783b10f0cdec88747fadae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 16:11:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9b0f5cdf98d4759a44a8cfc0080285c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c74535be94b649ae92758ed5f354fe48 : SUCCESS in 10m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/e3ab3c15a2f0418a92ff78f7e697ba76 : SUCCESS in 1h 06m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/074436858b5e41cab84a6609f391915d : SUCCESS in 1h 10m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/52d6ee4fc77b4df1a1e1945d1863ffc9 : SUCCESS in 59m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c056622652e482485961a869cd4557b : SUCCESS in 1h 11m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/55c60ddd95da433ead33aae0c209fed9 : SUCCESS in 18m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/49f6a1c2c61d40cf84b5e8a02e24fe16 : SUCCESS in 20m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9662a712f18e452e9e0ac65a50cce22c : SUCCESS in 13m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5a79296b120c42ab82fd1c38df20de48 : FAILURE in 5m 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/184e64e1b0c14d4c8da5971d2b897453 : SUCCESS in 25m 47s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/56c6160b175c43f0a69c3c28dae928fd : SUCCESS in 23m 46s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b0e103cd0ee9481ba83156b8c2f1a311 : SUCCESS in 34m 43s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/87ce98ef257f4d848524bd1907605ba9 : SUCCESS in 21m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3a78423840f54d658c875d99f7453c6d : SUCCESS in 14m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cd59069c6ba24127a9665383228394a7 : SUCCESS in 17m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8db269f900ff48cca60b6ba0bc24f73f : SUCCESS in 14m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/61417e9f8d9d45319a9a76365e121dfc : SUCCESS in 4m 11s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/52cee879f6e343db88ce23c977c4b869 : SUCCESS in 17m 18s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c69402794da94e48b905e1d78ac76e43 : SUCCESS in 5m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d53e361800c4467b242ffb7ed55b4ae : SUCCESS in 54m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3584a376ec4c415a89381ee46d8126c5 : SUCCESS in 55m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/314ff67a816e4d808bf090f5918209d6 : SUCCESS in 57m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d505219dea04e0eb3eb36d52352ffed : SUCCESS in 31m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67b848a2032e4f4b9e1e42592bb678b0 : SUCCESS in 2m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1a916075d17c4bdcb3b966d44d032aed : SUCCESS in 17m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b0c8635ebf54aa29fe72b2093338e12 : SUCCESS in 1h 04m 27s","accounts_in_message":[],"_revision_number":1},{"id":"d97188847377be8b401b1dfb12016411b9795471","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-05-19 09:53:15.000000000","message":"Abandoned\n\nsquashed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fdcb77266c385a8c45493d274f4cd4342405b9c5","revisions":{"fdcb77266c385a8c45493d274f4cd4342405b9c5":{"kind":"REWORK","_number":1,"created":"2026-05-07 14:51:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/77/987677/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/987677/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/987677/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/987677/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/987677/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/987677/1"}}},"commit":{"parents":[{"commit":"8b5bf592a289581861b6ad4f979827f30d5054db","subject":"s3api: fix inventory config modified_time","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b5bf592a289581861b6ad4f979827f30d5054db"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-05-07 14:50:20.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-05-07 14:50:20.000000000","tz":60},"subject":"s3api: fix inventory config modified_time - extra test","message":"s3api: fix inventory config modified_time - extra test\n\nAdd another test.\n\nAdd a comment re use of time()\n\nRemove unnecessary mock_timestamp_now\n\nChange-Id: Iacdbbcad1f75e96e700b9388fae33a74d00504e1\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fdcb77266c385a8c45493d274f4cd4342405b9c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fdcb77266c385a8c45493d274f4cd4342405b9c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
