)]}'
{"id":"openstack%2Fswift~987731","triplet_id":"openstack%2Fswift~master~I4722068bb7d44eae3b309c3ed8ae7d8da1b826ae","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-05-07 19:58:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4722068bb7d44eae3b309c3ed8ae7d8da1b826ae","subject":"container-server: add test for invalid s-p-i header","status":"MERGED","created":"2026-05-07 17:29:45.000000000","updated":"2026-05-07 20:00:39.000000000","submitted":"2026-05-07 19:58:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987731","meta_rev_id":"ee805eecac4bbffcb6f92e2938e3e9b461f2e525","_number":987731,"virtual_id_number":987731,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-07 19:58:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":2,"date":"2026-05-07 18:17:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":1,"date":"2026-05-07 18:17:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-07 18:07:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-07 18:17:03.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2026-05-07 18:32:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9e6cd2d56147001920f15936111cf72d7d67d6fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-05-07 17:29:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"020c0aa7dd03074ccd8b943a151038c8caae13da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 18:07:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7d3c757ca654f908bd3852355e7d884\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a336f0c7dc06459a903e8ee2fab8ba29 : SUCCESS in 30m 23s","accounts_in_message":[],"_revision_number":1},{"id":"de05eca58598433ae1067ad64d4ac5f5cf23f517","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-05-07 18:17:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8a69049fe65c3cc6a35a312c2f7920de485ed2c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 18:32:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63a5037c11f24461b8ddc50e17d7d4ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8caf7eb3646642c881519705ebd6d5e8 : SUCCESS in 10m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2f73c6cd30594fb3905bb92a5357bcc3 : SUCCESS in 15m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/84c90ac4e697493cae42e02dc58249bb : SUCCESS in 17m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d413f3ae5f1e418b9c1118a842b1c0ee : SUCCESS in 18m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6b6a7529cb9d4c88b608bcd46bb1f4f0 : RETRY_LIMIT in 1m 51s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b927bac4b19e45b4b70351b52bc2a92d : SUCCESS in 10m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/50fb2552737745928aaf1dc47001e275 : SUCCESS in 21m 54s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9dd2f0a05d5c4cf48b6a3c210af7e3b0 : SUCCESS in 28m 13s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/b7a60613dc38452b9c1750f45689639d : SUCCESS in 17m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9fb555e4204d48ae91f6b385e866976e : SUCCESS in 13m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2db74570cd6e4648afa07483a61d68a6 : SUCCESS in 14m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d636ebc22bee48a1b624cd3517d92973 : SUCCESS in 17m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/af576e05285649abbd652b926d337f0b : SUCCESS in 54m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a9442403772a4cc19e41d4d763d07d13 : SUCCESS in 57m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e42ed224ac9b4f678c62c6fcd9e973da : SUCCESS in 29m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e984736fa004414499de1c271a2a1610 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a3580c73a44d4a258aadc593787a25da : SUCCESS in 9m 36s","accounts_in_message":[],"_revision_number":1},{"id":"d2aa3062edfe4f698a0decadc77a7259a9ebdaea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 18:33:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b220472660b142719ed00968bc5f1145fe3a8ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 19:58:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24d82a5bb26d4d75ae9599a285ff60c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99ac0060d61b4db893f396ba40d82508 : SUCCESS in 10m 08s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c89846fbb7354287a315b09bdfaf90ab : SUCCESS in 6m 20s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/47fcbdacbb3247c4aa5509483fa9633d : SUCCESS in 14m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f67afc5c08324c359de2b52dee08efd7 : SUCCESS in 29m 14s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/27da4bd23f3d4f4aa6155a82daf28224 : SUCCESS in 26m 10s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/d83263098cc140bea9d18f595f2254eb : SUCCESS in 16m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2ccdfd9e828d4bb3ab8545b10021bd1b : SUCCESS in 10m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/48f0797506554265bbb970a6c95cebb4 : SUCCESS in 13m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a907c67e519b4be39e7e05a289b0ea66 : SUCCESS in 15m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bcc187d9bbf34f80a89270d50404afc6 : SUCCESS in 4m 32s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7bfae62be628465d9947980e800db329 : SUCCESS in 9m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6e5307afc15a40b999008c03c204d3b7 : SUCCESS in 57m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d6030fe112944ba1b41ed36acee10651 : SUCCESS in 53m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/068aa73852b34e9a988948ea0512f7bf : SUCCESS in 29m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25da1ed3fddf4c52a036f9575879b8b1 : SUCCESS in 4m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aebc9b29bef646a9b7bb696da27ff9e4 : SUCCESS in 12m 44s","accounts_in_message":[],"_revision_number":1},{"id":"11e1bf7b640864fa2e5ec4d370e845710dee269e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 19:58:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ee805eecac4bbffcb6f92e2938e3e9b461f2e525","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 20:00:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c365447fc664667acc3d99a672b0acf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5184e9c0de314c9db7c6e6253ffcf099 : SUCCESS in 58s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f1f56a10986d4a0c8fc45166cd107cac : SUCCESS in 50s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e3c9bb95be42b14b74a1add2a9eefd19f4ba14ba","revisions":{"e3c9bb95be42b14b74a1add2a9eefd19f4ba14ba":{"kind":"REWORK","_number":1,"created":"2026-05-07 17:29:45.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/987731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/987731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/987731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/987731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/987731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/987731/1"}}},"commit":{"parents":[{"commit":"43965416e7f09fdf8e2c1d62b01e5f74a26ab437","subject":"container: Split DELETE code paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/43965416e7f09fdf8e2c1d62b01e5f74a26ab437"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-05-07 17:27:54.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-05-07 17:28:44.000000000","tz":60},"subject":"container-server: add test for invalid s-p-i header","message":"container-server: add test for invalid s-p-i header\n\nChange-Id: I4722068bb7d44eae3b309c3ed8ae7d8da1b826ae\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3c9bb95be42b14b74a1add2a9eefd19f4ba14ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3c9bb95be42b14b74a1add2a9eefd19f4ba14ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
