)]}'
{"id":"openstack%2Fswift~988163","triplet_id":"openstack%2Fswift~master~I4b9b1f6440c5d22af396bdc4ea8360236e1b6a11","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-05-27 17:03:03.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4b9b1f6440c5d22af396bdc4ea8360236e1b6a11","subject":"s3request: detect incomplete chunk in ChunkReader","status":"ABANDONED","created":"2026-05-11 18:22:05.000000000","updated":"2026-05-27 17:03:03.000000000","total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"5f09381808e4d999e8acfa585002d2d8cd5e5c53","_number":988163,"virtual_id_number":988163,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-05-11 19:37:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-11 18:59:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-11 19:37:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-11 19:51:17.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"}],"messages":[{"id":"bef800b110cc46c36c181b8f3406d84bd5a1c19c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-05-11 18:22:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9ac76e85b9e525d373854dfdaa45c235fe1b4e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 18:59:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6061f07f4f641518b698a3d50a6d859\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5140df97b8914daabf0dba3681b94085 : SUCCESS in 35m 46s","accounts_in_message":[],"_revision_number":1},{"id":"a99834f97b3faac6d238aba45c74838592de8416","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 19:37:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f8d47669ffa4997a9b2ee99b9deab03\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f809d87eb6f4c389e7fb56731029d43 : SUCCESS in 10m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/211ac0be4e084de5b12ac56b54b5de53 : SUCCESS in 1h 09m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eb8a0454393d4b108928bee3260e7df8 : SUCCESS in 1h 03m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3053d929d12b40e9bd61bbcff6ea9423 : SUCCESS in 31m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8940830c4146429cb374e1ff28f57924 : SUCCESS in 1h 05m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/28c99ba348f14690806ac92e5b6f9a17 : SUCCESS in 19m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/817bc756d5684410a0c34b3a0234d2a3 : SUCCESS in 14m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/53d9059af5f54de2a914faec0525dbb3 : SUCCESS in 20m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f935f52de27f4cf3a689ff6d082099d5 : SUCCESS in 8m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5fe2525956a84a6595c1b818f5ae9626 : SUCCESS in 19m 08s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/77d0d69e401d47d8b06c6463f182406c : SUCCESS in 17m 29s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/490a99214ad64ecf8b52a83deb4b56a0 : SUCCESS in 49m 25s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/9f36d67ee54a4d98bea9d3ab72326510 : SUCCESS in 19m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d98b88a35a37490cb3b4b71614e161c3 : SUCCESS in 15m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7f5390b83d2542e7be5bdcd228c1dbdd : SUCCESS in 16m 03s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/42889ac317ef4845a1dc32f83cde93e6 : SUCCESS in 15m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3528add232a143b59834c417e0d74d20 : SUCCESS in 3m 39s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bd51f655ab7b49f8970d70fc4e2b4290 : SUCCESS in 16m 43s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2159ddb4fa4246799bb3b807372d0265 : SUCCESS in 8m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/030699ec67db44139a5f3287a1b235a2 : SUCCESS in 35m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/220bde38c5764d4cb1316292a6e3bdcb : SUCCESS in 52m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7662ce7861da48e0a263500aa8451d43 : SUCCESS in 56m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e9a5e63aba74986abea62f5b29aeee1 : SUCCESS in 27m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6230e21e8c148178c3fc08a22074736 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/027090b278024d428b47fd31e71fdbb5 : SUCCESS in 19m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78da578d0e504fb586295fb4d1add728 : SUCCESS in 58m 58s","accounts_in_message":[],"_revision_number":1},{"id":"65c41404206eda89ad91bb5d5d7552cce0d62e9d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-05-11 19:51:17.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5f09381808e4d999e8acfa585002d2d8cd5e5c53","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-05-27 17:03:03.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"06897ea36eee975d0452a618cf0750904c64babe","revisions":{"06897ea36eee975d0452a618cf0750904c64babe":{"kind":"REWORK","_number":1,"created":"2026-05-11 18:22:05.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/63/988163/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/988163/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/988163/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/988163/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/988163/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/988163/1"}}},"commit":{"parents":[{"commit":"cccb6809e662bcd3925565883e0fbd043a1e72c9","subject":"s3api: Add functional test for aws-chunked truncation bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cccb6809e662bcd3925565883e0fbd043a1e72c9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-05-11 18:15:11.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-05-11 18:15:11.000000000","tz":60},"subject":"s3request: detect incomplete chunk in ChunkReader","message":"s3request: detect incomplete chunk in ChunkReader\n\nI wanted to see how this would work out, but I\u0027m 50/50 about this\nchange and will appreciate other opinions:\n\nOn the one hand it seems appropriate to encapsulate the incomplete\nchunk detection in ChunkReader; as a class, ChunkReader ought to\ndetect short reads.\n\nOn the other hand, I could understand an argument that the existing\nfix is more obvious.\n\nChange-Id: I4b9b1f6440c5d22af396bdc4ea8360236e1b6a11\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/06897ea36eee975d0452a618cf0750904c64babe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/06897ea36eee975d0452a618cf0750904c64babe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
