)]}'
{"id":"openstack%2Fswift~988199","triplet_id":"openstack%2Fswift~master~Iceeffbf694b26e043b9ceb57eea94376bed41575","project":"openstack/swift","branch":"master","topic":"aws-chunked-recv-loop","hashtags":[],"change_id":"Iceeffbf694b26e043b9ceb57eea94376bed41575","subject":"alternative regression test case in TestStreamingInput","status":"NEW","created":"2026-05-12 04:02:33.000000000","updated":"2026-05-12 05:47:27.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"298f26acf85b8c091174d0f88ff004322ab58da5","_number":988199,"virtual_id_number":988199,"owner":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-05-12 04:50:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-12 04:50:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"564d83cb2873828de0ea29dde541f976d22fd3a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-05-12 04:02:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"192d87169b15ba4fce1f959baf41f22dcb12fde8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-05-12 04:14:06.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2e65dc2f61a6402a71b64d8a7521d5bb88879e5e","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-05-12 04:18:04.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9a2d64759304eb6946f0c7847a6c5bc74cbc8297","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 04:50:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b19524d7e35146e48d5a8d5fa7196f2b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f2440e3a19b410291bf513bed95efae : SUCCESS in 7m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e9348fba56a04347a77c452ca033407d : SUCCESS in 20m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/39e463fc48f44c1daab6b3248900b6a1 : SUCCESS in 20m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a5794c025a6645398b78a4351a05c2a6 : SUCCESS in 10m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/87036c279ff84e2485bf496ad6eb1fa1 : SUCCESS in 6m 45s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fe8b1537870b4b4695ed17e8e1a11e1f : SUCCESS in 25m 29s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6aed4799aae64e97be51d9d1944cac1b : SUCCESS in 16m 42s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/88111d62491843e1807cafb3aeb16d4c : SUCCESS in 19m 20s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/5fc0e71e9bd54404bfd94d8e5f1f33ca : SUCCESS in 18m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ee7d2789c5f7439596d0439b1ca575e0 : SUCCESS in 12m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d0810a9f763448c7932f5ce551a03588 : SUCCESS in 14m 45s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7393859b85e24618906a77199361694b : SUCCESS in 12m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9a32c9ef9b9f4964be9dbe500d1d2912 : SUCCESS in 34m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d7470db29610409c9aa7134217eaa2f3 : SUCCESS in 21m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4a9299e15b3475da2e38b35a1235378 : SUCCESS in 21m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0338bb2fcbe42b29a32e815766e4b70 : SUCCESS in 5m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2bfdd03a2deb4fbcb0fc11f360638a90 : SUCCESS in 14m 27s","accounts_in_message":[],"_revision_number":2},{"id":"298f26acf85b8c091174d0f88ff004322ab58da5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 05:47:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60d8f4bc79954e6a95c4a04614c0f7bf\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7e80229638df4cbabb2f83d5a97fd5fc : SUCCESS in 29m 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"32625a19fa1c2ff022969ab5856fa18108017de7","revisions":{"6c81ee7038fadb8bab400c5ff1fb426b3ce19762":{"kind":"REWORK","_number":1,"created":"2026-05-12 04:02:33.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/99/988199/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/99/988199/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/988199/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/988199/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/988199/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/99/988199/1"}}},"commit":{"parents":[{"commit":"5ad23e2f6b91f75a8fe0008c8dfc9fb3d4bb6ca3","subject":"s3api: Error on truncated aws-chunked input","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ad23e2f6b91f75a8fe0008c8dfc9fb3d4bb6ca3"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-05-12 03:58:57.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-05-12 03:58:57.000000000","tz":-420},"subject":"alternate regression test case in TestStreamingInput","message":"alternate regression test case in TestStreamingInput\n\nChange-Id: Iceeffbf694b26e043b9ceb57eea94376bed41575\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c81ee7038fadb8bab400c5ff1fb426b3ce19762"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c81ee7038fadb8bab400c5ff1fb426b3ce19762"}]},"branch":"refs/heads/master"},"32625a19fa1c2ff022969ab5856fa18108017de7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-05-12 04:14:06.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/99/988199/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/99/988199/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/988199/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/988199/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/988199/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/99/988199/2"}}},"commit":{"parents":[{"commit":"5ad23e2f6b91f75a8fe0008c8dfc9fb3d4bb6ca3","subject":"s3api: Error on truncated aws-chunked input","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ad23e2f6b91f75a8fe0008c8dfc9fb3d4bb6ca3"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-05-12 03:58:57.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-05-12 04:13:35.000000000","tz":-420},"subject":"alternative regression test case in TestStreamingInput","message":"alternative regression test case in TestStreamingInput\n\nChange-Id: Iceeffbf694b26e043b9ceb57eea94376bed41575\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32625a19fa1c2ff022969ab5856fa18108017de7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32625a19fa1c2ff022969ab5856fa18108017de7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
