)]}'
{"id":"openstack%2Fswift~988362","triplet_id":"openstack%2Fswift~master~I924ea2efe99ffa17eed7f7649b3fd2c685091fdf","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-07-10 07:37:18.000000000","reason":"\u003cGERRIT_ACCOUNT_6968\u003e replied on the change","reason_account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-07-10 09:29:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I924ea2efe99ffa17eed7f7649b3fd2c685091fdf","subject":"Quiet eventlet deprecation warnings.","status":"MERGED","created":"2026-05-12 23:00:09.000000000","updated":"2026-07-10 09:30:58.000000000","submitted":"2026-07-10 09:29:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988362","meta_rev_id":"3b42bccc2aa8f57bb4d95dedaa6bb720edc775a7","_number":988362,"virtual_id_number":988362,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-10 09:29:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":2,"date":"2026-07-10 07:37:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":1,"date":"2026-07-10 07:37:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-12 23:45:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-13 00:11:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-13 17:34:19.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"}],"messages":[{"id":"8312cc6d331503ed4c4364c1f2ec8904abb798f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-05-12 23:00:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22157ec847b886dcb914a139b2a9df1f00b753ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 23:45:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f869fe7c573d4d7c88e927663998eb58\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/60a10fd3022141a8bcdd157a9f874016 : SUCCESS in 31m 35s","accounts_in_message":[],"_revision_number":1},{"id":"773de322fc69d50f7a45797aebdca46e05078f9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 00:11:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/131abe6f5dc54b8b8fda00007e762c86\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5abf5f124a9446da9e2c6a249040b46f : SUCCESS in 9m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0a29ab80f6a43e2868db67d5f3e9e3c : SUCCESS in 8m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/11954c0c0a2742a5a85df89f3d5ab75a : SUCCESS in 41m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9ab178ebfabf4970993ff56f00425eda : SUCCESS in 1h 02m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a710a17a4e7846bbac665d45123dbe1c : SUCCESS in 37m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fcd63e1c10694aac935889600ea0754f : SUCCESS in 1h 05m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b1f82cf5cf9481aa315b3281c7b46f6 : SUCCESS in 10m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7b7a785bdf7e400faddaa2c5a28797b9 : SUCCESS in 16m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/078e1ee8cbf04deea179a7640bfccdf4 : SUCCESS in 20m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/82b24416e3d746a8a89c43c37b1a0756 : SUCCESS in 4m 27s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7c21e591f65841dc918c08cfae509a9a : SUCCESS in 25m 39s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c61cfc2c559c44d1878a95f02a27f4cb : SUCCESS in 11m 35s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/75662eee1949451ebbbfb7f13fe11f1d : SUCCESS in 33m 21s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/896d92398ab34ce3a252b666417c8419 : SUCCESS in 27m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3bb480cffd88443786d33d3ef4208ad1 : SUCCESS in 9m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/808f41b4f65a40baabdaed37a4f33142 : SUCCESS in 9m 48s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/66cf8ab3a38a4a19be54e6ab2a6383de : SUCCESS in 8m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1d3cb26893014f7a9ca68139500bc902 : SUCCESS in 5m 34s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/49651a6bc4da4ae290ef99054e521ada : SUCCESS in 20m 31s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/62ca504234554de7b276891e89ace884 : SUCCESS in 5m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/108a92113be94ae0bed386b1769618d8 : SUCCESS in 49m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5f54f0ca06a84571b8ff583edd2365fb : SUCCESS in 36m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0687227813624e5391a2b4f03ee97bdf : SUCCESS in 54m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e5ba0d0f3944768b18f34f5c92541a5 : SUCCESS in 29m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8596cd54fa7e4495b4a11a9207722c42 : SUCCESS in 3m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5eb3b918560c41c089bdde0b67a4a025 : SUCCESS in 15m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1d751937dfa4baa932f8cf6981d556f : SUCCESS in 39m 56s","accounts_in_message":[],"_revision_number":1},{"id":"44d24821d06610075605af173771fcb0477651c5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-05-13 17:34:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b667409418b00694af992572400979830d8892c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-09 21:47:02.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"205efc4b7dfca51d96bfa5bb139d4ab4bbf2c2b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 23:01:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46ff4f760bd14d05a761b1028034daa9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4228001cc13c4bbca19a00ba452707f8 : SUCCESS in 1h 10m 47s","accounts_in_message":[],"_revision_number":2},{"id":"8ada9ebfb399acc6c374a5bc7fca64a1114c80d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 23:02:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/148f1b4c52704fa288eccb2860663fef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac4a028b3f9d4b0b8391520f4c7ed85e : SUCCESS in 10m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bec65574f46499296688e1ed7a7e24f : SUCCESS in 5m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/7a1e210812da430481e798331193b02c : SUCCESS in 37m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/706f7f2846dd4ed2bdd09802340883b8 : SUCCESS in 1h 10m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cafb47129b7b4e19af5357472bf796e0 : SUCCESS in 54m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b3a03d4a7a44c2bbefbb649f635789e : SUCCESS in 1h 12m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/05774c51d82b4451870a6da90bd6e839 : SUCCESS in 20m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/789e6617c70f465b8943454ff47aac46 : SUCCESS in 12m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8e71ec2bd24c4439afe80d50cddd30ad : SUCCESS in 22m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5e350f36b09947a19d00838fa02149fd : SUCCESS in 5m 17s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/48e414982b7a4703a297c9b1c016a99a : SUCCESS in 24m 10s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b1149c71f3c5465fa0a332d25e844499 : SUCCESS in 21m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f10d9208795d4865bbbb9bbb05878ffc : SUCCESS in 20m 55s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/82fc4b9c1afc4c02868d65ba306c09b9 : SUCCESS in 10m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7fc4bcf8c1614c0abd83914453e2540e : SUCCESS in 9m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d3534c4f2133428586b1551862af5504 : SUCCESS in 15m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/25223e7108244b88948bb208981d1790 : SUCCESS in 15m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/51c63fac78d540da99bbb40d615a3031 : RETRY_LIMIT in 5m 42s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e869e800dd4a46e4a578050079485c7d : FAILURE in 23m 16s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9a8eb7cda96f49d3b56adfed9a6a0d5a : SUCCESS in 4m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/18d1c559f17348caadfc6842e745b2b6 : SUCCESS in 59m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1fd718765f824cd2a0a1b4cb0560bcf6 : SUCCESS in 52m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d829d329620f43e8981ec8bb9d5b5b13 : SUCCESS in 35m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1d8aa0e577e48d7b8c484091fa572e1 : SUCCESS in 29m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ced3bd89dd564e9f960f2c5d9c3acad8 : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2e20c938e15949478f17a01d643d4293 : SUCCESS in 17m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70b66fe4830a4742838bd78188a0656b : SUCCESS in 1h 01m 43s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"04314238a58f8db56692f3b03d8d11d9c6d02d92","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-10 07:37:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a616b34ad1b35a557e1a27f3d98c4b4759e24227","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 07:37:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cb50435ad856c7cfa7d3409cf80ad8aedf766c4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 09:29:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01eab2e555b84d5aab6c7e7ff4f0a74b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c530138b75734cb89cc977b5cc19b78b : SUCCESS in 11m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbce6eb650314dd8925b70a18998ce5f : SUCCESS in 8m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/b77d5e81d25b48b19e57150c38a56f22 : SUCCESS in 1h 05m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fdf28250059e4ad696bc37b6b630a93f : SUCCESS in 1h 03m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bd5402aa4c674266aab6eba8d7a9ced6 : SUCCESS in 35m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6681ceb08f2a4914ad43555330477a22 : SUCCESS in 46m 09s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/35b8e781bc4647f997873b2e2e314db2 : SUCCESS in 7m 06s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1188b20a86cb428583fb6a82109cd682 : SUCCESS in 24m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/825aff4ec4b5448a8cbc7c7476b1be6f : SUCCESS in 23m 50s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5f2103eaec1b4995848b4e0536d44422 : SUCCESS in 26m 49s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/049baf62735e4bb7a81cca6719241b1a : SUCCESS in 15m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e97afbe0e42f455eae1b7601c9dea01b : SUCCESS in 15m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3ee0e28e2c3549ada56c72d20b7a58b8 : SUCCESS in 12m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0a509e10e02b4107b89e015be2668d54 : SUCCESS in 15m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f99c5c48606c4f559cab2af3186f1792 : RETRY_LIMIT in 4m 16s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a269c1c3d954431e9e2533bd00712519 : SUCCESS in 9m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/131d8cae2bbd4a2788a626f307c427a1 : SUCCESS in 1h 01m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/641c802b53b244b79ad215797d3ab17c : SUCCESS in 52m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a4d804d8f16c4b83a65aa21b2440e62e : SUCCESS in 55m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1c55cc25fff49d283fdd4acd82b4cd9 : SUCCESS in 30m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b9e1b707f9847da9f5e58891bc2f656 : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c59b3e2797ab4f748e94fccdd086cc20 : SUCCESS in 24m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68791feecda5419e993db55c0af54cd5 : SUCCESS in 44m 27s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-upload-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-upload-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-upload-image.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"fd65b69a5514e3ef9e8dd86fb5be4d7e140d5e8a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 09:29:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3b42bccc2aa8f57bb4d95dedaa6bb720edc775a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 09:30:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c26191eadbe43069d33396ab6dd41be\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc49ad8d25074ba78f1b01575f50c628 : SUCCESS in 53s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/93fc9d06efa04ccca29e92d7a2899240 : SUCCESS in 47s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f18d0e76bab7661fc9c2847ef65b9e1a10257e4e","revisions":{"543c11cc3a5584c6ad7cb850fc64b7a4edb60c91":{"kind":"REWORK","_number":1,"created":"2026-05-12 23:00:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/62/988362/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/988362/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/988362/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/988362/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/988362/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/988362/1"}}},"commit":{"parents":[{"commit":"33967853276f0987a58e5b3aca450173adadc064","subject":"timestamps: check offset is an integer during construction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/33967853276f0987a58e5b3aca450173adadc064"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-05-12 22:08:15.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-05-12 22:57:24.000000000","tz":-420},"subject":"Quiet eventlet deprecation warnings.","message":"Quiet eventlet deprecation warnings.\n\nThere isn\u0027t much point to us letting these through until we have an\nalternative option available. Even then, we\u0027ll probably want to emit\nour own warning with more detailed migration information.\n\nSee also: https://github.com/eventlet/eventlet/commit/01a3da5f\nChange-Id: I924ea2efe99ffa17eed7f7649b3fd2c685091fdf\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/543c11cc3a5584c6ad7cb850fc64b7a4edb60c91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/543c11cc3a5584c6ad7cb850fc64b7a4edb60c91"}]},"branch":"refs/heads/master"},"f18d0e76bab7661fc9c2847ef65b9e1a10257e4e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-07-09 21:47:02.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/62/988362/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/988362/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/988362/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/988362/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/988362/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/988362/2"}}},"commit":{"parents":[{"commit":"8be4f02ef61a09b044965134b9b7f95a5eb88bc1","subject":"crypto: Fix listings when pipeline was misconfigured","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8be4f02ef61a09b044965134b9b7f95a5eb88bc1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-05-12 22:08:15.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-09 21:46:55.000000000","tz":-420},"subject":"Quiet eventlet deprecation warnings.","message":"Quiet eventlet deprecation warnings.\n\nThere isn\u0027t much point to us letting these through until we have an\nalternative option available. Even then, we\u0027ll probably want to emit\nour own warning with more detailed migration information.\n\nSee also: https://github.com/eventlet/eventlet/commit/01a3da5f\nChange-Id: I924ea2efe99ffa17eed7f7649b3fd2c685091fdf\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f18d0e76bab7661fc9c2847ef65b9e1a10257e4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f18d0e76bab7661fc9c2847ef65b9e1a10257e4e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
