)]}'
{"id":"openstack%2Fswift~988631","triplet_id":"openstack%2Fswift~master~Id36a681f512b0eb7ee70a4b645cace107038c199","project":"openstack/swift","branch":"master","topic":"p-timestamp-v2-fixes","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-05-15 08:39:30.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"Id36a681f512b0eb7ee70a4b645cace107038c199","subject":"sq: timestamps: add Timestamp.version property","status":"ABANDONED","created":"2026-05-14 13:56:55.000000000","updated":"2026-05-15 10:47:37.000000000","total_comment_count":6,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"9e0bd658787fb82cf558c630ed20441231bb3135","_number":988631,"virtual_id_number":988631,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-05-14 15:19:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2026-05-14 15:33:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-14 15:19:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-14 15:33:47.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"58c71bef0c6ed1d8aa207c8d80e1d4ff4e4a9508","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-05-14 13:56:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2e97c1c8b30b5a03413b3b28a4c742a12c876d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 15:19:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3de0a4aed63a423eacee01d7d9d46183\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe83ff4217a64b9c9969574463cebb9a : SUCCESS in 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/321f0dff3ce846d5ba3c187b0ecc3420 : SUCCESS in 8m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/4d893cf1ea814e0ab9113c7abf773d7e : SUCCESS in 45m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/854109a611c942d2b954b97512352355 : SUCCESS in 45m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5ec4bb5b57544a9985507496145bee52 : SUCCESS in 51m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0fc3a1a34f5b45ecae274861861baac3 : SUCCESS in 1h 12m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/295ba64458c542deb4ee0445ed762b38 : SUCCESS in 20m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4efd539852774d0c97468a17258780c8 : SUCCESS in 22m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0fc04d5efa8f4d528ea0b2a21e38e57b : SUCCESS in 21m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fcb673f573204d5fadecafdecffdc44e : SUCCESS in 6m 18s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/512baf93dc254c7daaf8d0b67bc78bcd : SUCCESS in 23m 13s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1bd7a66f5c964ae780dfdeba583e2fd8 : SUCCESS in 19m 29s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a5badf26b5c44daab7b45ab7fd6669e4 : SUCCESS in 35m 46s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/bd4e28cd3cf949f5b05c5d9557aa2eb8 : SUCCESS in 9m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e2cfa0e804ee4313b4f9ab419f1ae657 : SUCCESS in 16m 16s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b78e7571837c4e4ab91c380298c0e27d : SUCCESS in 15m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e78594896e354840bfd539df0565fcb1 : SUCCESS in 17m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5793b4581bf646409c588af1f049f5d4 : SUCCESS in 6m 56s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/949d9cb7f839412f9cb5f0f1b80da138 : SUCCESS in 19m 36s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e081bff2146b49318ade0605e75896f0 : SUCCESS in 6m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d57e5d8ea995407aa8460f2eff255255 : SUCCESS in 51m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/58c914be9b0648e78a1282a75f22bc86 : SUCCESS in 27m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/abd568ca681d45989fd31b7f677a0d11 : SUCCESS in 38m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64725edd381f4ec3bc0ec334c2a5e8fc : SUCCESS in 31m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6154075b50ee4781857448802b9655db : SUCCESS in 2m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/23e923cf94714de9821ee5216f7b89a2 : SUCCESS in 28m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b0d23bde1c945279ace75f100393bcb : SUCCESS in 1h 02m 41s","accounts_in_message":[],"_revision_number":1},{"id":"d2d1f6fc4b1bb743872797ca3267d3ab8a9e5629","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-05-14 15:33:47.000000000","message":"Patch Set 1: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0caff2fc83979b267bec58e19711b52db6760671","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 17:38:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/096b8072148d4b41a3583a5ef8c311cd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a8660efe0b84469596b5ee6acb2161f3 : SUCCESS in 1h 19m 38s","accounts_in_message":[],"_revision_number":1},{"id":"c8e6bb6f3b835b19a15019450f7e1e84cb75b76c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-05-15 08:39:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9e0bd658787fb82cf558c630ed20441231bb3135","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-05-15 10:47:37.000000000","message":"Abandoned\n\nsquashed to parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a20f20027d5529d242fceed9c6be632ff3aa89c4","revisions":{"a20f20027d5529d242fceed9c6be632ff3aa89c4":{"kind":"REWORK","_number":1,"created":"2026-05-14 13:56:55.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/988631/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/988631/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/988631/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/988631/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/988631/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/988631/1"}}},"commit":{"parents":[{"commit":"782c7e27bee3b9c0dc2b3ba66c87adeffbc01776","subject":"sq: timestamps: simplify offset validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/782c7e27bee3b9c0dc2b3ba66c87adeffbc01776"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-05-14 10:57:26.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-05-14 13:40:00.000000000","tz":60},"subject":"sq: timestamps: add Timestamp.version property","message":"sq: timestamps: add Timestamp.version property\n\nChange-Id: Id36a681f512b0eb7ee70a4b645cace107038c199\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a20f20027d5529d242fceed9c6be632ff3aa89c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a20f20027d5529d242fceed9c6be632ff3aa89c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
