)]}'
{"id":"openstack%2Fswift~988632","triplet_id":"openstack%2Fswift~master~Ibedc4a8191c4778a6b100f1e91ba6998df2e0f28","project":"openstack/swift","branch":"master","topic":"p-timestamp-v2-fixes","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-05-15 10:38:48.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-05-15 10:48:30.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ibedc4a8191c4778a6b100f1e91ba6998df2e0f28","subject":"sq timestamps: simplify offset property","status":"ABANDONED","created":"2026-05-14 13:56:55.000000000","updated":"2026-05-15 10:48:30.000000000","total_comment_count":12,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"87d155eeb2a32c68e057eb4b042bcd8a92dc9aa7","_number":988632,"virtual_id_number":988632,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-05-14 15:06:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2026-05-14 15:56:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2026-05-14 15:56:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-14 15:06:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-14 15:56:54.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"912840ff6d9169e6c7893368c2530f31ba45a825","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-05-14 13:56:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3318de03c57f8e25d978090275a4b2b82f9a3072","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 15:06:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d9ec630d77648d5bbe2cb36139425d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a4c329dfaa14226a259d42d9cf55d67 : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0417efd3522a40d19876769b9570a019 : SUCCESS in 5m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/0c6cbfff62d6494cba2c8c34b112e7b1 : SUCCESS in 52m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1c085129d702459a8dd8a8968d62e15f : SUCCESS in 1h 01m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c62c91fa96704756b739412592729bd0 : SUCCESS in 47m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b89a422d1af49eb8df6282a73f1c7c4 : SUCCESS in 1h 03m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a824febc064845c883ea6bd9f6e068fb : SUCCESS in 10m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f4740bc7345242cfb09d9f1de051fdfe : SUCCESS in 20m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/53e17f97619042298327b3fdf79c47a4 : SUCCESS in 21m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/28c2432596bd40e9955ec4b34c32906f : SUCCESS in 8m 50s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c7dcf8dbe5ad4bd58fd580b2da58d679 : SUCCESS in 25m 42s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d78a26d2571e47db9a0fee27f5ae9a59 : SUCCESS in 23m 15s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/dff5f19e6a5c445380cb4096e087854b : SUCCESS in 37m 34s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/ea838e7af1434d70b37fb1bb278b58d7 : SUCCESS in 25m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/07a85e082380465ea29b7e672002153b : SUCCESS in 13m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/22d956d5afc54b3e9078da94e0babc00 : SUCCESS in 16m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b1971cf5b0414013bfbf146f35833c8b : SUCCESS in 18m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c19f9981d5944ad5a77b59b520cc0ba6 : SUCCESS in 6m 52s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ea57dd42da4e4f3386104a1290b30064 : SUCCESS in 18m 53s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/41de8a74e2ea40248822ac6760118c77 : SUCCESS in 8m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac2f77cf230b4acb9619882a674f86f1 : SUCCESS in 51m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/768691672f0045d4ae7e8f5a29defb41 : SUCCESS in 34m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f83c823fb55b4d5a8fd645d69f109167 : SUCCESS in 57m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b3cf36997d340c886cc0112b840b900 : SUCCESS in 22m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7b521080aee429db2666a76ac354e5d : SUCCESS in 5m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5615c076486c4a7590b21c1560798dea : SUCCESS in 18m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/724dfa8fa87c4c5f9b86dc5d70388888 : SUCCESS in 47m 34s","accounts_in_message":[],"_revision_number":1},{"id":"da736e5acdec5cf64cea45deffea2d74c19a81ac","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-05-14 15:56:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0d163d768504ac0fe62f300da29357deab2bcf95","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 18:29:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65b061fec0f148f79fa494041f4c8463\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5a182bd0848646ca8cce08496577bce3 : SUCCESS in 53m 48s","accounts_in_message":[],"_revision_number":1},{"id":"d21826ebb7ed0d63215121a583b5e6103c4745b8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-05-15 10:38:48.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"87d155eeb2a32c68e057eb4b042bcd8a92dc9aa7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-05-15 10:48:30.000000000","message":"Abandoned\n\nsquashed to parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eae8b6b7a6b2c04cd857783262b1bc584bac8c62","revisions":{"eae8b6b7a6b2c04cd857783262b1bc584bac8c62":{"kind":"REWORK","_number":1,"created":"2026-05-14 13:56:55.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/32/988632/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/988632/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/988632/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/988632/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/988632/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/988632/1"}}},"commit":{"parents":[{"commit":"a20f20027d5529d242fceed9c6be632ff3aa89c4","subject":"sq: timestamps: add Timestamp.version property","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a20f20027d5529d242fceed9c6be632ff3aa89c4"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-05-14 11:51:08.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-05-14 13:40:05.000000000","tz":60},"subject":"sq timestamps: simplify offset property","message":"sq timestamps: simplify offset property\n\nFor simplicity, get rid of _offset in favour of offset. As a result,\nrestore the ability to set offst to the public interface.\n\nChange-Id: Ibedc4a8191c4778a6b100f1e91ba6998df2e0f28\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eae8b6b7a6b2c04cd857783262b1bc584bac8c62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eae8b6b7a6b2c04cd857783262b1bc584bac8c62"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
