)]}'
{"id":"openstack%2Fswift~988644","triplet_id":"openstack%2Fswift~master~I858854e9b221a90b7e943eaddd841e02affd67af","project":"openstack/swift","branch":"master","topic":"p-sync-fix-conf-path-test","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-05-17 19:31:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I858854e9b221a90b7e943eaddd841e02affd67af","subject":"container-sync: fix failing unit test","status":"MERGED","created":"2026-05-14 15:42:08.000000000","updated":"2026-05-17 19:32:34.000000000","submitted":"2026-05-17 19:31:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988644-p-sync-fix-conf-path-test","meta_rev_id":"4d3e0b2d2dd1c08663f28cf749b2714270e1a289","_number":988644,"virtual_id_number":988644,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-17 19:31:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-17 18:34:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-17 18:34:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-14 16:48:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-17 18:34:18.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"6dda662dcc1920190636198541ad5f6dde087878","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-05-14 15:42:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c6932816f58ae1bd9194d34e14dc44706925e6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 16:48:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/271e63756eff45478bb41319fcf15596\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8eb4e9aac814c9ba2a0e7a5967d5f7b : SUCCESS in 11m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/003ca47d870241c8b4760384662f604d : SUCCESS in 4m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/92dd4cfa67cf4e3ea4d570e9a330522a : SUCCESS in 11m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f04499550b3b4b20b99140fcfe402be5 : SUCCESS in 12m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bc763a9772d64c41ace60ede6b74d924 : SUCCESS in 26m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/baf102e335954a87bc69232ba0b63908 : SUCCESS in 4m 43s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7e4de5b418644566bb85b342ffb5ad74 : SUCCESS in 19m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ee85aaff7a7349de8303c42dc4ccd2f4 : SUCCESS in 23m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/04a49d2bf3da46f5b5efe234ca4538d9 : SUCCESS in 23m 08s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/203bf2fb84644ee4a4c10163392b5c2e : SUCCESS in 9m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dd7a7cb4364f4e08b31425252f4a917d : SUCCESS in 8m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ce13a8bcd49149198f62fac4e64c8ded : SUCCESS in 16m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/380649e5b1cf44a0803aafca80489faa : SUCCESS in 16m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/04ef70fd4be347379b74e87d96f5d31c : SUCCESS in 34m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1f3704739a674d4ca40a206f2a8fde11 : SUCCESS in 52m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/855962a06d2b40c7a520ba9c7d62973a : SUCCESS in 28m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40ab91234bd0452fa3e212a42933f503 : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/45069b51d5f6432a86a5c02e1ae308d9 : SUCCESS in 20m 26s","accounts_in_message":[],"_revision_number":1},{"id":"f17fe215a1baa0f230325c71757fcdcb86445682","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 18:53:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c42cdb2727404e01a25ccf11fcd53d7f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0c2331ef0a1d47f7a7c5724646a6abc8 : SUCCESS in 1h 05m 22s","accounts_in_message":[],"_revision_number":1},{"id":"fa9d76fb49db9cdc4cc542997a7fc8e2a20efde7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-05-17 18:34:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e91c251b4951d6082a456c0796d46994e67cc7ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-17 18:34:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7f41085528b8fa0cc8db9cc8ef23c1e69afab0c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-17 19:31:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa3d59c60ecf4fca9a01a10d55955869\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3048b989a97646a6affd2eca7d5c8ac1 : SUCCESS in 10m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/744cfdfd4fd7492db872dac2b91e18c5 : SUCCESS in 7m 06s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/1854701286f74d7f8dddb2fd08909e65 : SUCCESS in 19m 23s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ca5fdeb6d4e4480db243fdcda556f40b : SUCCESS in 20m 31s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f68bdef84109489ca70f3c813dfd508e : SUCCESS in 21m 19s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/65c1c1528b52488092ca342b37b15b33 : SUCCESS in 22m 50s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/474da0fde82e45719e3d66df3b2bd39e : SUCCESS in 17m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/497a82158f534a119ec71e853fca32dd : SUCCESS in 11m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ae96c144641047889dc1a91f34b7b925 : SUCCESS in 12m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/eb09dfb74a48479caa054e009d5449ac : SUCCESS in 15m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0c4c15a1e50b40429ae3c53cb1bc991a : SUCCESS in 11m 01s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/349f422cf4a04065b1c56f70d602f959 : SUCCESS in 11m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7159c9dfd5b24583a42ad89dd6e692a5 : SUCCESS in 34m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/250f5e82f5c4403c9941e81342f28f0e : SUCCESS in 55m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7bd470d850fb41379aee366c34087a2e : SUCCESS in 29m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45dff34df91c4e7e9573fa25bdabcf9a : SUCCESS in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0f1aabecb9fe467dae60e0012d2c72fe : SUCCESS in 24m 29s","accounts_in_message":[],"_revision_number":1},{"id":"1f96a565267508fc0680f443b074509e0a4bc05d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-17 19:31:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4d3e0b2d2dd1c08663f28cf749b2714270e1a289","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-17 19:32:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcee8a660f864f66a16c81608477edfc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e631123738e0461a8991ce69d79011c9 : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/2cc9afb9db244b54ad515c824280f522 : SUCCESS in 45s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"613df4d6025fec4b975d6434ddb934f418cd1c73","revisions":{"613df4d6025fec4b975d6434ddb934f418cd1c73":{"kind":"REWORK","_number":1,"created":"2026-05-14 15:42:08.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/44/988644/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/988644/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/988644/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/988644/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/988644/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/988644/1"}}},"commit":{"parents":[{"commit":"33967853276f0987a58e5b3aca450173adadc064","subject":"timestamps: check offset is an integer during construction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/33967853276f0987a58e5b3aca450173adadc064"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-05-14 15:25:58.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-05-14 15:41:14.000000000","tz":60},"subject":"container-sync: fix failing unit test","message":"container-sync: fix failing unit test\n\nThe RelatedChange made the ContainerSync daemon prefer to load\ninternal client conf from the default conf file, if it exists, rather\nthan use the hard-coded conf string. As a result, the existing unit\ntest fails if run in an environment that has\n/etc/swift/internal-client.conf, for example in a\nvagrant-swift-all-in-one VM.\n\nThis patch fixes the failing test by changing the swift_dir to the\ntest\u0027s tempdir, thus avoiding reading any\n/etc/swift/internal-client.conf. It also adds test coverage for the\nRelatedChange i.e. loading from \u003ctempdir\u003e/internal-client.conf\n\nRelatedChange: Ia12647addacbe56183b4548de68d252ee9485b1d\n\nChange-Id: I858854e9b221a90b7e943eaddd841e02affd67af\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/613df4d6025fec4b975d6434ddb934f418cd1c73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/613df4d6025fec4b975d6434ddb934f418cd1c73"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
