)]}'
{"id":"openstack%2Fswift~989785","triplet_id":"openstack%2Fswift~master~I148f00865495276b888ae6447fb37e78033ab43a","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-05-27 10:24:44.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I148f00865495276b888ae6447fb37e78033ab43a","subject":"sq? add to pickle tests","status":"ABANDONED","created":"2026-05-22 15:38:16.000000000","updated":"2026-05-27 10:24:44.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"6e67bf0e0e5412edc841b8e5018983f16a45cb1d","_number":989785,"virtual_id_number":989785,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-05-22 16:43:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-26 22:04:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-22 16:43:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-26 22:04:31.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"311be89093a4e5f96d250505ce55874bbe7ce3a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-05-22 15:38:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df208efeed6078ca56f553e27389dd9cd40d1974","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 16:43:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdbff6b855bc4b9ea11894d9a25eaf02\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04dc2d230b354e9390b60fabc53e7c2f : SUCCESS in 11m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f35f1c0cfde5443ba030e838d346ee80 : SUCCESS in 7m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3050c3cacb314133accc82259778699e : SUCCESS in 19m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b74e949c5b14466097eadb061b713cb3 : SUCCESS in 15m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2138ecd71a784d3ebd47173b6c4eb490 : SUCCESS in 23m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b28208ce9bf7459384e5316cd8dce977 : SUCCESS in 10m 08s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/647e75990792427498e35cbc047f5393 : SUCCESS in 19m 10s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b32b0b7102bf47f695b5e8beea35a5b5 : SUCCESS in 23m 22s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/21b6eeef06fa4ffcae4313cdd98af765 : SUCCESS in 30m 32s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/4e055fd9d51344ef9934acbcc2c030a9 : SUCCESS in 19m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8f1b3fa7cee54821833558fe8b4bee20 : SUCCESS in 13m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ebf4f2d3ec89496c9e7f88356a5b9027 : SUCCESS in 9m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e6013f428d2a45589f32aa55826347f0 : SUCCESS in 13m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c27313ce8b484c378369da6cfec283f8 : SUCCESS in 55m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5df22eba30df4a278bd54c116258f449 : SUCCESS in 52m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef1401fd289149e6b497e1e2dad75aed : SUCCESS in 25m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b71ccaf9f7d54887bd5e87d6b1219d58 : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6b55a900c0674b65a7ecab548e6bf171 : SUCCESS in 19m 22s","accounts_in_message":[],"_revision_number":1},{"id":"2f445e01697227baf80034b1eb5d04fa4be46add","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 17:05:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d61badaaed045808896ed93b872b4bd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7ca854e8f06542b6a1070327b9552d8a : SUCCESS in 56m 55s","accounts_in_message":[],"_revision_number":1},{"id":"21dff7d12c94abc8ad1417d1116356de6bb630b1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-05-26 22:04:31.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6e67bf0e0e5412edc841b8e5018983f16a45cb1d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-05-27 10:24:44.000000000","message":"Abandoned\n\nsquashed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"75bc71b8ae1877ccd2f7a0a68a3533d79fb8d0ce","revisions":{"75bc71b8ae1877ccd2f7a0a68a3533d79fb8d0ce":{"kind":"REWORK","_number":1,"created":"2026-05-22 15:38:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/85/989785/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/989785/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/989785/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/989785/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/989785/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/989785/1"}}},"commit":{"parents":[{"commit":"ca4d98bde9c4dad244d84fa8db429f1812668557","subject":"Add unit tests for hardened pickle unmarshalling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca4d98bde9c4dad244d84fa8db429f1812668557"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-05-22 15:36:20.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-05-22 15:36:20.000000000","tz":60},"subject":"sq? add to pickle tests","message":"sq? add to pickle tests\n\nChange-Id: I148f00865495276b888ae6447fb37e78033ab43a\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/75bc71b8ae1877ccd2f7a0a68a3533d79fb8d0ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/75bc71b8ae1877ccd2f7a0a68a3533d79fb8d0ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
