)]}'
{"id":"openstack%2Fswift~990358","triplet_id":"openstack%2Fswift~master~Ia9ea22d6bc5773147efdaf2a7334d53c2e895f0b","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-05-28 04:53:36.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ia9ea22d6bc5773147efdaf2a7334d53c2e895f0b","subject":"trivial: Clean up some bandit warnings","status":"MERGED","created":"2026-05-27 19:15:10.000000000","updated":"2026-05-28 08:15:43.000000000","submitted":"2026-05-28 08:14:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990358","meta_rev_id":"bfaa2e39d67e13b9a34f8fd7bc63615db8b8f48e","_number":990358,"virtual_id_number":990358,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-28 08:14:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-27 21:56:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-27 21:56:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-27 20:26:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"360530a0031d4fa18367bf7d971841a613b7cebb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-05-27 19:15:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a7286ae296e714e98257e801cf92f0a940902b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 20:26:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4cd13284583413588a639708f977e85\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10fc10c6bbc342519208b2f07c86eb1c : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24fbc3ebbab14181a710b431fef0da2e : SUCCESS in 3m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/31447039e44c40f7b740f51a6a06446a : SUCCESS in 31m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0c851587a29a4e2aace574fc69526b55 : SUCCESS in 1h 05m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cea28fcbaf444777ba2b27c2d9b9819e : SUCCESS in 47m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce33fb919bb9403bb903be6136edefd1 : SUCCESS in 34m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b6588addeeb45b583a954f097a1e65f : SUCCESS in 10m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/80d6f83c876f4713bc45c3a50f64f9c8 : SUCCESS in 10m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b88fcc2014dd417ea67113a043d5270b : SUCCESS in 22m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0c134164e61c4c55b0d30522fa4a350a : SUCCESS in 4m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/59d996531a554744a6967b830817eeca : SUCCESS in 16m 11s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/88b6c3aaec5d4e2ca4b5968dfb0c5fe5 : SUCCESS in 9m 29s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/018009932fc0469d845c7e9e8d75a921 : SUCCESS in 26m 26s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/738e0b9603a94eaaa9dfe307bb38cd3e : SUCCESS in 24m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/40905c16fe8f497db1fbd35b5184dc4f : SUCCESS in 16m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9b06e6aac44240e59960b96b8c72ecd6 : SUCCESS in 16m 56s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/88deef87981f4c97832f9930686a1481 : SUCCESS in 16m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bcc5b975f1f74eb781768a766912a2d6 : SUCCESS in 6m 10s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a95346ed5ee044c6b5917b1714a22569 : SUCCESS in 16m 22s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f39f2b7373344bd88b6bb0171e65f074 : SUCCESS in 5m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/70a4610c00b944c5b6dce794e95ef582 : SUCCESS in 45m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/265ea81ad4a04381b61683ed562cf889 : SUCCESS in 42m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/263ca548413847d097ff504d1d2858dc : SUCCESS in 32m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b58976513d047b3910188c2489eab6d : SUCCESS in 17m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81b7dd78e20141449846613acf235b68 : SUCCESS in 2m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3e4ca8b367ce44cdb9d6b94a3323b4fd : SUCCESS in 21m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bbd4da8841244f08921181e7a119262 : SUCCESS in 1h 01m 18s","accounts_in_message":[],"_revision_number":1},{"id":"ae22beaef2e79ed6e1ca9ced63e6828089547416","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 21:56:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa008594da754132a3551f5ed20e74c1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/42312c83272341d5b2bd98101b729fe8 : SUCCESS in 30m 27s","accounts_in_message":[],"_revision_number":1},{"id":"20ad5880b959ceefef9cdc1da82c596c96c62b19","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-05-27 21:56:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f9f2c7f8030459d7d08d8d7362abba3f2bc2b63d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 21:57:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ec9c80de164abc4f9206f455b0885e4b90928c1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 23:22:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fdbedff5e7ad47cbbaa1b7ef2ec6c429\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0209fbe065404cb6aa05933592f58bfc : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48a56d3e70354a7d83c0933033fdb861 : SUCCESS in 6m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/96f4fa3af91c435f9f393718732d4791 : SUCCESS in 1h 07m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8275b744ed994fe2805657288ec0cc5f : SUCCESS in 1h 18m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ae9e7b33d31a4dc19958fe61e7173db5 : SUCCESS in 38m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/188f52a3cbe14569bd08a08eac0519d4 : SUCCESS in 1h 09m 39s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c98030b6abe3432f81a9179fdebeef55 : SUCCESS in 6m 47s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/10aee084f8d0498e822bd7e47f00811c : SUCCESS in 25m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7528a2efee8b4e98a83bb1f99b19d9f1 : SUCCESS in 21m 07s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f2f808e1c8584f1987f9dbb1b2149fa6 : SUCCESS in 21m 16s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/411fc624c313479cb6c3f9dc7a831139 : SUCCESS in 10m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dc0f075964fc409e94ac35d992947e4e : SUCCESS in 7m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/84e628078f074362b985406d3dc886b2 : SUCCESS in 15m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/94e87c7339aa4bc5be329fe4ab3e3d36 : SUCCESS in 20m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/82a8b04310384ec58241f9618bf76ce8 : SUCCESS in 4m 15s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e00561f207804b07bc220d569eb92d21 : SUCCESS in 5m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5f8ce7afac304d3093a9f771ec6b72f3 : SUCCESS in 40m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2a16b0b126aa4b21882ed75b8885dec7 : SUCCESS in 54m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/876b083ce63449aeb18e0e91bce8fc8d : SUCCESS in 36m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/523188fafe954b2f8e65a791828b6cf9 : SUCCESS in 12m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0f1047b78724ec2bfe6053558574496 : SUCCESS in 2m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2c606a4b3953427b9c856e2d079dbb03 : FAILURE in 41m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8558ec80a5f048699f904f58c5103e77 : SUCCESS in 1h 01m 55s","accounts_in_message":[],"_revision_number":1},{"id":"a9de95eaecefd466479c3bc72ac2c222359ffdd6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-05-28 04:53:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1dca8aa10fc5205d4a1b74e53924a1f5eb6ad678","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 06:20:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29c9d717945648e0b01887d54615e995\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9e8f88521d446a7b0e20c704436ddb5 : SUCCESS in 10m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcfc4971e1354603ab4b477fcd067450 : SUCCESS in 8m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/ec94bd47866f40eaabd04c6cbedf80fd : SUCCESS in 1h 04m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b28e38b882564067b4c4a31a7799303a : SUCCESS in 1h 20m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/92511c3ecdb54702b701c6f79016369d : SUCCESS in 53m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a73d7727e4fb4c04ae2b103875c461bd : SUCCESS in 1h 06m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/685154c977a14e0e86ecbec85cbc58b5 : SUCCESS in 15m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2c931b47aadb4c45a5d565e3c2131e4f : SUCCESS in 15m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/29501e352e4c4d8da70613af4056c4d5 : SUCCESS in 23m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/12eb62c161204b5087f48cc8ed3cdfe8 : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ea147635f8924b3298bd21d1dc71000d : SUCCESS in 26m 39s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f9fb3a26443142cbb213695f9030f68f : SUCCESS in 21m 31s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/edc6545d88ea49fb95da958eb0b112fc : SUCCESS in 29m 51s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/728e5afd69cd4b18acc39de55dd76681 : SUCCESS in 13m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1fd927b88d9c4ea18a3a0fe9b9d4b4e6 : SUCCESS in 14m 29s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/73868417d01a4077b7b30ea716e0f3c7 : SUCCESS in 9m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/35bcebf852c44b2f9c3f36a5352a6232 : SUCCESS in 13m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a8c1fa4a7e0c4b23b14bea5e229ba479 : SUCCESS in 4m 04s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/87a1a4b630c64fbf85a3ff2611b51ae8 : SUCCESS in 15m 06s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1600f80913ac4432a500c293c1ad9c93 : SUCCESS in 5m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/eda015247ac94d9fb65d7fa52f0b4aaf : SUCCESS in 1h 00m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fc0da43f84324c6f8614e9db2c785e64 : SUCCESS in 53m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f98d14f0668b46bea3630573b53170f1 : SUCCESS in 21m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f485b3f253254dfa978d1b377345b10a : SUCCESS in 28m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6af2bb0a704541c3b34346a671389690 : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/46e6467ddb3d4bf7b7e7e66d1ba67691 : SUCCESS in 25m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/488d30a85a1d43aa9fab65d6688f9a4d : SUCCESS in 1h 14m 57s","accounts_in_message":[],"_revision_number":1},{"id":"d73af9bb6da96f94d60e08923a3436d5d409b53f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 06:21:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ff003e84dbb00aed3d60b2734a9515ee412a51fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 07:19:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d4e10f6f84c4a188b56aaf57e5d0d81\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3d8be408faf046c5af5671deff51c710 : SUCCESS in 37m 19s","accounts_in_message":[],"_revision_number":1},{"id":"19fd2cb21dc71f13feeff2b2b749502d327c710a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 08:14:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/649b55a9432243d392acda525bee64c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a81ececa73de45feac4f1a12262bf9ba : SUCCESS in 10m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d33708b931944b09acf0acb566b1e2d : SUCCESS in 6m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/77589775f913475c82f039615117dafa : SUCCESS in 1h 03m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a44b0b77d5594cd3a5bb25e447ae9e1f : SUCCESS in 57m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/42a974aaa5624025896db286616a38c3 : SUCCESS in 49m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d369ad9033cc4c64b5ed519e9795231a : SUCCESS in 49m 20s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/16d599b381c740b1af296bb209bc70f4 : SUCCESS in 8m 50s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7e8caad3c96248a997df68bebbe333fa : SUCCESS in 26m 05s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/bfc23729bbf74e369f8e2906769f2509 : SUCCESS in 9m 25s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/99ec4ca92e1240cc83baa0e1b9a6970b : SUCCESS in 13m 25s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/2afe937fd0b6469481c821bb66b60d3a : SUCCESS in 14m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/55e9c9d28fd242c5b2f05846580d4f26 : SUCCESS in 8m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e175b8d2823f4803924ddd4e45fed3e7 : SUCCESS in 9m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/77f44e1fa74a4bf181579ebddd3bc994 : SUCCESS in 15m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/35e39fea5f0e4d358b12204de1357df3 : SUCCESS in 4m 36s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/37abd4ed92a442c8925f1d02dc5cf4d4 : SUCCESS in 8m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fe2ebcc857e04b20af4f0471b431e71d : SUCCESS in 40m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/672bbde896b44a35b19d0404685bc03c : SUCCESS in 35m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/24e8a58636704adbac31ec2ff5ddcb5e : SUCCESS in 27m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/280cda4b3dff45f88b96ed2c4d30c539 : SUCCESS in 28m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a4cb4df3bff46a29538a98d07488df9 : SUCCESS in 2m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3fac0e66fb184c7faaaea39dfab58e85 : SUCCESS in 16m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86df464d57f24269a6b1d8488e929c14 : SUCCESS in 45m 09s","accounts_in_message":[],"_revision_number":1},{"id":"4b47e2688fc05744206f426096dfdb13be59ed76","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 08:14:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bfaa2e39d67e13b9a34f8fd7bc63615db8b8f48e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 08:15:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/311e36516bed4c34954cc01ebd76e816\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/585b379af6d44ae584828370c3ef3dee : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/fae4a25b7add4b2e98a3d99dfec1f5b3 : SUCCESS in 42s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"82b62b2a3ab099b822291548745d31ce9723838d","revisions":{"82b62b2a3ab099b822291548745d31ce9723838d":{"kind":"REWORK","_number":1,"created":"2026-05-27 19:15:10.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/58/990358/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/990358/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/990358/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/990358/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/990358/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/990358/1"}}},"commit":{"parents":[{"commit":"0da44e9ea2ea5fd3ded31941318be71c676feb2e","subject":"s3api: Add functional test for aws-chunked truncation bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0da44e9ea2ea5fd3ded31941318be71c676feb2e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-05-27 19:13:53.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-05-27 19:13:56.000000000","tz":-420},"subject":"trivial: Clean up some bandit warnings","message":"trivial: Clean up some bandit warnings\n\nSeen in CI logs:\n\n    WARNING\tTest in comment: bandit is not a test name or id, ignoring\n\nChange-Id: Ia9ea22d6bc5773147efdaf2a7334d53c2e895f0b\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82b62b2a3ab099b822291548745d31ce9723838d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82b62b2a3ab099b822291548745d31ce9723838d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
