)]}'
{"id":"openstack%2Fswift~990388","triplet_id":"openstack%2Fswift~master~I6c0a600937794ec41ab8c329dc6ea2dba9295b24","project":"openstack/swift","branch":"master","topic":"check-db-conns-harder","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2026-06-28 23:59:40.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-06-28 21:51:40.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I6c0a600937794ec41ab8c329dc6ea2dba9295b24","subject":"account: Split PUT code paths","status":"MERGED","created":"2026-05-27 22:23:38.000000000","updated":"2026-06-29 00:00:38.000000000","submitted":"2026-06-28 23:59:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990388-check-db-conns-harder","meta_rev_id":"2871b618c33b13886c19d782717263aaa89463ee","_number":990388,"virtual_id_number":990388,"owner":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-28 23:59:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-28 21:51:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-28 21:51:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-27 23:36:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-28 18:36:44.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"52a57e59ab44515bef257a27d3607e4895fa845f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-05-27 22:23:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"359936d54696a253962e0d7a773e5c5eec2c0acc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 23:36:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c15f4b9620b4b4a81423c6a0487e802\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4568ba4634714cfda0ebd37a10493057 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b9abacd3a614ce38cab57addd67d32d : SUCCESS in 8m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/823588d363ba4f39b9d99921c037a477 : FAILURE in 1h 06m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e0e92fa259424087b41f24ca70437f4c : SUCCESS in 44m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/558403fe7a7f4341948b4818a88a3fcd : SUCCESS in 1h 03m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/76542d96725a4e63aef33f6c129ad955 : SUCCESS in 1h 07m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/16cc7da7a5e941feb18ec9331d6c099c : SUCCESS in 19m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/20bda7e7d1a94768b56d8c087edc90a2 : SUCCESS in 17m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/365fc899541842fb964dee66f61c2562 : SUCCESS in 18m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/324a7cc82a934306a7d4151081b7fd01 : SUCCESS in 6m 55s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/11ed256dfe8648efbc803b9d24412556 : SUCCESS in 19m 31s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/edabaf54de674cff927293d8dcc1af9a : SUCCESS in 17m 07s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d16a2eec3da84751800de24c27df0c2d : SUCCESS in 21m 54s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/6c6d7a14cd364a9f943b6e0bb1a22a91 : SUCCESS in 20m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e035e9988e1340d080ba8eb7356f2185 : SUCCESS in 9m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/965858a48a134c19b2e4fd572aae41b3 : SUCCESS in 15m 37s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b6e1acba512a4eb49d578d7fbfc1bad9 : SUCCESS in 8m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dd462af057d447faaa43eaa93048eb8d : SUCCESS in 5m 00s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a0cf6ce8ef5e417ca7fa3f522647eef4 : SUCCESS in 20m 28s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f24f5c55e2a2491fbc0c1bbfac16de23 : SUCCESS in 7m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f030c3f4e3b48d08cf9629e654bf6ce : SUCCESS in 1h 03m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/180ddb6af7bc497495a0879638e08883 : SUCCESS in 58m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e0c0c9d3824947b6b8b875c02391576d : SUCCESS in 31m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d748b4d647144f0dbb68ae7d6a2365ae : SUCCESS in 12m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fed4ec3e72a1414e9e81eae1c3a88d83 : SUCCESS in 2m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d87ede7c87534c35ac4bc77d582c817c : SUCCESS in 18m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d6eee7cd1124afcaadbcd65ab74c735 : SUCCESS in 35m 58s","accounts_in_message":[],"_revision_number":1},{"id":"c40de6dd50b4845a6c42a919c25163c9e5fcb601","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 01:24:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2742dd47a024350937630c4b530e8a0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/caa8320d07064ba1a03455cefb8a872f : SUCCESS in 1h 12m 04s","accounts_in_message":[],"_revision_number":1},{"id":"c141ab8fbfda9d887230adc500778650d88b1b00","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-05-28 18:36:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"87e021a4ad2ca85ced2027f0eec2a34eef292608","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 19:49:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5464b04e85714faab416390f903814bb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07f337f483e445a989c6d20a0a94b5ae : SUCCESS in 10m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa8affe836da46d6942ec868d8b26cc8 : SUCCESS in 7m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/ae563f28ed5c4b6aa4dae40268f9259b : SUCCESS in 32m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3141d56ea6b14eb5b79437b60197fb70 : SUCCESS in 1h 12m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9bb92e5e58354783a54502de8bd38713 : SUCCESS in 34m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/721ce491894d4654a57fa7464d999d92 : SUCCESS in 1h 07m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/81e947d85f1041f698d762b19bfb8c59 : SUCCESS in 22m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aff7fee6d5c14a799b8ed56e718f045b : SUCCESS in 21m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b51670634ce4424a8528b100d0be956 : SUCCESS in 12m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5264f13cd542446a8b823097bdccd487 : SUCCESS in 6m 55s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/45db9d5984c04fa5ae2dcdff2e5e11cf : SUCCESS in 26m 04s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/00901a3e4702414992a7aa9adc5b65b3 : SUCCESS in 22m 59s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f7d2a920d4684e60bccd346c9d8e4f54 : SUCCESS in 30m 12s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/5093fc80ed0342a3a158ccf7f23ee0e2 : SUCCESS in 9m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4b586d1ef77c42efb7bdec20dd32596d : SUCCESS in 8m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c304adb9edf64b24883702a0f39c1bd9 : SUCCESS in 8m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4c05c70fb47f45c2bb8479238b79239c : SUCCESS in 15m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a1e34abc365145adb307420f1c4fef7b : SUCCESS in 4m 19s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3a4f47a9ca404d42b62586eaed822d96 : SUCCESS in 18m 34s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c181c2a24f03440f97710299c1d26206 : SUCCESS in 8m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c318bd750c247ef9d7a62cf295ed32a : SUCCESS in 1h 00m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/19ff2851891a443a8c4c6a5373df3baa : SUCCESS in 53m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6aafdbfd20b64c0a9290955fdc424c18 : SUCCESS in 22m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75f2f8fa655b4e99b81bb00582b956ef : SUCCESS in 11m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3205dd43c24248daa6b73caed384bc83 : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9604e4137acf45db8837bb817e7c681c : SUCCESS in 18m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59fe14121c1942b49353d5b439e02097 : SUCCESS in 1h 06m 23s","accounts_in_message":[],"_revision_number":1},{"id":"258d286ceff21fb5b54da7eaba34e0e06b545788","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 20:50:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74eb3ce5407647a29830d9641072f282\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1975edf495404fc69045e23e7cc16f5f : SUCCESS in 36m 41s","accounts_in_message":[],"_revision_number":1},{"id":"fe7d55582e5fff609916e275004bb675fcde099c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"real_author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-06-28 21:51:33.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_34930\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}],"_revision_number":2},{"id":"ff7f11feda998c788e2abd6011d2ca35653f1e41","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-06-28 21:51:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8d470e0a392436bd22c065247d2ee131e8fefe03","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-28 22:25:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85fdbcfd10d7416688394543f7f3a298\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b621d0f751fa435fb551d6d91f35b331 : SUCCESS in 32m 46s","accounts_in_message":[],"_revision_number":2},{"id":"f648366a822fe3299f3ca3edebba44fd8ed359de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-28 22:49:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0a76f38905a4c23b56680482aedcdf2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/141b3115430f4b4583bcd74ca07c1fd4 : SUCCESS in 10m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3553c2259f0402b86cb300782a5496a : SUCCESS in 9m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/38da6c0343334953834fde824cb8272c : SUCCESS in 35m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4b6d73c8835e41a9ad4a4d6d5f1d981e : SUCCESS in 33m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/049197c11b7d470ca2fc9166d0ed03a3 : SUCCESS in 38m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/67c4bd2456a948ccb6bb3ede8c8a7c93 : SUCCESS in 33m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f3567e26c28b4ea98532ee4953566de4 : SUCCESS in 21m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d431a51750784ba68c63b981e3bb4f04 : SUCCESS in 22m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/97d260d52d014669a9a4e8d499523fed : SUCCESS in 20m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f36b2146154a413f996926d24baaea6c : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dbc135415f29408e81aeab6d6847dbed : SUCCESS in 25m 17s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/859e72da3c75485598714f0c14df7693 : SUCCESS in 21m 54s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9234594d3fee44728471491f264c6169 : SUCCESS in 12m 59s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/68b9ccd705364a82b315290329ab7fcb : SUCCESS in 20m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/de362f7612b842aa81d5d2f48d6797f7 : SUCCESS in 14m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/048ac5b2ffc841a383508c8f98f99ba9 : SUCCESS in 8m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ac9503e7450c44f4a56b74381c4e0ff4 : SUCCESS in 14m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3848462149bb46b8a40559869668b116 : SUCCESS in 6m 41s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/12a8f1f19f984bae877ec6e4c96aa67d : SUCCESS in 17m 46s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2edb34f52194467c8494eaab0d471b3c : SUCCESS in 11m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f22ec14ad5d4f638ac6a9a75bfcd15f : SUCCESS in 51m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/14f38ecd01cf4bfb834f0f6efe7d3a24 : SUCCESS in 53m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b7ced846432c468bbd6fd068fea50cef : SUCCESS in 26m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d04ae738ac5d45eebc2d718d2f23e4ac : SUCCESS in 13m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9476b7143e89458382ddef2491a0f6da : SUCCESS in 2m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dd43b2bf297a4a5098620080d20a67e1 : SUCCESS in 8m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e85f269a1580430fadfec91d50194996 : SUCCESS in 52m 54s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/docker-image/run.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/buildset-registry/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .local/share/containers/cache/blob-info-cache-v1.sqlite found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .local/share/containers/cache/blob-info-cache-v1.sqlite found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .local/share/containers/cache/blob-info-cache-v1.sqlite found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"633365fc12000b84ccb0d6bc0b2375c1341eef32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-28 22:50:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0da489b78e9d89acb0e2da5cd1b923c5ab521193","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-28 23:59:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1530700645144d68dfe78008c3cf47c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b24ed90b00d549b4bf0042549e78114b : SUCCESS in 9m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3634f04ef5764f12a3e886e09ef948f4 : SUCCESS in 8m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/5e974ece7c784d9e97c19d68bfe066de : SUCCESS in 1h 02m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4a510deb34cd4f19ba3537eb8a8a5061 : SUCCESS in 1h 03m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/899a64e825cf48528ab8e5f565cebaa2 : SUCCESS in 53m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/75d068657e48486997033b609cfb9514 : SUCCESS in 1h 04m 22s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5ce329093b1f4b3193113b9df885981a : SUCCESS in 7m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a17dfb93ac5e4d7a9437f30e95c12624 : SUCCESS in 10m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/92e02cc4d27547f28df257a84a8dba76 : SUCCESS in 16m 55s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/56913b4852694778ad49a0446f86ba32 : SUCCESS in 11m 25s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/d6a0d10804ff48eba47f6ad0c177e14d : SUCCESS in 18m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c9860aaca5b04f5c9e54fc275348666b : SUCCESS in 14m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2c431c2d74704e4bbaf6c29e06df40e2 : SUCCESS in 15m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3772392262ad47fba12c2821b04ba93b : SUCCESS in 15m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cd1c4853c30d42e996790c8a8466cd97 : SUCCESS in 5m 43s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/160bd43336a244549446d126b6dd967a : SUCCESS in 8m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b349df64273c4230adc13f8e91b8cca6 : SUCCESS in 1h 00m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/39fae6a3b47c463084c908010f56e6db : SUCCESS in 54m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/394034ecaa524fbdb4ed0b0d9b52ac02 : SUCCESS in 52m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7d7ab9bc43a44f1b14973e245e284bd : SUCCESS in 24m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf8e18f6c790407d9387f91a4875d1a7 : SUCCESS in 3m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d29bd043236f4d2a93f1d923c63dad3d : SUCCESS in 9m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d6bc3eccde4f46e4a3c8659255d23f65 : SUCCESS in 56m 55s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/docker-image/run.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/docker-image/upload.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/buildset-registry/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .local/share/containers/cache/blob-info-cache-v1.sqlite found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .local/share/containers/cache/blob-info-cache-v1.sqlite found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .local/share/containers/cache/blob-info-cache-v1.sqlite found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"f768191ae6a49143b5453bdac9a4a3cbcb2a17a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-28 23:59:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2871b618c33b13886c19d782717263aaa89463ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 00:00:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ad7d637a56342ad8929a4b7efbad730\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ba6b9e080194ad8b0e09c5493ab569e : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f93747eba5e94b68a94bd21310eef9c2 : SUCCESS in 43s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b8f3c8b472bf00af594192b7697b420722192244","revisions":{"a5aed721b54daee42a732f5908f738ae9c7b6efa":{"kind":"REWORK","_number":1,"created":"2026-05-27 22:23:38.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/88/990388/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/990388/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/990388/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/990388/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/990388/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/990388/1"}}},"commit":{"parents":[{"commit":"05c11f44beef85a357ed5a2d98097da79a59dcdf","subject":"Close DB connections in metadata servers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05c11f44beef85a357ed5a2d98097da79a59dcdf"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-05-27 22:03:21.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-05-27 22:06:52.000000000","tz":-420},"subject":"account: Split PUT code paths","message":"account: Split PUT code paths\n\nAccount PUT requests behave very differently depending on whether\nwe\u0027re creating/updating the account itself or putting a container\nrow within an account. Split those into different functions.\n\nChange-Id: I6c0a600937794ec41ab8c329dc6ea2dba9295b24\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5aed721b54daee42a732f5908f738ae9c7b6efa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5aed721b54daee42a732f5908f738ae9c7b6efa"}]},"branch":"refs/heads/master"},"b8f3c8b472bf00af594192b7697b420722192244":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-06-28 21:51:33.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"real_uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/88/990388/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/990388/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/990388/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/990388/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/990388/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/990388/2"}}},"commit":{"parents":[{"commit":"667b0461361712ae4e9bbc3aea2dec89a0fedafb","subject":"Merge \"Close DB connections in metadata servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/667b0461361712ae4e9bbc3aea2dec89a0fedafb"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-05-27 22:03:21.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-06-28 21:51:33.000000000","tz":0},"subject":"account: Split PUT code paths","message":"account: Split PUT code paths\n\nAccount PUT requests behave very differently depending on whether\nwe\u0027re creating/updating the account itself or putting a container\nrow within an account. Split those into different functions.\n\nChange-Id: I6c0a600937794ec41ab8c329dc6ea2dba9295b24\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b8f3c8b472bf00af594192b7697b420722192244"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b8f3c8b472bf00af594192b7697b420722192244"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"05c11f44beef85a357ed5a2d98097da79a59dcdf","ours":"a5aed721b54daee42a732f5908f738ae9c7b6efa","theirs":"667b0461361712ae4e9bbc3aea2dec89a0fedafb","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
