)]}'
{"id":"openstack%2Fswift~991231","triplet_id":"openstack%2Fswift~master~I295a79983e6a2bb496a42b28deeca81de3114dfe","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I295a79983e6a2bb496a42b28deeca81de3114dfe","subject":"sq: defer extra timestamp collision scenarios","status":"NEW","created":"2026-06-02 21:59:10.000000000","updated":"2026-06-03 02:25:42.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f8b4f6007b332fcf685965b4c4114cc72b1cea26","_number":991231,"virtual_id_number":991231,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-02 23:04:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-02 23:04:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f3b8346fa13a8f05145a70e41f2d47ec6295d82d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-06-02 21:59:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f7e6b141c6d9e57ce5c78a2ccb97c6a0e543cf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 23:04:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b8f8e7ca334453b9548fd9f4d85fbee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6b113feb9ea4b5b8fc0ced02cbc08b8 : SUCCESS in 10m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad60126dd8154331bc0907b34ed4448e : SUCCESS in 9m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a976c9eb5b1c45349a33fe96ed0d9449 : SUCCESS in 59m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c729394666c4a8087437b5e73034c60 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":1},{"id":"f8b4f6007b332fcf685965b4c4114cc72b1cea26","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 02:25:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7362fa336f7b4994b50efac63d4816de\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d2ce976736a14cecaba512ae89eb6783 : SUCCESS in 1h 00m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"308a76464470be1c364f4d1d3b9b7cc763d8756e","revisions":{"308a76464470be1c364f4d1d3b9b7cc763d8756e":{"kind":"REWORK","_number":1,"created":"2026-06-02 21:59:10.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/31/991231/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/991231/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/991231/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/991231/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/991231/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/991231/1"}}},"commit":{"parents":[{"commit":"c189d64890ed12f193f7d29248a0315cafc7ab37","subject":"object-server: raise 500 on ec timestamp collision","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c189d64890ed12f193f7d29248a0315cafc7ab37"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-06-02 20:38:15.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-06-02 21:58:54.000000000","tz":-300},"subject":"sq: defer extra timestamp collision scenarios","message":"sq: defer extra timestamp collision scenarios\n\nKeep the FragZipper coverage and the overlapping data write probe,\nbut move the broader EC and replicated follow-up scenarios out of\nthis change so the linkat fix can merge with the tests it needs.\n\nChange-Id: I295a79983e6a2bb496a42b28deeca81de3114dfe\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/308a76464470be1c364f4d1d3b9b7cc763d8756e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/308a76464470be1c364f4d1d3b9b7cc763d8756e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
