)]}'
{"id":"openstack%2Fswift~991317","triplet_id":"openstack%2Fswift~master~I6f5fcdf802e5ea7b43ecac7f307b7f63df9369a2","project":"openstack/swift","branch":"master","topic":"p-ssync-rx-logs","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-06-03 19:39:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6f5fcdf802e5ea7b43ecac7f307b7f63df9369a2","subject":"ssync_receiver: uniformly annotate logs with path","status":"MERGED","created":"2026-06-03 12:19:03.000000000","updated":"2026-06-03 22:22:13.000000000","submitted":"2026-06-03 19:39:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"991317-p-ssync-rx-logs","meta_rev_id":"176c0eb7dede55997542f332d5b2f1965f62761e","_number":991317,"virtual_id_number":991317,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-03 19:39:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2026-06-03 18:22:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2026-06-03 18:22:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-03 13:59:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-03 18:22:35.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"bff0674662148b9ac7aa8ef8aadae8e960c0d460","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-06-03 12:19:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"381eec485ec1d8e918c47b19e6bcc52e935343fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 13:59:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17c9ce1454e642f1a020541b7c8af8b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c1b5d7b063d469b8caa4dbc9854b2e4 : SUCCESS in 10m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bb43303169f426190be221571cb759c : SUCCESS in 9m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/09f4f957b0874fb193a69c4d38ca0a35 : SUCCESS in 1h 06m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b9325deb3ce3424aa6484b75e0ed3dc7 : SUCCESS in 1h 33m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2ee99faa293948389035ad2a547460d5 : SUCCESS in 1h 02m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b9cff615aefc4b60b75fbbfa94850d1d : SUCCESS in 1h 10m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5a3b43914b364163915c18068de03c76 : SUCCESS in 24m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/85fa313c0194494393a15e68ca81f17d : SUCCESS in 21m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f84ba4a4fe1a4af0a5e7b41763a0682a : SUCCESS in 22m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/587f3e528d6b42fd8584e0a20462d9a9 : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fb69da214e4345a7a13a578a361508af : SUCCESS in 25m 40s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/de1a0df4f9e94a1ca8aec1fc79b6e9c9 : SUCCESS in 23m 11s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/669eb779e5364af59bf1b3c5361f7ed9 : SUCCESS in 28m 17s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/1f46a82abf9d493f8db016ed2aff47a2 : SUCCESS in 15m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d409a66b03654bec9df460917da167ca : SUCCESS in 15m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/03328ab3897944beb19b7350729ba316 : SUCCESS in 15m 52s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/eb062f5b4d854a0ca324dfefbd27e704 : SUCCESS in 13m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8c9cea92ce3e4ae3a7d4ac4121c90517 : SUCCESS in 7m 29s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/482c4096ef9d4d0d8586c33c13d36fcf : SUCCESS in 20m 49s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/782d56a89a02459db9f54d1dbf100115 : SUCCESS in 9m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/56e5b91cd302472ba20fcf491126f120 : SUCCESS in 1h 02m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ccb3f410025d45e3b244dc10c5823640 : SUCCESS in 44m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/93f059101b3e4804ad9278dee2f7854d : SUCCESS in 44m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b59ebd3f56514530ab0b7fd70aed1002 : SUCCESS in 30m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10b955c5043344fcb187e02556a6fa51 : SUCCESS in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/92ffc5ae87764019b39174106165861e : SUCCESS in 12m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6196e1fb3e73473cb8a9212be43d846f : SUCCESS in 52m 06s","accounts_in_message":[],"_revision_number":1},{"id":"788aff105e570ae39b2485c0afd9d4d0cc79b459","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-06-03 15:47:55.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"319c440369a09ca472cec7d1b39cc87f22b71681","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 17:25:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11f95706fb41404d944f7eccf9ce35f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/042b1a1654fd43108f72fe695d05f7b4 : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c28009b86b0c4da4914fa2268a17fa60 : SUCCESS in 10m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/0212ffe8b8c44cb2ab6ad60ce2ea49ba : SUCCESS in 1h 08m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8ff11940e9b240a3b948b3de260d0026 : SUCCESS in 1h 11m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c5ee08594d644ddaa9d475b2b9e5548b : SUCCESS in 34m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/09a6ca87f36e43a9b104cd921fcfed0e : SUCCESS in 34m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/de79e6c2a458481c817c9a62f4f12b8f : SUCCESS in 22m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1148002471b044919ea60041228e8d31 : SUCCESS in 21m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2041dbedba26490d9907b13fcebd596e : SUCCESS in 19m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7779c95155d64aea9944496af01aa409 : SUCCESS in 8m 39s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/98456a19bce248b2a9e2cd0ac57a1bbe : SUCCESS in 14m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/89f03c8e80774318ade07dc8f4df3dbb : SUCCESS in 25m 05s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0e646f0fa594433c8c5a3ddc4842161d : SUCCESS in 30m 59s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/6bbd1885da5245c5afd95ed8994436bb : SUCCESS in 21m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3966759377664adbb659a8eb5ef5ced9 : SUCCESS in 8m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ff9371ce49b24e6899918c8d49e12632 : SUCCESS in 17m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4e92b153eaea418a993d72892d1fccaa : SUCCESS in 10m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3bac1d76a0cf4242a90ad260aeab1f0d : SUCCESS in 7m 36s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/391727c5a59841279f7db31fecc71d43 : SUCCESS in 24m 41s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/56701dce740f4767a32484142d5ee4a7 : SUCCESS in 9m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/559759994a2541e8b6d7c1f8c51dfcd0 : SUCCESS in 57m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f959fd54b361499d9c34a629c9481cb1 : SUCCESS in 54m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e52602974a5245e1ace8a97b50ab87c1 : SUCCESS in 1h 05m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92eb2706b4594ced9e5c740a35c8eb57 : SUCCESS in 12m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b86a029209840c9854cd5474078f8dd : SUCCESS in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a49a2a4541ec4e55ab826c660c091d74 : SUCCESS in 21m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c871e7011b84b89aa363a0e0dabeb65 : SUCCESS in 58m 06s","accounts_in_message":[],"_revision_number":2},{"id":"acf9352ef140a168d91a5fcc384e3e76f20d17c1","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-06-03 18:22:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b5c17abd4b10e2fb01fece1494c9d4c839655041","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 18:23:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"87b3b59572da2f849121cb63ee6982f6d382c828","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 19:39:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bab9493111054ae08373f3c9ec865928\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9995b790d03453b87b7dddc30a7a7b5 : SUCCESS in 10m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5709ae1ad613475783d9dfbc133e7a64 : SUCCESS in 9m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/93b23bc05a9a42318206c9fcd9f978c8 : SUCCESS in 1h 10m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d4a6460964fb469eae12ae6b15a7a930 : SUCCESS in 34m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/551a3a1e73fb4895a000fba34596da54 : SUCCESS in 1h 08m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94955225ee394875bd1ca87cc75a9332 : SUCCESS in 50m 01s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b1e1339a5cf247a7a83ed5ee8b2da26e : SUCCESS in 7m 53s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/acb66a73b1cf410e82337a5362ba9b03 : SUCCESS in 25m 15s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2d2e4e94ec994c3ea52a6279177afc00 : SUCCESS in 9m 58s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/06835c27972b49779d6b0f9e98c9edf4 : SUCCESS in 12m 15s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/158d374dad744e11ad400e6d1bc7bb64 : SUCCESS in 20m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/26432a24eb02432f9f8dc2fa89e41b34 : SUCCESS in 16m 29s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/775161cc28c34853b65cc5b412b59b3c : SUCCESS in 13m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4524b2b38d284e58bd6d9bee4674a341 : SUCCESS in 18m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d3363a02a7c546f9a95dde6e655cb6bb : SUCCESS in 6m 46s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/baca8cfe4e694f88ba7b0ec70bec0de5 : SUCCESS in 8m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3362a8c2966e44598b8a0cc971f4b595 : SUCCESS in 59m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/984de67861494d54aa46c92c8eb037bd : SUCCESS in 38m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ecbae1c461c142e2bb333954ca7171fa : SUCCESS in 59m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2e64cb42a234b7f9dd1aa3658e82909 : SUCCESS in 29m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c1eafac765f4e87a558007a37c287cd : SUCCESS in 6m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9cf90e815332452888e7cd0c71b8fecf : SUCCESS in 20m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/787bd1eb48e24cc4925a2ffc5f9a0be7 : SUCCESS in 1h 00m 07s","accounts_in_message":[],"_revision_number":2},{"id":"dee446b071932e32ce7aa4109e9e7dfd862d8743","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 19:39:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"dead18205bb8cb641935e9e0042605023a7fb065","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 19:41:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/799d4195ca3d4b10811480e9b5c7c531\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/167d61e0b6254a0db24768a324002378 : SUCCESS in 1m 05s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/931292660de14b66aa4ef2d3d0109e2c : SUCCESS in 1m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"176c0eb7dede55997542f332d5b2f1965f62761e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 22:22:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1841e974fd934b8e910efff62684d616\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8dfe7deef7dc4c6e964e2f7d1125ef11 : SUCCESS in 31m 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"00e93a641af0f70a06a031613b8166a0108dfd4f","revisions":{"1ef924f25e141b512a7d73024c6f80a20b8ef113":{"kind":"REWORK","_number":1,"created":"2026-06-03 12:19:03.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/991317/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/991317/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/991317/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/991317/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/991317/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/991317/1"}}},"commit":{"parents":[{"commit":"f52988bb3f3bca035ea913643d71114d43ef43ac","subject":"Merge \"Reject oversized chunked S3 XML bodies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f52988bb3f3bca035ea913643d71114d43ef43ac"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-06-02 13:49:53.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-06-03 12:03:43.000000000","tz":60},"subject":"ssync_receiver: uniformly annnotate logs with path","message":"ssync_receiver: uniformly annnotate logs with path\n\nIntroduce an SsyncAnnotatedLogger class to ensure that all ssync\nreceiver log messages are annotated with the same prefix, which\nincludes the SSYNC request REMOTE_ADDR and path.\n\nApart from adding the prefix where it was previously missing, log\nmessages are not changed.\n\nChange-Id: I6f5fcdf802e5ea7b43ecac7f307b7f63df9369a2\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ef924f25e141b512a7d73024c6f80a20b8ef113"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ef924f25e141b512a7d73024c6f80a20b8ef113"}]},"branch":"refs/heads/master"},"00e93a641af0f70a06a031613b8166a0108dfd4f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-06-03 15:47:55.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/17/991317/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/991317/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/991317/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/991317/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/991317/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/991317/2"}}},"commit":{"parents":[{"commit":"f52988bb3f3bca035ea913643d71114d43ef43ac","subject":"Merge \"Reject oversized chunked S3 XML bodies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f52988bb3f3bca035ea913643d71114d43ef43ac"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-06-02 13:49:53.000000000","tz":60},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-06-03 15:47:55.000000000","tz":0},"subject":"ssync_receiver: uniformly annotate logs with path","message":"ssync_receiver: uniformly annotate logs with path\n\nIntroduce an SsyncAnnotatedLogger class to ensure that all ssync\nreceiver log messages are annotated with the same prefix, which\nincludes the SSYNC request REMOTE_ADDR and path.\n\nApart from adding the prefix where it was previously missing, log\nmessages are not changed.\n\nChange-Id: I6f5fcdf802e5ea7b43ecac7f307b7f63df9369a2\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00e93a641af0f70a06a031613b8166a0108dfd4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00e93a641af0f70a06a031613b8166a0108dfd4f"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
