)]}'
{"id":"openstack%2Fswift~991443","triplet_id":"openstack%2Fswift~master~I288837ce96888e3426042570fc183ebc3e88de39","project":"openstack/swift","branch":"master","topic":"p-timestamp-offset","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-06-08 15:31:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I288837ce96888e3426042570fc183ebc3e88de39","subject":"timestamp: increase unit test coverage","status":"MERGED","created":"2026-06-03 15:47:18.000000000","updated":"2026-06-08 15:38:21.000000000","submitted":"2026-06-08 15:31:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"991443-p-timestamp-offset","meta_rev_id":"471990495f9ed12e8ca51a424160ffe22cc45a41","_number":991443,"virtual_id_number":991443,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-08 15:31:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":0,"date":"2026-06-04 08:44:00.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2026-06-08 13:08:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":1,"date":"2026-06-08 13:08:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-03 17:17:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-08 13:08:41.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"}],"messages":[{"id":"6c040d08c12a6f19b3ffdcc20486c285f8b6372d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-06-03 15:47:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c0e54bcea8f6b1d4b4351025341686c1c4a65e9","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-06-03 15:56:48.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f893bafae7bd848228ebffe614f1e33f8787f87b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 17:17:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/038f53a93a614c0f9c10d12b3a9f4dbb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0597363cbdfa449d95a7839d2bec1ee8 : SUCCESS in 10m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec422990c68f41fcbdfc044b44d4d250 : SUCCESS in 9m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8adf1582f0e744c185247fb3b6e816e2 : SUCCESS in 21m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/418f1adfcef04365a097827939abd9a9 : SUCCESS in 21m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a896cabcee6742f7be5a31dbadd7ac74 : SUCCESS in 22m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5f37aa17c6c44e66aaf3b3f02a8e947c : SUCCESS in 8m 37s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/34c9767052ad41bfbc92149b7c6fa0fd : SUCCESS in 27m 28s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/890ab3fd9a2d4743b5f6f69e87cce63d : SUCCESS in 23m 48s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/50654c95295e4b60864546c740209aef : SUCCESS in 33m 07s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/66189885a12b4d8da6656bf087108fc0 : SUCCESS in 25m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/51ad7b0d6c81486cacae341c9ef93a2f : SUCCESS in 16m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a9cb54f2d57c4fce8fb44745a84f0791 : SUCCESS in 15m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a8578c10498749178ba32c50830e1534 : SUCCESS in 15m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a49604d2c6454067a8f40a06d864feb7 : SUCCESS in 56m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4e047e9701bf4ca7ab86f86e11b05d3c : SUCCESS in 1h 01m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49077d3fcab64a638d4cd76c445d3313 : SUCCESS in 28m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/774d31b2471d4197aa85844ac5d73826 : SUCCESS in 5m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5f4df23ef6b849e3a56a3244162b3fb8 : SUCCESS in 20m 40s","accounts_in_message":[],"_revision_number":1},{"id":"304d03408a4b4305dfb14831281e931802c45eae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 19:21:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5217ab9916d4ad4833d0001d911f35c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/50c570ee94524d908d12041cbbda53b4 : SUCCESS in 28m 33s","accounts_in_message":[],"_revision_number":1},{"id":"93d1864575db5a9145bccf285b462591f73bf5e7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-06-04 08:44:00.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"70f7bf5ad666633186f69e7574f8ee9f4861fafc","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-08 13:08:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"044f7f2754102514cef27e528bb3f76eb82cc9d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 13:09:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"271d938116d771c3fe86f74bd85b4e0f8636bdbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 15:31:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e384c68b2aa34e38b6d8714d52e6867a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b2e48d376ec496486fcc8b823902fa4 : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afe31ca4b4ab4683bdab7a67c70fd4d2 : SUCCESS in 5m 03s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/06460f0703394c6db8e45cf1073e18ae : SUCCESS in 4m 41s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/437bc0ceb1ab45dfab3bcfbdfd914d23 : SUCCESS in 14m 20s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/dde3b18fa61d4356a1981ae835e0b6b6 : SUCCESS in 23m 42s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/33b7fdc88c9143ea8e4f5e211955ed3d : SUCCESS in 32m 19s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/34e5aa9442714a8290ef0d68395ce83f : SUCCESS in 21m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6d8cc38e2ea647ff9072c58cc5e44f7d : SUCCESS in 15m 02s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7f1a04223bb94917bf85ead2854680ed : SUCCESS in 17m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b8f0f52ec00b4e36ad3bf4a8099d8f86 : SUCCESS in 14m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eb0eef04bebb4ddf8baf86060fdd1155 : SUCCESS in 6m 24s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/08aa2eef1b6e45db9231bb14f5c39c40 : SUCCESS in 11m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3be7dbd4e9d24ee081135500a66ed89b : SUCCESS in 58m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/834e886a575f4237b34a8f531fb928dd : SUCCESS in 53m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c9c631fe41445e6a9b1f4723c7f598f : SUCCESS in 30m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/698e66ae42984b99b7de786fdb6a31a7 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/96b0d5bbf6a5441aa894a7d2b1db2852 : SUCCESS in 20m 33s","accounts_in_message":[],"_revision_number":1},{"id":"37539af17f3b8be52b187a9ac3757888de8f1502","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 15:31:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"471990495f9ed12e8ca51a424160ffe22cc45a41","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 15:38:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2fbe4466e2045f1bc1d9e293e71e886\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ee601aaf7304991aeae88930c603767 : SUCCESS in 52s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/cb2087891b854edb9af32106f8c88e96 : SUCCESS in 1m 58s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cb8263e9858021349d5376d32387e5b6ca72db96","revisions":{"cb8263e9858021349d5376d32387e5b6ca72db96":{"kind":"REWORK","_number":1,"created":"2026-06-03 15:47:18.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/991443/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/991443/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/991443/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/991443/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/991443/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/991443/1"}}},"commit":{"parents":[{"commit":"6e6010253b285269daf711896781c9ad27b84800","subject":"Merge \"Add x-amz-server-side-encryption on s3api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e6010253b285269daf711896781c9ad27b84800"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-06-03 13:48:05.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-06-03 15:46:35.000000000","tz":60},"subject":"timestamp: increase unit test coverage","message":"timestamp: increase unit test coverage\n\nChange-Id: I288837ce96888e3426042570fc183ebc3e88de39\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb8263e9858021349d5376d32387e5b6ca72db96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb8263e9858021349d5376d32387e5b6ca72db96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
