)]}'
{"id":"openstack%2Fswift~994452","triplet_id":"openstack%2Fswift~stable%2F2025.1~I84b0b6878f0baef025000d047aa705ad241a3a13","project":"openstack/swift","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-06-23 20:02:00.000000000","reason":"Change was submitted"}},"hashtags":["CVE-2026-50221"],"change_id":"I84b0b6878f0baef025000d047aa705ad241a3a13","subject":"gatekeeper: Block X-Container-Host and other update headers from clients","status":"MERGED","created":"2026-06-23 15:01:00.000000000","updated":"2026-06-23 20:04:15.000000000","submitted":"2026-06-23 20:02:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"994452","meta_rev_id":"6666375f6d81d3facf4b608ff649968d5f9dca7e","_number":994452,"virtual_id_number":994452,"owner":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-23 20:02:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-23 15:21:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-23 16:38:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-23 15:01:00.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2026-06-23 15:21:24.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2026-06-23 15:56:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-23 16:12:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a5234341dbbd46751cc3d5e179121ec7b3a10d51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-23 15:01:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d2d410eb4cd1c7027fc826bd7c8b2e253069ea18","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-06-23 15:09:53.000000000","message":"Hashtag added: CVE-2026-50221","accounts_in_message":[],"_revision_number":1},{"id":"2eedcbafbc7810b7ebd3292896ea0570abe57768","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-06-23 15:21:24.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cdc5b96c12a54f39e7e60e192aaefde0c9abe31a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 15:56:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bf4d42633fc469b854fc4fdc459e885\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/acc091da989f4ac990d223c67800d839 : SUCCESS in 52m 07s","accounts_in_message":[],"_revision_number":1},{"id":"4600b3f7f26802ac0b5ae26e2eee48ef7bdbc895","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 16:12:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a539bdf1f9434035af28ce67a319a837\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eead2a90e7045b3a2d4555f2a37de8c : SUCCESS in 5m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cdab8f97c2c45c1b45d4e0e5847cecf : SUCCESS in 8m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/af724b30a18b4ee7aff22c19195eaf66 : FAILURE in 50m 04s (non-voting)\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/677ba4f30db74f35b0ffd911f766f303 : SUCCESS in 58m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22c1df0e61bf4f30a80e36e620aeca47 : SUCCESS in 1h 00m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ab29e1441cb2402e918950841f216b5a : SUCCESS in 17m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5a4f9fe83cbb4511951c941e8dde9761 : SUCCESS in 21m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f5b06e30f26347989df41e99ec5fe77b : SUCCESS in 21m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6ff9b9e1a186456c938f8bc6004bdfce : SUCCESS in 7m 05s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5008c44a3b7f4d7791fd3a2c6f495e98 : SUCCESS in 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0299eb26f2c44f3a9c001bcb381695ed : SUCCESS in 21m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/271497b781fb458ea625ea9072bea826 : SUCCESS in 19m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/91567c724f624f298371dd78c068d8d7 : SUCCESS in 15m 29s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/039f29f807274aec8a8959fab7ccc451 : SUCCESS in 15m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6f2ae904eeee45939e32c9106ceda9c7 : SUCCESS in 16m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f6dff0a1bbd741f099df1e6b74c98c56 : SUCCESS in 9m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e5371f6047ae44fe82323082f2459461 : SUCCESS in 19m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/585d7be3f99740d784c8cdbe6cd47e8b : SUCCESS in 8m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/388aab9230724d76850cb550dbb6c5c7 : SUCCESS in 1h 05m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4447d16e899446df8209d9eb6c36d723 : SUCCESS in 45m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7ea4bca0a3044acea1f2391d5be3547e : SUCCESS in 52m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14aa245a314e476382ac5ef8cf0b2533 : SUCCESS in 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99dd21bbac5e4843a81bd923bbfa870d : SUCCESS in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6807ccc1d61849da8769508f19b205fb : SUCCESS in 9m 01s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2aa39f53b1184125a78ed6b661b9cb29 : SUCCESS in 56m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc1b7a9470f844e896d7dbb56d3f8ce2 : SUCCESS in 32m 33s","accounts_in_message":[],"_revision_number":1},{"id":"905a9950e6c0965394f54fa49f7af56300263b4f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-06-23 16:38:18.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2a57f7c364e152cda0be7969ed5c90f2edb579c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 16:40:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"97755e60f7ae733a367281c54451e8e2762cac45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 20:02:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd0ad1e6c3434c4ebcf00958b629481e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21b53698745c4f66b4e474832eb0ffb5 : SUCCESS in 10m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c6344c38b48474cb884d4f5d4987833 : SUCCESS in 11m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d24352c860884fdf98778f6de57afc8c : FAILURE in 38m 12s (non-voting)\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bb408fc443c0483992e5ee36bff09e23 : SUCCESS in 53m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e32ca36633724cc9960f56dacfb38d9e : SUCCESS in 55m 09s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/29c680b841de4e7794abbbbe9485b2fd : RETRY_LIMIT in 1m 02s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3cccd80f640b40a48b9f7f5c9bdcc454 : SUCCESS in 2m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d5f377262d834c77966e87e14bbd96e6 : SUCCESS in 20m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3ca2c095b4bf41a49cd7e03133a6e013 : SUCCESS in 27m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c78692edc71b48d68cd3e9e4f019b2f3 : SUCCESS in 13m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a479ce89f0314ae3856e9b45db8e850f : SUCCESS in 15m 00s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2b5d250f8036434080f4e8a269afb1e8 : SUCCESS in 18m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7843169fef1944f3b1fb6cfefefef04d : SUCCESS in 5m 30s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6ed9e1a2ff3e409a802938271488635a : SUCCESS in 10m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/982bcdc5bc044bd9863099a4404f3364 : SUCCESS in 53m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b9d59edbd0414f6b9111dfd7ec37b48b : SUCCESS in 34m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/99a8e1de07a34627a916b3843688e603 : SUCCESS in 38m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/085cfe144a51455ebfb54f6fab899fba : SUCCESS in 1m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a68e0bed2144434983c51d22b6a98d03 : SUCCESS in 5m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/270e2fe017aa45a5bdfcb124ec7ecbe0 : SUCCESS in 12m 55s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39a42821d191419399e32d2136bc0228 : SUCCESS in 49m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1756429d478544e285f8b9909616a600 : SUCCESS in 32m 47s","accounts_in_message":[],"_revision_number":1},{"id":"3f4ec86c47b7a18aacf08a200308797b2ac6192b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 20:02:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6666375f6d81d3facf4b608ff649968d5f9dca7e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 20:04:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a725e32d27247cdb9fbecd449d70321\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8651aeedf80749b482bba7be6b0bc5d1 : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/746f0e5953c14a7d956d619c9a3c62a7 : RETRY_LIMIT in 5s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b6d6b50e1e315735b6f88c71665f62d64d639441","revisions":{"b6d6b50e1e315735b6f88c71665f62d64d639441":{"kind":"REWORK","_number":1,"created":"2026-06-23 15:01:00.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/52/994452/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/994452/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/994452/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/994452/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/994452/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/994452/1"}}},"commit":{"parents":[{"commit":"ff2dcbb6b9d1ff2830ee63cf4453af0b084b15f1","subject":"Merge \"Harden pickle unmarshalling\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff2dcbb6b9d1ff2830ee63cf4453af0b084b15f1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-05-12 21:57:34.000000000","tz":-420},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-06-23 14:54:32.000000000","tz":120},"subject":"gatekeeper: Block X-Container-Host and other update headers from clients","message":"gatekeeper: Block X-Container-Host and other update headers from clients\n\nThere are a variety of headers that get used by object-servers to\nupdate container-servers or by container-servers to update\naccount-servers. None of them start with X-Backend- (they predated that\nnamespace), and previously were not blocked by the gatekeeper. As a\nresult, a malicious client could direct the object-server to send\nupdates to unintended locations.\n\nNow, block all update-related X-Account-*, X-Container-*, and\nX-Delete-At-* headers at the gatekeeper.\n\nFixes: CVE-2026-50221\nCloses-Bug: #2150261\nChange-Id: I84b0b6878f0baef025000d047aa705ad241a3a13\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n(cherry picked from commit 75b050fdb1d1a6096f3765faa3b2059e28065d13)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6d6b50e1e315735b6f88c71665f62d64d639441"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6d6b50e1e315735b6f88c71665f62d64d639441"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
