)]}'
{"id":"openstack%2Fswift~996143","triplet_id":"openstack%2Fswift~stable%2F2025.2~Ia614e1e3827bbec6d147835df33ca198603e61be","project":"openstack/swift","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-07-07 21:34:38.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ia614e1e3827bbec6d147835df33ca198603e61be","subject":"CHANGELOG for 2.36.2","status":"MERGED","created":"2026-07-06 17:35:55.000000000","updated":"2026-07-08 03:35:26.000000000","submitted":"2026-07-07 23:40:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996143","meta_rev_id":"98016230a472941ba641b3da1dd41be4e6eb788e","_number":996143,"virtual_id_number":996143,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-07 23:40:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2026-07-07 21:31:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-07-07 21:34:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-07-07 21:34:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-06 18:18:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-07 01:21:57.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"b7fac4c668983cb7b1b3b6478092822fe69646a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-06 17:35:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85e5604eb13da3907a65688f4260d51a129c8cfa","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-06 17:39:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d2d3bd86ad570c3ce65b5648ef3b672c0ae0989b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 18:18:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e2853fd49144fb5923e596dbf864673\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d5cd65395634917b10b45cce8809419 : SUCCESS in 13m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d31925453d02430cbc66cc20d91d939b : SUCCESS in 8m 37s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2cd4300c4efe425096e91b4bd6a515fb : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d8be40e1c8f9416aae0f7cfd824e5f89 : SUCCESS in 25m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9de33d314e26433ba7034e10342253cc : SUCCESS in 19m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6e14f4f06b4745809323fcb85d90c2e5 : SUCCESS in 31m 11s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2344a71eb6da4ee9a914f89e9bbbf3d2 : SUCCESS in 32m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa790f8a07544dd5936fe4a1b7f36fe3 : SUCCESS in 12m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ebf43ddd7924615bc23d61ee5238896 : SUCCESS in 3m 32s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"c93e5eb35a244d3cbacfbaf66b1434fbc2413935","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-06 18:35:24.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"3569c43c91359109d79a224cb21483cb5a36bcaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 19:16:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17b816953bd5411fb799cf6a280ef3ad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bd0131d847f4058b33e5b1c904935cf : SUCCESS in 12m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6e1d65636604534b82b87e39a9def24 : SUCCESS in 6m 22s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5b53db50fd5043b18dfa91150d92efd1 : SUCCESS in 7m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/02fa7a4dfb2b4c6d83c7f347d33e045a : SUCCESS in 25m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3bea32e1d25e4313b60a3ac1451a686d : SUCCESS in 10m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4ccfc43b9ebb499d8e5751a90c8ddcad : SUCCESS in 27m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ac5bd99dd3c441409679d710f9524c17 : SUCCESS in 31m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9273507a1e8646428fa0375c88e10869 : SUCCESS in 29m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51b716e4619a4fd8b8b6aa2b04ea7d04 : SUCCESS in 4m 48s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"1d1633fe234bd968367679b4e715392a816adb2b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-06 19:41:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cfb0fc2f1d41ee5ef3e9691b6b7e76d82c66f0c8","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 21:07:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02fc4c75166f4aec92da3408634c963e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/339786ac04024f0d834bcd5780e210ae : SUCCESS in 1h 02m 17s\n- swift-tox-func-encryption-py312-arm64 https://zuul.opendev.org/t/openstack/build/ac77cec503214cd4bfcacfd15a9b9669 : SUCCESS in 38m 44s\n- swift-tox-func-py312-arm64 https://zuul.opendev.org/t/openstack/build/7d9dd78c44e84b3b8605bf19e28ef664 : SUCCESS in 36m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5cb0670e80f04a2c8f6f6c03e8eb5706 : SUCCESS in 10m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0c90e5fffd934c22af634ec4e0a2bac9 : SUCCESS in 10m 10s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3dfc129c740e49c88090f8a575badf24 : SUCCESS in 26m 27s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/b501bda7d71f422c8d2adf9501885213 : SUCCESS in 10m 16s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c043a2a46e747b593522f0c2b464b23 : SUCCESS in 19m 05s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/31fc871264a44e7792d07ab01c9e8a23 : SUCCESS in 10m 53s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/02d900653c75410d9c9c1013a0304278 : SUCCESS in 10m 47s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/b5db91276bf34188ba34939285dc5199 : SUCCESS in 9m 42s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/7c5440ba39444dfcae84131fb55d8ee5 : SUCCESS in 19m 12s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/15805cc9df614a428f59e913c6996b4b : SUCCESS in 21m 48s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/ac743a8711564d6296cd0c164ed5b861 : SUCCESS in 10m 22s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/d8542c948b644608a629bb47e91403e3 : SUCCESS in 17m 10s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/83a2010ccf4b440393dccba79b450e73 : SUCCESS in 10m 43s\n- swift-multinode-rolling-upgrade-dalmatian https://zuul.opendev.org/t/openstack/build/7ff3a90a2f6447e09d19b813f743942c : SUCCESS in 10m 47s\n- swift-multinode-rolling-upgrade-epoxy https://zuul.opendev.org/t/openstack/build/baf9bec4844843908d791c1ac90790ae : SUCCESS in 18m 42s","accounts_in_message":[],"_revision_number":2},{"id":"765c9df2d2424c2f09b526f4e9f6b59c078b0dc0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 22:16:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90636628377e4dca8b23f8a5e869ac73\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0121d2b86dc34af5928e3b27a9b06f57 : SUCCESS in 34m 03s","accounts_in_message":[],"_revision_number":2},{"id":"b7b2a7879ad8de4a59701aa767fd504f2a48d2f6","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-07-07 01:21:57.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"14c257808ee026b876c96534bcdae2e462a9614c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-07 01:42:54.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"6c71c6e7fc618590547d78e8b66a3dacb0553b13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 02:07:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a0143df4425443db63b4e4f8117b06b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/079bab0ca05d4b6bae48ba0f699f6397 : SUCCESS in 9m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/092e3b4b84a245b8bd5ff4f384219179 : SUCCESS in 8m 28s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fb889366e8704722b9fdc087a7d1f1d3 : SUCCESS in 4m 20s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9dc789f3fe4c42308443fab80c8dfbdd : SUCCESS in 10m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4f8336641c88405cb868568a4a6ccb07 : SUCCESS in 8m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d2c71a9610664f619c4190c1e9aa8526 : SUCCESS in 22m 59s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/fcfa24b5099742afbe796d2496201029 : SUCCESS in 11m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d5d7b3b4a2e4973aa7ce13717e88932 : SUCCESS in 10m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c58997c700e44fba8be407da67e34511 : SUCCESS in 4m 02s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"5d90f9cbe375025880b020cc24e07ef871dfcb40","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 05:24:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60290b4a7c2e4c13b01280b269d74167\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2bd5ec3515cd4b4496cf9b7961e0c5b2 : SUCCESS in 1h 11m 51s","accounts_in_message":[],"_revision_number":3},{"id":"add3002c967c6e7f79dedf078c909171067f427c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-07 19:34:22.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bdbd7a0b250c9b4a461468876bf85dc5ebc3a535","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 19:34:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"320335fa3d40bc2bb5d2074f273572acebf55945","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-07 21:31:19.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"d12186b577198deb991b8149813928286930d2e5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-07 21:34:38.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e90710f1448db52ada1005c9033154b2c5d7307d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 21:58:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/198dffa183eb4c6c9f3bd4fe716838ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebc124fee1674c31beb608c6293161d7 : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ad37fbc36b5442c84b48ccf4b92cc5a : SUCCESS in 7m 55s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/687851c7088a40cfbd167c447da135e2 : SUCCESS in 4m 18s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8d6a7862ba324b529f82c0ba042a74de : SUCCESS in 18m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0c612ed3ac2a4a8fbc29fa184a6fd120 : SUCCESS in 18m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/401d1646c6c04d7e9fe443a4c6f64465 : SUCCESS in 23m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5629583b29f2416fba6d381152f76980 : SUCCESS in 11m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb56688806204947a8335d6e1bae2665 : SUCCESS in 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8c290de75bb47d99860c17b25f65a2b : SUCCESS in 2m 26s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4},{"id":"34608b0d1e4afefa5dc67e96f9fd86043e068474","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 22:26:39.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"758b32f0a462b83d9152fabf5a34f42f47cbaa53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 23:40:13.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77577cf7182e4165bada3c5646f02412\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aefece3fe62e4e0e865299f2cd3785a1 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b8ff5a88f904cea92e5b50186ce7e7d : SUCCESS in 10m 50s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0790e909f5d3422192d444f5d478ec1c : RETRY_LIMIT in 19s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c7ba3223b84d47f8a9f04b53e52b70ac : SUCCESS in 21m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/882bb5c790bb4bc9b011dbc53630aa0e : SUCCESS in 8m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c967489150e64f89ad70bbc74707914d : SUCCESS in 28m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/aaeabcd252f444cb9f0d651e8daca1eb : SUCCESS in 10m 53s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a8e4e8b3813c4da1a19a881d9499e64e : SUCCESS in 15m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/655842d746ff415eb6519a376c8f5d16 : SUCCESS in 15m 45s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b799c0343baa4c79be0333503b202c0f : SUCCESS in 15m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5d35f98b96274bca8de26433880e1389 : RETRY_LIMIT in 6m 25s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4c739649f085418e85b6d18d80d3e4b6 : SUCCESS in 8m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89cde42c85164657beeaaca220322497 : SUCCESS in 27m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b15296623ef743e285ee0f77b0c8dd52 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":4},{"id":"97d58afe9cbb4b89baa007fd8ea3b3e812aa40d9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 23:40:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"28a85b3c3af5eb32241713c1c8a2412c92fe680e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 23:43:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20efddc9283c4dc0ab78cee8f4973b9c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d308914b7df48e892e52b7c77680fd4 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b6a8eaa3ff84dce80b0ffc697a108c4 : SUCCESS in 41s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d927c2657d2d43248230e90d2550ba2f : RETRY_LIMIT in 5s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"98016230a472941ba641b3da1dd41be4e6eb788e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 03:35:26.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17d7f8fbbfc74da3a5c8d0cecf9547e1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bea35c90a37142d189954f4520f904d7 : SUCCESS in 57m 27s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"bbd9c5e012d9f4821d67b9dfd538180de63e733b","revisions":{"e55c46540742ee511338c70af5652f90e0fd93f6":{"kind":"REWORK","_number":1,"created":"2026-07-06 17:35:55.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/43/996143/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/996143/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/996143/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/996143/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/996143/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/996143/1"}}},"commit":{"parents":[{"commit":"f8c92d5027229796162f79893d5b117f440661ec","subject":"gatekeeper: Block X-Container-Host and other update headers from clients","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8c92d5027229796162f79893d5b117f440661ec"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-06 17:35:36.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-06 17:35:36.000000000","tz":-420},"subject":"CHANGELOG for 2.36.2","message":"CHANGELOG for 2.36.2\n\nChange-Id: Ia614e1e3827bbec6d147835df33ca198603e61be\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e55c46540742ee511338c70af5652f90e0fd93f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e55c46540742ee511338c70af5652f90e0fd93f6"}]},"branch":"refs/heads/stable/2025.2"},"83b28f18fb78a79d5be5f20a273b34bf534259d5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-07-06 18:35:24.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/43/996143/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/996143/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/996143/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/996143/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/996143/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/996143/2"}}},"commit":{"parents":[{"commit":"a0ab476057de035e08ecadf3cebd09ac5d793f0a","subject":"CI: Fix swift-multinode-rolling-upgrade-dalmatian job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0ab476057de035e08ecadf3cebd09ac5d793f0a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-06 17:35:36.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-06 18:34:56.000000000","tz":-420},"subject":"CHANGELOG for 2.36.2","message":"CHANGELOG for 2.36.2\n\nChange-Id: Ia614e1e3827bbec6d147835df33ca198603e61be\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/83b28f18fb78a79d5be5f20a273b34bf534259d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/83b28f18fb78a79d5be5f20a273b34bf534259d5"}]},"branch":"refs/heads/stable/2025.2"},"99b429bc91256bdd15924b1c0e70cbc386d5d6d6":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-07-07 01:42:54.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/43/996143/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/996143/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/996143/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/996143/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/996143/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/996143/3"}}},"commit":{"parents":[{"commit":"f8c92d5027229796162f79893d5b117f440661ec","subject":"gatekeeper: Block X-Container-Host and other update headers from clients","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8c92d5027229796162f79893d5b117f440661ec"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-06 17:35:36.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-07 01:42:44.000000000","tz":-420},"subject":"CHANGELOG for 2.36.2","message":"CHANGELOG for 2.36.2\n\nChange-Id: Ia614e1e3827bbec6d147835df33ca198603e61be\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99b429bc91256bdd15924b1c0e70cbc386d5d6d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99b429bc91256bdd15924b1c0e70cbc386d5d6d6"}]},"branch":"refs/heads/stable/2025.2"},"bbd9c5e012d9f4821d67b9dfd538180de63e733b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-07-07 21:31:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/43/996143/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/996143/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/996143/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/996143/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/996143/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/996143/4"}}},"commit":{"parents":[{"commit":"d1dd2a08954a1aef93e4482c5ac3dc4b4d6a0260","subject":"CI: Make failing CORS job non-voting for now","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1dd2a08954a1aef93e4482c5ac3dc4b4d6a0260"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-06 17:35:36.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-07 21:31:09.000000000","tz":-420},"subject":"CHANGELOG for 2.36.2","message":"CHANGELOG for 2.36.2\n\nChange-Id: Ia614e1e3827bbec6d147835df33ca198603e61be\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bbd9c5e012d9f4821d67b9dfd538180de63e733b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bbd9c5e012d9f4821d67b9dfd538180de63e733b"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
