)]}'
{"id":"openstack%2Fswift~996153","triplet_id":"openstack%2Fswift~stable%2F2025.2~I2480dd38be2ee78131d841a5cb6cdf54ec915466","project":"openstack/swift","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-07-08 15:58:41.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I2480dd38be2ee78131d841a5cb6cdf54ec915466","subject":"Fix rolling upgrade jobs","status":"MERGED","created":"2026-07-06 18:35:24.000000000","updated":"2026-07-08 20:40:24.000000000","submitted":"2026-07-08 18:31:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996153","meta_rev_id":"01a0e677dc74110b2ad4a53bcc66cfa733883a46","_number":996153,"virtual_id_number":996153,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-08 18:31:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-07-08 02:57:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-07-08 02:57:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-06 19:53:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e4aa1885b239fc9eabff1e2ea6a62ab8c8ba4073","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-06 18:35:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8ebcc4800588a70406b9ab617ff40b819d0517b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 19:53:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7768212456724eaeaceedb1dc4cc0b72\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a29fb6671918478b9c8d27ed2f534cac : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b5692708c054afd97fd21ef983cf73b : SUCCESS in 9m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/fdf5d99de75b4a90b8e44d219f63fa17 : SUCCESS in 1h 08m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/30366b9720054b12a509ef4050120d5c : SUCCESS in 1h 04m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d94f45ada5e4f8dba4ddb4f8e468543 : SUCCESS in 1h 05m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/24c75d47cbef47a8859f808f0368628b : SUCCESS in 14m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/15f1195b75454d9b89497e56b48339fe : SUCCESS in 18m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5e8394832aee40dd9ff677204ed3edb4 : SUCCESS in 12m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c3b46e8b40df4aa986d3588aa96e576d : SUCCESS in 6m 31s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0661b8be0dda452295dc4ad3bf424105 : SUCCESS in 28m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4de93c4b911e4edf8bff3522e6fb19d1 : SUCCESS in 22m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/975274e4a240465f96ff648b6502aa47 : SUCCESS in 38m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ac5effe506984e7ca774ccb4d3d9b082 : SUCCESS in 12m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f11e5766c64c470cab9cc3d93b982605 : SUCCESS in 9m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e8588c6c2d2d4a1aad27cec83eff0e7c : SUCCESS in 14m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/84d9512fff8d461f91c417b0c8a5388d : SUCCESS in 9m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/217ba65494164a8ab9d2aa5480559969 : RETRY_LIMIT in 5m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/85b1b8a83fac4b3a99d7ecde3d11a4cc : SUCCESS in 21m 42s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3ece92da248e4fb4b4b8b5e5c0bdc241 : SUCCESS in 8m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f3b3d28402d449cb3f58c3f3fd20a49 : SUCCESS in 49m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b9f19143e8c34d04963c7a09d152fe68 : SUCCESS in 27m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/540550c190f948c4bcd8ffa661937857 : SUCCESS in 26m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5844e1d1bacb46faba90ef3ad0d2469f : SUCCESS in 28m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4491e5a205104c648f86b96696bd146c : SUCCESS in 2m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c01eaa31e0334d8fa6929493538f223c : SUCCESS in 19m 19s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93fdf4274f914c9daecd6753a7b0ca0a : SUCCESS in 56m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce090aaf9c5a4ed28692351861b41eec : SUCCESS in 1h 00m 32s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"70fcfa9db6a84707e2724fb681569b0a5d7cc132","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 23:03:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7126a5f4e1a403a842563d77a3aa297\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3eb32bff360847b6a6820b3881925f4b : SUCCESS in 1h 13m 49s","accounts_in_message":[],"_revision_number":1},{"id":"261cc826efdf6e52a6bdfc5dd1ab088c340e2a22","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-08 00:08:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9a7f2c3cf4e3fbb9c4392014aba24fad69bd3669","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 01:15:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/744edab9be6a419aba8a5d986e92c8a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a24c3bcec7d47ec90f493bf0326240c : SUCCESS in 9m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f834bdf497254db0a4c5a0cedcc2abfc : SUCCESS in 8m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/d2a78801e0c74aa99d5408ffd70972ad : SUCCESS in 1h 02m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0860b4070d604b7cbae682fcd1644e00 : SUCCESS in 54m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/06af5bdde7bb4a33b4cc7955001fca26 : SUCCESS in 1h 01m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d01af62639444179f8a2c1cccaeedcb : SUCCESS in 20m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c601d254685d46289e95c0ae488928b9 : SUCCESS in 19m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f616cf233af24cfcb87d33c00ebccd0b : SUCCESS in 20m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8ad3788d870d4efda3ed0bd749d680ef : SUCCESS in 6m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1067468eb47b4e4cab80fb56b90f483b : SUCCESS in 24m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6dbcc9453df44e5b94eb343e47500ef7 : SUCCESS in 24m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cd74ad6de4c8419b8ff9ba8985d2a506 : SUCCESS in 28m 36s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c284a31eac354484860b6f09693790a4 : SUCCESS in 29m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/868883f0f0d849f29c4d45934477a2f4 : SUCCESS in 14m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/da857fa70e6e4cd9b266e49bed65de96 : SUCCESS in 15m 37s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/bbe5b358c05f4593b78ca14bf1fcedb1 : SUCCESS in 14m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c9d26ba1aa1b41f787453fca1c61e35f : RETRY_LIMIT in 5m 14s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2f8a81877b6b4b34b702ea4ecc766879 : SUCCESS in 16m 54s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1838414abfc041d9aac5ba101f99b6e2 : SUCCESS in 8m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/acdcfc6fc1b943079c6eec60a1cb6bc5 : SUCCESS in 54m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d4fd6369f0af4be3b799875dd7575139 : SUCCESS in 47m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/07d6714e55754939b455db25587e3f98 : SUCCESS in 47m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/082884f57bea46b99b179ccd9a5b46f8 : SUCCESS in 31m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37ed19bd0c1f499fb1ead657adea9f06 : SUCCESS in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f68bc979519d42168db993119d936a4e : SUCCESS in 19m 23s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0606870f4204386a0205459293efc2c : SUCCESS in 57m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b615338282274ee593dc1392ed7127a2 : SUCCESS in 57m 47s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"7cf429879b7f781e1e70c407a8125f8331829d5b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-08 02:57:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"db95545deb99f260f024aa7317d56f552519897c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 02:58:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"430eded22499198381f0bae496f15a2d88833cb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 04:56:06.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/68794566643e4f66b9174bbf3a451d76\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64cde8ab2c2147cbbf88fd3f6994c50b : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5a31f24690942ee9cb03dc3bc20b513 : FAILURE in 15m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/a6e98401b54546c59231dd1f36a29606 : FAILURE in 19m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ef2c5e1171774d4db227cdffd72cef93 : SUCCESS in 32m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37541cef69064d95a65814e043a02d33 : POST_FAILURE in 31m 30s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ca1ed8ba7d274a809e3cccca0496ec08 : RETRY_LIMIT in 6s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9e1ddd9fd9ea4544aae8782aef0da15c : SUCCESS in 10m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/25c59f5f40ba4a128daa36434fcd5a66 : SUCCESS in 9m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c2a116ab9788412ba9b534249c3503ad : FAILURE in 24m 18s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d42ae5da4132418c89083ebb2c49447c : SUCCESS in 21m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2f373318ee1343ebb0a2bd7bb8a138e0 : SUCCESS in 8m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/659fef8bcc2b423987676360608c734c : FAILURE in 15m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/42acebc46f3b4ad686d0338d2894e80b : SUCCESS in 8m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a77ce436650a46679a482d47944784a3 : RETRY_LIMIT in 4m 23s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2576425d9f7c486ab849915a0c4b4efa : SUCCESS in 4m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/70a07f24137e4343b0584a8ff046fee3 : SUCCESS in 37m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5b7b4536a0f94f4690e4600d3f0eed55 : SUCCESS in 48m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7bb6311f3d8f4ae295ba87928a6d341a : SUCCESS in 29m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94e62dc4483043c99b19cc85775324e9 : SUCCESS in 10m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff210aacf8fd407eaae884b01831a6b2 : SUCCESS in 3m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/de49a30ab89f44eb9fa6c116726055e2 : SUCCESS in 19m 09s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f0a6647f395482e9b12aa03bb089f9d : FAILURE in 28m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf9a84a3f0c541ddbcda884c035c037a : FAILURE in 26m 10s","accounts_in_message":[],"_revision_number":1},{"id":"7d9c22867d2b625b60e54afa3069e74c2ac4763d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 05:54:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/817af336929d4189a164b379afaa5c37\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dced72921f5c45cc848c033eddd2e72c : SUCCESS in 1h 02m 25s","accounts_in_message":[],"_revision_number":1},{"id":"7fa046657d6a3e713c64ca3c67ff302c336cf9fc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-08 15:58:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"835023fe4f47dcd63aab5f3d374d8d9d2723b913","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 17:13:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7099431d399e46279338353a11f89641\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a681c6884ea2417db9df5d8427446786 : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de5b34d76f88491fb7b27397c2b5accc : SUCCESS in 3m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/42c664df5a9546fc94766c4460406ae1 : SUCCESS in 1h 04m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9e4abbcb8f69456f965e5dfb54264351 : SUCCESS in 1h 06m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8f012e75afb40c99bdb1031bf4dc439 : SUCCESS in 1h 03m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c82af2c238a949a0b8c83c995e1f17dd : SUCCESS in 11m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/42c3cf4618254043a212cccc29042321 : SUCCESS in 12m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a19338984fe24d3081447975014d6093 : SUCCESS in 22m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3b1682a226cd4e2e9eb7d2ac6cccf7af : SUCCESS in 6m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/94cf63fdcb954913b50d2b2a0d8d9970 : SUCCESS in 11m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3e106ea928ff4f84a7ea7549593b15c8 : SUCCESS in 22m 18s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/45d3a9d1846f47148e4d0272b09337e6 : SUCCESS in 12m 50s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5da998ac6a434ee9ad27f1fd22182bbe : SUCCESS in 29m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4cd3a0e6199c45c4b95d0a0f463dbf9c : SUCCESS in 14m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0b927249837344b3a36902a0baf5b650 : SUCCESS in 15m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7bc19e4886774f7fbdf02815dd22dc73 : SUCCESS in 11m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ed76ea8a5a374f60822d4e2bef6f6f14 : RETRY_LIMIT in 3m 24s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1edab9d8c3cc46e1831836cbd8d2ea33 : SUCCESS in 20m 16s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ca2add407e29472897cd106d88cfae80 : SUCCESS in 6m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5f9ef0b8cc084739b5e5fbb426eff6c8 : SUCCESS in 56m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/71574217e49d45fb85d811db8c5659d2 : SUCCESS in 27m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7b1d7e52aee54dcd90e945afc519e7d9 : SUCCESS in 26m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d14c702f06b4596877a88b3725cdab6 : SUCCESS in 27m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6050279b77204256a21f1dc8faae5e8b : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a16df6dd337646c2ba8a656dc106fd2d : SUCCESS in 10m 22s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43d07810864b4980b54dbbb819483bc5 : SUCCESS in 48m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7aaedb3a3c541bab6659d104c62e1a6 : SUCCESS in 1h 03m 01s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"701c094a58339cff002021317e689896fb3ad227","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 17:14:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cc25d668f941c52d3e028e5ae66e34faec72bcd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 18:31:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a19feedd8ee47b0aec09f3c5f9a3287\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d70c4884d7a4502b0a8a1b6a4acfd8d : SUCCESS in 9m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24afe5d9d9554ae0ae18c1bbb5ddbfbc : SUCCESS in 9m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/e36a536a935542969f0290f40cb041f8 : SUCCESS in 1h 11m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/45e79a7db3fb4da8805e55b0da453ebb : SUCCESS in 59m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a74c6f1f43084d8a8d59fb3f542043d2 : SUCCESS in 45m 39s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/939889c7901c45b0845ec2c727c968f5 : RETRY_LIMIT in 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/696d7bfc257345dfb904096200cf1e03 : SUCCESS in 26m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f7b7a81225484ed6b541d26149f4460c : SUCCESS in 20m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1caa420b27df409aa389af7dd5fc3349 : SUCCESS in 27m 25s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/02558fc751f6400bb55d6d4bf7352315 : SUCCESS in 30m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/489d6caee71b4adea6a88e35d25055f9 : SUCCESS in 13m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/008c6e8f5fa840b4b32e027888b5502b : SUCCESS in 14m 45s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f452330d371444ef82451d0ccd9709c9 : SUCCESS in 15m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d65034ecc82749949b33823831af314e : RETRY_LIMIT in 3m 41s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d4802845e12647aeafcda19ae025ce0a : SUCCESS in 8m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/abb353b565e946e3808999acc71ca53f : SUCCESS in 53m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0f31c2c630fe4da7a700f901b1c0edca : SUCCESS in 28m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4095000af4dc40cca1959307719ddbaa : SUCCESS in 38m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e1444bbd00349fbbc46d2882b7fbc74 : SUCCESS in 22m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e798776c5b6f48dba940585ca06fea02 : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/909a33f1a61b497facf9f24671bedb0f : SUCCESS in 17m 48s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04a412213c544d9a99138ee0749710bc : SUCCESS in 35m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8fb005b6a9a4b96bd07f7170706171e : SUCCESS in 55m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ef893835c4d76f97586ab460ae408de9e7faf354","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 18:31:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b5df00f082a6ff6d6882794417069e642d4a3b5d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 18:35:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/315c040a9334433b9b69442fa8e108de\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/225531740856426494c93cd636ae7947 : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/9812e7fafc784d2890c9732ac17ab02e : RETRY_LIMIT in 7s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"01a0e677dc74110b2ad4a53bcc66cfa733883a46","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 20:40:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0aed2ad9f78c4b72a8f231cda8e77d93\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/de4e1e7bc5ad4f04bfd4a2f275199392 : SUCCESS in 32m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0fd83e8da46fcdfc4a5771f2c19ca185464d5327","revisions":{"0fd83e8da46fcdfc4a5771f2c19ca185464d5327":{"kind":"REWORK","_number":1,"created":"2026-07-06 18:35:24.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/53/996153/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/996153/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/996153/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/996153/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/996153/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/996153/1"}}},"commit":{"parents":[{"commit":"f8c92d5027229796162f79893d5b117f440661ec","subject":"gatekeeper: Block X-Container-Host and other update headers from clients","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8c92d5027229796162f79893d5b117f440661ec"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-24 21:35:48.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-06 18:34:56.000000000","tz":-420},"subject":"Fix rolling upgrade jobs","message":"Fix rolling upgrade jobs\n\nNot long ago, tox dropped support for the --showconfig argument. This\nbroke a bunch of things, so Zuul responded reasonably: check the version\nof tox, and if it\u0027s new use the config subcommand; if it\u0027s not, continue\nto use --showconfig.\n\nHowever, we used to have some pins on tox versions. When tox realized it\nneeded an older version, it\u0027d make a new virtualenv, then re-run the\ncommand with the old version.\n\nSo:\n\n  - CI install latest version of tox\n  - Zuul sees \"new tox\" and tries to use the config subcommand\n  - tox sees the version pin, creates an \"old tox\" venv, and re-runs\n  - old tox interprets \"config\" as arguments to the testenv\n  - tests fail like\n\n      The specified regex doesn\u0027t match with anything\n\nNow, force the old version of tox for the old branches earlier, so Zuul\nknows how to do the right thing.\n\nRelated-Change: https://review.opendev.org/c/zuul/zuul-jobs/+/979136\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/881597\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I2480dd38be2ee78131d841a5cb6cdf54ec915466\n(cherry picked from commit 08f1eb9c9fdd90a1b4dc43fd5039c7b49f379e07)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fd83e8da46fcdfc4a5771f2c19ca185464d5327"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fd83e8da46fcdfc4a5771f2c19ca185464d5327"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
