)]}'
{"id":"openstack%2Fswift~996177","triplet_id":"openstack%2Fswift~master~I261642065d8905a41dd3c590c3b4514573556134","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-07-17 16:19:37.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I261642065d8905a41dd3c590c3b4514573556134","subject":"CI: Make CORS job voting again","status":"MERGED","created":"2026-07-06 20:51:37.000000000","updated":"2026-07-17 18:28:03.000000000","submitted":"2026-07-17 18:24:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996177","meta_rev_id":"76b61ccea2e62c0790898307c5148d40a3cbc4ba","_number":996177,"virtual_id_number":996177,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-17 18:24:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-07-17 16:19:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-07-17 16:19:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-06 22:22:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fc8733efb323263f29240c96bf320362ec7abb72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-06 20:51:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1281c87b225248a06e567eca424e96b8afcafba6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 22:22:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c36b415faab44e53b17d512367971d0c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f49e2449b1244c19059400a1ceb14a0 : SUCCESS in 10m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79d0cc335a484bd29582536d1c55a01e : SUCCESS in 5m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/bef0b813711b4e0b9cf666ecdbee88a7 : SUCCESS in 1h 02m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d999f50f4717484db1fbfede118ed1cb : SUCCESS in 32m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/51cb06efe5694eb7beb68ded7d41667e : SUCCESS in 47m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7e4ad4f80c248308dfbafe021c09b60 : SUCCESS in 1h 22m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a15bb065463f4a46945c3430476d3519 : SUCCESS in 20m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bb579792c23449349e61120c177b18e4 : SUCCESS in 17m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/23b946a448614362a440897bd6936b73 : SUCCESS in 18m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8bee9c124d47455ca0b762641987d3ec : SUCCESS in 4m 57s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/62ebe0a873474f73a56bc044e4ba49b0 : SUCCESS in 25m 50s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b086232f087c4fbfb1b9863b2f5c8785 : SUCCESS in 12m 25s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f2c8eda39e9345a4bbac35d191761614 : SUCCESS in 11m 17s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/8041618aea4a419b8af9657c9fa1a613 : SUCCESS in 17m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fe139f3643b546e98485b4712ae85ec9 : SUCCESS in 14m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/54761b64e44242a9aa5237043d0b9f49 : SUCCESS in 8m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/35ef3ed3d4414c5e960f4c6d7588659c : SUCCESS in 15m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/63ce049e1a9b4096a65d386a5f658aca : SUCCESS in 5m 53s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9cfa14ebff71480d958dfecc3585001e : SUCCESS in 17m 51s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3e3513fde6e247ec9f421f1aa0dd4adf : SUCCESS in 9m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a1cdc94e5cf43f6bc477e9383e248d3 : SUCCESS in 49m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/109f2a2b09f242928bf89fca1ccedd41 : SUCCESS in 22m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bdd38dd695fd4fdf80eed2b26a5ccf53 : SUCCESS in 23m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ffc513d734b4faeb66dd80f2bf15d24 : SUCCESS in 10m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d32318064f5a40619a9e2d6b7caf0a9b : SUCCESS in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5202051c5a964ecb99c4480f2141d4d4 : SUCCESS in 19m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2eeb1cb39768478d99bc00243b6e7a75 : SUCCESS in 1h 03m 38s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"4873578fe3cbd1d5f850f6bf05b5bb8cd26d4f86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 02:41:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/481f7b16e2dc436f9fc3bdc78a9802bd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/74a76f0d0fc94b5da6f82c5e5c337e48 : SUCCESS in 1h 12m 11s","accounts_in_message":[],"_revision_number":1},{"id":"be1a700cfa3cfab83167935f937aa84823aa09ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-09 18:18:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"49d8aa3baf72d3d363d3689cfe339e5ff09ffca7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 20:50:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5727bdc6b57d46c79ab17c5fecc480fb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cc1c3c5dfdb4706bea2933d2c35f388 : SUCCESS in 10m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86c0c0c4726a487680e2fe3676f9c539 : SUCCESS in 7m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/eb3046ad5f4c4635b00ed64d1d82b5d1 : SUCCESS in 40m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ea30a567ffda4f52bdd47a18f0cd3cc0 : SUCCESS in 1h 03m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/044fbe626b384c9081611284ff125b1d : SUCCESS in 1h 00m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d4d4a135f68c4076b481c24e238f1ca4 : SUCCESS in 1h 05m 16s\n- swift-tox-func-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/93f041bbecfd4a35af0cdad6bacdfc8d : SUCCESS in 32m 58s (non-voting)\n- swift-tox-func-encryption-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/3e0f62e41ff448db80e58c7b5ff59a8a : SUCCESS in 30m 44s (non-voting)\n- swift-tox-func-ec-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/fdbc318a688b44fd85d59b7d04249699 : SUCCESS in 21m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3a95a6f35edc4814bddf2fd59dd6c1f5 : SUCCESS in 7m 08s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cd4a5f81c6fc42428b5351e0001e769d : SUCCESS in 16m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d0fea509f1c94741914e143602989dac : SUCCESS in 20m 46s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b78f31b681294935a6297b7238b4a51b : SUCCESS in 38m 27s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/d46ead1cff674f7985ed019181bc52fd : SUCCESS in 20m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7385848343ca4ac4b2dc16431b68ec22 : SUCCESS in 17m 03s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8dc05f58f2a3477f86f6cd0aeea52304 : SUCCESS in 9m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/20ca2e74a63e45b795d43e5c4e785819 : SUCCESS in 9m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e8562c548ba448ab88d2954ef96218e8 : SUCCESS in 4m 14s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dc34ba062c904034adb7c3e4925af802 : FAILURE in 7m 56s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/80c3c27730314377b66f7757466ec216 : SUCCESS in 10m 31s\n- swift-probetests-centos-10-stream https://zuul.opendev.org/t/openstack/build/22199fe6bb9e4a9e8e4671f8fffe1f56 : TIMED_OUT in 2h 01m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a01ab9e8cea7450d9f2c20ba5354b9f5 : SUCCESS in 30m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cb301ae8fdb842caa84a7c6244775ce2 : SUCCESS in 54m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20eb3c6291f642aea5e33d8859902e24 : SUCCESS in 23m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08c33ffb0cb14e089a2ba35c3f8104bc : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9b41e2b2bb5a46e4b1d4d54be63698ed : SUCCESS in 19m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e8b18b931834f4faf2c852773000d28 : SUCCESS in 50m 46s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"39cefc2f6ce0e92564217f425cd0e9165349bccf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-09 21:47:02.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"6c87233dcb472bf4cc9db3d084c5949084649648","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 23:02:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/451a3d54dfa54132a02065f78c4d420c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/21de8cb05bfb42ef91524192ce376e01 : SUCCESS in 1h 11m 43s","accounts_in_message":[],"_revision_number":3},{"id":"68f643fbbad0202c7527e13e3a1311f8dca8e5ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 00:09:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37769a3c36d9424bb5f251d135156c0e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2e6ae4dcb494a409676cf79f3feb394 : SUCCESS in 10m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3966679d751c42caafe3a11b5bbb7e5e : SUCCESS in 9m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/a0742b44a9a540a49d77afaa430fbdb9 : SUCCESS in 1h 07m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e33826e9fbe04cb7b62d3b1948569c5d : SUCCESS in 1h 06m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/43fc58a311e24e1882350aa5d3aa04d9 : SUCCESS in 48m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d7ca7a7c9a5436c81c040f6562db641 : SUCCESS in 1h 08m 24s\n- swift-tox-func-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/db426c8e1fc44642aaa5607cb2005b18 : SUCCESS in 24m 11s (non-voting)\n- swift-tox-func-encryption-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/51c4854beb7b464ca7077d7d02464a0d : SUCCESS in 23m 21s (non-voting)\n- swift-tox-func-ec-py312-centos-10-stream-fips https://zuul.opendev.org/t/openstack/build/dc79f719134a45dbbe2634cfcc30fc9f : SUCCESS in 32m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/89e3926c3d4c433eb359f4c62b590623 : SUCCESS in 7m 36s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/32381ac41ce840aea3d588d9b50a5fa9 : SUCCESS in 26m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d1748282f1ec4ab386c3aedb17a84054 : SUCCESS in 9m 29s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/644bb61e372e431fa5d7614d76819505 : SUCCESS in 30m 08s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/896bba26570643aab1897823773e1ae0 : SUCCESS in 21m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ec9eec5a7c274bbbae4bd89913e3f9b6 : SUCCESS in 14m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5402c7660c454c849d81a330a30a768c : SUCCESS in 16m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/14bc5ac25c17498389f51e3c4cee8aaf : SUCCESS in 15m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2a0d4508018545cdb8f0994c32c73b72 : SUCCESS in 6m 27s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4bc533c64ec741739f797205833af6ac : FAILURE in 11m 03s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/18768e3d43d248dc97a381fccfe535d8 : SUCCESS in 9m 49s\n- swift-probetests-centos-10-stream https://zuul.opendev.org/t/openstack/build/9812ae77ca954c9e9eb64fe11ceb5a88 : SUCCESS in 1h 17m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/612ec7e89390491b93ac68378107d480 : SUCCESS in 52m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ee7c139b883c459291d3c50c32f9fce1 : SUCCESS in 54m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfb8642d2f874218b8fde2befba8a865 : SUCCESS in 28m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fa877b16f014b188c8b5eeaebe78812 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ae2f472504644b0895972c0268ac0c57 : SUCCESS in 21m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0922e932e24d42ec9c69f5e66f86d2cb : SUCCESS in 1h 02m 18s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py312-centos-10-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"f23f6631fd7f44163f7899507dba0442c6b0ced2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-16 23:00:11.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"87580b68b0273b7541a7c295d5c8518e89f9ef21","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 23:31:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75c9e44794144d7a857d527674b4b76c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/80a77ea92335494989145a40d8c38347 : SUCCESS in 28m 57s","accounts_in_message":[],"_revision_number":4},{"id":"ca1b0c59662b4fa9f97d7c4b9c1f280f21f09663","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 00:14:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36d748535a0348b4897264f32a381b2e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9188661d57e4b5aaa50fd97f015ef9b : SUCCESS in 10m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b942efc19e484e1aa9138bcea6dc8912 : SUCCESS in 8m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/4aaa52a17ec548409ed9e6aaa9ec40cd : SUCCESS in 1h 03m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1f5947ec8ad64d09b685ae5722956ac1 : SUCCESS in 1h 08m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/775ad29ada504abf985554a4ff6de58c : SUCCESS in 40m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe931d100ed74c2881e6013e394c5dc5 : SUCCESS in 1h 02m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2f5394fb7c9f4279911ec23d0c4383ad : SUCCESS in 19m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/607a1e507e5c47458553537b77629ace : SUCCESS in 20m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1eb577265e884bcbacd7c34cfb567f40 : SUCCESS in 22m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0a4b48cfb6cb4040be1c6c3dacb2d21b : SUCCESS in 4m 22s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ec6c79b59c14443c95ddf70277ccac55 : SUCCESS in 10m 30s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a08f09b31e62475a8ae60fae0f331ac4 : SUCCESS in 16m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7a65a1233e724cb3994213f04ddf62f8 : SUCCESS in 24m 34s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/3194326acc6f4fb78e22094d7daf6628 : SUCCESS in 18m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0bf8dfc97380435fbdb89327b374a159 : SUCCESS in 16m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/655b932e3a4641039a15f50595c4386a : SUCCESS in 13m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1127926669434937951f9a1714407a37 : SUCCESS in 15m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/68283a849eac4a00b18dced236a5b1bc : SUCCESS in 6m 22s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7bd257ece8b44fffb2407084e4c11c07 : FAILURE in 4m 57s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7aeae543fe7544f9819cfa89f3eeb9cd : SUCCESS in 8m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee01a385be6d40369458d107de2b2eed : SUCCESS in 58m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f42147fad9e2458a8a78321c82f4666a : SUCCESS in 55m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dee684ffea2f48fe9bfabeec13e37050 : SUCCESS in 34m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e10900a1e3434c00a57111d2ac985791 : SUCCESS in 29m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70ac76e46f594c189b1ca6e1b9c05720 : SUCCESS in 5m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2e7d0fcb30944bed90567842296cac06 : SUCCESS in 25m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bfd2ec2eb2514843b80326143e09e018 : SUCCESS in 48m 45s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4},{"id":"9cc570fe99c322b52ed6fd0ceb347d25568c3a25","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-17 16:19:37.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"9021112a1a0f69bc61464f117efdca2eee89c921","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 16:20:35.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"140adc5325d385e4773e40800d909d487ab7cb6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 18:24:41.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f470be38f47d4c28bcca223cf945f558\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51433c1ab93046a88eefafa8bf5a7795 : SUCCESS in 9m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/949255ab5549460d944e673c66e5dc21 : SUCCESS in 10m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/939c373bbea44a2185d079200eb11cdb : SUCCESS in 1h 03m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/56eca39dbcf3489f9052a0e6461ce98f : SUCCESS in 1h 06m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/651e6127c06c431fa17d60c00257c052 : SUCCESS in 50m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ca36d1c07a0499bafc157560655c933 : SUCCESS in 54m 41s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a4091d9c94634bcb9643f2a30f1c054f : SUCCESS in 7m 17s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ae0389327ca84edcb5daec7bcf219d8c : SUCCESS in 21m 04s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/88335169061c482fab44ae8b0b19d772 : SUCCESS in 24m 46s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3c037d3665344f698e97c411e90baa8a : SUCCESS in 39m 42s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/729f824306284e9897f41d6e2a999830 : SUCCESS in 20m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/52daf89c942e45f6a138558bc070d66a : SUCCESS in 16m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ba5691fdc9bd4548a00dbe41db97f470 : SUCCESS in 19m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e6cab504699b474b8707179638c177b0 : SUCCESS in 16m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f2c70f43b08a4bf3a55fde42654c8e97 : SUCCESS in 7m 03s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f28c1a67a53a4494bce2b107a11adb71 : SUCCESS in 10m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/02ec6942a87249ba852665111efb0c36 : SUCCESS in 1h 13m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/00cab9fe7787446aaad635c4abd7a59b : SUCCESS in 1h 00m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/766d96a0f52d4c9898881c035c9d0a60 : SUCCESS in 54m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4e660daa5c74c76a89120926ee26da3 : SUCCESS in 30m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21a95c56a75a40e192e5b51a8f900242 : SUCCESS in 5m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/20bfc891e9594072beaa10abb8d0fef4 : SUCCESS in 23m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8250a3c7dfe4fee85d59e20debdfc54 : SUCCESS in 58m 09s","accounts_in_message":[],"_revision_number":4},{"id":"caf84ef54187f6a3b8cf3914af5677b4a891c792","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 18:24:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"76b61ccea2e62c0790898307c5148d40a3cbc4ba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 18:28:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ac3e42d43ae490bb3891a8dde6f31b3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d0231fa82d947a68ddac09c5d1696c9 : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/1a4bc0621f49400292d0a16420b64bc3 : SUCCESS in 44s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4ea341a46a8f8b2c95cc9942332add810fc687af","revisions":{"f329b9fd979648d23500533dad2eea8e59457db2":{"kind":"REWORK","_number":1,"created":"2026-07-06 20:51:37.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/77/996177/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/996177/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/996177/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/996177/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/996177/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/996177/1"}}},"commit":{"parents":[{"commit":"a26ec30926c82063585cd0b752d558ab0fef0a96","subject":"CI: Make failing CORS job non-voting for now","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a26ec30926c82063585cd0b752d558ab0fef0a96"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-06 20:47:59.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-06 20:50:58.000000000","tz":-420},"subject":"DNM: CI: Make CORS job voting again","message":"DNM: CI: Make CORS job voting again\n\n...except, we probably don\u0027t want to resort to skipping cert validation.\n\nRelated-Change: I4bbfb15abe5a0b95e8f53bbb8e394846756a2238\nChange-Id: I261642065d8905a41dd3c590c3b4514573556134\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f329b9fd979648d23500533dad2eea8e59457db2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f329b9fd979648d23500533dad2eea8e59457db2"}]},"branch":"refs/heads/master"},"f7cbf2c1a1f51d5ee5ecc173402ae076756f006d":{"kind":"REWORK","_number":2,"created":"2026-07-09 18:18:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/77/996177/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/996177/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/996177/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/996177/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/996177/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/996177/2"}}},"commit":{"parents":[{"commit":"bea9ba99507fcf1caf75675655382364af1639ea","subject":"CI: Move (most) CentOS 9 Stream jobs to 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bea9ba99507fcf1caf75675655382364af1639ea"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-06 20:47:59.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-09 18:18:00.000000000","tz":-420},"subject":"DNM: CI: Make CORS job voting again","message":"DNM: CI: Make CORS job voting again\n\n...except, we probably don\u0027t want to resort to skipping cert validation.\n\nRelated-Change: I4bbfb15abe5a0b95e8f53bbb8e394846756a2238\nChange-Id: I261642065d8905a41dd3c590c3b4514573556134\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7cbf2c1a1f51d5ee5ecc173402ae076756f006d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7cbf2c1a1f51d5ee5ecc173402ae076756f006d"}]},"branch":"refs/heads/master"},"bfd3332792ec948ffa610571d086b4e398a2c00a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-07-09 21:47:02.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/77/996177/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/996177/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/996177/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/996177/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/996177/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/996177/3"}}},"commit":{"parents":[{"commit":"caeb1a3de8e00480b748feb3c13c36881821c2da","subject":"CI: Move (most) CentOS 9 Stream jobs to 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/caeb1a3de8e00480b748feb3c13c36881821c2da"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-06 20:47:59.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-09 21:46:55.000000000","tz":-420},"subject":"DNM: CI: Make CORS job voting again","message":"DNM: CI: Make CORS job voting again\n\n...except, we probably don\u0027t want to resort to skipping cert validation.\n\nRelated-Change: I4bbfb15abe5a0b95e8f53bbb8e394846756a2238\nChange-Id: I261642065d8905a41dd3c590c3b4514573556134\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bfd3332792ec948ffa610571d086b4e398a2c00a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bfd3332792ec948ffa610571d086b4e398a2c00a"}]},"branch":"refs/heads/master"},"4ea341a46a8f8b2c95cc9942332add810fc687af":{"kind":"REWORK","_number":4,"created":"2026-07-16 23:00:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/77/996177/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/996177/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/996177/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/996177/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/996177/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/996177/4"}}},"commit":{"parents":[{"commit":"977d7c021ebe93b5e82ffef88a64381da7faa032","subject":"Merge \"Remove gettext from bindep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/977d7c021ebe93b5e82ffef88a64381da7faa032"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-06 20:47:59.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-16 23:00:04.000000000","tz":-420},"subject":"CI: Make CORS job voting again","message":"CI: Make CORS job voting again\n\nOpenSSL got fixed.\n\nRelated-Change: I4bbfb15abe5a0b95e8f53bbb8e394846756a2238\nChange-Id: I261642065d8905a41dd3c590c3b4514573556134\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ea341a46a8f8b2c95cc9942332add810fc687af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ea341a46a8f8b2c95cc9942332add810fc687af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
