)]}'
{"id":"openstack%2Fswift~996179","triplet_id":"openstack%2Fswift~stable%2F2026.1~I4bbfb15abe5a0b95e8f53bbb8e394846756a2238","project":"openstack/swift","branch":"stable/2026.1","hashtags":[],"change_id":"I4bbfb15abe5a0b95e8f53bbb8e394846756a2238","subject":"CI: Make failing CORS job non-voting for now","status":"MERGED","created":"2026-07-06 21:19:06.000000000","updated":"2026-07-07 00:07:46.000000000","submitted":"2026-07-07 00:05:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996179","meta_rev_id":"4ffdd7faf61b9dc602f01805cfb44021a7e3a58e","_number":996179,"virtual_id_number":996179,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-07 00:05:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-07-06 21:21:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-07-06 21:21:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-06 22:54:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"527a08333fd447de0e42f888270c9d5644894b12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-06 21:19:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9a7e043058916415657c47d80fbed0a51937fb7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-06 21:21:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fa1634d01a451a7d282aad3e21efac5b41ca8af0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 22:54:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fb22f6a32de46c494a92ae69b55083c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8750bd793d2b4da988d0066f883de50c : SUCCESS in 9m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea77ba64383b401da770e153ac50394a : SUCCESS in 9m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/f43b46290eeb48e793e0959764fa4c11 : SUCCESS in 1h 26m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/33ee0b2ae78c4043bc3f0f24f3ee1d3c : SUCCESS in 1h 08m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fe51578784294ca69688ee1e8f40e5b6 : SUCCESS in 30m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/170617c59dda4dd28f0e21d1fd29560b : SUCCESS in 1h 06m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba509440bf3a4319959bb771d2bd4aa3 : SUCCESS in 19m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/afc08c13992947d6a3d07f5897b940eb : SUCCESS in 10m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/10ba76e60d3b4a188fffc28bf40fc28c : SUCCESS in 14m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d6139273d2c44830b1e1ed551353bac5 : SUCCESS in 4m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0be130a79c604ff6ade1c8da40e4c3bd : SUCCESS in 25m 31s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a50e5bd2ee554b468febbc1bfd5d2ef4 : SUCCESS in 11m 16s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8b7c8103ddda4447b52b90228dfab4b1 : SUCCESS in 31m 26s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/37ca28c9657e44e29de5eb26b480aab7 : SUCCESS in 21m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/58d2a66f2cec485a9af6aef64b3ae973 : SUCCESS in 14m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/97ef8e7e3f6f40bb94663ccfa598982f : SUCCESS in 9m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4681ea7b214742368c321ba168f1cc9c : SUCCESS in 11m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f1a212d1bc9e4f08b7c2db8315d70725 : RETRY_LIMIT in 3m 53s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1bf4f81ee4cc4b7cbcd6fddf1250b691 : SUCCESS in 19m 37s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f1d7dafacf054069ae08404ee2244ff8 : SUCCESS in 8m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5efebbb6e56645ed90d8d22f68db9d0c : SUCCESS in 1h 00m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/af040bb42d5548c5a05ab114ebd67d14 : SUCCESS in 53m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/837da394618c496ebe81f35d8e030f60 : SUCCESS in 56m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc381d53a1ed40d39f900891c3606f52 : SUCCESS in 28m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65607b90bba04b7fbf6afcd24bce269a : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/27cf765155db42cbb969908118520ad7 : SUCCESS in 11m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/572c87cd53234de7bcd60b9424834fff : SUCCESS in 1h 11m 20s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"60b25d107ae5b0287e518aef3a50f2140b9e1858","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 22:55:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5035d092794f6a947e3262b594d0a563007db46f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 00:05:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a4138088a8746afa782acdba8de8cb8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8e08b781e2c42c295313b333b39ced0 : SUCCESS in 7m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6080071494841648ffec2cea2085cf2 : SUCCESS in 7m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/19b53877c4a34d88badbfec301a00dda : SUCCESS in 1h 02m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f60e0a996aa545ac9def9f635c4c09ee : SUCCESS in 1h 04m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/84c352321dce43f68a0f1b0ddd94b839 : SUCCESS in 31m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/654f619c497f489a89d5d6c7bb80d4f4 : SUCCESS in 59m 40s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3dd91b8cfec34015b8dc391c91eddba8 : SUCCESS in 6m 33s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/15af1bb84833426ea717b377880607cf : SUCCESS in 17m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3a086c9a848149d896b245b8ed1c0dfa : SUCCESS in 8m 58s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/69f556917c1849688d8b2f0166c22033 : SUCCESS in 29m 00s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/782aebe744b6491ab46421caddf62d2f : SUCCESS in 20m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cb77f9006fbc449ebf8d59119b225ce7 : SUCCESS in 7m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/850b2d7135824f93af2b780096b1eae8 : SUCCESS in 12m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/596f733c87ee40618a183383e4cca561 : SUCCESS in 15m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/791d0b9f804b46119950d27a70f4a290 : RETRY_LIMIT in 3m 11s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/bdf522bbeecb439597389dd9852cac76 : SUCCESS in 7m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2126d8ddc7f2400e97b87c411c390a15 : SUCCESS in 59m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/31eeb6d199d4449fa63acbc6142447c9 : SUCCESS in 51m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ad430e0557f240b5aa7c6847df13c0ab : SUCCESS in 22m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf5c1debbfec4d23b5addd971ef0e8ac : SUCCESS in 18m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c09f82ee0a844ba1a345b5c603005867 : SUCCESS in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/345353043ef14e2d838beaa7cad99539 : SUCCESS in 10m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cdec9581bb4c44dca6b45344765bc2e3 : SUCCESS in 30m 36s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-upload-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-upload-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-upload-image.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"bbe6d6cc81fa3701dcde17f3b3b41d6989b9008c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 00:05:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0307f5556e0d8d3dfff92e260f22667c15f0a830","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 00:07:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb7c3acb5ba3445981820af18bb590fd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b6b7602e0b94d2eadcf3941928a801e : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d434c302d2f94da88667cf964534040f : SUCCESS in 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4ffdd7faf61b9dc602f01805cfb44021a7e3a58e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 00:07:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/628fec62796442bd9702ce01fc7dac84\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/821dc73cf16f4ad1a6867f5d3949588d : SUCCESS in 57m 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"62e7ec1be033ce0ae026687ccb3065b32c0a0a81","revisions":{"62e7ec1be033ce0ae026687ccb3065b32c0a0a81":{"kind":"REWORK","_number":1,"created":"2026-07-06 21:19:06.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/79/996179/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/996179/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/996179/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/996179/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/996179/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/996179/1"}}},"commit":{"parents":[{"commit":"51051ff3a0528bf996006e2c2833bd992da3ac3e","subject":"gatekeeper: Block X-Container-Host and other update headers from clients","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51051ff3a0528bf996006e2c2833bd992da3ac3e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-06 20:46:00.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-06 21:17:44.000000000","tz":-420},"subject":"CI: Make failing CORS job non-voting for now","message":"CI: Make failing CORS job non-voting for now\n\nRecently, our CI machines started using OpenSSL 3.5.7, which can raise\nNOT_ENOUGH_DATA inappropriately. Fortunately, it only affects one job;\nmake it non-voting until there\u0027s a fix in place.\n\nSee https://github.com/openssl/openssl/issues/31807\n\nChange-Id: I4bbfb15abe5a0b95e8f53bbb8e394846756a2238\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n(cherry picked from commit a26ec30926c82063585cd0b752d558ab0fef0a96)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/62e7ec1be033ce0ae026687ccb3065b32c0a0a81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/62e7ec1be033ce0ae026687ccb3065b32c0a0a81"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
