)]}'
{"id":"openstack%2Fswift~996555","triplet_id":"openstack%2Fswift~master~Icc5e0496b951d0d408048fdfcc30ecfcb302423e","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-07-13 07:49:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icc5e0496b951d0d408048fdfcc30ecfcb302423e","subject":"tempurl/staticweb: Make listings with ?inline work as intended","status":"MERGED","created":"2026-07-08 21:53:03.000000000","updated":"2026-07-13 07:50:49.000000000","submitted":"2026-07-13 07:49:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996555","meta_rev_id":"7687fd2e5acf60a0997eaea7b5215c7fba79f50e","_number":996555,"virtual_id_number":996555,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-13 07:49:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":2,"date":"2026-07-13 06:36:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":1,"date":"2026-07-08 23:41:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":1,"date":"2026-07-13 06:36:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-08 22:51:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-08 23:08:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-08 23:41:28.000000000","updated_by":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"reviewer":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"state":"REVIEWER"},{"updated":"2026-07-13 06:36:31.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"}],"messages":[{"id":"7fe99aaf7a36ccb9537e92188793332022fe85e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-08 21:53:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e62b603ebbc7656664cfef010c611dea05864c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 22:51:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa948d18272e4296b333ea38ecd72b4d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0d834ed54d1648e89c3a67ca769cc645 : SUCCESS in 35m 17s","accounts_in_message":[],"_revision_number":1},{"id":"ddcfdfcc414060a1a5f762dce07693507c4339f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 23:08:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/47c8cb16c9f549d7b01e3312cf42dab0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61f10d68724a46c6b7b5ba951cae7c2a : SUCCESS in 8m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5c0aa09d9834ab8b16b2ec0a96164ca : SUCCESS in 4m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/c48dadf4a13641efbb2be2343c249d42 : SUCCESS in 35m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/54df040d528a4578a0dd6c9dbafa27c3 : SUCCESS in 56m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f9294689a3ac4676b3c7f68b1445015b : FAILURE in 36m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d6c024c783984430ae8deba09f64c737 : SUCCESS in 1h 05m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d16ce0da5e25434b91ad7c5030c0e3fd : SUCCESS in 19m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df85a13e127144b4a31f3daf5ae758da : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/80ebb5ff7a8647d3a4b471c5bcb0a6b3 : SUCCESS in 20m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2e69f5cc332d41389c4fb0d20ae0fc82 : SUCCESS in 7m 16s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f8a722c919d2439a88c47e1bf5b4b9e1 : SUCCESS in 27m 08s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/97ac5d22b45a43c795b69a9f89535046 : SUCCESS in 20m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/05472fdcb9a14a1cb16191e905a73eb1 : SUCCESS in 24m 14s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/d2049468fb3d4589ada65d95d87fb0db : SUCCESS in 10m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1cd366916e474435a8b989ed333cf9f3 : SUCCESS in 15m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b38f38ef654b4ef589123272a2a3e78b : SUCCESS in 15m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4729cf2ce7fe4c56a32979ebf17fc70a : SUCCESS in 15m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4d6dc744bdd543759eee8ce92a870514 : RETRY_LIMIT in 3m 52s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b96e2b2574164381b601f091cf4925d9 : SUCCESS in 19m 52s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a20b947ef80848e1ae5097d106eae0f4 : SUCCESS in 9m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/eadc4423e96f4468bc0edba22c48f7df : SUCCESS in 1h 04m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6c3f4dc774714a7bb16486677eb66eba : SUCCESS in 51m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cccfc4546a39433494039e04210a6783 : SUCCESS in 38m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2eeb81e676da4198872ab50b14cbb349 : SUCCESS in 22m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba2e280aa5d94dccaa29097c9409085f : SUCCESS in 5m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/450b1624279c44e1b8e4298228310305 : SUCCESS in 18m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69a4e5cebb2a4134b250a189aa660e2a : SUCCESS in 1h 00m 52s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"c2e152fae072a7ffa74d0c5d61cf6f0299211f8c","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-07-08 23:41:28.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bdb68253728b29cd788323755155bf36613b6a21","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-09 16:06:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"950a28c8dd79a7a654b1d3f7d5ee3a1089f58e0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 17:22:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8fa97d9f0367471ba46db21e3c9ad2cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dd55a222996448986ff53b78f0e8ed5 : SUCCESS in 10m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0d579cc221f492e92d5608530e75008 : SUCCESS in 7m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/af409cfe619745669e6763088643cf05 : SUCCESS in 1h 09m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0916341d76ea4128880bdf6c6f360a2b : SUCCESS in 1h 06m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7ba4326b71b64137a331b009c291007e : FAILURE in 9m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d35b10b3b9c442bbbfe590f151054be0 : SUCCESS in 56m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/86f27a8f59d54c4598f7fab9f61b4c46 : SUCCESS in 17m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fcac926e0d7f46f485ed0dcf948c4f2e : SUCCESS in 21m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1677f4350c894c8c83afe6112fd709f6 : SUCCESS in 17m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ed5da9ededf042ccbfb8aa326780966d : FAILURE in 5m 31s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/64f0cf09519546e7a94f48827f08bbc9 : SUCCESS in 25m 01s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/492af0032a29452b9aabf5b4324b151a : SUCCESS in 20m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/09a1de0a9c7e48dcb93cc91a0b7d0234 : SUCCESS in 23m 44s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/b4684ea58435491b9eaaa34d6b36b3a3 : SUCCESS in 21m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/95a9b8efef7d4b8ba8dc29d03d9a8b06 : SUCCESS in 16m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1c38e3e7230d4a4ea74184f10ed2a823 : SUCCESS in 13m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/75fcfe78c8c547018ddebebdfa0ec4bb : SUCCESS in 14m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b5f827471b3b4c4c83f059339690737c : RETRY_LIMIT in 6m 24s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7d36136533ba4ddfa766a211edb1a365 : SUCCESS in 21m 26s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1f80cb75928c499a89bb4eddcbdf9170 : SUCCESS in 11m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f8f50e35f4584b92b1fb1b633db915d1 : SUCCESS in 1h 03m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9783018d41074542a77b8455fc9d9ede : SUCCESS in 56m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c58e8ebdde2c4e5d8905db057a4c4229 : SUCCESS in 40m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab3efca4777c419bab1bd8bdfc150667 : SUCCESS in 29m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a299caf4a1e464ca9ea0e4061dcda2a : SUCCESS in 5m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5c2aa83aeee54676b9d308896dcaccc6 : SUCCESS in 26m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29993c5c232044978766bc70c4868315 : SUCCESS in 1h 00m 54s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"5b0dfcb8d5058a8f81c0a80832bfa622bca7b0ea","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-09 18:02:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d51c790fecce6fef2cba87b640eab973a01f38a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 19:30:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6938f05f416a4d74b8e68b365eeea36b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ccc217e2c8d400d9db75db2e0086ad5 : SUCCESS in 11m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63e06a8dcf114ffe99bf4e60f574876d : SUCCESS in 10m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/25fa091d19b6443c935e8bc2d0de6474 : SUCCESS in 1h 08m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fc892b9ebb3341bf90ddd74111486694 : SUCCESS in 55m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0f78415cc1364cd6adde283b3da5ffc4 : SUCCESS in 55m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e92114c5ea904b98ab6643c038ec3fbb : SUCCESS in 41m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f30e92a9c6fa455b9eef3eb3a805c2d0 : SUCCESS in 21m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bc980b74f7614981b77f3ca836436371 : SUCCESS in 20m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a7f13aab9f084765a1f92fec6dbfe910 : SUCCESS in 23m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/28a4b36375be44bfaaf006774738e96a : SUCCESS in 6m 50s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b7435e3a8a674f67985fabdc26fd30d5 : SUCCESS in 25m 30s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/621e7c6a67aa4fff9fc5e8a57a47a9b6 : SUCCESS in 23m 47s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0d9d8d5440eb4a6e818b8cb2aadbd05f : SUCCESS in 31m 16s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/21844d4e78164742a5dbd710f0967666 : SUCCESS in 21m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5013f9768ba0430fbcc6b835aca74687 : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d9f77c00682c482ab2a0062f26c16327 : SUCCESS in 15m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/344a863400424d02a16bf8cf0aada52e : SUCCESS in 16m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e6a6b3d41a3e4281b88039bd98c98075 : RETRY_LIMIT in 5m 53s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e12c68625f504b6cb60d06cd20a626dd : SUCCESS in 19m 24s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5cf16ba6b53c491caaa7676e34590c2b : SUCCESS in 9m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f9cfaae31c494dfda3420c22db43f0ad : SUCCESS in 1h 06m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/95198031bde34b26b5f896dfc44032a1 : SUCCESS in 57m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cc8f0b7a0e6642829beb3e6384a81553 : SUCCESS in 56m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ee775ac603a4f77a959b2b5a73a3db1 : SUCCESS in 29m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71c735b38a524b599bad05a249989c3a : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5239494a68fc45dc9662183f0a448662 : SUCCESS in 19m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/443f3d48c6c24dd7ba27f9b1cefee28c : SUCCESS in 54m 41s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"c2b388ddf977a99cc5526a9e1070600e125e22e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 21:09:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18a93719cab143fca7317bf2c370f189\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1da6dc9b0e2546ec8b9be84c0c286886 : SUCCESS in 26m 40s","accounts_in_message":[],"_revision_number":1},{"id":"0b1652e2c5670086f71c0b9bb8752f07acc13d6e","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-13 06:36:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"39fcb06a86fb0025c455a06c23567579aa76b35f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 06:37:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b3f24506cb2098424dc9f16f4d5daa4a2a808f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 07:49:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa43008bb71641e9b44e0c2c1976dbb2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/066ff210e0cf4fe38a8075db98d7b211 : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a65ed222b0e5450db59fdba55ac5a138 : SUCCESS in 3m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/0afe528602b54f4cbfd929f7ac59b592 : SUCCESS in 1h 02m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d0fab62dd2974df98987c7d96a955c1a : SUCCESS in 1h 01m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3c4c2c662d1e45749b5e70a90a214f82 : SUCCESS in 54m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58cbc3658c274742b3b9bc2b9679de28 : SUCCESS in 59m 26s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6ea921da7c0b4e84a27370c08964ec24 : SUCCESS in 6m 33s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c3f6b036e53b43b681e5bf0d26ad7286 : SUCCESS in 26m 11s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6fbf5192dbbe4a07b04e84d49960094a : SUCCESS in 17m 11s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4d1033c6aef14c7894499f552dffec04 : SUCCESS in 30m 37s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/fbe6ec307b284c8e8611c4377fc06b0c : SUCCESS in 19m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0fead1b8ef264462804ca9b6dd2cb6ff : SUCCESS in 15m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fbe46c0439d242ccb31b50505e1dd103 : SUCCESS in 8m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f95bfb3c2a524392ab87473196c296ee : SUCCESS in 10m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6f303bdb070e48aebc1b36e2fe01319f : RETRY_LIMIT in 5m 48s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d214d6478dfc4b1cb7f5469324330c7b : SUCCESS in 8m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5cff28f0c4c741a1879aa610dd149431 : SUCCESS in 52m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7657f0809c7a49469075159d47322f68 : SUCCESS in 48m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/637e0f70b66d4a948f5058c78fde31e2 : SUCCESS in 23m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e57aebe9b8f4d23858ca3ce64e0f562 : SUCCESS in 29m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12a164d7e6f843bdbb6eb9ca77d8709a : SUCCESS in 2m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c21c6e3e72be44019a265bb0653d1ab9 : SUCCESS in 26m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ac7cbbf500b4a169b6d469f43447428 : SUCCESS in 1h 06m 54s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-upload-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-upload-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-upload-image.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"af72e502d14a06ad79a48ccfa923a49df6278621","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 07:49:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7687fd2e5acf60a0997eaea7b5215c7fba79f50e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 07:50:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d27d0c7c7594e2cbe6733ffcb77678a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3eabe6fc2d2d4d4f80a1234c6dbe36d4 : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e279399784c64183aa99650168afcd02 : SUCCESS in 48s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"911191aab0662e00a7c386e5a40f72caa87521aa","revisions":{"911191aab0662e00a7c386e5a40f72caa87521aa":{"kind":"REWORK","_number":1,"created":"2026-07-08 21:53:03.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/996555/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/996555/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/996555/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/996555/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/996555/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/996555/1"}}},"commit":{"parents":[{"commit":"9f5881d5143fd06a328bde2713cd9c61df579289","subject":"AUTHORS/CHANGELOG for 2.38.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f5881d5143fd06a328bde2713cd9c61df579289"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-08 19:04:46.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-08 19:05:29.000000000","tz":-420},"subject":"tempurl/staticweb: Make listings with ?inline work as intended","message":"tempurl/staticweb: Make listings with ?inline work as intended\n\nWe have code (and tests!) in staticweb to preserve `inline` query\nparams when generating URLs for the listing, but tempurl wasn\u0027t\nactually sending it!\n\nFix it so users can have media displayed in-browser rather than always\ndownload.\n\nRelated-Change: I7949185fdd3b64b882df01d54a8bc158ce2d7032\nChange-Id: Icc5e0496b951d0d408048fdfcc30ecfcb302423e\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/911191aab0662e00a7c386e5a40f72caa87521aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/911191aab0662e00a7c386e5a40f72caa87521aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
