)]}'
{"id":"openstack%2Fswift~996577","triplet_id":"openstack%2Fswift~master~Ib43b5f3424002ec8b82af04ece6e585ae63207c2","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-07-09 15:28:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib43b5f3424002ec8b82af04ece6e585ae63207c2","subject":"crypto: Fix listings when pipeline was misconfigured","status":"MERGED","created":"2026-07-09 01:47:01.000000000","updated":"2026-07-09 15:29:52.000000000","submitted":"2026-07-09 15:28:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996577","meta_rev_id":"1023293a111ebb33378a5cf463188a898aa3f91f","_number":996577,"virtual_id_number":996577,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-09 15:28:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":2,"date":"2026-07-09 13:33:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":1,"date":"2026-07-09 13:33:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-09 03:10:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-09 03:26:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-09 13:33:57.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"}],"messages":[{"id":"d1f78a15ec094dcddffa784c2ff4a4769908f9a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-09 01:47:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5879b3833b232e18c6f67e745e35f72b710e1e9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 03:10:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adde723d33cb4039a388f28a0c70bfe9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/425a28c54ab3476688db1ee00f5d9996 : SUCCESS in 1h 20m 17s","accounts_in_message":[],"_revision_number":1},{"id":"ec7a0d65f8388774768ae7aae5b36bd183eb5000","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 03:26:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff4d21902d154356933eca438736bdb9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e1a9f84c89a4c35869f7bcbac4adbb9 : SUCCESS in 5m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/879a02698ba14440aa3340de14e29e32 : SUCCESS in 9m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/9d01a42b623945d4b1b1aab61c0d71b2 : SUCCESS in 1h 11m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/be9bf7ab39c64e3e836994d9406aa403 : SUCCESS in 1h 05m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d4f2e649598740fe9424aeab73ab4f02 : SUCCESS in 1h 05m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31f1b2a1077d4e168cb307d513c63d5a : SUCCESS in 1h 08m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/49c682a9fea94698af3db0552bfc3d42 : SUCCESS in 12m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/864858a0452b4aa19f67e863546101fe : SUCCESS in 12m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d70aeab646424051b844904085b9e110 : SUCCESS in 26m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/46ee8d14ae1247199204c25ff299f2ce : SUCCESS in 4m 06s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3526690fb5e445b48974d327c83bd861 : SUCCESS in 24m 10s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/33ec81012d8b45e5b0deac639128c0d7 : SUCCESS in 24m 22s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/eef19bdf488c4a27a536d76ceb2cbc79 : SUCCESS in 32m 19s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/0f47d07f54e24a368b1e532b41556b81 : SUCCESS in 23m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a1931e758c6c4e53bed526df37609f63 : SUCCESS in 17m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d65f716e013b4077997f9d1cfd65dffc : SUCCESS in 16m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/279f9c1fb458492b897c31ec30dfc03c : SUCCESS in 8m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b84ae0362c7c4d12b67da58703aa3de2 : RETRY_LIMIT in 5m 23s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/31b68a9ddaec4d939ff544526b33fb48 : SUCCESS in 23m 48s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/573ddfabd91a4298b799009a02ba3258 : SUCCESS in 9m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1fb0a8a064f7425ebf81835fe153a3bd : SUCCESS in 59m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6fbd0a1ddbec477781989d23f5ad4885 : SUCCESS in 1h 00m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fa94fab29fda4ef599ee0219c08d254d : SUCCESS in 56m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7093d24743246b8ad90b55e1ab1aef5 : SUCCESS in 27m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8a0617b9b274030808d0622f6a11cfa : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/41312ea8c44340c99425a2894f306f40 : SUCCESS in 26m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d006e9226b4418189d2baa8a26d7266 : SUCCESS in 1h 03m 11s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"417bd29c9ea08c0634adbf3ba9c50e9aaffc712d","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-09 13:33:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"76cf5d303bbb486bb0d6161631406e9e235da4c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 13:34:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a74f84459e137b93b623022f5c0174dcd376c536","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 15:28:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed1fd2a9d2cd4ef38473aac8e73750c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e0eb75f318548acb0519d3beafd59f1 : SUCCESS in 10m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8aa72cddf0bf47b88f7921b168492987 : SUCCESS in 10m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/fa49073d816143a9a72f75926950fae5 : SUCCESS in 1h 02m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1e87a9fc05074fd0b32d028f84470be8 : SUCCESS in 1h 10m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/90f8fc7557904e8187734a6b9403513d : SUCCESS in 1h 07m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f5c192f9cbc47d18e64309fd04090f1 : SUCCESS in 1h 08m 42s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/25e109d7f22a44c1a5711cd3165e306b : SUCCESS in 6m 45s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fabb64d618a04b85a51c1ac27d930bc6 : SUCCESS in 26m 16s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/aceae7784b574a21bad791e47351ec06 : SUCCESS in 26m 54s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ea8855a8319b4d33adad3b5062b6efe1 : SUCCESS in 31m 10s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/3a1d03bb3c5641c287d272bf5e1f289f : SUCCESS in 21m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fcb1f04b2ae84573bc2e2f9fde41c4de : SUCCESS in 16m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/52bbe7f10ca545b3997d859ea181141b : SUCCESS in 17m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ebbfad6f307b4ac8baa0fb1d1c9f477c : SUCCESS in 16m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/249180c4f30d43b895e2c63a9ce4019f : RETRY_LIMIT in 6m 37s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6a33377976cb4d06af442849bf147f41 : SUCCESS in 9m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ba18eb1f899741929536160c040e7a98 : SUCCESS in 1h 04m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dcb8cf01537c4bcbbf0eccb16e36d6f7 : SUCCESS in 1h 01m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/73f710f44d9a4ce0bea954a6822b8045 : SUCCESS in 46m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8938bfe967644857aa3510368c5538e9 : SUCCESS in 31m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f33df9cb9b8c495fbcbf01b62e483a0e : SUCCESS in 4m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7e588d859a5344cfa77ef80c134f2732 : SUCCESS in 21m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d4f4480661c427684483cedf8bdd496 : SUCCESS in 1h 00m 29s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-upload-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-upload-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-upload-image.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"146dd494adf4d64a09f0fb7b69f47cf4768b8d8a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 15:28:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1023293a111ebb33378a5cf463188a898aa3f91f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 15:29:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88fbf886eb404b2e81291138f5f0cb74\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60d24e4a28fb45169b398e081c61ae0c : SUCCESS in 51s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/1212dd7d7b414fac8150e9da2d68c796 : SUCCESS in 52s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8be4f02ef61a09b044965134b9b7f95a5eb88bc1","revisions":{"8be4f02ef61a09b044965134b9b7f95a5eb88bc1":{"kind":"REWORK","_number":1,"created":"2026-07-09 01:47:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/77/996577/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/996577/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/996577/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/996577/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/996577/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/996577/1"}}},"commit":{"parents":[{"commit":"7c5408008a56c9fe6363175f2665cb7b5229c86f","subject":"Forward-port CHANGELOG entries for stable releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c5408008a56c9fe6363175f2665cb7b5229c86f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-08 23:23:26.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-09 01:46:53.000000000","tz":-420},"subject":"crypto: Fix listings when pipeline was misconfigured","message":"crypto: Fix listings when pipeline was misconfigured\n\nWe\u0027ve seen reports of folks getting back 500s when trying to list\ncontainers where some (but not all) objects have the wrong path in\ntheir crypto meta due to pipeline misconfiguration.\n\nThe trouble comes down to how we cache container keys for the listing.\nThis has been present since encryption was written, and for good cause:\nderiving a key is assumed to be non-trivial, so we don\u0027t want to go\ndoing it thousands of times for a listing when the result should always\nbe the same. We\u0027ve updated how we do that caching twice: first when we\nadded support for multiple root secrets, and again to account for\ndifferent metadata versions.\n\nHowever, we also need to cache by (container) path. Otherwise, the\nfirst container encountered in the listing gets stuck and the next item\nthat uses a different container causes a traceback ending like\n\n  File \".../middleware/crypto/decrypter.py\", line 420, in process_json_resp\n    new_body \u003d json.dumps([self.decrypt_obj_dict(req, obj_dict)\n                           ^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^\n  File \".../middleware/crypto/decrypter.py\", line 438, in decrypt_obj_dict\n    obj_dict[\u0027hash\u0027] \u003d self.decrypt_value(\n                       ^^^^^^^^^^^^^^^^^^^\n  File \".../middleware/crypto/decrypter.py\", line 138, in decrypt_value\n    return decoder(crypto_ctxt.update(base64.b64decode(value)))\n           ^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^\n  File \".../middleware/crypto/decrypter.py\", line 440, in \u003clambda\u003e\n    decoder\u003dlambda x: x.decode(\u0027utf-8\u0027))\n                      ^^^^^^^^^^^^^^^^^\nUnicodeDecodeError: \u0027utf-8\u0027 codec can\u0027t decode byte 0x9c in position 1:\ninvalid start byte\n\nRelated-Bug: #1729921\nRelated-Change: I1a6bcdebdb46ef03c342428aeed73ae76db29922\nRelated-Bug: #1813725\nChange-Id: Ib43b5f3424002ec8b82af04ece6e585ae63207c2\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8be4f02ef61a09b044965134b9b7f95a5eb88bc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8be4f02ef61a09b044965134b9b7f95a5eb88bc1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
