)]}'
{"id":"openstack%2Fswift~996824","triplet_id":"openstack%2Fswift~feature%2Fmpu~I2efc165b5764ad25c69b7507566e8fb1e532fb88","project":"openstack/swift","branch":"feature/mpu","topic":"merge-master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-07-15 10:18:05.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-07-15 11:33:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2efc165b5764ad25c69b7507566e8fb1e532fb88","subject":"Merge master into feature/mpu","status":"MERGED","created":"2026-07-10 15:37:18.000000000","updated":"2026-07-15 11:35:26.000000000","submitted":"2026-07-15 11:33:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"996824-merge-master","meta_rev_id":"217bd50028c81a6d277a5043f8c5f7dab904cd02","_number":996824,"virtual_id_number":996824,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-15 11:33:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2026-07-15 10:18:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2026-07-15 10:18:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}],"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-10 17:10:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-10 17:18:21.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2026-07-15 09:02:04.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"CC"}],"messages":[{"id":"88f3762c6141d115975117b162944b1c22c2b613","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-07-10 15:37:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6064fa30ddee8adcadd7e0a996d702c5dbe7de4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-07-10 15:38:40.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"aaf37d5590cc91324602928dc87f3adbf52a3261","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 17:10:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/21e1043613f04b7287586e1b2e3d895b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60ca3c9dd2534cf1a3059f40c4669db3 : SUCCESS in 10m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76d01d0bfce147838ea7e814aa90325e : SUCCESS in 8m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b90f05dfd8dc49e98d4b7bf849b1d72b : FAILURE in 1h 27m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/91b77f06b30c4337a26b421d176f7428 : FAILURE in 1h 27m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5db1f0aa2bf240698c39944a1af8fae3 : FAILURE in 1h 26m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c6277462cb7142558e53314f0d525e72 : SUCCESS in 7m 25s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7a5a101761d74728baac328555ad7e37 : SUCCESS in 27m 33s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/be61ad52ed7843fc8a4693425dbd6d7c : SUCCESS in 19m 29s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e98b1af066b24c4fa434515db440dc50 : SUCCESS in 30m 48s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/f3f56066eea94549a6c11e3bea911a58 : SUCCESS in 20m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/df86f92b5938431b9416c0ca8e6a33a5 : FAILURE in 17m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0d83f97c4d6a4b118074e4366252485b : FAILURE in 16m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cf5ddc577fe8490cbdfd69d72fcb4541 : FAILURE in 19m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/56ece72a1d3944be9e53f0e8383e6f30 : RETRY_LIMIT in 4m 22s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/085aaf536467465499bb08fbfe1ab983 : FAILURE in 22m 27s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/509a8ca48a334a4e99b7e123845bc4cb : FAILURE in 11m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/35c6e0c6f04646a09389f8d1948ff54f : FAILURE in 3m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/600842aa4a4d4d218a4ca44798ee0fb3 : SUCCESS in 30m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e7e82054979418a9ff18b3308bb9d50 : SUCCESS in 4m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8a89aadd16c14de1a439650792add280 : SUCCESS in 10m 45s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"d4f72565db63118ed430c165efcc69f06d54a95d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-10 17:18:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f7f53a14fdf4d621aef471c5ad275f661ee9aa2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 19:04:10.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d27b26386774c8cb8177bb4c487bb40\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a24209c658a4f9a873e9949670c70fd : SUCCESS in 9m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b37b3fb06fc2411a9b2c981c8566d0ae : SUCCESS in 9m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7033cecadd58495493effdb696013f4f : TIMED_OUT in 1h 31m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1096ccfdd0a642eabcbf2093e33c68d3 : FAILURE in 1h 30m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/68db21fc3935420aa49fdd8a5985277e : TIMED_OUT in 1h 31m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/32d11e2fb3f048a2ad277c869354e648 : SUCCESS in 7m 18s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/62b112e8b0bd48b2ba17127021367d4e : SUCCESS in 17m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9e16d510a84c43b399a786e6cca54601 : SUCCESS in 22m 42s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b6da03e1970c4e60ad270a8bd3571b6f : SUCCESS in 26m 25s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/7c964b5f3e5947d5ace32423c4ad6c87 : SUCCESS in 21m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ea09cd1f10994ac99d52b389e0359fe0 : FAILURE in 16m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1e5e58c4e9f944eeb6b287e8b9caa42e : FAILURE in 13m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d4e2277dd762420bb9e16692020e8c46 : FAILURE in 16m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/860471b540104079b4f91e927d235816 : RETRY_LIMIT in 3m 27s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e4ef6ddf1a574d99a880acba6ff37731 : FAILURE in 21m 35s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/bf5227607aa247f58fe07da22f41bae4 : FAILURE in 9m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/45ee6598004546568e2b75344bd1fea6 : FAILURE in 4m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd128c3f51fe4e679ce54dbe8537e81d : SUCCESS in 29m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd32598cd0c24fae987c2c822be674f6 : SUCCESS in 3m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b331004d1dc74b33bed087f50b30a832 : SUCCESS in 18m 37s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"73a2c7a816e70f6a35309b638252e9611e86088a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 23:15:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/157f3b0f88a74bb6a6c55768b3b38119\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/64a0499a3b5f45578f92dbab19dffb9e : SUCCESS in 1h 14m 45s","accounts_in_message":[],"_revision_number":2},{"id":"fd897475aa31ec3bcfb62548b7f322e6d8534e23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-07-13 11:54:56.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"94af4cfa028e61bd216ac0cda36a436c53fcf8b3","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-07-13 11:55:24.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1223601786645a8b8fb07929cbd857c453bfaab3","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-07-13 11:58:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e66aed5fee0000e8454c67c67fe83a9fb0bb2657","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 12:10:33.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fb086b4b1534030ae3b9a8ee684e704\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b6a070b1e2924c99bfea45c82320e48e : NODE_FAILURE Node(set) request 38aa74d7a0f244e4b7fa1800e23b8c81 failed in 0s","accounts_in_message":[],"_revision_number":3},{"id":"89dad100c3f51a57709ece2460c6e2dca2ad8322","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 12:42:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a63899166aa43b793016d726f8330f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e07417acb03046a3a9f0de8294008afa : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9de95c43eb141f48fe3833e36654313 : SUCCESS in 9m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4470aa81795f4c28a56c82240e569f98 : SUCCESS in 13m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/772ea527995a47bc9c60fba92cdbe558 : SUCCESS in 23m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/01afa79127024da9b50ec7efc7d3e03c : SUCCESS in 24m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1ba6858a02db48a69f8e4f6faa894136 : SUCCESS in 6m 27s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c3a44167f9814198b81781bc76c7ac9b : FAILURE in 23m 05s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8bdc955bfe2840beac39e0e49e2b8d4c : FAILURE in 9m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a77cd67d77ba49089c4e64b0e98515e2 : FAILURE in 15m 32s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/dc050e56d68148088e1d092449e4da66 : FAILURE in 23m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/16e1fbd81bb34662bf1dfcf7aa2a0cba : SUCCESS in 14m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b4f52cbe5b1f4d6a8fe7f9f856b6431c : SUCCESS in 16m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9dea366127d4408b834e02825f7aa120 : SUCCESS in 12m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1868001674854dfaa5555021170726e3 : RETRY_LIMIT in 4m 11s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/343e8188d8974e8e9ac9965b019a1742 : SUCCESS in 12m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/65d5eed515604020b667bb529a1e418a : SUCCESS in 13m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/54d8360ea8fb45658bae19af5af980bd : FAILURE in 40m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94c3687179e0498ea7df0d9c65830d6a : FAILURE in 14m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61e855a28097490c9ed566eda5c68465 : SUCCESS in 3m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/17bd9c59e5844ef18b1263f9fd2c80af : SUCCESS in 25m 06s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"e766cba22a2fc9c4963337f302da8a912e9341e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-07-13 14:49:20.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"d6ea79c6f43a2f280ad2d4a75e744d318be92f61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 16:11:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63ebe6d5556745389b39a05b74526103\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/be1f05ed48484bfc89ac5c3fe8d85be3 : SUCCESS in 1h 16m 05s","accounts_in_message":[],"_revision_number":4},{"id":"cbb9f088d9981aedba1290ef5f2c15e1deb7519d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 16:16:26.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f91a8aaf9384ebdae9af6e1b81a4c42\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cb00365386b4571bf23dd546d686359 : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea36c6061ace41c1bd21b665644a2d47 : SUCCESS in 6m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5819dccb97784ecbbf5c8dd05ea6c03d : SUCCESS in 12m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/418b6ca4d82e4c64a112b804bf5fa3a8 : SUCCESS in 24m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/db97172c42ec4aa8a16ea03bc2669a82 : SUCCESS in 24m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c26ff4fbb911472496dc21e0a556c419 : SUCCESS in 5m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b0eea89a45cc407280cf7eb765e7ec6b : SUCCESS in 27m 00s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2f42c2a5959443a0828e5bc37f5ffcf5 : SUCCESS in 20m 03s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8a6777dcce944b54ba1400c2ea9046c5 : SUCCESS in 18m 10s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/ce5c35b1e48f4cecb193ed267c02a41f : SUCCESS in 23m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f6fb93eeaf214d009913a6b6f4cd1f3b : SUCCESS in 8m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0cb29953457a42aaaf89728c3c7f99e7 : SUCCESS in 9m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d83a4f82d0da464a90b2bdcf7c9425ff : SUCCESS in 17m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dc9ef0b056f8432fa10d9da703edaf2b : RETRY_LIMIT in 3m 29s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/56035c285f174e52b6eacd89a185dace : SUCCESS in 25m 42s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7d709a17ed4f4081bbc13a01abb10400 : SUCCESS in 9m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7730f95230cd4a9a9bc3fd229b38be76 : FAILURE in 1h 06m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48c4f49d38964c1ba6cbb8b808ea70cf : SUCCESS in 26m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6b5bd89297146c18ed31d817e18f6e0 : SUCCESS in 3m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ba69cd335e2d4929948bf09ed7dc1e00 : SUCCESS in 19m 17s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4},{"id":"fe0a4df2fcb2b3f627d899a76afb4b9335d78d0e","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-15 09:02:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"036026720dacbdad38522ddc4c0941d48514f562","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-07-15 09:08:01.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"7b1ee8f5508e6c197b57728e2e3c0071acbf7093","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 10:14:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f722bca6ecd4b6796d2a8b6adbae863\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54053ca833c04d41af4db8eeffae74be : SUCCESS in 11m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06a0b26c67274628a8afabab48cb995c : SUCCESS in 9m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4da59d6d24ee44a1addcffd2b95c36be : SUCCESS in 21m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba06e78659864f2ba11a80f48a38797e : SUCCESS in 22m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/028009e2c1424966bbe633b989548d23 : SUCCESS in 25m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d5702fafe0ca4e499e9a8fc72fde1b12 : SUCCESS in 7m 17s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e3092cf204474f96b3b1429f32cd9c19 : SUCCESS in 30m 30s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/512836296cdd4102aaed55cfc46d3a86 : SUCCESS in 22m 02s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a76f2bc6359443d087c6f4dbecb4a1e3 : SUCCESS in 36m 11s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/192c53d6b7ef41729e7c1118fc8b0344 : SUCCESS in 22m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5b129546ac0046eb9ee4888a2bb48dcd : SUCCESS in 16m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bff95b6ab50e4124b354d40e339a1584 : SUCCESS in 16m 47s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/43641c7c9a29414da1529def30ac4991 : SUCCESS in 16m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/280f805353834357b33e6c5ebec3f578 : SUCCESS in 6m 42s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9c3a4ccabd644c1a8efc54c75b711078 : FAILURE in 7m 48s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/820f17d04644411bb8d19212d29672b4 : SUCCESS in 11m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/56f880fa0e574526b1ad9212e677fe3b : SUCCESS in 1h 01m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb7024ed75ee422f951ab420b27335ff : SUCCESS in 29m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28918c0673c34bc0978c6a22ada223b4 : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e94685a30236449fb5a23762d8f3900a : SUCCESS in 18m 08s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"b1db35ca9f1723fa9826a244d2d5bd80934ed9ff","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-07-15 10:18:05.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"e04d78814125c0250c9eff99d2ef0c7ab724423b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 10:18:47.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b8c04c626ab806ae699bb325ebc2a133edd0187e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 11:04:42.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c326aa6d98004f09974e14c0d9a5ade6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e1e56f818fa24df4b50a9aacf1bc490c : NODE_FAILURE Node(set) request 3074457bbcfc4a1285ef383044ca1e47 failed in 0s","accounts_in_message":[],"_revision_number":5},{"id":"27d4e15d3dfaa7f0e875b8c7fa5469584e7933a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 11:33:18.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/177a2d41f0d9483394bfff7faebfd462\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6875c1060bd4439e8b77b363323b3ca7 : SUCCESS in 10m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba770b50a0034bd2b29a63b9bc163fcc : SUCCESS in 8m 28s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/39fa2e550a744950b2ff57ccb99c77ce : SUCCESS in 7m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/37f4011a4d4d458c8619a2d9f9ad017a : SUCCESS in 25m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/69ebf1c5bbe249c4b213e576cb0acef5 : SUCCESS in 9m 26s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e4d3948b5c424dd5b21a480b39f9d8b7 : SUCCESS in 13m 11s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/5028673ee47643beae01bb4b68b77a70 : SUCCESS in 10m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5fa11163c6d24ecd851829b16d98a623 : SUCCESS in 17m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/04a0f1418dff4456b051da7a6c532607 : SUCCESS in 18m 45s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7a5d36c8152740f6bad4325628207b95 : SUCCESS in 17m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bdb6e1f1df9e4260b4b023a74c3a6fb3 : SUCCESS in 6m 26s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e62f7f47535f45839b2d0f4a4057c42c : SUCCESS in 9m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/49c430d489d24354b465cebbc49dea33 : SUCCESS in 1h 09m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f2f9f8b61444cb6b605c3535bc331dd : SUCCESS in 33m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f95b239a8c3f4bcfa232b052ea274720 : SUCCESS in 3m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/081926ef4cd840bf8b4503f0edf7e8fc : SUCCESS in 18m 56s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-upload-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-upload-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-upload-image.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"adf0d1a6b78797697a3d55a140ed148efe8334cb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 11:33:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"217bd50028c81a6d277a5043f8c5f7dab904cd02","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 11:35:26.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c3cd16c37c64f039e78cd743520e198\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdd7bef7ac214f80b76b909e554bdbea : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38a4bac61a834346aa65f28f548d99ef : SUCCESS in 42s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/617c57286b6e4698a16f217e26bac753 : SUCCESS in 47s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a991fdf18b15af3835ccc8abcfce2bd8cb862ffe","revisions":{"f65efac20e681807674c77257dab1fc9fcc211db":{"kind":"REWORK","_number":1,"created":"2026-07-10 15:37:18.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/24/996824/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/996824/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/996824/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/996824/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/996824/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/996824/1"}}},"commit":{"parents":[{"commit":"d599203e9218ec95f34846736042a41f45300ce9","subject":"Merge remote-tracking branch \u0027gerrit/master\u0027 into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d599203e9218ec95f34846736042a41f45300ce9"}]},{"commit":"f18d0e76bab7661fc9c2847ef65b9e1a10257e4e","subject":"Quiet eventlet deprecation warnings.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f18d0e76bab7661fc9c2847ef65b9e1a10257e4e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-07-10 15:35:17.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-07-10 15:35:36.000000000","tz":60},"subject":"Merge remote-tracking branch \u0027remotes/origin/master\u0027 into merge-master","message":"Merge remote-tracking branch \u0027remotes/origin/master\u0027 into merge-master\n\nChange-Id: I2efc165b5764ad25c69b7507566e8fb1e532fb88\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f65efac20e681807674c77257dab1fc9fcc211db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f65efac20e681807674c77257dab1fc9fcc211db"}]},"branch":"refs/heads/feature/mpu"},"406188621d2b792636ecf9408d7d0cd1e7a98e5a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-07-10 15:38:40.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/24/996824/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/996824/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/996824/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/996824/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/996824/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/996824/2"}}},"commit":{"parents":[{"commit":"d599203e9218ec95f34846736042a41f45300ce9","subject":"Merge remote-tracking branch \u0027gerrit/master\u0027 into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d599203e9218ec95f34846736042a41f45300ce9"}]},{"commit":"f18d0e76bab7661fc9c2847ef65b9e1a10257e4e","subject":"Quiet eventlet deprecation warnings.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f18d0e76bab7661fc9c2847ef65b9e1a10257e4e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-07-10 15:35:17.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-07-10 15:37:44.000000000","tz":60},"subject":"Merge master into feature/mpu","message":"Merge master into feature/mpu\n\nChange-Id: I2efc165b5764ad25c69b7507566e8fb1e532fb88\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/406188621d2b792636ecf9408d7d0cd1e7a98e5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/406188621d2b792636ecf9408d7d0cd1e7a98e5a"}]},"branch":"refs/heads/feature/mpu"},"d05ff787305c931a75b3dd93c3a2ba66a8aab508":{"kind":"REWORK","_number":3,"created":"2026-07-13 11:54:56.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/24/996824/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/996824/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/996824/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/996824/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/996824/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/996824/3"}}},"commit":{"parents":[{"commit":"d599203e9218ec95f34846736042a41f45300ce9","subject":"Merge remote-tracking branch \u0027gerrit/master\u0027 into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d599203e9218ec95f34846736042a41f45300ce9"}]},{"commit":"f18d0e76bab7661fc9c2847ef65b9e1a10257e4e","subject":"Quiet eventlet deprecation warnings.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f18d0e76bab7661fc9c2847ef65b9e1a10257e4e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-07-10 15:35:17.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-07-13 11:27:57.000000000","tz":60},"subject":"Merge master into feature/mpu","message":"Merge master into feature/mpu\n\nChange-Id: I2efc165b5764ad25c69b7507566e8fb1e532fb88\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d05ff787305c931a75b3dd93c3a2ba66a8aab508"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d05ff787305c931a75b3dd93c3a2ba66a8aab508"}]},"branch":"refs/heads/feature/mpu"},"5055839cdc614ddbec5afc063a7577d77635ba3e":{"kind":"REWORK","_number":4,"created":"2026-07-13 14:49:20.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/24/996824/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/996824/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/996824/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/996824/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/996824/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/996824/4"}}},"commit":{"parents":[{"commit":"d599203e9218ec95f34846736042a41f45300ce9","subject":"Merge remote-tracking branch \u0027gerrit/master\u0027 into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d599203e9218ec95f34846736042a41f45300ce9"}]},{"commit":"f18d0e76bab7661fc9c2847ef65b9e1a10257e4e","subject":"Quiet eventlet deprecation warnings.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f18d0e76bab7661fc9c2847ef65b9e1a10257e4e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-07-10 15:35:17.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-07-13 14:41:31.000000000","tz":60},"subject":"Merge master into feature/mpu","message":"Merge master into feature/mpu\n\nChange-Id: I2efc165b5764ad25c69b7507566e8fb1e532fb88\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5055839cdc614ddbec5afc063a7577d77635ba3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5055839cdc614ddbec5afc063a7577d77635ba3e"}]},"branch":"refs/heads/feature/mpu"},"a991fdf18b15af3835ccc8abcfce2bd8cb862ffe":{"kind":"REWORK","_number":5,"created":"2026-07-15 09:08:01.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/24/996824/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/996824/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/996824/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/996824/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/996824/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/996824/5"}}},"commit":{"parents":[{"commit":"d599203e9218ec95f34846736042a41f45300ce9","subject":"Merge remote-tracking branch \u0027gerrit/master\u0027 into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d599203e9218ec95f34846736042a41f45300ce9"}]},{"commit":"f18d0e76bab7661fc9c2847ef65b9e1a10257e4e","subject":"Quiet eventlet deprecation warnings.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f18d0e76bab7661fc9c2847ef65b9e1a10257e4e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-07-10 15:35:17.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-07-15 09:07:16.000000000","tz":60},"subject":"Merge master into feature/mpu","message":"Merge master into feature/mpu\n\nChange-Id: I2efc165b5764ad25c69b7507566e8fb1e532fb88\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a991fdf18b15af3835ccc8abcfce2bd8cb862ffe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a991fdf18b15af3835ccc8abcfce2bd8cb862ffe"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
