)]}'
{"id":"openstack%2Fswift~997312","triplet_id":"openstack%2Fswift~master~I0461eacb50b01e73367f8eb607adcbc605dfa8e7","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I0461eacb50b01e73367f8eb607adcbc605dfa8e7","subject":"sq? add better parsing for label values, key, and metric names","status":"ABANDONED","created":"2026-07-14 23:24:15.000000000","updated":"2026-07-16 21:10:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9fb566f9af06735553a5481edc8315257e480acd","_number":997312,"virtual_id_number":997312,"owner":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-15 00:33:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-15 00:20:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-15 00:33:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5c8f7a5b8334f3db2d5a810c5631a6c71905767e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-07-14 23:24:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"873248911ad10e2fd33b928a17cacc0b00f798ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 00:20:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dceb6645361415597d4797a801b1a38\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e1e779fc31d64e3b8d3cf5368b238e04 : SUCCESS in 53m 56s","accounts_in_message":[],"_revision_number":1},{"id":"000a4b7f5e2a23cfb2260f7eaaca0b42b152ca06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 00:33:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84ac86806fb14b0b88013527aa26f6e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46c51f50058e4f2aa3e81fc5f422aab6 : SUCCESS in 10m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eeb05e2598344159b8b7bfab27af7a45 : SUCCESS in 7m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/caa9ff34f22545609c86db8944b2a321 : SUCCESS in 1h 04m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7d7e23282cad4ad68c560254481ecabd : SUCCESS in 1h 03m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/189feb3b9a5047339ad558b783983a24 : SUCCESS in 52m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d3ba72c032b94ac7a9e8a3890b7ddc27 : SUCCESS in 56m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fab732f98ffe4ac89538697a4bf67cae : SUCCESS in 19m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6763b49671f741a8b44e7e6f740b551c : SUCCESS in 16m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b35dde78f1e44e7a6db26e24d923fda : SUCCESS in 21m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/87eb24e2afbc45dcaff329111605ac41 : SUCCESS in 6m 03s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e7c9e65a1c644ea79e4a14021eea2ec2 : SUCCESS in 10m 17s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d34a82e07d7f4004bc2360ef15d80a92 : SUCCESS in 21m 25s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/242d745b290f4bad8d45b091a86c1cc0 : SUCCESS in 31m 09s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/0c6c6b32754240f7844779e6ae889af6 : SUCCESS in 16m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b0814f09759044bfb49067121b4fc55c : SUCCESS in 14m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/58943d5faefc409ab036a452075cf98b : SUCCESS in 14m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d22de9bb7cc24b8bae02f8f8013f0ccb : SUCCESS in 15m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bf4ddb11f2124aa98cc9940293de47dd : RETRY_LIMIT in 4m 49s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/07c33cdc5ca54591a8a8b8a8323fa396 : FAILURE in 4m 52s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1bb319c9489f4d6783d3afa8ba8ff429 : SUCCESS in 5m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/db80765ea6ba4a2aaa59c909b99c4b66 : SUCCESS in 50m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d752834f20b84c0e914b1fe1f1bfc7d6 : SUCCESS in 57m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ef925cd0c95f478a9824149dc72b7eae : SUCCESS in 49m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42a37a2454be479a8f1d25a81e97c507 : SUCCESS in 14m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fb62f5b5ce947ce978704dd9cba010e : SUCCESS in 2m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6d122067114b47b7b5328414e614a674 : SUCCESS in 18m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/416651a07d8a498d8619a800d732296a : SUCCESS in 39m 25s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"9fb566f9af06735553a5481edc8315257e480acd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-07-16 21:10:16.000000000","message":"Abandoned\n\nsquashed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"25b1c090d5e29a20a45f39f989fb662d1ed8be96","revisions":{"25b1c090d5e29a20a45f39f989fb662d1ed8be96":{"kind":"REWORK","_number":1,"created":"2026-07-14 23:24:15.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/12/997312/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/997312/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/997312/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/997312/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/997312/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/997312/1"}}},"commit":{"parents":[{"commit":"b7e62281fb7e807745056fb6f58f589d19327a13","subject":"add PrometheusClient for file-based metrics export","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b7e62281fb7e807745056fb6f58f589d19327a13"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-14 23:23:06.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-14 23:24:01.000000000","tz":-420},"subject":"sq? add better parsing for label values, key, and metric names","message":"sq? add better parsing for label values, key, and metric names\n\nChange-Id: I0461eacb50b01e73367f8eb607adcbc605dfa8e7\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/25b1c090d5e29a20a45f39f989fb662d1ed8be96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/25b1c090d5e29a20a45f39f989fb662d1ed8be96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
