)]}'
{"id":"openstack%2Fswift~997325","triplet_id":"openstack%2Fswift~master~I48f34077e4ec867a4d5bef0ef3b77b6e3ebb66ac","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-07-15 13:03:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I48f34077e4ec867a4d5bef0ef3b77b6e3ebb66ac","subject":"Remove note about old pip\u0027s behavior","status":"MERGED","created":"2026-07-15 03:14:44.000000000","updated":"2026-07-15 13:05:21.000000000","submitted":"2026-07-15 13:03:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"997325","meta_rev_id":"d9a614e21cd33c2148717ddeb32a8518900b0444","_number":997325,"virtual_id_number":997325,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-15 13:03:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":2,"date":"2026-07-15 11:56:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":1,"date":"2026-07-15 11:56:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-15 04:29:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-15 11:56:33.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"}],"messages":[{"id":"3dcb2f426c9079a672f3dcf0a0c29f29d8711077","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-15 03:14:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d8e00e38c9d6b7ed724e14a0e6d3b232e24f478","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 04:29:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22d273cfc9c84320a184c0e783070f5c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/fa9258b2bbbe4c6db18e6a0ce5953ba4 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5df7bffbda2449789061757f24f06568 : SUCCESS in 5m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b4192a4a458e40f2a40401134d354e9f : SUCCESS in 2m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d5f1408702244b0a278ccf82b8b401e : SUCCESS in 9m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/bdb216b369194fb99d3314f26f88a471 : SUCCESS in 1h 03m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0efff61b1a06453db724ed310b36ba60 : SUCCESS in 47m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bdc77f4cefbb43dc82e7b75abc0cca8d : SUCCESS in 51m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51320ed37b824764a5fa79f895dbeee7 : SUCCESS in 36m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/677ede5fcda34b4f9d5fc860228f43a4 : SUCCESS in 17m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e894a2ae95dd4663b93674b4c1ab5d66 : SUCCESS in 13m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0261d0db87274ddd934c130b6d579ee2 : SUCCESS in 17m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2a236f3dfde44bee881d8f114cee5b61 : SUCCESS in 9m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/359f4a0e68cd476fa9e51e065699ff00 : SUCCESS in 20m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/85a921a187f5487499f0959cae31ce7d : SUCCESS in 21m 29s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0d480f9f62694f39a0311380032aa37a : SUCCESS in 30m 30s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/e258722ed58447eca355c523429fd1b7 : SUCCESS in 21m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c39ea6d0df024352af2c467a536b2558 : SUCCESS in 14m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c64aa3d7b0c54ee1b32395dff28f139c : SUCCESS in 15m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/76b4a834979d4765a78c19ccd5b3cf63 : SUCCESS in 16m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6595a883a51a4c31a63444c4e40efbdb : SUCCESS in 7m 04s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e850da382c3c4e838e85e1cbe2d45049 : FAILURE in 7m 33s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7744422a0e69462095b05ece76342f6e : SUCCESS in 9m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/966a3fef5ab846268b5ec15cfc53f6ca : SUCCESS in 1h 05m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4d8a3316f3ed4190944e88dc4ba602a2 : SUCCESS in 29m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5cd0c552653041a5ab2a50b14e9aa826 : SUCCESS in 22m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39644ce4be4b4bb39504e8c01050302b : SUCCESS in 14m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/295e78c441f74868af509d87d5abe4ef : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b4401dd4702d498486b744db824f5aec : SUCCESS in 9m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a85c83fc33d74a3ca48b871e963b25b0 : SUCCESS in 25m 29s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"2f55bf97e6f62c00ca4c73c81a3ef499649de447","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 05:15:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7cfbdf834044cbdb52076862cf278b4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/31f03e937f594a8e98ab20daa980c0a7 : SUCCESS in 46m 06s","accounts_in_message":[],"_revision_number":1},{"id":"823455d0198b255570e120215774b7ecdb857bf4","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-15 11:56:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bc9185d11919f600d580512836e1be3f370adcfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 11:57:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"695caad5512e369edb24f64d0039bc95b3734217","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 13:03:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c7aefae3108427b8176d154d1694180\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cdd486bb716b4201bcb63d3e09e471fa : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7014c95757ea43c6be86906f96b7fd0b : SUCCESS in 10m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b69b9fbf942846758f283d1e0967f8b1 : SUCCESS in 2m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7e4a020c1f14cf7b25456c6b39c2034 : SUCCESS in 8m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/d3a22e31689e4ec19b0123fce26f89e4 : SUCCESS in 1h 01m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/01b1574f59744a25a76230fa69d47ebf : SUCCESS in 29m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dc8a51fa39ed4a87b7932ba393eccd0d : SUCCESS in 27m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee2ff88b31734735a8da29cf1596066d : SUCCESS in 1h 04m 13s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2cb6ba14f3e54f559ddd8976eca211c2 : SUCCESS in 5m 11s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f72d2fc439bc40ff95823f640184b6da : SUCCESS in 12m 00s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e16baa09f42d4513863f46c5616d3c83 : SUCCESS in 20m 57s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2119f33c87c346d699c8a3c065e605d4 : SUCCESS in 30m 58s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/0a54022d2bc14c6383c7609355bb94cb : SUCCESS in 20m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/51c7f43ce2e84042ac6d5a83b50722a6 : SUCCESS in 14m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6fb5f354e255441e825932d7e5a6a3a5 : SUCCESS in 15m 13s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0c93b643132744a0835a102e696dbb80 : SUCCESS in 15m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e6fabf3170df417bbe0fa77ecd436494 : SUCCESS in 6m 28s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/44e4a96e51a143a1b9978fedea709cb6 : SUCCESS in 8m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/14c9b12cf979499098ab657f2e9041d9 : SUCCESS in 38m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/97922047422848c192feb1ae618990ba : SUCCESS in 54m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/746839964bd642d7a840b3a2665352dd : SUCCESS in 22m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b62858ed6fb4b4d89403941770a5a42 : SUCCESS in 12m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80f3d5ff398f4ae0a94763512943ba8b : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8464c8ea9e284a1eaa1750c4a4b94c55 : SUCCESS in 9m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c615e61f0074511aa5e358478d74f65 : SUCCESS in 38m 33s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-upload-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-upload-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-upload-image.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"e7afe6709682d5146979cb43002fd2bf5cda03e0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 13:03:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d9a614e21cd33c2148717ddeb32a8518900b0444","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 13:05:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e61f8b8582fa4b3da0a3cc43b4169306\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/1b4cf7e07f464157be528357ad48ada1 : SUCCESS in 44s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d9778b1facc48cd803bf0e47f1f9ddd : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f06b3d52148a4a738349f53f63512baf : SUCCESS in 59s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/3b1a39a28ecd498fb798d631fe5c7105 : SUCCESS in 55s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9f6709d3292b8ca40c13294bcae7de1d2f7d8093","revisions":{"9f6709d3292b8ca40c13294bcae7de1d2f7d8093":{"kind":"REWORK","_number":1,"created":"2026-07-15 03:14:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/25/997325/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/997325/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/997325/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/997325/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/997325/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/997325/1"}}},"commit":{"parents":[{"commit":"36332a5dcf6523e3aa6ca1ab7c5a751f4dd66414","subject":"Fix unit tests with keystonemiddleware 13.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/36332a5dcf6523e3aa6ca1ab7c5a751f4dd66414"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-15 03:12:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-15 03:14:23.000000000","tz":540},"subject":"Remove note about old pip\u0027s behavior","message":"Remove note about old pip\u0027s behavior\n\nRecent pip can resolve dependencies correctly without requiring that\nrequirements are listed in a specific order.\n\nChange-Id: I48f34077e4ec867a4d5bef0ef3b77b6e3ebb66ac\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f6709d3292b8ca40c13294bcae7de1d2f7d8093"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f6709d3292b8ca40c13294bcae7de1d2f7d8093"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
