)]}'
{"id":"openstack%2Fswift~998072","triplet_id":"openstack%2Fswift~master~Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297","project":"openstack/swift","branch":"master","topic":"s3-checksum-persistence","attention_set":{"38496":{"account":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"last_update":"2026-08-18 22:32:30.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-09-16 15:00:30.000000000","reason":"Vote got outdated and was removed: Code-Review+2"},"39146":{"account":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"last_update":"2026-08-28 13:06:12.000000000","reason":"A robot voted negatively on a label"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-07-21 22:18:37.000000000","reason":"\u003cGERRIT_ACCOUNT_39146\u003e replied on the change","reason_account":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-09-16 15:00:30.000000000","reason":"Vote got outdated and was removed: Code-Review+2"}},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-09-14 08:25:20.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297","subject":"slo: Pass segment HEAD headers to manifest hook","status":"NEW","created":"2026-07-21 00:00:43.000000000","updated":"2026-09-18 21:36:46.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":56,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"b1ed478cabbea2232e92b36a1ec127d1b3884650","_number":998072,"virtual_id_number":998072,"owner":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-18 21:36:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}],"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-21 00:43:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-21 01:15:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-21 18:12:07.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2026-08-07 11:24:17.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"CC"},{"updated":"2026-08-07 15:18:45.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2026-08-11 12:56:21.000000000","updated_by":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"reviewer":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"state":"REVIEWER"},{"updated":"2026-09-01 15:09:54.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"},{"updated":"2026-09-11 17:54:40.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"db516c5bc66509ef9c4a4a790e3737a67b9eb899","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-07-21 00:00:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bfdea72261ba02c292dc8877ecdc7c0e0812ea1b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 00:43:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a0f7e090cf74fd788e64d1818f125fd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8d90c8d96e7a4124be7d0a07e399f255 : SUCCESS in 41m 38s","accounts_in_message":[],"_revision_number":1},{"id":"482ea0b32f620a263f185d514f707e0846293f9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 01:15:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d31d11399912437ba3e36e1db1eda34d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e8c98b38ce44ccfbffcb4d0fb0f5e48 : SUCCESS in 12m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e6c8bc7de9d4c70acf8bdf055a432c2 : SUCCESS in 3m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/80e882515b614f74a4bcad94fe99030f : SUCCESS in 31m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/652e3f1534cb4b5da9cf6dcc6637e3a0 : SUCCESS in 1h 08m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d6dcb4f0eaef4694ac7a8cef00232782 : SUCCESS in 28m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ccbe428151940aaba85858867dd6121 : SUCCESS in 1h 01m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cf6682b0f55d4e10b2ef2f0b2b2c6b0b : SUCCESS in 19m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fb30ecb1d99d43849ca13338daaf6d9a : SUCCESS in 18m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7e267d5ca4f54fa0a7ca78991fb307e0 : SUCCESS in 20m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ebf9dba3270a4dbaa447fbb528feef5a : SUCCESS in 6m 45s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a43192db0c0b4c599d480b766cba7863 : SUCCESS in 25m 04s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/18852c239fd24074957c4b27ea9849f4 : SUCCESS in 10m 12s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ba5e137144eb44a3ae5f88ac7f1490b5 : SUCCESS in 31m 13s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/df880dfe11a045a5a056622d61cd0ea2 : SUCCESS in 16m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/df677cfb8ce64722b2696f7e72b88a25 : SUCCESS in 13m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/222cb11c435b448388398ef0dd161c61 : SUCCESS in 14m 41s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4586316f65664958847f2efd9507aa0b : SUCCESS in 15m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cea0b084c472446c8a7b3ac11fb5930a : SUCCESS in 6m 38s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e3f32f6ac5bf48e9bdf2929e8ade894c : FAILURE in 7m 32s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ab1d917da8e54452b6d455c5c7c1bf7a : SUCCESS in 11m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c6e6dd6285a4048973cac231c75712d : SUCCESS in 1h 00m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/46771135c6c24e37a8b0313a572c21ef : SUCCESS in 22m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/12b0a15135e04d41bbdbec98e88b849b : SUCCESS in 22m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e65b72ff2e2446acb27b1d117f47cedc : SUCCESS in 10m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5054fbb052be4000937551b9cfe4d209 : SUCCESS in 5m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e6d5526c5481456193127d9629da2fd5 : SUCCESS in 19m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d21f4a5909141bca90fd59458a3f73f : SUCCESS in 41m 01s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"c28dcbb3727f9ba58601ba7e9385582da533a326","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-07-21 15:58:44.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"894f1e3f3f0994d251fdc523fb01c37823236218","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 17:12:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fef04e08c7746bab624ff089cf751fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23dacbbc893c46f8860e216687a6ebe2 : SUCCESS in 10m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbc6663721af49899406b36530819dc2 : SUCCESS in 5m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/6171da87a7474d82ab8c4b062444e056 : SUCCESS in 1h 05m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9472b1369cc54a7db21a1c183883ceff : SUCCESS in 1h 02m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/043d39449d3e4471a0863e7c6b0f2745 : SUCCESS in 42m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58fc7eee48d2441baf612fec48b8b218 : SUCCESS in 37m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/27a3e9be652d4ca9a4b7c7d5ea2b112e : SUCCESS in 10m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c8aeb3fe3e8487eb292be3f3f10b181 : SUCCESS in 11m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3463f621457945ebad34ac4d7e0cb202 : SUCCESS in 21m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b3066193f08344f880dbe9427d67c1bc : SUCCESS in 7m 34s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/969be425889a4deb9aedc4449073595a : SUCCESS in 25m 46s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/99fad24e0c9b4760ab2613ea1e230766 : SUCCESS in 21m 45s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ed8e45a80aca4db68fd2c825c13622d2 : SUCCESS in 40m 00s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/499fa74dd4b6445ab31b24606609068f : SUCCESS in 21m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5c8b93da771149f497515e1bbdcb46da : SUCCESS in 15m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/beb96c2b00944b948e4e2287c674b062 : SUCCESS in 13m 47s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b616afb6b49a4528bf388acb6ad9de0a : SUCCESS in 17m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d551ba0153254ec7928db6f7d371b7d6 : SUCCESS in 6m 22s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/90ea6eb0ba2a4870a8ffc2220b9f9133 : FAILURE in 8m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8971c20d52cc47d6b16ae37bc333cbc0 : SUCCESS in 9m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7152bade8b944bd5b4324f174d76d2f8 : SUCCESS in 1h 02m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dab21994562c4faeae3cbb4c3d26c2b7 : SUCCESS in 34m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9c3f98390b7c4dad83885b837420fb1e : SUCCESS in 22m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41fbf088a5ff4d75b2176d6f33a09b79 : SUCCESS in 31m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fdc3d720bd941e1b29e814918bde62a : SUCCESS in 2m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ad6ea44f4737476c83c7247f3ee2aef4 : SUCCESS in 28m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2924e8f067984951ba9c8aa0703e52a3 : SUCCESS in 55m 41s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"0ecb92f145809f2dc170341316f39a56eeffbbb4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 17:44:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8eb4f6b38ed4208a9e35a7fa86d4cb9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/055f3687d47d426aba067b6e6a43488b : SUCCESS in 1h 03m 51s","accounts_in_message":[],"_revision_number":2},{"id":"b4e8d74eaa8ef81ac5f045d69dd1eb1adc5ca563","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-07-21 18:12:07.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d76eb3f174396b5b1a56fe24d7c39bc43947bc3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-07-21 22:17:56.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"f77f5d8aa794296a718d2885cf991f8eb42d3e20","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-07-21 22:18:37.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"676dbfeca7ee1d28fd709f34c197b1f505adbe88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 23:10:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/051cd4e31739441f8b6bcd08cb9e4518\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/65d44e28961647a78886efe93c2b7967 : SUCCESS in 51m 26s","accounts_in_message":[],"_revision_number":3},{"id":"1e8c02d874307d742b23a869984e5da6c00b6058","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 23:24:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4660dd0a6d4e4471bd6763995d11e8b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4565cc790f0a42469b1482fb75759ea5 : SUCCESS in 8m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1541ba314b34e4ab477035bd233f287 : SUCCESS in 4m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/2761e8bfd84f44dbbe0a7162411e3a78 : SUCCESS in 1h 06m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c99d9211f5ac461ebab2352425a9f886 : SUCCESS in 1h 03m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/32abbf8997cf454198b6dcd72edf462b : SUCCESS in 31m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5df4499866f74979b96bc43420264798 : SUCCESS in 1h 00m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8430084e3aa84302bb1588dc545d20f5 : SUCCESS in 16m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/194831969ef745f588171d3f134888ac : SUCCESS in 12m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1df18d4777e34ea1ba6e4ff5740ec847 : SUCCESS in 17m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ae50f694d8fc45cd85c3efdc5c3e35e5 : SUCCESS in 8m 48s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fe379d0a0aa5471daa242d59fc99595f : SUCCESS in 24m 35s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c7e810e7281d4528a98782b833ee1dd5 : SUCCESS in 18m 44s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/354fc2210dbd4b38b19b019fbe5e68af : SUCCESS in 14m 29s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/dfaf6bddb59a452783373902bd1eac43 : SUCCESS in 20m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b2b8901ebf854ab09988acce112355cf : SUCCESS in 9m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f220768fab594dea893cc7a013ce4e2a : SUCCESS in 8m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f92173cbf372444db1eff96c3e9d4d0b : SUCCESS in 8m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c933ce42e7c6479ba1f64b5b1c4c5517 : SUCCESS in 8m 06s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1dd7e78a4535444b9fe3c47ce2c78f1c : FAILURE in 3m 56s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7a1ab3fa2bf84280bd91518f6a7a6eec : SUCCESS in 8m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1738be5692df46ee8b2a79cb9b87c13a : SUCCESS in 57m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/093809bb159649e1be41425836038168 : SUCCESS in 56m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b04cd5c234844776ab4976c4522d77d9 : SUCCESS in 49m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/868f9a37e6964ce99d0a85d6c327310c : SUCCESS in 28m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9742bf1a3b534d7c83f9b6dc898b9b00 : SUCCESS in 2m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dd98a54c13ae417f8118a75ef44fe5d7 : SUCCESS in 19m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69904079ddae4fa6b9d3d897a082c2c7 : SUCCESS in 35m 59s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"0621538b86ce92b9d18b5b61e3d3f52712688ff8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-07-22 16:40:54.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"4b40a4159fcd36f9f20c8ae5eb3d159ac10a6085","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 17:30:41.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d50d32c602144e388959ed63c2369e5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/21fa479a51cc45f5958f3c5399f0363d : SUCCESS in 31m 54s","accounts_in_message":[],"_revision_number":4},{"id":"b37b4341201151857faf87e8b12b7c45579b1922","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 18:04:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/37538c48ef0b4f03b27c0b942921e8dd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cecafb6158f46b38d85eaac4ee52a46 : SUCCESS in 11m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/382b42e8376b4d6c83394cef2846e357 : SUCCESS in 8m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/9e7667eef128493080f1ffbb3c7fde06 : SUCCESS in 1h 17m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/66b1a736d87d472993f91ccc90b68280 : SUCCESS in 1h 13m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/63f6407ef86b4f93a04122c76c78423d : SUCCESS in 51m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc3fdb536c59448dbbc61c7960fd4154 : SUCCESS in 1h 00m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2846631c62634c46ae3dea2b5148ae77 : SUCCESS in 22m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d929bf5a14fb4f6598e6ab6e6a9f56d8 : SUCCESS in 20m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9d224980ce95445eadbdd29ee1e67f0f : SUCCESS in 24m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2c6e65f9f77642dca5c52adbd6993380 : SUCCESS in 6m 27s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1f1b9326833a4529acdd5bf9bfa7bf70 : SUCCESS in 27m 39s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8c6feb8bf2d64b969b2d01a944c9b2e6 : SUCCESS in 23m 43s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/675d0b4bb02b4cfcb6275a9deb8383f3 : SUCCESS in 31m 30s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/04a61f0461784e0cb480dab61c2e857c : SUCCESS in 18m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/14ab54d4fecd4983a9f476e3b8999819 : SUCCESS in 15m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3654f50aeda7400692093973552f85b1 : SUCCESS in 16m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/132ac4f49023413b8df82051c8165bc2 : SUCCESS in 21m 00s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9faa48ce9584c13a0ef437ff3210e45 : SUCCESS in 23m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1317acddf938447dbeab90a111ffc6ba : SUCCESS in 6m 55s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2b9802c37fa547429822be5506e2d6de : FAILURE in 8m 15s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9141d5dfa38c4127a8cc6928b6c23fe4 : SUCCESS in 8m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc3ec84ffbeb4a8d8f5daab22e674868 : SUCCESS in 1h 05m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2712fea0f89845ba9dd19face4e58b72 : SUCCESS in 55m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/de9fc9d6bb6546e094cd833bd0727d79 : SUCCESS in 57m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc161b5fea78478a8ee072b3959a770a : SUCCESS in 31m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aef17c2569d943958cc32cfe0e9ca9c4 : FAILURE in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/93c647711ad94c21a26fcb53a362d488 : SUCCESS in 20m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e23c7c4be914c6b8178946895f07185 : SUCCESS in 1h 05m 00s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  Comments left for invalid file swift/common/middleware/s3api/controllers/obj.py","accounts_in_message":[],"_revision_number":4},{"id":"44b991812dd1ae9afec81a0a54503e88b418f4ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-07-22 18:26:15.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"927d1eede9e7e633717b33a54016ce0bee0d2972","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 18:56:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d12cf627ce64992973bd0c3bf6dc5d8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ec49f19858cf407eb1299a375b75063b : SUCCESS in 28m 48s","accounts_in_message":[],"_revision_number":5},{"id":"202e71c41cd8d06d97747bb3022a26db279b7a83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 19:34:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40251ea60b2345b08f8b5706038f9cf4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3199f091d82f4e6097efc76359f00cbe : SUCCESS in 11m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e170fe5c208b4120a66480b4a41521cd : SUCCESS in 8m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/f68ad182895b4b1b8d1da94b8be69ac8 : SUCCESS in 1h 03m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ad07112b007842abad455d0733dd8c0f : SUCCESS in 1h 01m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a037e5f7f4374ad1ba234a9f54e3d5f5 : SUCCESS in 50m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e39bd02495014a129c11011c8609d8e7 : SUCCESS in 52m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d9ce2167aa444e6ea71f38c39043065e : SUCCESS in 22m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e9fabdc7038c4cae949e2863ab4cd112 : SUCCESS in 22m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a67b266e7aa9470c9782fbbb206cc1c8 : SUCCESS in 20m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/585340003fe34d9597a44afa87ff85b1 : SUCCESS in 6m 13s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/45b27ed9f5e7430881351468002bbc00 : SUCCESS in 23m 43s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/81f545771fa54fafaf51edd38785b2e0 : SUCCESS in 22m 00s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b285698743b44f7da94d67df1ab10349 : SUCCESS in 22m 53s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/7df35996400b47e6826e402d2364129b : SUCCESS in 22m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b7c723b2b7a94445853ec93d9b42b3d1 : SUCCESS in 14m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2c46fa65a7a2499c8169387cee8eb25c : SUCCESS in 15m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/64d56d4d3dfe497db70253147f54a76f : SUCCESS in 16m 28s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/270f0517ae9c400baeb35352d3aa47d2 : SUCCESS in 17m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3124ac189772498388e0a881c61ffa3e : SUCCESS in 6m 49s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/be4555eaa8c34e639d786760db1ef0c0 : FAILURE in 8m 10s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/56aaa4160302469bad91c2a6ef86733d : SUCCESS in 8m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c973ffc190cd49aebca900a1ca306d8e : SUCCESS in 48m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ea623598f61043cb9ca228d481ec8341 : SUCCESS in 58m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/37c76bd42d154c0b80b1614144af5897 : SUCCESS in 56m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3630eea7b9e4c579a79ea51ef94fe81 : SUCCESS in 29m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3ccc908b58246bf869e079406a48b71 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/261952b890b0407792ca53fc32cef080 : SUCCESS in 20m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7662f9c2d78e483a96ffe9c29f7184dc : SUCCESS in 59m 52s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"218959deac896ee4e5a3a8db5bf63513bffb12d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-07-22 22:10:24.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"54889dbaf02cc15d4d2a901bd59953bd8328f686","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 23:12:34.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ef5f8dffd4b4daab20436c4f89165e5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f5007904d7504f67b00b4abf79d6c667 : SUCCESS in 1h 00m 23s","accounts_in_message":[],"_revision_number":6},{"id":"64615bed615b27ac0a136cf6b32b74a92b5a1060","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 23:20:04.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f6d470a20a74d2a87de53113e8a659c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b84732676bba47499d73d8affcb9658c : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69b90edbda2b484987bb2986c0ffb16f : SUCCESS in 8m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/46f4975785834995a5822f7a7acdfe37 : SUCCESS in 36m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f4a42f5409cf4d7c84274f245ca75749 : SUCCESS in 1h 04m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8b39d232c3574839a3bf24d0315909bb : SUCCESS in 31m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8bed6622824f4d61b8644dfc9fd4f02f : SUCCESS in 39m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/64f701f27bce4fa0b8ef0262e405771f : SUCCESS in 16m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/74b77b9cbca34841b6438e753efb1b20 : SUCCESS in 20m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4cb24439161e4d2c865520773cdc6517 : SUCCESS in 11m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e683d6bc41ab4f7dae6caf811611d88f : SUCCESS in 6m 16s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b82e5d026ba0463f9c448ab8cfb837f8 : SUCCESS in 28m 30s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5b0c947486a2461bb5b9136de5e2dc3f : SUCCESS in 9m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/525689aa86b74cbd911f220c4a5b2919 : SUCCESS in 30m 31s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/6566462f6edd4900a38e54836a274ce1 : SUCCESS in 24m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4346541136c344d4bdc13b3aa9eff09a : SUCCESS in 16m 02s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9ce8b494b32f4afea56d0801fdd73d42 : SUCCESS in 18m 00s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/181ed5bc78b44fb99b70e7bc57507ccf : SUCCESS in 14m 29s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/f298b3aaca7647858a86b9584c7ac861 : SUCCESS in 9m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bd631e17d21c4b6f8ccf95cda2fba847 : SUCCESS in 3m 48s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/997b048b5ac04c7ab5178ee0b464da57 : FAILURE in 4m 14s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4ff36857f4e745beafe5c6c8cfecb06a : SUCCESS in 8m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c8111a20b494b078cb62d1570c58626 : SUCCESS in 38m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/be3f524b902448bb9e8196be0eda8140 : SUCCESS in 56m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1c1d609a92db4075b12eb7896ba25894 : SUCCESS in 51m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa3411a2f975432f8fcc8259e5bc404a : SUCCESS in 22m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2f529ebff9d4441a81daf3f94f0c50c : SUCCESS in 3m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dcb2cdbfaa13432785683150eb462393 : SUCCESS in 11m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b2a1e9b8d1a4790af7bb1e6a47449e1 : SUCCESS in 27m 13s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":6},{"id":"fe8321a3d2a769f1ea9faa8da1168df523b268de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-07-24 19:00:34.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"2edecc01080b7d2f3aead219a51a75c091b9c04c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 19:49:36.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4713b91c1cc849158438506360bd72c7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/11667b407b474bc6a97055945e72386c : SUCCESS in 47m 19s","accounts_in_message":[],"_revision_number":7},{"id":"d84073fcbf8261d7e71c51f45b414c514da3de7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 20:11:17.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6c0f247592342dfa709c9539d08d709\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6369be4350b64ec995ae5fccac117ee2 : SUCCESS in 10m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3be148936ba6416db96d6c4e93ce4f09 : SUCCESS in 6m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/5e78339a73c84a0197f98160584c398d : SUCCESS in 59m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8de90889099b42e3a53342d0d1be78fb : SUCCESS in 59m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6ffb06a8bad34874a7ab888574d9e390 : SUCCESS in 33m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a0a2da1401141cdb91b87ed5aac67d0 : SUCCESS in 59m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/74f9eee72a6e49ad84cdf737129be901 : SUCCESS in 17m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ffeee664d8ff45d68a8eadd1088ae801 : SUCCESS in 19m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/380987a2eb7a45b28181fe319221b921 : SUCCESS in 22m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f4e99a87f3b642b391dae469b12fd53e : SUCCESS in 6m 57s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/24d632d51e7942b4aabc5a29f32184f5 : SUCCESS in 27m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/13e2fd51cca24d9db841a5e030522e96 : SUCCESS in 15m 49s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/97bcc17a05a7446f91dc90e03940ffca : SUCCESS in 30m 45s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/5159fee6db9a4cf88ab047084232a55d : SUCCESS in 16m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/69248f5ff89c4a56b60e7e3c53df405d : SUCCESS in 16m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/eb5ac67051064815a30ab4f44eef3ced : SUCCESS in 15m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0cd1c5136b5f4a36acd62a81c68e308c : SUCCESS in 14m 22s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/334d3eb3eb744411908c3d7fd1e80cda : SUCCESS in 19m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a52c2291eba340ea9809ec76d4089eb4 : SUCCESS in 9m 32s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f0671ea05ddd4fa99b7d8f8ca6f5b312 : FAILURE in 7m 40s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a9a82803cedc4fc2a0f979eb41995815 : SUCCESS in 9m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/deba75668fd44894ba88d81554dff144 : SUCCESS in 1h 03m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3a2521c8e4454f1ab68f7873e24a3cdb : SUCCESS in 52m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3fc4cfa4e29a465ba4d0bd394119bc56 : SUCCESS in 39m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e472cda881d4d3c869cd106a9ea5878 : SUCCESS in 29m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/965843bee15e4cca9d033b25e1d66de9 : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2659d3e8e29946a2b4324c9757a46135 : SUCCESS in 18m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ab1d82bc2034a47bee23caa47e0ec67 : SUCCESS in 40m 45s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":7},{"id":"4b4d4a64e37dbe6070056026ffe455538b296f6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-07-27 17:04:37.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"256a86518a1c2e996c7ce6fea2ab9bba6b526b27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-07-27 17:08:09.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"94efd0caf758facf1bbc1f4fff5987dbd1f9daf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 18:24:53.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/441fe60b9aed44e289117c469e868cc9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5888557977d64308a38c6c0de5a7fee3 : SUCCESS in 10m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9084c42ac66458fb02ecf80604f7394 : SUCCESS in 9m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/3489f3f4aa7c4abeace954d65acdb64d : SUCCESS in 1h 09m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d338a0934133454f8291d98234a0f002 : SUCCESS in 1h 11m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6c7b1c9ea57740f6bae3578c0cd8968c : SUCCESS in 54m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/76a1ea8ee26542a59d42af65e50d23b3 : SUCCESS in 58m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d6e675a9fe34d14ac69b300d2bd445e : SUCCESS in 19m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7747728ca666483bb7366c4c35b2f622 : SUCCESS in 22m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58e20c775ccc4b3ea186ecac94caeb43 : SUCCESS in 21m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/368662a0aea1461b9665db68233ecafc : SUCCESS in 7m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e1576f48742a4423aeaf71e5515cddb3 : SUCCESS in 26m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/fa179bc343ab4bb48e3469d98ed8c16c : SUCCESS in 22m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5ba91685f9e642fabdc2583bc7670e3a : SUCCESS in 37m 15s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/7776ae4f7ed5444882a9472091f49521 : SUCCESS in 22m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/18807d48dbdb450387a4a4ef76d6f5b9 : SUCCESS in 15m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/44c904e7181941e88bf488fbc93d3198 : SUCCESS in 15m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3d1a84f72a824ca8abb483e0fe539340 : SUCCESS in 13m 17s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/c36bf7b231f24e4c8555b29121812731 : SUCCESS in 16m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/60439f8977c64e0dba9654597e11b971 : SUCCESS in 7m 29s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/21110d65a185410db31fd393ce8c89c1 : FAILURE in 9m 51s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f3d144e10bae4fa58cf9311d2a08c009 : SUCCESS in 11m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d9afabe673344dabb029357928eff216 : SUCCESS in 48m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bf34d6ebdf144ac29b32c8f7f774af76 : SUCCESS in 46m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fbfa459129d14b33b2eb67e133cf8794 : SUCCESS in 52m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e816824c35b54e708780ce1a3711e259 : SUCCESS in 28m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c62f6ae2e1ed440b85fc12684999c635 : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bbbae8b9c8254100958074d3f2dbb2c0 : SUCCESS in 21m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d621a4bddd6f4c168c510ef620e61eac : SUCCESS in 59m 30s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":9},{"id":"c3f1416ee4668caeab429252532735448053d8f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 20:50:22.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c53e59d44e2b4547a273a8407211aa80\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/15114444250648da87fcc765e76de167 : SUCCESS in 58m 29s","accounts_in_message":[],"_revision_number":9},{"id":"45b5516944f0811bfccc02bad029fbdbf6c419b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-07-27 21:16:18.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"669ea63869da664c434c881f86ab7add411ab0c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 22:25:09.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8c2c1254b7845eba8dabb34f568762c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95b92fced76f459c9e80ad7ebc5ae101 : SUCCESS in 10m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00c04644e04c4349840dbd0bbb48a9c3 : SUCCESS in 8m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/7bef535f047f4a7c8008246c080a3eae : SUCCESS in 1h 03m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ea3eac1c068344d6877bd713f6492785 : SUCCESS in 1h 00m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1b68d6c1e0db412cb7d18d9d5627ac7d : SUCCESS in 49m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d20e025cfdc342fa9d6e4b5ec7f16f60 : SUCCESS in 59m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c43f520832424048ad93bc081a87dcbe : SUCCESS in 19m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9d130488089e44c8821163b6c9abbbdb : SUCCESS in 19m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8b8b8a3ff2904bee9bde5f8b42f04640 : SUCCESS in 17m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8d769015be834fb69811fdc0f2a44d77 : SUCCESS in 8m 31s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f449f4f843b24a61bbe459f6088134f6 : SUCCESS in 23m 17s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/cb4d2178eff14f7692c7fe5833bb8ab7 : SUCCESS in 19m 45s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/443ce2f1b10f4b089ee27f4eceec3145 : SUCCESS in 30m 27s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/e66b4a948e8047f9944ec4775b3272d1 : SUCCESS in 16m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4ccdccdb235a46258c6f3cb0e6205135 : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b52f90f1f522444580536551b3494e03 : SUCCESS in 9m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e5883255a87247fd9754df80ffcd8300 : SUCCESS in 15m 34s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/77778a08697a485da6afe51b509aac5c : SUCCESS in 17m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1a2b0bbf463e430b84c48cd56bd1e941 : SUCCESS in 6m 29s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/76f7a24a542143108f707e68f5ad5c1f : FAILURE in 7m 03s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/07ebc62c6cde42f88f8d69d570a556ca : SUCCESS in 8m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3efba2e3011b4c409ff2f90b845548ca : SUCCESS in 1h 00m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/029bb075fac747a9b652a51c2177550c : SUCCESS in 33m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c673ee2f5097444eaed7676a5c8204ba : SUCCESS in 53m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/533fbbf39d284c9a9614092ef790973b : SUCCESS in 27m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ab011e11f1645aba360d9065abd6161 : SUCCESS in 5m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4baa8cf523ae48ec8ad4e49d9c271f09 : SUCCESS in 18m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3891aa16daab439b810dd19dbad094fe : SUCCESS in 1h 01m 20s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":10},{"id":"d4e8f1d61407020d11cd3778a0a8ac1fd1e3f9e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 23:16:57.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e31745653fe5423cb7ebc74e5cf6f5e1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/380f2875d9c14fa687ef7c2f132b550e : SUCCESS in 58m 21s","accounts_in_message":[],"_revision_number":10},{"id":"1e9634561da900c0428d18d27ad3eee33ff70fc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-07-28 17:29:04.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"c96bdec2e669eef705a1c82022f12c25035550a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 18:38:54.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0916544f0b7414ba75dc22b2f864d9c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a61fb9fef44a425eb935b8758a7fe657 : SUCCESS in 10m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d001619ab3a4b2e84eb9292981d6520 : SUCCESS in 9m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/0f4106b3a3994b7fab5871ed7ef0f955 : SUCCESS in 1h 05m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cbe2cb075d6f44bf9a9802000da839c4 : SUCCESS in 1h 02m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/835028d41699417a9b20ef397e23d42b : SUCCESS in 58m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c77f41086b114f4a9a52373945262579 : SUCCESS in 59m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4340977e043342f9bae9800f7af494bf : SUCCESS in 10m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e8088200708c4222a2fb846fbe31d705 : SUCCESS in 39m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d05cf06177648cdbfcc1d0109d54f6d : SUCCESS in 20m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2c9cbd437beb4f2697a2e627d523f1c2 : SUCCESS in 6m 50s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/43534650b6cc44db81e8d0733f0ac8b8 : SUCCESS in 22m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ee6031d862ec4a49b0cad65b26f445fd : SUCCESS in 9m 50s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e68282cf9a1e4b20a3d148c62313b36a : SUCCESS in 13m 59s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/3e5bba5eac8a4fa985111dad77d3e2f1 : SUCCESS in 19m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8e00ae84d450452785aeb49f89b39a73 : SUCCESS in 9m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6556333a54e244ecba39d366af5722db : SUCCESS in 15m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8789f4a37d0943358e1b14487e9167d1 : SUCCESS in 17m 11s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/d9f8988da383400a9c4cc26cbb71223b : SUCCESS in 17m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/29530b2a0c8b408f96a721a2c0d2f749 : SUCCESS in 6m 15s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1c7d632c354a42e4a1c728942e764bf1 : FAILURE in 7m 21s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e98f267ad5ee467dba507f34a2fb0c31 : SUCCESS in 8m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea345f90bc7c4db3aaa03c437de33804 : SUCCESS in 56m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c1b00f4e8d194a8db83a4b83c5e7cd80 : SUCCESS in 52m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6504cd0acf3e45639943c07381ab24b7 : SUCCESS in 55m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8953fbf7f2404fe1b48b1de20c88f3fd : SUCCESS in 28m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cdbf5e45caf40439d87a5b00d7e4618 : SUCCESS in 4m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c5e5873060dc40449c1f952b5ba38888 : SUCCESS in 11m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/901c18138d684294a200fb041d052fae : SUCCESS in 1h 01m 43s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":11},{"id":"005ad7bc9432cf4bacc06721c322827ab5aa6d32","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 18:48:40.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/524b5b374ec346c68c078ce1c2aced09\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e4c5b6d0c4124fa3aba8a95e08fee456 : SUCCESS in 1h 14m 48s","accounts_in_message":[],"_revision_number":11},{"id":"3398939811eab06cbbae2c48265977de62e3bb51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-07-28 23:23:20.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"78e2b96d8e48ebb9c28e4857df74663a05d18b9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 23:57:12.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/156afa9604444657af017b1418aee96c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7df54393e40b4112baad94bb488355f3 : SUCCESS in 32m 04s","accounts_in_message":[],"_revision_number":12},{"id":"c2bc2d8b1b9d86376edabe58c33fe6e9d414565a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 00:27:42.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e7ddfbe6f0f4ad495fc024e0255bad0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d81658df206748c58a8ba94433a650d3 : SUCCESS in 9m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e00eb29373b4cc4b1497a2ad9bd38d4 : SUCCESS in 8m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/765a583ed34e4cac82408c600d562ad2 : SUCCESS in 48m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/47f3c6107e714732adb5e7b14e33c5e6 : SUCCESS in 43m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5a1c05a03b034d56a756dc0f6be71bb1 : SUCCESS in 48m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3e4b51d4e60411f8e8e7f17b4ebd03d : SUCCESS in 43m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5c84ad87c3424c7ca9b73fd3b0e61b3d : SUCCESS in 18m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5da49cb2b4fd4f74896250178babab5c : SUCCESS in 19m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4649510d4ceb43bb9d5394165f923ea2 : SUCCESS in 20m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a9d371f64ccb411796e670846a3a57cb : SUCCESS in 8m 55s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b4d98180936c4a378dcc5da754baa8fd : SUCCESS in 21m 32s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/591937ee792d4481a9ae821f05e1d379 : SUCCESS in 22m 38s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a3eba213d3f8434c993a005d3132e0a2 : SUCCESS in 11m 37s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/d833d0eac28543cb826010f5d5c2a508 : SUCCESS in 21m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/260d28e8cdfc4c52901cbd15c3e3a4aa : SUCCESS in 14m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/43f0773a5eaf4a32b5405c756ad863e3 : SUCCESS in 15m 00s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e755abcd1f134c0ca4b13c83f19e15b5 : SUCCESS in 15m 26s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/eb41ead2506546dcaa86db7b09113946 : SUCCESS in 17m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/efcba8326c0349d889f07a9e70481151 : SUCCESS in 6m 08s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cf1da6e234224c6599f15cc0108465ce : FAILURE in 4m 57s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2f6bc8ce23d54c8489ed27393b104c2c : SUCCESS in 8m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8119fb259ce64e0eba891d3a892fd1f9 : SUCCESS in 58m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/55b9d58211dd43fe811e3aa0ed8ebb58 : SUCCESS in 34m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/53add66da65d49b0889004c9b5e9f13d : SUCCESS in 56m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/770361b22551406a90fe36339925db9a : SUCCESS in 28m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7d1a9680c4441639db3a0d692de2bb5 : SUCCESS in 3m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/030d3a2835d94555ac1622e648397016 : SUCCESS in 25m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bade1373fee14fd28d139227f8295dd5 : SUCCESS in 39m 44s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":12},{"id":"0a1d9db3c3d132b87b20c8e39c0fb8e6e1b950d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-07-29 16:20:11.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"a10b7c5c52df45c094f6900450d1c40b014932ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 17:38:09.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79754483067442cb8c20cf4a4bfcfe63\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d70bc223e424468be8536a2c3d739cd : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b739a6849765485d9e9adb1c91df1e73 : SUCCESS in 6m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/32bff7cd9d4d4e6a9ab3e1f998a1de0b : SUCCESS in 1h 12m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/505e5559817048a6a797ba12a863e9ff : SUCCESS in 54m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/22ca9e1521a0473282286056ec60f9b0 : SUCCESS in 42m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/766760e51e544012b0395d0f89026d71 : SUCCESS in 1h 01m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f0008b1bd71c491a8d36d003aad08aab : SUCCESS in 21m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b16503f07a9b459a8497bf61e6b7b360 : SUCCESS in 14m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a6887580bf640c0b7f9da8f5095dfe1 : SUCCESS in 14m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/02a041f48f36454497d076e610dcf307 : SUCCESS in 5m 39s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/febae93ad12e470990144b4dc3174b4d : SUCCESS in 26m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0b253e76834d4db1a0a93a619b0cdf09 : SUCCESS in 18m 43s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e149aa885f4646fa81f6c13d3b1526ed : SUCCESS in 25m 23s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/923454972b054c20926f67e6358d0dd2 : SUCCESS in 24m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0ae4f9ea5f9f4633a3d431234f194c57 : SUCCESS in 8m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/92e0281620924e4b96fa31bb5a60d93f : SUCCESS in 10m 52s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ca54012f55f64bebac30b4a3a130cfd7 : SUCCESS in 11m 25s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/6bb489793293457aa99ce2239ed27c1f : SUCCESS in 16m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cacb325d132d43e58c23286539117d7f : SUCCESS in 6m 19s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/af32a50ccd124b25a488071350f8fadd : FAILURE in 8m 35s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5fb06a9fd3ed4d4f8b5d71216fcbdf12 : SUCCESS in 9m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/75ccc8a02c8c4e2da66b96868a9ac229 : SUCCESS in 36m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/05899f8337464f709d0e4c7706d5a8aa : SUCCESS in 32m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/58df84516bf143ad8696712bec652835 : SUCCESS in 32m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10e69e1b50a04c1c9c4acece9871581a : SUCCESS in 25m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/239bdc98767a4516841a01a0e070e710 : SUCCESS in 3m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6285280de4914305bacd51220589dda6 : SUCCESS in 20m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4168afd027f4b6bab3b1a44af8f3672 : SUCCESS in 42m 05s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":13},{"id":"9af71beb782184fa1085a685383bae9456cc454d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-07-29 18:20:47.000000000","message":"Patch Set 14: Patch Set 13 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"bbc649f9149cdbc770187f360b5ea85010d8e80f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 19:28:48.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/439de651350f4fee836305bee51d35e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dba6d239a944402f882365c06f7b3efc : SUCCESS in 10m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/094999a9527e4343844ed3f412308486 : SUCCESS in 6m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/848b20da58b44d22af5e7fe7d42cd837 : SUCCESS in 51m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6724d8f0755942a9a03be5c53e775318 : SUCCESS in 42m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a83c0ccc18d74cfeabc5f84c47021f54 : SUCCESS in 29m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a0ab9189f994f2dbad606a6625a6a05 : SUCCESS in 57m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a984c8a9cde24c71a1d286e7fb7f9508 : SUCCESS in 19m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/011c616209594793ae16d4f1d2941840 : SUCCESS in 17m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2c659b9a680b43fdb43cc9a1cf2c1e29 : SUCCESS in 21m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bf6537e68d304336b7822d767930b05a : SUCCESS in 7m 45s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e6ee7846154244069ebcf2516da2f0ca : SUCCESS in 28m 04s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1a0fab0dca1846ad929dc56fc9f7bed4 : SUCCESS in 19m 12s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/cae1303fef524438996ecd2cec2e3f84 : SUCCESS in 31m 15s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/6748cb052e83479a95a32c3cdf19f989 : SUCCESS in 22m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0d2c82a4b73d4ad180ecd9fb651652df : SUCCESS in 14m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d96711427304478a91bedf08ea7b9f91 : SUCCESS in 17m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a18041a3c88545ec9d4d609f5f0eaaf4 : SUCCESS in 16m 03s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf51cca3457b429685429955ffc1b123 : SUCCESS in 19m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e54aff6eef7746c099f2a901a7d757f7 : SUCCESS in 9m 14s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5e15ca4bf748483a889567fe484a06b5 : FAILURE in 8m 27s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4448d59d33174bf8ae49726063cd2224 : SUCCESS in 8m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b981a3719f0e4f1cb88aaeaa0be9152a : SUCCESS in 1h 02m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0144bd89f3134aa4ac47984e3f5b2103 : SUCCESS in 54m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/19628c4cb5434c8bbb0e6bef7ce09e99 : SUCCESS in 56m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad8f85f6a70742bebf4162a0440f8623 : SUCCESS in 16m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb6d6453c77f4099a09e4e5fad39b0c7 : SUCCESS in 4m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/38ae5abd3f994768a979260e74269532 : SUCCESS in 27m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ec79180481846f8a64b7d72df662800 : SUCCESS in 56m 39s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":14},{"id":"883c3887b5da478f3309e55cd5061eab2ca58fe8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 00:41:19.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8b77b0046d947899d2e6811caf9f8c5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a19b1ff5178e4d86b8f32ad194964f64 : SUCCESS in 1h 18m 08s","accounts_in_message":[],"_revision_number":14},{"id":"dbb8a2570a1eaf3ea6b8bba1279773ae64986d1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-07-31 19:55:34.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"8e1bc0a6174772263fe1c24f115473598322751b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 21:04:54.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5162c418dd044649417a8b2a10965ef\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/50526d16c0b14b8790912957a4f104e6 : SUCCESS in 1h 06m 54s","accounts_in_message":[],"_revision_number":15},{"id":"bc570b48445f78e5f4939adc01f289567e296ba5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 21:06:36.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fc20be92dca4786b22bd474bb39490c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0da019f863cd4cc1a36a1057b4a7a18d : SUCCESS in 8m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c8725c12ac14763a8424b10f98ac1e4 : SUCCESS in 7m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/03830b136b5c49138f29fcd24d75f023 : SUCCESS in 29m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6bd68d44d138485bae8d68c12ddf5269 : SUCCESS in 1h 02m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0bf6c85d96cd4249ba78a5db4a816208 : SUCCESS in 48m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e0c4462140a4b0981b78a0b94a4e922 : SUCCESS in 31m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/495fd3fa6e174c949bf5e01205460595 : SUCCESS in 20m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d18ca55a9554bb49058cb477d44c111 : SUCCESS in 21m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/03df1128aaf6446fa0dd442a21274ad1 : SUCCESS in 19m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bc7f0a70b66c4a45945b9e4e02349279 : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b4f14b70f6f44a009d6a9b05a7f1f0dd : SUCCESS in 22m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b22a740273b64f6a8ba2c55454c3cc16 : SUCCESS in 22m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2dbbfd1a126b4df08921e98dbee9bba1 : SUCCESS in 11m 26s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/590309bf72a14d2cb0c37244846debb5 : SUCCESS in 9m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5983d76dce1b422dbc64510b09a15ceb : SUCCESS in 12m 46s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0516cf8ceaae4d52b4539fa448b5f494 : SUCCESS in 9m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c9cfaa6872e84edaa26581840143f304 : SUCCESS in 16m 16s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/aba2b6d12292488595943760ba81d12f : SUCCESS in 18m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d31377df14204bb598757209fb66aeca : SUCCESS in 6m 18s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2e60fa2528894e59ad53ed68cf8f2f72 : FAILURE in 8m 31s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/710fe8e32095462f864a66bbe0f36dce : SUCCESS in 8m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1bc98aeb924e4695aa778131abfca432 : SUCCESS in 1h 03m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/073e638089a547f99b36499b27d58619 : SUCCESS in 52m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/68af80cdfe984dbe8dc057c25a39e445 : SUCCESS in 33m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2960938b5cf4c4bb9dc3186ba4b33d3 : SUCCESS in 22m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff89092ecd5c4857bd7cc820bd9b8ead : SUCCESS in 4m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ae925190676e4bb7a8e0b5baf51cfc1f : SUCCESS in 19m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4f02a4a467d489388975786d2849815 : SUCCESS in 58m 08s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":15},{"id":"2410fc30207a97893069d0deca344ce91b15e2e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-03 18:38:22.000000000","message":"Patch Set 16: Patch Set 15 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"cc6ba204e29fa1ac5a36f9316493750c1c8ab7d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 19:37:13.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da7577d4f51f42739effe01f8655e9f2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c1861f5381c0447282de56b5d13e61f6 : SUCCESS in 55m 41s","accounts_in_message":[],"_revision_number":16},{"id":"4af4314ed0958957bef9d0986f6ce1a8e8acdf48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 19:45:46.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f08afa83f6274ef5acb8d7c202485d0c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da9de72e572b4cb883b0e7b361600279 : SUCCESS in 10m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62392cefe43e4a7896e4c59cc1f1148b : SUCCESS in 7m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/a6e235bdd00a4185916e161b0cb748ab : SUCCESS in 36m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b835b26cb1774070be68dc9c5cfcb174 : SUCCESS in 48m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/39c6c3464e59427883823946512e3455 : SUCCESS in 56m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3ef408131b54f698a0fef20c4399aa0 : SUCCESS in 1h 02m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e2ecb2ca839d4e4fb3a5225f968145f3 : SUCCESS in 17m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/70b1136f25f14fc39636625ee83b3541 : SUCCESS in 12m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/afb3e56c8b9f40aa93f6487001e1b1bd : SUCCESS in 15m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2b97d2e4ead144958fed2e141c906f2d : SUCCESS in 6m 00s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e718bfe77f2a49ce8da957b574aa554d : SUCCESS in 26m 17s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/da460739fbe24950af5f35eb4184f6d9 : SUCCESS in 12m 25s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/12dd869ac89e410d98b1c8c383a777d0 : SUCCESS in 31m 51s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/107426f6dd6541b5a5674c1c99f409d7 : SUCCESS in 14m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a53b853e36384c67826f1c4cd36218c5 : SUCCESS in 14m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4bc0e5ac25934f99ac7133fbda71d6e0 : SUCCESS in 15m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d4af40c4fbe5475381bc861d1dffafcc : SUCCESS in 19m 30s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/16155654be0b4c92a95e7e807d96f419 : SUCCESS in 16m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8fce4b9d116449d78dccd3bce37596b2 : SUCCESS in 6m 15s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cff8c97086ab4039a896781c0ff4226e : FAILURE in 11m 08s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7217fe08789549e0b5f048efa7d85dff : SUCCESS in 8m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3dfdfe82a15c412caea385e68e6a1ab6 : SUCCESS in 1h 02m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8d70c4f33a1244188eba863abed224f6 : SUCCESS in 33m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ae46d9b4650e4e43bd39a53d015301a6 : SUCCESS in 29m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e781091ff7a24009bf01960f0f7f89d4 : SUCCESS in 14m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dae03bb376fa4eb99422bafd67079487 : SUCCESS in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2c0141b6627c4f0387343790179d3e36 : SUCCESS in 24m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/192f1045745a464cb9dce91cf9cd6661 : SUCCESS in 52m 12s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":16},{"id":"165ee736af02c1a044d5a2498e9d98cd2153d74b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-04 18:18:30.000000000","message":"Patch Set 17: Patch Set 16 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"44b30916515dc94bf1475061dcbe3fc15154deca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 19:32:32.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a8509ae57bc4990827092e31df6d3e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fcc709d66a5482f8a19b979ee36a2e3 : SUCCESS in 9m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96cdc230473144da8cde94ac86443b3a : SUCCESS in 8m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/a87678d8a1444966b41c714762d3a43e : SUCCESS in 35m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fb57756c8fcf435eae323af118e7caf8 : SUCCESS in 1h 07m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2a0df36d85804a9f85fe2a21e939ea1d : SUCCESS in 32m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/607796f73198409b8f2162bf6d9ab6a3 : SUCCESS in 59m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/14ec3379cefc4340bd7e5690c9e737e0 : SUCCESS in 17m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3aa54f6c0bd543aabb774c2dd4914e56 : SUCCESS in 23m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7661afd3aea94e82aa041ffceba5be6a : SUCCESS in 11m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/58269eb6e10b47619e44f8328f70b87c : SUCCESS in 6m 50s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/88b57835b382469d8ff22c97ce9a39c1 : SUCCESS in 21m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/82e9a6cf1c2047ac967f9da5da75a5a7 : SUCCESS in 16m 46s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1b0f974140b940288812b4a894870d7a : SUCCESS in 31m 17s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/a955f93088444c0fb51b3a065df016da : SUCCESS in 16m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0b50682f86e54a96b8912bbf4e65cb78 : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/125560babcc644ba94b20911b3c95ee1 : SUCCESS in 16m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8a677a58448b4264932d36e38cfb25a6 : SUCCESS in 16m 41s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/00438b5d41854695bf421913c16ad747 : SUCCESS in 18m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3e85be6f7ffe4391a48be69156f73825 : SUCCESS in 6m 08s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/15c8462f0137497f8314775efe9fe506 : FAILURE in 7m 47s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/710c31dce11c47818f9c5022ad8b379d : SUCCESS in 7m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/164dfbacb2a94c4d81b7e2743c8f708e : SUCCESS in 1h 01m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/62654f0fe0c84fb980aa529b392fc5a1 : SUCCESS in 56m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/08cfe316dc5844dcbf621e2906190fa8 : SUCCESS in 22m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05291ed257064b8182645952a3a7ce4d : SUCCESS in 29m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f99b1cdec0084fde9e8c6c356708f43b : SUCCESS in 3m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9d5daf2dc40443eba6b822b1a3d408bc : SUCCESS in 20m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dfdf8b6b185b45e2a08802c40b5e0a4b : SUCCESS in 53m 37s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":17},{"id":"b43f0a61985a13b35e0c7c3799c3bc462611b8cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 19:33:41.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3945c8a748374858825b931ab017fd27\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a41488824636484e95fa087160088f32 : SUCCESS in 36m 01s","accounts_in_message":[],"_revision_number":17},{"id":"f2cf64a7781880f64b20ca831ba147744920589b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-04 21:14:23.000000000","message":"Patch Set 18: Patch Set 17 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":18},{"id":"4ed1cb61ad9bffaff4d02823dbd17749e34b8e56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-04 21:14:56.000000000","message":"Uploaded patch set 19: New patch set was added with same tree, parent tree, and commit message as Patch Set 18.","accounts_in_message":[],"_revision_number":19},{"id":"a24c17cc9150ab25596df00d9771c2ec468bd73f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 21:52:00.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c775e0f79c54f308b2f00b31f8f630a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f1ead49e3237473d8938a8336872e97d : SUCCESS in 35m 45s","accounts_in_message":[],"_revision_number":19},{"id":"4140b4fd6089e2b1e959425fd4b2972e5d25c5c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 22:26:02.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82a9754935874bb3b9b23d7acf3d8e0f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/873c8607d9764e89a06a4bbb90e3f82d : SUCCESS in 12m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1943d80ce0c487e9c0c6009d5b0035d : SUCCESS in 7m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/f77c5d3585dd4e62a7e5234c2c5731f7 : SUCCESS in 37m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/aea651a1b409446998d9a3e84bc88671 : SUCCESS in 31m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b14234950cb147fdbd19537e7db0e9ec : SUCCESS in 26m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/118f1aef7251491998f92fe9c841bf02 : SUCCESS in 42m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d2cdaaba57c407db3d23c5a1a890610 : SUCCESS in 20m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/883c9d56617247df8e4d9ccb9cb58e53 : SUCCESS in 23m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e6a45af7c5cc4433a50682852cb55d1b : SUCCESS in 17m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dc60933c5631449cacf78ad2608e8a3f : SUCCESS in 6m 55s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/252ba729ee3d4c6ab09982e81eb2cce3 : SUCCESS in 27m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/68181161bb3140b691d8aeca11558c8d : SUCCESS in 11m 22s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/dcc26ccdf0344de79c98c6cd282be795 : SUCCESS in 26m 19s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/f4f962f501e640a6bcfe932c0444deb7 : SUCCESS in 9m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/aba994875d0c4bc49ec97feefeb84a2f : SUCCESS in 15m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/dca96dd48b0f4c748c02760756d27f1d : SUCCESS in 9m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e89d255e6b264b80a0a28ec6cc64c0b3 : SUCCESS in 10m 13s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc60b58e78d44fff91e0404f1301f1db : SUCCESS in 17m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/897490fcfb3742ffa60ba3834a8013f9 : SUCCESS in 5m 45s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/da027d676e8442adab4a7917dbe37316 : FAILURE in 7m 09s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3466eac69d8f4169b3e0482bc0b7f5e1 : SUCCESS in 8m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1ae7a1b7bb3f453a8fa8a88142bcbee8 : SUCCESS in 54m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fd9e27632449420b90f9782de4d2bdc8 : SUCCESS in 59m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ceda4c211d4f453b964d2057d0e3d116 : SUCCESS in 56m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c68c9b2fd1040209a586312ba6257c6 : SUCCESS in 11m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f8aa55561c848f38c648f505c56ffd4 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ccbe7cb33bbd4f76864db2506b38da78 : SUCCESS in 16m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce538d71955d461ca67aba457f9278ab : SUCCESS in 55m 55s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":19},{"id":"4491141ed51bc241070456a0081d959bed997805","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-04 23:42:26.000000000","message":"Patch Set 20: Patch Set 19 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":20},{"id":"3c8c3d7fd60559a6dedf47f2d371a963ed61dd47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 00:16:56.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c7a0ec91e034534abd57ddaa3d3285c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c59d97140e3540caad53bc3cec177aa5 : SUCCESS in 32m 58s","accounts_in_message":[],"_revision_number":20},{"id":"06cea4d472cb2af07a949dc0d35c808a6434d964","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 00:55:49.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f606e05578b54ad5b63dac4a5572ec15\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4f5589f96784a94b3284319ccc61385 : SUCCESS in 12m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9347365d9a94241931fae62d7c8124b : SUCCESS in 8m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/3d0a9964d92a4393abe0de07da21938d : SUCCESS in 1h 02m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/30bc8aae95f9498da8cb665c992acc14 : SUCCESS in 36m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2fd2481ca75a42628d9fd1ff44ec11ca : SUCCESS in 32m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16a64cfdc48e4ec9ae4a8cc1bf4259fe : SUCCESS in 1h 07m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a22e65a1df9c4cc4b1b23e52381e8171 : SUCCESS in 19m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b24fc08d6fed4268b97980e4da615b48 : SUCCESS in 19m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c2f41070ff4949d1a16115ebdf132db1 : SUCCESS in 11m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f127f03969be47198208a4ce8b524821 : SUCCESS in 6m 42s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/551f50c0aae542589a28ed15c7bf97ab : SUCCESS in 10m 32s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8cdaf3564a1a4ca797ff508c4e17faa4 : SUCCESS in 10m 41s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e350706d83dd43d8b8a55402e01b5ffd : SUCCESS in 14m 18s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/4c59e0ec10cd42b382c15ccb025600fc : SUCCESS in 17m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/19c1145d7a5f4a62801d62d618d96917 : SUCCESS in 8m 16s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1b22db32503c46de9cb7fea1deeba45d : SUCCESS in 14m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/329bcc2a21214ae2b118aaef22b5cb44 : SUCCESS in 9m 26s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/9e02905b7417406f8731956363b879c2 : SUCCESS in 16m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8decee2fe9654f3996f147d461c1c342 : SUCCESS in 5m 13s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0c8a8c24e30b45deaae7dc1cc3e44856 : FAILURE in 8m 02s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8a8f71d420ec414bb22df1de4dd15aea : SUCCESS in 8m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/02a26206eeec4a99b0b1f1908e3f9d29 : SUCCESS in 36m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e27afc6265284b198b2d8e25820190d2 : SUCCESS in 52m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/47337fe016ab469eb2e5d40043502277 : SUCCESS in 56m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f31f03b887c7448f954fb625637651cf : SUCCESS in 28m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/050cb519febd4187bd33a925bee6fc6c : SUCCESS in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/776f0fad9fc043968c51dff034999008 : SUCCESS in 15m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ad292e0f0c14f74be5a214b1eddfdbc : SUCCESS in 55m 10s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":20},{"id":"cf1a4202fb2976b93c4f29830a83f411d722c31f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-06 22:27:20.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":21},{"id":"d2e565c239325bee80089db3b0f4343cd55e9168","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-06 23:28:35.000000000","message":"Patch Set 22: Patch Set 21 was rebased","accounts_in_message":[],"_revision_number":22},{"id":"32d33c192f72c8d6816ed52d5a9af3e0bee6c14b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 00:27:42.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92fcc653f3f843a6a20e8968f3123dd9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/371ab29295f74b9eb9114c3accad85a0 : SUCCESS in 56m 42s","accounts_in_message":[],"_revision_number":22},{"id":"af727ddc3848ea3b5af5cdcb4ba6ea03da92bfe0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 00:47:15.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1b16547dbd443a28c845a7801683067\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b2fb731445643e1b7cfa5430c5cfd89 : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e116bb3075f14babb3fe8039fa4f64aa : SUCCESS in 9m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/aa3b1d198b95425f873008e49ad76147 : SUCCESS in 1h 11m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bd432656d18b442395e4356c7bec2b1e : SUCCESS in 48m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e4c3ec3abcd24f3e84658a9d1d1f66e3 : SUCCESS in 52m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2bc70cbf13bc4f4d84a4332d31071523 : FAILURE in 58m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e260800dd5334aee9489ffefa1b05a5b : SUCCESS in 15m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fcbbfc9356a34ccf822638b9db6cbd45 : SUCCESS in 18m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6c82af5c77ab4c3ab44db70f8d3c3dcf : SUCCESS in 16m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a2dd9d18617b48e4bd1a2469c1ae5a4d : SUCCESS in 8m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c9165490d40444358776e32f4dd4f34c : SUCCESS in 27m 33s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3d07acccceb841e5bcecd30d78fa089e : SUCCESS in 17m 12s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/38e729a7f17f45ceace5db6eebf379e4 : SUCCESS in 14m 04s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/4691bfe2fb304c72ab94fc1a55604358 : SUCCESS in 18m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6cc00595607e449ab33d2ae03bd2417e : SUCCESS in 14m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4efabedabc4c4c04a7cfbb43ef56d5df : SUCCESS in 15m 45s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/67551c6598ee4c7a91b4cd5c0bdeeed5 : SUCCESS in 9m 44s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/e62d898e93e5461c9053194d6c441f55 : SUCCESS in 16m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f7f3d4bfb5364b4aba721d27e05e3d38 : SUCCESS in 3m 12s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/31cce6220686410e90c8de9bee2e748d : FAILURE in 8m 22s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b884dd8aabc847b6805d2c104e90e163 : SUCCESS in 10m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/027c881bd77c4765bc74d621006fba0a : SUCCESS in 1h 00m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/90df979ec4de4f3ca88f0173c55aef31 : SUCCESS in 51m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1bd3da6a551d4430a38500ba3305a422 : SUCCESS in 26m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0654ac3bb6a345028d404a3514122126 : SUCCESS in 30m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2030e61b764148f894c8e591dda457dd : SUCCESS in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ec119e7e76a04075838a0fa5f41d4479 : SUCCESS in 15m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f34ca51124f04dce9c8a478ad7305a97 : SUCCESS in 1h 12m 20s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":22},{"id":"db3dbfc2bd44e6ab45546a7c67e5a20553c3c4bd","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-07 11:24:17.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"8e2ee8a65df296d227ed17d90bb753a73ede09c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 12:43:10.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc9323d82091407eac3cedb009feef66\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/370a43feb3284e0cb0f702f0c011064c : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8238d7e6336415bb5c83475f75e74da : SUCCESS in 10m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/9fa0c41cd34c44a3a625ffa01e9e5cb1 : SUCCESS in 1h 04m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0bbed82847e943e88b9f1b883f4066f0 : SUCCESS in 43m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/625795a0399f41e795ddd47f2e5f51f5 : SUCCESS in 49m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47e4476b4898426685a56fe8ee366b78 : FAILURE in 58m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6f2a046ccb094ace9124844bd29b41f4 : SUCCESS in 12m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/174a6c189ee44191bb955d65924088b4 : SUCCESS in 20m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d92be2869bc246fd93bca6d06be638de : SUCCESS in 21m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/241c870decd041008f7c7693f2dd2426 : SUCCESS in 6m 53s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bbfbdc5b519f4bcaac533ff7e7690939 : SUCCESS in 12m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9b0998096118416c9dc2d2c34c06841e : SUCCESS in 18m 05s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9bb5aab7c2fe4f598aa476dc2e94416e : SUCCESS in 18m 00s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/d4a6a58bfbeb428a85d0f35e6186db58 : SUCCESS in 10m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6396d3f0a99e42de9405a92e5ecb5527 : SUCCESS in 14m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/97e479153a924c638a35dff4b2e5fa68 : SUCCESS in 15m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4be96dec66cb482880e0778209ae7ac9 : SUCCESS in 16m 21s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/11fc4bc6ac6c4591a5aa5f404948380a : SUCCESS in 11m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e8a00b0c254e4faa872df291802a42c0 : SUCCESS in 6m 44s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a7ebb2ff6f1a4336841ab68ac1f60bb1 : FAILURE in 8m 30s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e3d6721881074d7c88e84697bb5caa04 : SUCCESS in 8m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/45d7ae8322724bc0b2aabdc97c771bbe : SUCCESS in 39m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8b5481c677944a8387f3c791c4fb4032 : SUCCESS in 29m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8df132f91dd7435ea74634347bea4683 : SUCCESS in 53m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec1d4fd5a78a4163b4ab48c1f906e4c4 : SUCCESS in 20m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68f2b262b12646f189fbb05e1bb93557 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/03aa487985f84aecb867885f4ba3dd2d : SUCCESS in 17m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3b3eb0ab44047fd956ba4ac574ced09 : SUCCESS in 30m 26s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":22},{"id":"d8a46c3039a0e04da3c49adafa0e56386344bdd6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 12:48:58.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ea8c60bc19740ce91061497dc88e1e9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8202ac40f6ae41e9b006677abbc4a2d6 : SUCCESS in 1h 09m 08s","accounts_in_message":[],"_revision_number":22},{"id":"e5f667d222585ec8c00fdb10b51d8a620ab02bc7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-07 15:18:45.000000000","message":"Patch Set 22: Code-Review+1\n\n(8 comments)","accounts_in_message":[],"_revision_number":22},{"id":"85fe9cc009ff38d65bc2d5f9dac566fb6e38bbe0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-07 18:32:32.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":23},{"id":"0d4e9153005b8f062a6ce0a5caa606fa84e460ba","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-07 18:33:30.000000000","message":"Patch Set 23:\n\n(9 comments)","accounts_in_message":[],"_revision_number":23},{"id":"65e2dc7d665ca654fe6cd549d27dc213063eb51c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 19:45:54.000000000","message":"Patch Set 23: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/428b42e57ce649a2bf1b9f127c5ddc72\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/491068a17d8744a387e172fad0a565f2 : SUCCESS in 11m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e4ab6229ce14212a5c1347da53576e6 : SUCCESS in 9m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/4610e94eb05c4eaa826358584a645e93 : SUCCESS in 54m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a6c82f5a1de645b5a50844b9c8e5e704 : SUCCESS in 1h 04m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a7dde6c07bfd4ebeba4fb1a190db4433 : SUCCESS in 52m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/39a63d035bd44eb08b54577c274a5d11 : FAILURE in 1h 02m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3c58227e59de42379aa925bed214c6e6 : SUCCESS in 20m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8648361184374581b8345da7a923eaf8 : SUCCESS in 20m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6c9c74b4fb9f41518cdcef7faca3d95a : SUCCESS in 21m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5951d7ad2d394582bd5006c262e89e58 : SUCCESS in 7m 36s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7c1c85d9869740f0b9331f850ca86c6a : SUCCESS in 10m 14s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/be65fdd7bd17448b80c4f9a98f52bf99 : SUCCESS in 18m 03s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/df4cc2b3eef64c26acf9fecac63348d5 : SUCCESS in 18m 37s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/0cf82b90db1e4c958c2dc71a05a7de30 : SUCCESS in 9m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c0fbd272019b4955be390d930cb7b31a : SUCCESS in 13m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9b32c3e400c442c689dabea98cbf3ca4 : SUCCESS in 8m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a9f2276c028a4b9da8aa90dab7df175a : SUCCESS in 14m 52s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/02f84668e3cd4f50a37fc84b1c0b07c6 : SUCCESS in 14m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1e61bb12317349dd96b7e9b751747df2 : SUCCESS in 6m 15s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/734f5880b30b4b21a197b11d9bf2da53 : FAILURE in 9m 48s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8e70bef76825455c95c73aef78363326 : SUCCESS in 8m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fac6e08bd23a4f279b62359e3616f0b3 : SUCCESS in 1h 06m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cb100f841296473f83978a7faf1815e9 : SUCCESS in 57m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d0f722c1b9984247847ab3ab9323329d : SUCCESS in 58m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/762d7bb3e2e0434783737c729342d248 : SUCCESS in 11m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f896be805e8d47b8892f87fd51ddecae : FAILURE in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/750d861f5a194f129bd3edc85ee647d4 : SUCCESS in 19m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/921cdcd513204b80bf6eada2515125a1 : SUCCESS in 25m 38s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":23},{"id":"cc148a15255fe67a78c3e1c3dd9ee14cd0bb6894","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 19:46:54.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7b67809b3b44a1eaf7a9fae0fa3d7ec\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/311b3d514bb44a6984da513c2f719f7e : SUCCESS in 1h 08m 10s","accounts_in_message":[],"_revision_number":23},{"id":"a510686dde9e41678a31dc5c0cad1d02b6bf4ee9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-07 20:16:29.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":24},{"id":"0a98fdc9668f20992f401a2cac1110b4f8261824","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-07 20:16:48.000000000","message":"Patch Set 25: Patch Set 24 was rebased","accounts_in_message":[],"_revision_number":25},{"id":"0b851c9c1b0e26a807c5d199193cca93211c94f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-07 20:54:24.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"8c5b8f02cf6894bf9f449088577850c956563967","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 22:09:07.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07515810bc354d0799a5c1a463cf25ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0321af564824dcc90c87184b7616aa9 : SUCCESS in 9m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d229254002454434a8d4e6a8d7544ec4 : SUCCESS in 8m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/d6d5d084cfeb4018868d072efa083c52 : SUCCESS in 56m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d2eaad1519d14d4d9a4a115180ab6210 : SUCCESS in 1h 00m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d91713d9e3c247558427a1c24943cf87 : SUCCESS in 37m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03230b6af72a4835b45edc44ebdad5e9 : SUCCESS in 1h 07m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e005f73b41274de79fea17ec79b91457 : SUCCESS in 19m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/20e2233079b24570b3a35c71be267f59 : SUCCESS in 21m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6caf924c5008482fada616e67fab9760 : SUCCESS in 21m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/36e98013c4984d2ca08f77ec84364ab7 : SUCCESS in 6m 41s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7e2b03196508402b85107760f92c7b76 : SUCCESS in 16m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/dfe5faab05744f63bc44c97059974c41 : SUCCESS in 22m 35s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5a0b7d0d41954694bcb8b3a3d50f652f : SUCCESS in 32m 12s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/6976a91839fa4167a8d58f29c7c887d2 : SUCCESS in 21m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/46a7455e54a4454d9af4739107a49166 : SUCCESS in 16m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a061c6439f1349639cfaebd965cf1d76 : SUCCESS in 11m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7f8bf2d32ed946cb8b1cfa92dbe5a8ee : SUCCESS in 17m 12s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa3cb7ddfe754a5e9d2ddf22c42f8328 : SUCCESS in 17m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/56f1cb84a8c041afa1b93d00ca7501ef : SUCCESS in 6m 23s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/76358c5308cf46479d104ea79d6f4248 : FAILURE in 4m 28s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1a2fc96372e74ed2a2bf93cdb9257a0f : SUCCESS in 8m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/72abb2c8589045b2b52ff0551a8b7893 : SUCCESS in 59m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ebf982d03598451c9d359fb08e6b653d : SUCCESS in 56m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f85fadf3bcf840ccba370121c2e6cf0a : SUCCESS in 52m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/720628debcbd471992e4b54efa60658d : SUCCESS in 27m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31c23677f4bd42548a213ec77e13ef99 : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6839b54781f74691b9ca3223f3ad1eb1 : SUCCESS in 20m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2195139471c243a6b6eea0fcea84abb0 : SUCCESS in 53m 49s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":26},{"id":"a796e6aebf5d80a41b7add2d79226058c5d57304","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 22:09:37.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21fc25abe7004d789d2b42f9cbc81dfa\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b7896a70c96a4ac78ee8b77e72407460 : SUCCESS in 1h 12m 40s","accounts_in_message":[],"_revision_number":26},{"id":"852df0b18983754e64f3288e031138c0b731188d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-09 23:52:46.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":27},{"id":"78226678f12b9c5ed280f3379d98eac199acc509","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 01:03:56.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8993bbba70ba4673baf30693ee391c5b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2518b4d056144fdae27a9e940269fde : SUCCESS in 9m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76d45d3bd6744d8eadb232544d55e1d1 : SUCCESS in 3m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/5e7f4af1ef164bd7ab4e49ec2f302bae : SUCCESS in 51m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8cd637e11dbe4c66ba22c6af21367c62 : SUCCESS in 42m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bdd1796fcbb948fcbdd1766f1df586b7 : SUCCESS in 52m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/228ca636cd3248c389d02f62ea5e0fa0 : SUCCESS in 45m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/350764e4bcfd4904bdaf850e356214f9 : SUCCESS in 11m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8d356c95dfeb43dcac3a5fa704337ef4 : SUCCESS in 11m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4e7dcf67939749f38a2682400ff0f214 : SUCCESS in 21m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0b2b90eb6117409eafc033b603ea88f1 : SUCCESS in 6m 40s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b96e4437d1a247668858d6124dcec6b3 : SUCCESS in 25m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/296bbbd506e0443ca1c82823f71a6f4b : SUCCESS in 21m 54s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f61266d9ca10403195ad2da040014fa7 : SUCCESS in 11m 57s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/2584e05143d44c26b49fc707a692b588 : SUCCESS in 18m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9fe5fb630eb9488f98645a0834233f5c : SUCCESS in 12m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3bbfd956ca2a49c1ab069dc51f9937b9 : SUCCESS in 12m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9271ae09f302453fb8a70c4dc241fecc : SUCCESS in 13m 53s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/ccfa4cd7045448dcb2b0afc595802c3f : SUCCESS in 18m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/52ecee34f826422b94b1b502a76f2237 : SUCCESS in 9m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3f7bded302c8495a8e652b86bfc87177 : FAILURE in 7m 11s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b113f41f956f4c7c8b3b6577cb473b5e : SUCCESS in 8m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca16d24ccbda48fa96733607c6a1c97d : SUCCESS in 1h 01m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6b388c87de774dd08db5d57d2e000c72 : SUCCESS in 57m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9bf458d2a9ee4dd1bb2ed70636942d7c : SUCCESS in 51m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e50166f779347e78fc6fbc7bebfeab1 : SUCCESS in 14m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64824d3d005c4e2b81d4200bde3bc901 : SUCCESS in 7m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/768308892e11402f9d5dea1d09177ac9 : SUCCESS in 19m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/423c663ac77c41d4bbe7222e6cac32cc : SUCCESS in 1h 04m 23s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":27},{"id":"6140c012957a09db651ca76cee1dd905f4d17248","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 01:12:34.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f90e4cf6af2426ebe2d4ca03d66ff08\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/418075de9c874c8090b07339e8b2df6f : SUCCESS in 1h 17m 38s","accounts_in_message":[],"_revision_number":27},{"id":"3468d78f6dd4fccd1693bf0f5a57b92924eba20a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-11 00:51:03.000000000","message":"Uploaded patch set 28: Patch Set 27 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":28},{"id":"946d34eb8430490da2c6f908e045229aa7335095","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 01:58:16.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2824c70d86df49e3b3226c34a96fc573\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/608181f900de46619c9adf35468e3dd6 : SUCCESS in 9m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7aae6076e644b9b816bd3b45b7e7070 : SUCCESS in 8m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/bbae548d4bd5490685024479d16adc68 : SUCCESS in 1h 01m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0a3b7124e44d46649c3189bc3feff273 : SUCCESS in 45m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/501b3de52d2b4ba38777b13f5df61b5b : SUCCESS in 48m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/991199fd898b4f6fb345b2046d1491e1 : SUCCESS in 27m 19s (non-voting)\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/64de8972fa87459bba1ff0b2cff50389 : SUCCESS in 15m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e2f32afd13c242eba9ecfeecf3b5c0a0 : SUCCESS in 19m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/92036615c57a4176bbe85a8b23c5f422 : SUCCESS in 11m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/208383d878a44d4f98cbd06ba1c35505 : SUCCESS in 6m 43s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/455b2e81f93040d4a7ebd52493dcb9e9 : SUCCESS in 16m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/67250cc818634f6d9791f44982f60a5f : SUCCESS in 21m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/886b2c7a82e34402bee9e037102832bd : SUCCESS in 33m 29s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/748a2313c74642c89dfedb01e1dd5081 : SUCCESS in 10m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a71ba87095e94235bcfca2de46351594 : SUCCESS in 8m 16s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/adc3895ba488450ca89d9cc80c2d7a27 : SUCCESS in 14m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/340e2168c94a457a8d0562a38790c0c0 : SUCCESS in 15m 18s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/34a04cf096af477f981c86788d284a1f : SUCCESS in 13m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2198d554d9264dd9bf2f2ac68a8620e3 : SUCCESS in 6m 11s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/27d0a1cb293640cab5a7edfbf3bf710c : FAILURE in 3m 49s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/bf55f7de9bad4d399b8e8478b74e05b0 : SUCCESS in 7m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/52692c603d2c41aebbd74bf3bfef7317 : SUCCESS in 1h 00m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aa52f039741549638b36fbf3a10a4211 : SUCCESS in 56m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6665a5ec461f4e368747c5ae6fd3b519 : SUCCESS in 56m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc8bdd01aa6b4ab5b4adbb78f7209416 : SUCCESS in 21m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fab3e1ebf9246e4b41e3c310c3f1083 : SUCCESS in 3m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4f30de4b7e2d4ec4b9d58ea49c6faa52 : SUCCESS in 11m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f55186153da4211bd1f643fe1bbf08b : SUCCESS in 53m 52s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":28},{"id":"175ae2611693cb7181515c5379325f0783f1167f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 02:10:20.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e506bc8fc8ff4ebc94ab6a3e1ed411a8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3340af3731e84b9f9097ad986607d131 : SUCCESS in 1h 16m 59s","accounts_in_message":[],"_revision_number":28},{"id":"947490223ff75b56d30447b24dc9ec8cb6920d46","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-08-11 12:56:21.000000000","message":"Patch Set 28: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":28},{"id":"94b90373e258747753ae841a7a4052c71e0031d1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-11 18:17:38.000000000","message":"Topic set to s3-checksum-persistence","accounts_in_message":[],"_revision_number":28},{"id":"0d86c070b84c2d4adfacaca4a40710b445220d79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-11 19:39:36.000000000","message":"Uploaded patch set 29: Patch Set 28 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":29},{"id":"e42d35171b7b8b82a61099a1ee3b6cf61b77bf0b","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-11 19:40:22.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"4e28a2a0ada38945d6ad54717610e5525a802a66","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 20:26:12.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d10e0eb829be428b94546970d720f861\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/70b4f75f12b94fbcb39d0ad8d40c5af4 : SUCCESS in 37m 31s","accounts_in_message":[],"_revision_number":29},{"id":"5a635c2c033e1c4e7f0ed0ba95797f3a03bfd936","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 20:51:54.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa1b4f23b59041ddb584391e4b439229\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2be1200c8bf747f48d4129885b4a30c4 : SUCCESS in 10m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39a216fc3f9a4c8cb347ad3272237d68 : SUCCESS in 5m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/525a255a670d4265b31385448060baab : SUCCESS in 1h 06m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f6abd5115ea4413f9ed54b76ed73cf0a : SUCCESS in 36m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/59a510a5fba0409e83a0572842751cde : SUCCESS in 53m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c7f39b8c481240adbb9a3ff12f480baf : SUCCESS in 59m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5cc6167f2c744dbcac3c463bf21aa7c0 : SUCCESS in 12m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0fd61fde6ea24ccba01701cef9bd708a : SUCCESS in 21m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/874c1a42e49040c6847b3cca48bf8aff : SUCCESS in 25m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9ad693ebc41845b792ae37e2774f79b4 : SUCCESS in 6m 38s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/feb8fec5b5fb4d1fb7dd08a25e7b994a : SUCCESS in 20m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/82f12f6c1add4ae58a485d1e4490980e : SUCCESS in 17m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1e37a306cd4b45a5a9c4a47f50f7cfa8 : SUCCESS in 23m 25s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/64253280cc0743e2b80216a745e34926 : SUCCESS in 20m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/093e8d190e70484cb6166a32fbd01836 : SUCCESS in 14m 44s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/08764ef36bb64316b60008ce721b7c64 : SUCCESS in 14m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9b2f131836684564ba0496eeca24ab2e : SUCCESS in 15m 09s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7be89cee8c54349b70ac92a1fb846e1 : SUCCESS in 16m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ce1dd50404ba4b7a8c8c4f8b1bebd553 : SUCCESS in 8m 07s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3d9f65988b8f4168861f730d0c3974ee : FAILURE in 7m 27s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/87fa135a0f3943bf92f1645ddb1f50d6 : SUCCESS in 8m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf3253953a5946e387a2a7b2b152a9d6 : SUCCESS in 59m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2d9b5eb9c62447ac9296ec004700ec52 : SUCCESS in 36m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f191e3e47f2243b08c0591746f0de9b9 : SUCCESS in 1h 00m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f4a271b888043b5b97f8ed3116c94ae : SUCCESS in 15m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2757c83998da4e59a56ec7ef4a21762c : SUCCESS in 5m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/80b49649a537433eb9d1f88f640af532 : SUCCESS in 18m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4ae3663ebf64f84a1f7961fa0bf2100 : SUCCESS in 1h 01m 17s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":29},{"id":"44c23fb18940279f41cb60691d9a2b21a2d0be17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-14 23:08:13.000000000","message":"Uploaded patch set 30: Patch Set 29 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":30},{"id":"6f816861c75f031e7bd575abf0b8fafce2683c40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-15 00:21:59.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd30a726680b46d9b51233d00e98bab1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a541d9d65c914ddd808dbbd33be27139 : SUCCESS in 9m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64d4997092104bec859f894dc40995ac : SUCCESS in 9m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/4f092becd5c4415985ff55a4d02dd770 : SUCCESS in 1h 07m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9f90fdadc6fd43e08c8f80ed93a8d814 : SUCCESS in 58m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3b9fc5b8475d42adb69adff9aa811475 : SUCCESS in 51m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db52750705824242a53c9378b318f010 : SUCCESS in 57m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9629eb7bf55b4f4383ef0ef4870019de : SUCCESS in 10m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/01f26170e32c44d896ca175770c52eb3 : SUCCESS in 22m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7bc578ef25e24d0db533871ac0f64461 : SUCCESS in 20m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0e923619ecb4491e82b67ab578e2f140 : SUCCESS in 6m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/557a96b2fa714b92a53c0b33d98cef96 : SUCCESS in 18m 55s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/741bdb4429b846148559ea08977ae828 : SUCCESS in 21m 25s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5332cea0a87a4737b08ad01c2994a384 : SUCCESS in 30m 13s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/1b551d9d7d8944df9cbeb36795c9d69a : SUCCESS in 21m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2a80ea0605944f678250fdbe5d2ecf6a : SUCCESS in 14m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/91623466fccb44548b182cc64d2dbda6 : SUCCESS in 9m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cdf7a383305844f68f5329c4d002f677 : SUCCESS in 14m 49s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/f368d86a751e4d61a1f6500339af6742 : SUCCESS in 14m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e8dd469badfa40e4bc4b50556a62e25d : SUCCESS in 8m 04s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/13bedce555a34fe6969d246733260e74 : FAILURE in 7m 42s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/bfa2d92de1114f0e91f7a72dbadedf05 : SUCCESS in 5m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/dd759646e0f24c24b65e484dce543c6d : SUCCESS in 50m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f801787c9b1b40e4abf11a07d1288952 : SUCCESS in 32m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/90a1b2b1623a4f5e9cc1e786c107dd99 : SUCCESS in 33m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b927c85e9bab445c9b8de74cc91f4bb3 : SUCCESS in 13m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f09d72a403740708f25d3f6dc0808d9 : SUCCESS in 2m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1f918c2e49314087b948b638674d5103 : SUCCESS in 19m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b881d166b80c41f98f68870f2ff573a6 : SUCCESS in 40m 39s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":30},{"id":"3d7456417f5467e3031c25603d4ff9eb54d5cac6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-15 00:24:29.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb699f553408488285c18a586a75cb27\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fb771ca8afb54612b6aa3f2f2e86f962 : SUCCESS in 1h 13m 43s","accounts_in_message":[],"_revision_number":30},{"id":"e5dec088b476b161495a2e3021e149e01007f282","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-17 15:09:49.000000000","message":"Uploaded patch set 31: Patch Set 30 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":31},{"id":"16451a8b68ed5192d9cc2a90d48fe9df7c8c480b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 16:23:03.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c66d4e241c06408b85342e6e9907554d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3533c8c1e0ff43b2af853ce51d6daf90 : SUCCESS in 12m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5cefa9da42849888093b5ff7254f217 : SUCCESS in 7m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/3bda5923ce88471d93ace48a1d73b483 : SUCCESS in 1h 02m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/073d12d2b9454d3ea960a2c1ca155f07 : SUCCESS in 1h 06m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/829670c869694e35ad10c4241d5b7039 : SUCCESS in 41m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2c419ac7c834b01b05eec8fbb94dc41 : SUCCESS in 29m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/38bd5b6686a245ce85f72193aed562e6 : SUCCESS in 19m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/46082011fbd442c3b3ebbb0a09a2cf70 : SUCCESS in 21m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e4201233a2b14b4fa0479ce1a009e99a : SUCCESS in 16m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d5f1ac8846a441d1aa710884a86f48b1 : SUCCESS in 7m 19s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/375225a344454c84a9cc74603a43ad4d : SUCCESS in 20m 33s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ed4d537328d94db4a31127a70c0d730a : SUCCESS in 16m 43s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f367c689035541c89117e5f696d6b6b5 : SUCCESS in 11m 59s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/c0ba886149064a5fa0974460f02b344e : SUCCESS in 21m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/69f49a2c4b714c3e908f0b59b1d33ba0 : SUCCESS in 15m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c85aee86ac18431b9f62f9ae5157b794 : SUCCESS in 15m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d8516dd0d84945c98df3d3c19956edbe : SUCCESS in 16m 36s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/2461d0cdb7ed45589da034fe90bca391 : SUCCESS in 16m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8be3cfeead9b4c039754976e1c8dc569 : SUCCESS in 6m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/59b5793f93134c94a196d64a7fc8acc0 : FAILURE in 7m 36s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/41e073c238554e9c866389edc7f2fd84 : SUCCESS in 8m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/806ef0d37f16420898ed93fee8c083fd : SUCCESS in 1h 04m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7cf202911fa742a7bc3f6bf9a3928d9f : SUCCESS in 31m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9d4168cc5abe44e18fc9e2d9a3347f76 : SUCCESS in 49m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/209f89e7326f46eaaf20d86d985b64de : SUCCESS in 29m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/428aedbd800c49debb9a55d107655997 : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ef32d667344c47adbd49ea279c2d4f20 : SUCCESS in 18m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/efaeec2ebfda4aff91996bbff5fa8545 : SUCCESS in 55m 16s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":31},{"id":"a76ffa970c2a953304c46469b40bc9f52c19da1f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 16:32:43.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/432df497a5c04288b9f7cd80e51f291d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/18daed0eff364b17a3e73ad1cc6737fd : SUCCESS in 57m 09s","accounts_in_message":[],"_revision_number":31},{"id":"c2228dc284957e884d9906b134b2dd5dd33c6dff","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-08-18 14:25:11.000000000","message":"Patch Set 31: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"122d0b7fbaa67d7650434c7fd852d4e0766dbb27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-18 18:51:39.000000000","message":"Uploaded patch set 32: Patch Set 31 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":32},{"id":"5e3c2e80f0a51d53cf17bc018f4630fd3f945602","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 19:03:30.000000000","message":"Patch Set 32:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6412e12b1da64f639b538f8bfcd3be41\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9f761b83471f4e31886668be0c79d4a7 : NODE_FAILURE Node(set) request 3967cd85323c416e8277b3cd00eb303b failed in 0s","accounts_in_message":[],"_revision_number":32},{"id":"193d4b0495fbe99c3c4c411be58c53a55b26a878","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 20:08:25.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a77218010de40529c31ba4a2076b5d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29a11b4fc9304f10959642e66b69a136 : SUCCESS in 9m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35770e2e6d834d7b97d250fb2b97d3de : SUCCESS in 8m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/c4ec428103494d75a36a9f2ffbc112b5 : SUCCESS in 1h 09m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/95c61d57d57b4a9eb3cab3f34de256c0 : SUCCESS in 1h 05m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/699ce65360ad4a78bba7a3acbd41e1c9 : SUCCESS in 37m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b31905aaadcf4d4880a1ef0dc73a70cb : SUCCESS in 1h 02m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/60a3782f991e4efdaddf5921833af8c6 : SUCCESS in 17m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d7398b54b6904ee390234ca5d4514428 : SUCCESS in 20m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/504ab1244de94560b387f2f198dff910 : SUCCESS in 21m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dba59167e641477cb4c511be4435bb29 : SUCCESS in 7m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/674ff01168f84a308da63e50a7d69fce : SUCCESS in 25m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/461bd45f6d1c4212ba9e6a3f17feb01d : SUCCESS in 11m 29s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/36ee90be1c60458a8e0c91fd2e4da6ae : SUCCESS in 13m 27s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/cd070eee20b0413e85dfc91012caaaaf : SUCCESS in 19m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b1d577af2bbc45c1a21784bda871b6e5 : SUCCESS in 9m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f0b6e91ed1ff4e329fe4b80cf1ecea5d : SUCCESS in 9m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e9fe2a458b244bd1a9aa376bb1f537a4 : SUCCESS in 13m 42s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/d64c823ff9664bfe935ef690b2c10b14 : SUCCESS in 9m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a75646567d90450594e359b257cb4969 : SUCCESS in 7m 05s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fea3fe0a69d84e1ab25da66774998ca2 : FAILURE in 5m 04s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/651daa79993d40ef8b40b44658e60f4d : SUCCESS in 10m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d2ecbca7da614f25912dbf25e4be4f58 : SUCCESS in 37m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3132ca40ef6b48ab934756dbf43424e8 : SUCCESS in 23m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/08598b5a66f34ba9bd0bb7770a742ea2 : SUCCESS in 56m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e3e62c42cb34391bdcb7c77ace30ff8 : SUCCESS in 26m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63e009c47a034e6b9022848b80b630ac : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/321d066189d44518b607d450a6dd4b7a : SUCCESS in 17m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf26eeedb5164c5ca4f73fa7cc6dc93e : SUCCESS in 25m 45s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":32},{"id":"b7b2f6e5745a22fa05caac311be8df08adaa1e96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-18 22:32:30.000000000","message":"Uploaded patch set 33: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":33},{"id":"dd677c599c6f27a8525483c685e77afbbe4c5d6e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 23:39:55.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8008f4f398884d738ff4512e1a6c6569\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/062348939590443da8abe8f83b6f1ae1 : SUCCESS in 1h 00m 00s","accounts_in_message":[],"_revision_number":33},{"id":"e568d558bef8de893925177bed3f74a9e6779975","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 23:44:23.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38f9b74f69ef405c985de9e0d912866e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dfed75cd7b049f1914e544485aaabd6 : SUCCESS in 10m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1cf0d966c8934859bd14b234b3976c86 : SUCCESS in 9m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/f4d976c460c247f69063e580b90d79cf : SUCCESS in 1h 05m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3cacf21f8d54482a885e71415a316410 : SUCCESS in 1h 02m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/33c59ab60b2f492292135770aec22430 : SUCCESS in 55m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69e871a1befb42b5a188f1a415a655f3 : SUCCESS in 1h 01m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1b8cc193d9fe4f939fb622b70c35d461 : SUCCESS in 18m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/25a11617ef8f4f099a6d6c38de1ee2cf : SUCCESS in 20m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/911fdaf19e914c3c8bfb1a802903840f : SUCCESS in 20m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/56b10c9e7b8e40408a891d1aef79956d : SUCCESS in 7m 23s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4189b8f8a2084b52aa7da6c6c89ece1f : SUCCESS in 26m 37s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2dc0fecf8aa24d8d997016e760dd909e : SUCCESS in 18m 00s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/bb29c97423c24075ba06f677c64c15b5 : SUCCESS in 27m 10s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/ef249be648884610afad47f8e5277866 : SUCCESS in 20m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6744ebf961fc44498558e06349aa8e2c : SUCCESS in 15m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c8db5dfac1b64ba7b2e7dd59d7f2f943 : SUCCESS in 14m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ee649cd3c2e14488b36c114dcca9b922 : SUCCESS in 16m 30s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b14d054997347b7b842ff6da869ddbc : SUCCESS in 19m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0cc905fa15b949228359d52858a14a66 : SUCCESS in 6m 33s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e8267f7b21a74738a46a31d972121255 : FAILURE in 10m 06s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/607ce2250b3a4c1892636872c5921b32 : SUCCESS in 11m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e465d0cfc04441b28c2d6554c3c98156 : SUCCESS in 1h 03m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c0417fe192844ab1bafb5a96fcb3a4e9 : SUCCESS in 52m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/38573f0975b2416c8124166c4fdbcc95 : SUCCESS in 53m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/682a4274547046429780a19b6cef110f : SUCCESS in 28m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c478493f090142d6b20f363194318a89 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/05007904ce494582ab30406be1718542 : SUCCESS in 19m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59b7617a7b52488c874dfcb9e0b02b3c : SUCCESS in 1h 01m 06s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":33},{"id":"6d42905c3cc76ce541e9675513e61cd93cef2e02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-19 22:49:20.000000000","message":"Uploaded patch set 34: Patch Set 33 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":34},{"id":"3055991eaea288b14ce62b6c4ac5943121ac61e7","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-19 23:44:40.000000000","message":"Patch Set 34:\n\n(7 comments)","accounts_in_message":[],"_revision_number":34},{"id":"280446c91c15828eb23ab45061b7d2b28de60d5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 23:51:41.000000000","message":"Patch Set 34:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/952abedf556f4b628b5686ab7db11894\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/653356af4d754f1885967bfedb55be56 : SUCCESS in 1h 00m 13s","accounts_in_message":[],"_revision_number":34},{"id":"3c1929d4108e15d15f5e6fad5eaa5f9b24efb104","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 23:58:38.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aec754d5d7d9413490f992d683a76c11\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/550f879443784308a934eaa9746e1fba : SUCCESS in 10m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b554281c4224ca2bdca52bb96f47ac2 : SUCCESS in 8m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/8e2b6786daec49e0881a2cbf7357a384 : SUCCESS in 36m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cad7063853a34322a8c8a4a92fd1824d : SUCCESS in 1h 02m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/469e17750dfd4b15981f955cad63eb08 : SUCCESS in 54m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01e43daa533c4074a0e942d2fa957d67 : SUCCESS in 1h 02m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/253f3319433a4b8c9b37279ad77b111a : SUCCESS in 19m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/23abf47fd8704c6db308691ade550a90 : SUCCESS in 20m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/932b3afe66ee4fc1836ab4d04264c9e9 : SUCCESS in 10m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4653768f094546ac8ed28b9a344eebe2 : SUCCESS in 6m 24s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/634b06068ea24be0be9d2feb8941165d : SUCCESS in 23m 35s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c2181a53fef145bc97c67eaa2d6cfd6e : SUCCESS in 9m 33s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/24a7967e7bb94dacb2bc51fcba142cda : SUCCESS in 12m 09s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/db2e3ed5d8fe42e99079b6edb5ae6757 : SUCCESS in 14m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7911e8ae0a76421e8a56f2dce04cc15c : SUCCESS in 9m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/793b614b83cf494ebd81e3d520b63844 : SUCCESS in 12m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/01299831063d400e9e4525c26d40bab3 : SUCCESS in 16m 26s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/78bbbdea83034267a4151c752d5c9996 : SUCCESS in 17m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/839e04ee497b4635956905b7291ff05d : SUCCESS in 6m 30s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9ecd95c2df0f41758b134bcf1014019b : FAILURE in 9m 05s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7b2fa732a2dc46a79f21ef29bc57f0ca : SUCCESS in 5m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b1e5f037b2c49d8bd7cc8b76cc30c03 : SUCCESS in 51m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a0a3317a292c42e6b3be83cfcc84066d : SUCCESS in 56m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/49d6ba7a03264836a5037c46e56d97e7 : SUCCESS in 52m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23c1738e259744219ae63f31390c640d : SUCCESS in 30m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb5b64edd1c74ab89259797dae8f7763 : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d7b71b9b0a8a4d08b4ff6eceb8f349b4 : SUCCESS in 11m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97251eeee29846ca993ec3bdaebed9b0 : SUCCESS in 58m 05s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":34},{"id":"a3607818eb15d265bbde56b2a89d4ae5715aca95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-20 15:48:41.000000000","message":"Uploaded patch set 35: Patch Set 34 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":35},{"id":"ae17f0879cd9826fe6946106db5d6a22c56c1315","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 16:43:14.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93273db369df487899119eea484a146d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4635da19476d4ddfa98c33bfca5a265e : SUCCESS in 29m 47s","accounts_in_message":[],"_revision_number":35},{"id":"cf5271e3429da0c9e4740f9ab2b4322f0af3ec89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 17:13:29.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c2bb7f95d2b45a1a6aceed74f35f80f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6995700c8cf4753b1ae64d4decfb22e : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b87a54c35e34deea5d56f8529f9ba45 : SUCCESS in 5m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/75d896dea704498983dc4d7b991bf900 : SUCCESS in 1h 06m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3dcc56304ec2472ca80eae48038836d6 : SUCCESS in 1h 04m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6835f7dc2a9640ab8462399f4f05434c : SUCCESS in 1h 06m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c66f3f71b064154910157566fc0e55a : SUCCESS in 1h 01m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1304617612f54e23b2957f8cb3f94e5b : SUCCESS in 11m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5f26d0b1b43647c79a61f61176ef4fc4 : SUCCESS in 12m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/66d08ac9223c4b38869dfd892f44c037 : SUCCESS in 18m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/404d57ea70d741d095aed98532795920 : SUCCESS in 7m 20s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/039c70d4d182400e8f8ef8bb6946fdb8 : SUCCESS in 28m 53s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/19003392175441a180d2130ebeebffcd : SUCCESS in 22m 29s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/bdd9896f0b1d4589935470514472ea90 : SUCCESS in 33m 29s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/4aca2fa6fae94a31aebf753d4285586d : SUCCESS in 10m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/21f91c22f95a4a158b9cae4c9d0c794a : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ee90bf13fecb432682aa70f80ea959f2 : SUCCESS in 15m 15s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7b3fa2c737c94d5fb29d0fab1d10de42 : SUCCESS in 16m 35s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2c87286947e4bc4807ac3014e8a0549 : SUCCESS in 17m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f5ddefc71b54450f815aa3d8b3e98158 : SUCCESS in 3m 51s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a9b5ac23572144ce8935634ed7c2a8b3 : FAILURE in 4m 14s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b916d4a4920444e2ac67d9842e11b191 : SUCCESS in 10m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7504cecdb2b42c0bc7e1727e19601fa : SUCCESS in 1h 03m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4ed7934d9b394cb2a56d27d04004f4c1 : SUCCESS in 26m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3ddc7a0be73b43c5bf05f5e6a56ece9a : SUCCESS in 58m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0202874bdcd145f2acefc7ef8ecb235f : SUCCESS in 29m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04908b645e484815bde78ba5bed7c513 : SUCCESS in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bfff0aac5b524ff39a8688551703111d : SUCCESS in 20m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f1d7f11e64c45d59987669b5e1d9eb0 : SUCCESS in 29m 05s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":35},{"id":"632311906963384c69c22e736307b38aed23f86d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-21 21:22:06.000000000","message":"Uploaded patch set 36: Patch Set 35 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":36},{"id":"2f73ab623c5efa7b021c717b4cb405c2a41e1e00","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 22:33:28.000000000","message":"Patch Set 36:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/086781f350b34c7eb294b13eaf7ef9c3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/442a0063214341649fae4f846b1edba9 : SUCCESS in 1h 06m 27s","accounts_in_message":[],"_revision_number":36},{"id":"46e46846abacf5e3ce27e29d3cf489d0dbbe1a33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 22:34:15.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc813cc152f94f6185e2243d5c0b1345\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d00a135ba6f4e78abf146443503516b : SUCCESS in 10m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ae7ec45eca54afeb6525b488262528d : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/606f2993f66c4b9e866a583bddc34838 : SUCCESS in 1h 05m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/abc03b7286d74094a6dd3932e1388a7f : SUCCESS in 1h 01m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4b6878c0960a4e738528dce35181183b : SUCCESS in 44m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/371e16d882fe4c358a0513c3d1a591e5 : SUCCESS in 57m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2e0d411f76e4463b8505b56771acd7c0 : SUCCESS in 20m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f4ac83c231044027800b930e0956f5e2 : SUCCESS in 11m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/102cbac27a6e4cfbab2dfebb52d5583c : SUCCESS in 11m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9ab8e8c4b9774b79bdcb54122b13979b : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f435c19d7dc04f47bc96c1e7f3e81d81 : SUCCESS in 26m 13s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b1686180c6fb49d0b8dbf89dc385ab6e : SUCCESS in 21m 10s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2b25cd0bdfa044bca2d9a6258555d458 : SUCCESS in 31m 14s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/a1a80bbe3aba437e9d8709fecabf25a4 : SUCCESS in 22m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/081457d93efb406699c823ee51d626f4 : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e50f3db74f784a7a91c711656c1edcce : SUCCESS in 13m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cbd035bb21b543db9b50ac79ad09218f : SUCCESS in 15m 16s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b62da6dc4a440a1addd51ac4ed2df9f : SUCCESS in 17m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/915db3f2560a4d04968fa40ba109c0a8 : SUCCESS in 8m 12s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/724c9561e1a44392ab8497bfd2446c05 : FAILURE in 7m 43s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d2a5e0078aa3436393143fac12247a34 : SUCCESS in 7m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e50d46cbb9f457e802e3fe2d33ebaba : SUCCESS in 38m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0a5d8737bd734ad492ed2380c0da8111 : SUCCESS in 49m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0e096d839b8d4976afac71fe25c93514 : SUCCESS in 56m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99b9b80ad6af4e4f8358d1c4e6113be4 : SUCCESS in 28m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc7aab89cfa94e129b80f923fed7b727 : SUCCESS in 4m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c9562e9a8b894c29a12599ac782bc475 : SUCCESS in 18m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30029d704708459c890aac8d7948c3e6 : SUCCESS in 42m 00s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":36},{"id":"7423e3acaaf8946e0417e9e9bd7ae136031b374f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-24 15:47:51.000000000","message":"Uploaded patch set 37: Patch Set 36 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":37},{"id":"74d95c4170125a29abb5deaebf2182d31b4cf250","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 17:20:07.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e2f52dbc40240c6a31500a2079dc7e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0112c383de2c4e37a3e30735825b9530 : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bfd1fe13df7b4584aab18c1875ac31fb : SUCCESS in 10m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/bf9ca4a06fcb46d59472bd301c5732ad : SUCCESS in 32m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/92c8c79398484b1d80685c17d5c8a91d : SUCCESS in 1h 04m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/527eaf2fda5846bf8d4f1b8a0add3256 : SUCCESS in 41m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0242a963a51c4aa59db00680a668cb9b : SUCCESS in 49m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cfee4673bbf6498bb5a1c93e2711c7d2 : SUCCESS in 20m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/24884ddaf88144bfa29230a560d40ef7 : SUCCESS in 22m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c0cb2382547c4615a0362b2d6ccc28b0 : SUCCESS in 18m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8a6e8708add449cf80c7d85d357e799a : SUCCESS in 8m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5a2783fd5b61478a89552d5349c1e22a : SUCCESS in 21m 17s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/eb41a65b9d3e4a13b8a13b926df40686 : SUCCESS in 12m 26s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e97ad04246aa4e1094283ee318790e3a : SUCCESS in 33m 39s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/389a7bdba0c74e79baa160a838283302 : SUCCESS in 22m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ce426d323dc2487c86ccd0b04948dd9b : SUCCESS in 12m 16s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bcf1718a0d7d4f479e62f6e78f10332b : SUCCESS in 17m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6eeca7a406234e879044069a23a0bb87 : SUCCESS in 10m 19s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/d85b183df06a4da7be5d4211e65fc9de : SUCCESS in 9m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/84dccc1206264d3397d0e2ef58532c6c : SUCCESS in 7m 11s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/134351a432024503a865fcba7c241fee : FAILURE in 5m 28s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/619f28810e364b8aa34e7bfccd2de00e : SUCCESS in 6m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9bc45790a30f457aa29372a993a6520e : SUCCESS in 57m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a5a5ed99402f4f5a9cb8cf6602636e40 : SUCCESS in 53m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f36fd391fc1040e0a51660fe1c5becfe : SUCCESS in 38m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed1dfc535df143f1bfb2a4870d982324 : SUCCESS in 30m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3df47382a0864e80ab1e1bdc2f6b95bf : SUCCESS in 6m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2a1e9e8a892947c098c9ede03f96d33c : SUCCESS in 21m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6bf375e505714fe2b0e0677969f9eb0d : SUCCESS in 31m 34s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":37},{"id":"9c2b9a9edc6973c0351bfb71e2efc5bd8b09f5a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-24 22:45:24.000000000","message":"Uploaded patch set 38: Patch Set 37 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":38},{"id":"8f1c840c548185714141af13e9cd9bca8666a227","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-24 23:45:12.000000000","message":"Uploaded patch set 39: Patch Set 38 was rebased.","accounts_in_message":[],"_revision_number":39},{"id":"2cddf92e83701e7059c1e719ae82b03ba4828fd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 01:00:08.000000000","message":"Patch Set 39: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8db869018ba4f598d7efdd11697be44\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4efe1d09097d44fe86dd6f82f5b080be : SUCCESS in 9m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75871b3878d146daa8a662da13ae412c : SUCCESS in 6m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/54e09023690a4630972a16332d557159 : SUCCESS in 1h 08m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/33af3085b7eb4f36a88106868e62e812 : SUCCESS in 30m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7dc7396710e44ded87d1ad57a0480793 : SUCCESS in 27m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/29d8f01bfe2a47acb6c79e6cb0c48565 : SUCCESS in 1h 04m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4a7e61df46714dfbb860999ab44ebf12 : SUCCESS in 20m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4f32634a8b56476c90cbacafaf23ef6f : SUCCESS in 12m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b5087deed5784fab97bdedc592255683 : SUCCESS in 17m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/66268e26fed343bead0eaa111a82d072 : SUCCESS in 4m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ec5ad8c79d4041cfb1147f3ad847da53 : SUCCESS in 13m 34s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/33ab75f6521e489c9ca60ef2b2ceaea7 : SUCCESS in 11m 54s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d03379599f0e410f90ac958abcc3c2db : SUCCESS in 25m 11s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/41fadbe1153c4594b974252d62b49ba9 : SUCCESS in 16m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5bc2081bea724b0e88afd46283676711 : SUCCESS in 8m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ba0599d28937481baf097fc114f6e447 : SUCCESS in 16m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/411021c6b30141afbd9820a18ddb092b : SUCCESS in 8m 59s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/c284a52071a344f88a01b95a49f6418f : SUCCESS in 15m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/44f84370e7064ae7b8b787eeb6d38e8b : SUCCESS in 6m 36s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9c57e94086084572b9f6a8c58caef031 : FAILURE in 9m 52s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/12d2e83ffd7b48d7abd8751e6520dab4 : SUCCESS in 8m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/37e6f8762ee04e3b9d86ae329561e757 : SUCCESS in 1h 00m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/047be2ee1cb749d985c02cdae7e5f7fc : SUCCESS in 52m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1dd6ad6a13c2456e99a7cf33f0e633a7 : SUCCESS in 54m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc7d25f2a53d4d9da4babb5921c1e7a5 : SUCCESS in 11m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7cfb0feacb4409fa5c156c45432e4d5 : SUCCESS in 2m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/58d1548b50d94b23bd5aaea208ae6e5c : SUCCESS in 17m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6cbc2c8447d459fb59c7730a15fb0f9 : SUCCESS in 1h 03m 11s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":39},{"id":"bb26fa220e4580b78928eb5aef698d8aecd8bc19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 01:06:34.000000000","message":"Patch Set 39:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06ae2c571c41421b96bee60c593a0f75\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6e7c3ab505be41a8bba8a42511f8a11c : SUCCESS in 42m 03s","accounts_in_message":[],"_revision_number":39},{"id":"ef1ad10183f827260a970166e7b8921dfe326632","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-25 20:01:23.000000000","message":"Uploaded patch set 40: Patch Set 39 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":40},{"id":"df719e19b12ef2f1ad08c84ef5de80034384e3bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 21:13:12.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1c3961979ea48a7bb35cc43ffc01f58\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3016009ffb1e406c8917795268625ef1 : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b787f847e4749389311c10bce85f79f : SUCCESS in 6m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/a4e69d4c5f2b4a62a7511a79f3ac7aa4 : SUCCESS in 43m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/89ce5a81e34e486ba9494833a07ade19 : SUCCESS in 45m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/582dfa7cc79d47c1a0bb922d2200e854 : SUCCESS in 35m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2eef739c982841feb8e0559a7d84ce01 : SUCCESS in 33m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d46d9fd7d7c4edab69ec5006c9706b6 : SUCCESS in 21m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b143f39f06a41b4bd2db349e77a27a7 : SUCCESS in 21m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d984da5b751e4831944116989e6591ae : SUCCESS in 20m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f0d160207216494aa55223fad70e7a70 : SUCCESS in 6m 09s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8ae725e4b6b64cae81ee9c306b9798cd : SUCCESS in 25m 22s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7991a2311ec048ffa4686ae986b89a7d : SUCCESS in 23m 45s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/64a2bd76507740909c1b8cc0350eb7ca : SUCCESS in 35m 05s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/35d8d9221dfc4955b0d2cc14697078cd : SUCCESS in 11m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ac869ceee158481f9fe5ef80eb22dac5 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/549b98d3d51c437794ea6c1c168d9222 : SUCCESS in 15m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/df331e54184b4911a162ec63687f5bce : SUCCESS in 16m 02s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/c126d5cc1fee43b09dbf8af2d9b0b4f9 : SUCCESS in 18m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c576de03c52d46a288e063a27a08c426 : SUCCESS in 5m 37s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b631c463d22547ba94fd16f0bd140c7d : FAILURE in 7m 58s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6d12b8d573384b2f889eae97a4430d28 : SUCCESS in 5m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4de03a57573b4459b8ab7958b68891ed : SUCCESS in 1h 05m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/939f0852a6a74e0aa2a18ea763803736 : SUCCESS in 37m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0c88f525868d41c8bef5b16e36629f38 : SUCCESS in 37m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/903f2efc945b4f8fbd58a0c164fd04fc : SUCCESS in 21m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90a5631c27ef48aaa36852fedb39051b : SUCCESS in 5m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b94c5b7964a94796b5d87fb26c1d9dda : SUCCESS in 20m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04c5eb363ef5481f992a8359c748a6bd : SUCCESS in 43m 28s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":40},{"id":"1f46cb2a5e542162b626a440c948c0d5d6c288e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 21:17:50.000000000","message":"Patch Set 40:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94c0b72f15e14b36b87ea9c4f2960bc6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/45241b62cfd44fbe974d3f074072c906 : SUCCESS in 32m 43s","accounts_in_message":[],"_revision_number":40},{"id":"dda701b29a0899e8073dba95e2bd025b5e3ba5cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-25 23:29:14.000000000","message":"Uploaded patch set 41: Patch Set 40 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":41},{"id":"8cbbf8635f32edfd44d7354a38da07fea150f018","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 00:50:42.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a0ef4a6b93c402885b94bc9d3ea24a0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33b4fc04e19a424a96d2b800ee1bcf5d : SUCCESS in 11m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b509b0992d5d470c8f228f47bcb5817e : SUCCESS in 6m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/45a6924d20874b8e9d3d12f99421e591 : SUCCESS in 48m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eef325ec6138489d983e8ff6b054f447 : SUCCESS in 58m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/baa35d18497b49af912dedb37e61e24a : SUCCESS in 43m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e2872a868c84c5f8cb8db7744393674 : SUCCESS in 28m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/79d53bc4aa6a45b5b1579bf27789bbf3 : SUCCESS in 18m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5e0a317f459447c6ae396dfe8bffcfa7 : SUCCESS in 21m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1af63e39c7d049f3979a304ac692dac5 : SUCCESS in 17m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/567728731d9f4d688a91ef798c52510b : SUCCESS in 4m 36s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0da140ced4874ca682003e6e45338cf2 : SUCCESS in 24m 22s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f30a4bb61eca450da2bb19ba1a2b8ee9 : SUCCESS in 10m 05s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f81bf39fa73a49da89a73be7ed172bce : SUCCESS in 27m 19s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/75f66da028b948ce90302103f16b10dc : SUCCESS in 19m 57s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dd9817d7b3f6491b8c9ef814f99dd221 : SUCCESS in 16m 46s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4b3642ab183e4eb9a5c034b0e5372cfd : SUCCESS in 17m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/00f798f4aa004f8f94c636b255696bdf : SUCCESS in 16m 25s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/5303c0dabea14353841f876a3bcf5dd5 : SUCCESS in 11m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/75290851addf45e7897c9682f2b47706 : SUCCESS in 8m 45s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/56a754fbfa0b4df794943a0ffba46363 : FAILURE in 9m 47s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/df896c17b7414c739acf94f4c7ffb5af : SUCCESS in 9m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5634069d3e5e4bef8086a0ac25562740 : SUCCESS in 1h 04m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/90862c173b0b497ca8b01e581645bab9 : SUCCESS in 35m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/02d401a2cb6b4de1862c4432ca2b38d1 : SUCCESS in 23m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b57ffc8a6414c2ba02edc949ce4b78a : SUCCESS in 28m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af1c28f6c59d416d9d2af6630ed9da95 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c174c4530c684eef99a3d5ba04086675 : SUCCESS in 11m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b316860bc294898988c852a5f60e737 : SUCCESS in 54m 05s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":41},{"id":"53b797043f32b867892d3125678f2ac4d62b87fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 01:51:18.000000000","message":"Patch Set 41:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c4d53c69db74b84a6dbf10a28ca5fd9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b34da4c33cb44403abc94501e43e42ae : SUCCESS in 33m 43s","accounts_in_message":[],"_revision_number":41},{"id":"0442ad8ddd1a72cf1f6165ca00d32fe1b681c78d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-26 02:01:11.000000000","message":"Uploaded patch set 42: Patch Set 41 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":42},{"id":"ef476d7dcc214e7497d1ebb0fd26ed03373fb888","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-26 03:28:02.000000000","message":"Uploaded patch set 43: Patch Set 42 was rebased.","accounts_in_message":[],"_revision_number":43},{"id":"3c9b381ce6aa063fa4f4b66f65eaca0c8a6ec16a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 04:50:08.000000000","message":"Patch Set 43: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90d25728fba64ba1b7251f87bb574148\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a88171c60c1c499ba992eff5e15a0c13 : SUCCESS in 10m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/794a485c03a245d683f9e518ddbee7ac : SUCCESS in 8m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/62e2a5611efe476ebd69222d52c69935 : SUCCESS in 1h 05m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/15669a6464ed48bfbb8b375c0354a7df : SUCCESS in 1h 00m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0d2b1ffe337d4d8eb2d366a0b1979399 : SUCCESS in 29m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87255af3d1aa4affb86ce4f740011304 : SUCCESS in 1h 00m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/941a211d2ee54106bb9e9d8f3ef3c998 : SUCCESS in 20m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/91db9f1f2dc74dc0b597b08de9e515cb : SUCCESS in 20m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2db499f47b7147879ab9973c8d186f1b : SUCCESS in 19m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/508181123e4e4b849ab29c6b050f5742 : SUCCESS in 10m 09s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4d99794529b34041850de4dc9c57ae7e : SUCCESS in 20m 20s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ad9fbd787d4c4f02a03eb8f2c80d51ad : SUCCESS in 20m 16s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e181823738cb4041931e5d6711956713 : SUCCESS in 25m 34s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/2d42527122d94a67b3c0c20d5ed8ef47 : SUCCESS in 21m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d368f32f27ac4886892bc22031d2e28f : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ee75e184f3454eebbdfd770672e8595b : SUCCESS in 16m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f0d6335e1a8c4d7c9c6b059b01dbd1e2 : SUCCESS in 16m 41s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/a203f2fa610e4880aaec030b7d2ddadd : SUCCESS in 17m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eb17e5f805a14d9d9c85b0f53f15e737 : SUCCESS in 6m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d68aff118f2740a693c6ec07c9ca30f7 : FAILURE in 7m 45s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5bb28c7a703a4326923ac9a4bc868d38 : SUCCESS in 9m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b00f53416c774bed9d846cde15107975 : SUCCESS in 1h 07m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/306eec0cd04846ac8d7e7bfad6f39b89 : SUCCESS in 58m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3933e5baf242487c861ed3b35f11b034 : SUCCESS in 38m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d6d16f6a6874c75a0b16c2766f4f8ba : SUCCESS in 25m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6388274e0634c6cabe7b05abb928a31 : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/121d5b9485434844ad1c70e047fc6311 : SUCCESS in 18m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3661bba0c71f409da09395edc59e2970 : SUCCESS in 43m 23s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":43},{"id":"aabf2fe73b2ee34570169c3c9837233c47c3a3c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 08:39:37.000000000","message":"Patch Set 43:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d91ea09f9d045b38b9bb482f9e080cd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7de23b70571145e9a019a0164732de72 : SUCCESS in 1h 09m 08s","accounts_in_message":[],"_revision_number":43},{"id":"66bc60ba85890c5babc90889c9f45e738e77d464","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-27 05:56:10.000000000","message":"Uploaded patch set 44: Patch Set 43 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":44},{"id":"a1ce978a5ecfa84821586f89896595416adc77eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-27 06:36:39.000000000","message":"Uploaded patch set 45: Patch Set 44 was rebased.","accounts_in_message":[],"_revision_number":45},{"id":"a4d9d037eb9581308956976d9ca6cddd131d9c69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 07:54:09.000000000","message":"Patch Set 45: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/529027bbf2224685a5037ea9a71c2234\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0da1d21151c543dd987b0a52ee113758 : SUCCESS in 10m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/658e5730d223464fb0455e1ab5662b61 : SUCCESS in 8m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/de76f3b828d04ae5b13ab24f5abd7037 : SUCCESS in 1h 08m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/80c96b52264741dfa852cc03776a1b6d : SUCCESS in 33m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/33811124b05b455288bd14f18b1b9cd5 : SUCCESS in 48m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f2d55c29cb74d99860946e249f16722 : SUCCESS in 57m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5635dc67467c4a93ae63aee8f59f2416 : SUCCESS in 11m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7a4d9863e5124a37be826b1462c7b9cb : FAILURE in 6m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c3d462ca6b4448c88e46c39a3dee27b1 : SUCCESS in 22m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b466e7171a4d4a39baa0c3cc06500180 : SUCCESS in 6m 49s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/72f783178091485aa37ecc85f12368ac : FAILURE in 19m 27s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f002392483dd4958842081c6e6ce9d79 : FAILURE in 17m 45s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/40a414b849324c3991364af131af7817 : FAILURE in 21m 25s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/2db5ed9ebe6e44378dfc4e8def5baaf2 : FAILURE in 15m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/204d93b713584609a0234dffddcf758c : SUCCESS in 14m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f91bf2b2004e4c6c806b7bcfc2d3e095 : SUCCESS in 15m 52s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a41e64bf499346139c4cec35be48377a : SUCCESS in 14m 08s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/1d27014177424d8c924c8c7f9291402d : SUCCESS in 10m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/32fae7b3148e4691b28affb75dad16ac : SUCCESS in 6m 42s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bbf457f0ebb545628bf48dec00a2a24a : FAILURE in 10m 36s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/affa795c2af64f6db6c15fa50c715268 : SUCCESS in 11m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa51506830204723b5adcc6ea2e8f208 : SUCCESS in 1h 11m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/de10c4dae47542da9a1b8dca03e25163 : SUCCESS in 54m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c75684d7f190402da4f5813c38ab893e : SUCCESS in 38m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92ef2ad9619c4b10b2beed834d28da38 : FAILURE in 12m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80d35ec71eca4dc2828da652c3025ab6 : SUCCESS in 3m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/636c911ae9bc442e8f4121743ca1c1a0 : SUCCESS in 19m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2e4a8a92942438983530e06140bc99d : SUCCESS in 1h 01m 00s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":45},{"id":"0d2c8a085f7aedf9371b12e24d619a136e7e7e29","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 07:57:58.000000000","message":"Patch Set 45:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/210816dcfa43410ab35380ca929b841f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a3faffef06d54a73981e71aba4251386 : FAILURE in 56m 02s","accounts_in_message":[],"_revision_number":45},{"id":"90a93bf3bab6f41d342349644a0425731d6243b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-27 13:56:13.000000000","message":"Patch Set 46: Patch Set 45 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":46},{"id":"e918a4d274f845169e1ad18859045529fb2c9528","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 15:19:25.000000000","message":"Patch Set 46: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e367a779ee64b90a11205264f49b063\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9766e092a36d453886484a80b829fd1d : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/391e33b697134d6b84503d95ea4d8ed7 : SUCCESS in 4m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/f39d0e1fc45c4e7da5113b798bc6987e : SUCCESS in 1h 12m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a271eb03077f414889f3dfd2d7407e35 : SUCCESS in 1h 00m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/52e2221814014fa09f49e50fcfd1b0a1 : SUCCESS in 50m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7681248a39754b0e9b75ff3becb694af : SUCCESS in 1h 14m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/84604af9a91b4d99abf85f98855a0752 : SUCCESS in 24m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cfbaf61c32c24a69bf36b4c3707a6efa : SUCCESS in 19m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/403754a8329e42e2bbd3ea00ed2aa129 : SUCCESS in 22m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ebbd518b36f644bd897985315aeddb44 : SUCCESS in 8m 05s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9f0500ff95714f0eb451dacfd9a4f069 : SUCCESS in 27m 11s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3548e1bb7a854413bf9cac485062e94b : SUCCESS in 10m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a0a26bde142b4d7baf7f3669a21edcac : SUCCESS in 40m 57s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/94cf6f218f024bc79c05ea402d7787cc : SUCCESS in 22m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5a932e03138e42fa873502cd89c45e2a : SUCCESS in 14m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1c5ea599ab604ad0be1776e62933a6d8 : SUCCESS in 16m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4fe52762f3564a8ca131127c73a82363 : SUCCESS in 8m 53s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/1991e519fc48478f9e7e9c7861701a5e : SUCCESS in 11m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/67505c9dbf634a8cb3bc631a736d7992 : SUCCESS in 7m 13s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c96dbcd8a21a4a89850d0b3393ceb662 : FAILURE in 7m 03s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d52a6376b67c4b9fbbf096eee4873516 : SUCCESS in 8m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9aeb58ea5fce4d5f8482264061725b8e : SUCCESS in 38m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fafa33df91704ca2aee86a2c9374c479 : SUCCESS in 1h 05m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c3b625d7af8e4b91bab85e18ffa34172 : SUCCESS in 52m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01520698e2954cfdb7f82392e5b50548 : SUCCESS in 12m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6783a44bd4ee48b58065481abf536f18 : SUCCESS in 2m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/46d407534b89481380cfa4193f6ede86 : SUCCESS in 27m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5605490d31b24e8fae7a6ec266eb40a1 : SUCCESS in 28m 08s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":46},{"id":"0fa2b1dbf2b7ce31f47ee8fd3e4be9fd3e155cfc","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-27 16:33:01.000000000","message":"Patch Set 46: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":46},{"id":"48a70b78fde691513a28de6638cf75aceca69bc9","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-27 17:26:47.000000000","message":"Patch Set 46:\n\n(1 comment)","accounts_in_message":[],"_revision_number":46},{"id":"77d708ec00de0a2ea9a13c2ebead7f84f9b93a6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 17:30:10.000000000","message":"Patch Set 46:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/273b39adb5ca4a6cb536c101d3b6ee3d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5e75352084194f38940dcda5112b5c72 : SUCCESS in 1h 16m 52s","accounts_in_message":[],"_revision_number":46},{"id":"091b6d89cc50adc6184b66587508a0ea18dbb36d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-27 21:57:30.000000000","message":"Uploaded patch set 47.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":47},{"id":"da302f58621b83843683f3c3c8519256a4783f72","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-27 22:02:11.000000000","message":"Patch Set 47:\n\n(2 comments)","accounts_in_message":[],"_revision_number":47},{"id":"72b60624435b91e0094535d27e399f43f03c3232","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 22:48:29.000000000","message":"Patch Set 47:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19408389b60f434bb19b6324ce77683f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1eb342008c4849298323d95da4c76dd3 : SUCCESS in 46m 52s","accounts_in_message":[],"_revision_number":47},{"id":"60b08bedd8e20a5a6bd30333a1a4604b7c92c15c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 23:06:51.000000000","message":"Patch Set 47: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/04d06f5953cd4e1fac58ab23c82d6dee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e93970c333c4119994b0911b12bf51e : SUCCESS in 5m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0519bf744a24219ac0deab669ed0d04 : SUCCESS in 4m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/c90623281cf34bc6a02906b4aea33dd5 : SUCCESS in 1h 03m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/299dcf5e77e647e7890906d9fa0e16b3 : SUCCESS in 47m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/777cc791e41d4e329541ef27c0c7c67a : FAILURE in 13m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0430ed3c898c4bc6a27a1c7210242859 : SUCCESS in 59m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65bf0cb6345f498d8c3f82f1b9174394 : SUCCESS in 16m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/868970ad27fb43dc84ba13a78af384b6 : SUCCESS in 15m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/85a7c609eb9f4698885207eee204db64 : SUCCESS in 10m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9a5451f7a16b44b1aebb335a67915de8 : SUCCESS in 4m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fd7851da5ada4569a187036c586f1858 : SUCCESS in 12m 11s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ae1d741fc0f344b28bc129749bb6d8e5 : SUCCESS in 9m 04s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d096f38bda6a4846ab79f270f2aa6b5e : SUCCESS in 20m 40s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/f29d99b644884108828f3558e7303125 : SUCCESS in 10m 01s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b2c2f52550404885b329240d72a16546 : SUCCESS in 12m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7e27a15b65d14c88a532400a7fb2d488 : SUCCESS in 15m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fb8bd30d6f5440989a905f1474c6d1cb : SUCCESS in 15m 51s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f5c41d5ad724b63af512d9a8aaf8ad7 : SUCCESS in 10m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f20f8d84cd1d406395476e6f9ec52588 : SUCCESS in 6m 33s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/433a9067149943febedaefd7941df5f8 : FAILURE in 4m 47s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/cf77754da7aa4ac9819d4d45be490154 : SUCCESS in 8m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d63a17bcfd45491b93ad353db0f1e801 : SUCCESS in 1h 00m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a511d1424baf447092cf0c53f8a5baa9 : SUCCESS in 48m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ceed5c47274243e4bbac6d42d7763132 : SUCCESS in 25m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1a77bbed29f490b808d880d1693c5bd : SUCCESS in 12m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba19194b733b404d8cd7d421b829bf80 : SUCCESS in 2m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fc42cab7874d46b988c5c9776f9e629e : SUCCESS in 18m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a24a3d0b93b43b79178a9f017e28184 : FAILURE in 11m 58s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":47},{"id":"8e5f4a5a6fe39bcce782197b8a95e5773d7d8692","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-27 23:18:02.000000000","message":"Patch Set 47:\n\n(1 comment)","accounts_in_message":[],"_revision_number":47},{"id":"f3710f16c0c6de139f327027e562011abd907781","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 00:06:25.000000000","message":"Patch Set 47:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31d08bfa5c2d4dd1803f70054f24c0f1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2736675f85354ad78aa23414822a46d3 : SUCCESS in 42m 15s","accounts_in_message":[],"_revision_number":47},{"id":"449196f7afb0e4e88da0ddbdc2ea3ed253faf023","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 01:01:05.000000000","message":"Patch Set 47:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d43a5811afd4ba69bcff6a182558522\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a140a711f5704e27b1f2dbc782e89760 : SUCCESS in 9m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbbc619ddc69411a91b6305643c492ca : SUCCESS in 9m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/413d7ec8e473477cbc36a376af7ec632 : SUCCESS in 1h 05m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2a10a1d63fb64511976119c364ec9fc9 : SUCCESS in 1h 06m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fd1093520b144a639011cfee8c6a463c : SUCCESS in 49m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b33911d2b134574a65fcc693ade0126 : SUCCESS in 51m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/90ff84d5e14f48a29ded9cde17b43dd9 : SUCCESS in 19m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2f2d05960c964896b6de148f333da534 : SUCCESS in 15m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/603070b5f99b44a18e0e0dff819ca61c : SUCCESS in 22m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e89ab1dc6a9742f093d607ed8154fa16 : SUCCESS in 6m 06s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5b3f5563ae2947bf98967ee445ce3f53 : SUCCESS in 10m 15s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/bac871461acf49499a49cf7dd8922c95 : SUCCESS in 9m 11s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5ca87e8e1e05417688a069e2658ec2d0 : SUCCESS in 24m 15s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/e36d108d2b5e48c994eac07b35ec1ccb : SUCCESS in 19m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/aac644be6f2e46669c524d4e3b7b4598 : SUCCESS in 14m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1263d746bb634d858b89ce54eff24971 : SUCCESS in 17m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ad5f5050189440f19652c8d9fabcbadf : SUCCESS in 14m 58s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ba73a920f7340be93e5163d348cdd92 : SUCCESS in 16m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/722d39c67bd744828b4e9dd4995b413e : SUCCESS in 5m 55s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cb26a89283534c3ab316d292c135a938 : FAILURE in 7m 15s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/bac1b0f5c93a4a88a919418c07a10746 : SUCCESS in 8m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/276a859365f14209a558193a23c4d61f : SUCCESS in 1h 01m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/45d3ea2d3fb04c369e344db60c7f375f : SUCCESS in 53m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/083c8916cfaa4ef1922c5a0d254f15d8 : SUCCESS in 51m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5aecd42a498b447c99f8340bc35deb8a : SUCCESS in 30m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ad252fbfc1a49f9a0f2c77704098fdc : SUCCESS in 5m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/130b60838a32414399feccffe6f9698a : TIMED_OUT in 1h 31m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59c6b33746c14384ac388b19345e6169 : SUCCESS in 53m 31s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":47},{"id":"bc118d5eeae8a28874889f87f83d2f42ad1a9198","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-28 02:41:01.000000000","message":"Patch Set 47:\n\n(1 comment)","accounts_in_message":[],"_revision_number":47},{"id":"f49d7959d27a727c580afc5fd98937916bdee914","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 04:04:29.000000000","message":"Patch Set 47: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4eb5dbe5abd249e298dbbf780c3573f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/593c272eeff940dd89f3a6a7c72e6b99 : SUCCESS in 10m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa662bc25f0047159e08a6f2c16dec5f : SUCCESS in 9m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/0e5453f9114a44f780edf2487129cd7b : SUCCESS in 1h 02m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0878e19c6889412384aeecb85d837c9f : SUCCESS in 1h 03m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cd352dfdfe9445f1aaa8308a3837c01e : SUCCESS in 59m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c59438e4a26248d48c636ccb9ea226f7 : SUCCESS in 1h 00m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b053ec17ac664511b603c98e4977669b : SUCCESS in 19m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/44f73e22da844536ad02bc9388e89ca1 : SUCCESS in 10m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/137d97a6162d4bbbac45c49e80a3d334 : SUCCESS in 12m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/25ce4d01c84a42f996fe44096774712e : SUCCESS in 5m 42s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4f4e0bad7f734e0e8e08f03b665f2d91 : SUCCESS in 27m 15s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/364d57c587e441dfa2e93e6b7e34578d : SUCCESS in 22m 48s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8a8e464df87e445c95a177ce455d58ac : SUCCESS in 24m 38s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/27dd37dda1e14173a973e450b07d4ec2 : SUCCESS in 20m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c9e60613f7004ab4b88e1701de784fe2 : SUCCESS in 15m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7c6a1736f0744e10a769e637a2c872b8 : SUCCESS in 15m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/85ca12faedfa42aa8acecfe1ade41f75 : SUCCESS in 16m 59s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/d491483f2ed6487e88d0876fc18a06b8 : SUCCESS in 10m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ee216e79133a4af9b8bc9c5673837cf3 : SUCCESS in 6m 45s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2ee0a5ff375b4f3c97e711235895ce4f : FAILURE in 9m 01s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5eee923675c44fc6a54790d77b3987d1 : SUCCESS in 9m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a230f4d14aa46ccae706a47960eda5a : SUCCESS in 36m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/95658262f82c4d129ad3d27da109dfb7 : SUCCESS in 51m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b244c188e36143098653be2992de94a8 : SUCCESS in 55m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3abdbd392c6a411eb3149bf6a1013090 : SUCCESS in 28m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bd8f4b49b924eca8fbce994177d94f2 : SUCCESS in 4m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9c80d2fb01094e74b8eec8abc347e5b5 : SUCCESS in 20m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4186988ab0c4a6ba8e3c8fcb1691b90 : SUCCESS in 54m 50s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":47},{"id":"06ee370123e80b517cbc9b61e71a75c7d7e5378b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 05:18:42.000000000","message":"Patch Set 47:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73a7d2df2d9746dd9d4e84f6689c8d3b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cc58e5600ebe478083a5d3555b4e8c8b : SUCCESS in 43m 03s","accounts_in_message":[],"_revision_number":47},{"id":"dcac468e74154c45b11a03f76e9c6045af222931","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-28 11:59:22.000000000","message":"Uploaded patch set 48: Patch Set 47 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":48},{"id":"bbdb067d692dfca6c1542d17d34ad2b77c2e101f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 12:34:47.000000000","message":"Patch Set 48:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68e6b2c4e9294fc79dd79fe53293c27e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dc10860d3708471c8a9ac9d83cbf0083 : SUCCESS in 28m 57s","accounts_in_message":[],"_revision_number":48},{"id":"2dec110edd1e4b8d5601368dffca8758802518ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 13:06:12.000000000","message":"Patch Set 48: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/18777fc439764678b3eaf0b47f687613\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b81d70ef067468f84149c2a07011244 : SUCCESS in 9m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29e21c9a7d6b4f44a92c74f38b297afa : SUCCESS in 8m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/871d0dfe890f4c2189e6c25022c71ddc : SUCCESS in 31m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/48392f66ec0b440c9c6f7addd7b64f56 : SUCCESS in 59m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9d86ae133e914dfeba71a3d9c4ed6a8d : SUCCESS in 59m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee5a01e2b01f406ebeafd894ddace7d9 : FAILURE in 45m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3ad75df137e14fd4b9c76e15fb8a9e39 : SUCCESS in 20m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0958d7f046ed404285218b8dbc2e3a89 : SUCCESS in 21m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/85d90e91ca7a4069ad6d1c7dc315d209 : SUCCESS in 22m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1c9fdf878a9e4938a6b6b96c66b94bcd : SUCCESS in 7m 03s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8ba9330c7f144a288e2c4a1256b7ecc8 : SUCCESS in 11m 11s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d97b7093c45c4002af5ed76672fad1e2 : SUCCESS in 23m 16s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0ef574a257144fad8c20d603330402a6 : SUCCESS in 13m 36s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/164e04c5fb714a5b96f4ce1dd22f1816 : SUCCESS in 21m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8bb3ea5345cc471b90f29d1b56d5a83c : SUCCESS in 14m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/425787b57ea646fda994610e57088a83 : SUCCESS in 15m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b3b707e5006b435cbce8f488e6252567 : SUCCESS in 9m 13s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/02be4ce119f54cfba5bd1db8c01091c9 : SUCCESS in 15m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/43acceee49db4c26a4414ab15a7eea17 : SUCCESS in 6m 58s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/672a2749409e4cda83744632d586522f : FAILURE in 7m 47s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/977b3ef420e34fd296393ac6410eda42 : SUCCESS in 8m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7888a069d22c455cbe646da13f27b4c5 : SUCCESS in 1h 01m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/df1235e13e5d4ca2ad5f8d9632a12fd3 : SUCCESS in 26m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8742b668c0ba441dab2d0cbe607b39f2 : SUCCESS in 25m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74ca7028a31744189894ca6ceb437cf4 : SUCCESS in 12m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a89242c6a05c4c8fb655ea0f1bcbab0f : SUCCESS in 2m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dc5acf88b7ad46f0bd3291b213505396 : FAILURE in 19m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e5f7f0920fb4f3197daae4d20936792 : SUCCESS in 55m 30s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":48},{"id":"e43ff897b9a0c839283be2d84c400636ff43c5b0","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-28 13:17:55.000000000","message":"Patch Set 48:\n\n(1 comment)","accounts_in_message":[],"_revision_number":48},{"id":"cb86dbe919e018ce373f12713de22cd0eab04095","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 14:33:27.000000000","message":"Patch Set 48:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ad8067792e541aa84b6695e1d38e461\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/358e6e36dcb34e1bb7f7b4b85fa884e6 : SUCCESS in 1h 12m 46s","accounts_in_message":[],"_revision_number":48},{"id":"6c721102482fa2b91341879565ca65f9c76f6203","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 14:36:32.000000000","message":"Patch Set 48: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/731148f5c5ed4c3e92d4d86908973670\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e60fb02bbc2a4f0cb96d2c3ecfd755f3 : SUCCESS in 10m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1de6fd1fa2424ffd97f470862ea274a0 : SUCCESS in 9m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/44fe1cbc92e048f7b865b3ba0b4eb132 : SUCCESS in 33m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/19a827cf3e6a4b4aa58093ba35ee8efe : SUCCESS in 1h 10m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6994c90db6d5457db8fd367754b9f294 : SUCCESS in 59m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bbc6c004c9284cce957c4a1fb14200df : SUCCESS in 48m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/be0c23900c22421f95d657f44bc06d98 : SUCCESS in 19m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9a339f52342d45c8bd472c101cfd7f40 : SUCCESS in 11m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7b5787a5841341e8848e8c52607a2d08 : SUCCESS in 11m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8c7ba1389e104ecb91b9a05b1d2181b9 : SUCCESS in 8m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f04cfad570154a76a53dd5ed5fe4d57c : SUCCESS in 11m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ff30ff244a11421c929a9cfc1cb2cb74 : SUCCESS in 15m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8f364b736a674054b489da18657dbaf1 : SUCCESS in 33m 48s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/100b6880ad394ebe98d73342bb2abad0 : SUCCESS in 9m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/65916a8f43104b19ae852f7f8ac3a9b6 : SUCCESS in 17m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4c316c4d11844d2cb7ae5fccdb4dc0d7 : SUCCESS in 15m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2b88fab3b6454a72a7b4a2692542a836 : SUCCESS in 16m 54s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f5b11602491469192ccba622c640980 : SUCCESS in 19m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/aad3896ba55645368e85dd709f1e96a0 : SUCCESS in 6m 14s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/42216abd00644075895739c42734714e : FAILURE in 4m 17s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d0235ced8bad41dd88d43a41b3f6ea0f : SUCCESS in 5m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8eb24d29fcf46d48f0bae4163ec4cf7 : SUCCESS in 40m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ea171c78c2094176924b1902809b3237 : SUCCESS in 25m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d96be63205174e84b606e1c86909bacb : SUCCESS in 37m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/363ae73cdd004be78b19c02fc901d8c0 : SUCCESS in 22m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d92d10f9d234d3a829b6a4211b117dc : SUCCESS in 2m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/58d292c31f614c36b9fc15412a14a404 : SUCCESS in 19m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22f20e3cdb214d90a20bbda1c66533f1 : SUCCESS in 1h 01m 54s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":48},{"id":"7fd58b4c54cab756b5f8cf8cd9f2c42349a50728","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-09-01 15:09:54.000000000","message":"Patch Set 48: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":48},{"id":"bea5c3fe3e1fafcde9be7dcd28df250f187fe17e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-09-06 15:46:35.000000000","message":"Uploaded patch set 49: Patch Set 48 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":49},{"id":"4f6ab03d965728616968da805121706d9a2d3529","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 16:38:28.000000000","message":"Patch Set 49:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d77a659356a148729f058b423560d130\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/89325c81cc6f42e2b7dcb7d1b8a535f6 : SUCCESS in 49m 35s","accounts_in_message":[],"_revision_number":49},{"id":"8ec342a258ce6bf82aadfdcc9732410fcc16d0a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 16:51:42.000000000","message":"Patch Set 49: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f96afa9e8bb4853b3bc34c098692c1b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7093820b11d3476282f4a97eaf6b46b6 : SUCCESS in 9m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3daaa15e056842f38d5cff04843abcf7 : SUCCESS in 10m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/59320074673b423691c5df3fc49cbd9d : SUCCESS in 30m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c4dda61d4431406a8d6071ada4e84de5 : SUCCESS in 45m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0755b50d6b164b40b96e8e94aa0741d1 : SUCCESS in 53m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb9c9445d6144844a6ef2abd20676cb3 : SUCCESS in 28m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/90b852de713542e6a9af267f4412a349 : SUCCESS in 21m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/342cb18ab29648f79f8320f9f5a3795e : SUCCESS in 21m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d1ab57fbbec34b95b1b7e0432fa74572 : SUCCESS in 20m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3eeac7fba5954ba190c6ee5d76f8e4e3 : SUCCESS in 7m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9ee85d786e5245cca6c8d79ac11607a5 : SUCCESS in 27m 14s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3c6cc475f7404d01b8c5a3d397e13e77 : SUCCESS in 15m 37s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4f168e4323b741819e54f776f60cf39d : SUCCESS in 31m 07s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/f6b3459fa8e148b9af30a2607cbb351f : SUCCESS in 20m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4edd3166832647f2b07d91fed7c5bb4e : SUCCESS in 12m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ca99647ca92d4bb0b281c7e20c99068d : SUCCESS in 9m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/36bcf750fa72427aacc39f39e50f2383 : SUCCESS in 8m 45s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/b45cdd30e94841ce83cb976e24404009 : SUCCESS in 17m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e808f3daad5e4c688808c781fd7b2557 : SUCCESS in 3m 42s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cde409cd82124dd2a651f8af49032dbb : FAILURE in 30m 23s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ee9b0fd7dc85487f887fdb1c90d446dc : SUCCESS in 11m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf9d0f02023a46378638775ce9d87b1d : SUCCESS in 59m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/242b5c00f427468abed67ef8cad9a1f0 : SUCCESS in 35m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0f5e1ca99b6948b39568380ca4e2bc52 : SUCCESS in 50m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6962bb2e1a1c4f9fa6ca71ce064e157b : SUCCESS in 30m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a226c9dd1bf4a018383a36a02333d72 : SUCCESS in 5m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/86c9a252cf8e41e8bc6844a2da70af5c : SUCCESS in 27m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/32ea488e54da4aea88e1f697b1501e2f : SUCCESS in 42m 39s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":49},{"id":"c6bd84efc049e18f9b9a4ed44e24365b4baf383f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-10 22:53:39.000000000","message":"Uploaded patch set 50: Patch Set 49 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":50},{"id":"f737961f0d985474c38b6b08d695a0e80e554588","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 23:46:03.000000000","message":"Patch Set 50:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d14bb6388f66489ebb6233b4abc7ed61\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/541e04991e45435b9b69111a95a1224a : SUCCESS in 50m 28s","accounts_in_message":[],"_revision_number":50},{"id":"e05af385cf4254f9591b58f8132eb31406733994","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 00:06:17.000000000","message":"Patch Set 50: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3110b38444094f59bfdcd647f6d8fcf4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7070cda0934541cab3f18d2ae4ec3090 : SUCCESS in 11m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d30934d716b4d029c89ab4a25a42f39 : SUCCESS in 8m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/04e752c779aa409883924a54e4a05228 : SUCCESS in 1h 05m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5f52bee0dc7d4ff79ca5867350c80216 : SUCCESS in 1h 04m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/851a820ffecc4e238514b5afad8f0fd7 : SUCCESS in 39m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ec93edc88e242b9b50e37ba32a35aad : SUCCESS in 1h 06m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1bbeaaffaa6245d4b4230e2f9d509cbc : SUCCESS in 19m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b9de2325eee54e989552769d7a883d49 : SUCCESS in 15m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/add781caf0874059bd5b45a7e0269a76 : SUCCESS in 23m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e230ce1fa3f7463f868d371383574106 : SUCCESS in 6m 28s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f26b0f24ded5433d9590a44dfe84239a : SUCCESS in 26m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8c02e4be8f624b6ebfd4e8032db94e67 : SUCCESS in 22m 04s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/dd447a806cd4440e9fb3cd0cdc26ee13 : SUCCESS in 30m 50s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/203b51f346854b708a089096767acff6 : SUCCESS in 20m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1f40488da2044ddc9726c64aa5ebc412 : SUCCESS in 12m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8ae3f7f8cf2f49eab5b20ed6debb8223 : SUCCESS in 15m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ace4857274394e5ca85b9f4ba8fcfe48 : SUCCESS in 15m 17s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/5fedf9f7a4a546bd8ba4df220c9de60a : SUCCESS in 17m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7ed0a2a67a8343228a1e591cd31684d7 : SUCCESS in 6m 51s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3cc4524bafe3468f9b0417117301f360 : FAILURE in 32m 12s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/eee030df6d7142079ef457dae717bfc2 : SUCCESS in 9m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f60bf2996ad94618b9e9ad93e31666a0 : SUCCESS in 1h 07m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e2ff196d36a347a896c0db42b05f8c70 : SUCCESS in 1h 00m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3255acc4a2f7458bbee46bca73ba8176 : SUCCESS in 51m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9fcf3f9a37884316a154b4f329df3019 : SUCCESS in 28m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c599acc39f494eb485dd7f12285b66e3 : SUCCESS in 5m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f848373501574ff69f5c6358cf9988d6 : SUCCESS in 18m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f477733d2ef74f1b95f9f1c9e3dd10f2 : SUCCESS in 57m 24s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":50},{"id":"f71e49c71e6e4786b578fad2919016bb7e3a93fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-11 15:21:51.000000000","message":"Uploaded patch set 51: Patch Set 50 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":51},{"id":"8dc2dd1e006dd002398f17e84182deb4b60835e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-11 15:34:30.000000000","message":"Uploaded patch set 52: Patch Set 51 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":52},{"id":"dbdaceb280a64fed34e7afea80b8accb3970f4bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 16:12:43.000000000","message":"Patch Set 52:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc88acc3d22946049cfd919f94e760d9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/27e854cccfe44a769786b3b3c8b1332f : SUCCESS in 35m 35s","accounts_in_message":[],"_revision_number":52},{"id":"acac5266d77ec9295fd0e14bec716fabf179b90a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 16:53:30.000000000","message":"Patch Set 52: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf5c2447876b42ec96760d0c61ae9e18\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94bdc7c4c44549d8b4ccd1fe7a1a1847 : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58316dd44d02431f9d71b3236e58663d : SUCCESS in 8m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/730f0731f65d4d7a93b6f5e82a7d4b87 : SUCCESS in 32m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/16b7c582d612443085550872a62b5efa : SUCCESS in 1h 09m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/78ba9e45b9d2498187bb602b5af003f6 : SUCCESS in 42m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b6dbc47805e347fba075f23be480e327 : SUCCESS in 34m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5e8b73168c684090815856eec3de7feb : SUCCESS in 12m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d96ec34b8f7a40e7891da7fff56a6313 : SUCCESS in 21m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d69bf5447f714afc9efe30a1f9c6d9db : SUCCESS in 13m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8e2abfdb55ce45c3bd5dffcefcb26654 : SUCCESS in 7m 37s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/aa64efde14194cbe95c475c4ecfd55a5 : SUCCESS in 25m 14s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/42c1f2f268894d2aa69a1c8d6881b106 : SUCCESS in 23m 15s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/950d2aa7363844599d382f8f3f2c187a : SUCCESS in 23m 21s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/0f27e563b0aa4947a4c371b3a5fe8ee3 : SUCCESS in 22m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a8ce57705a6b4ab293857e0ae39bdd7f : SUCCESS in 16m 23s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/154d63c882784b9781c5a8301cac3ac9 : SUCCESS in 16m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b1afb17b3c2b4399b575a6181b776948 : SUCCESS in 14m 38s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/476f549c26d441939e852fe33459bc50 : SUCCESS in 16m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a0e048f50099482f91034cb0dc2e5e46 : SUCCESS in 6m 46s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/45c6797f61fc4d28bf65bf812557c584 : FAILURE in 25m 32s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7759a98e8a3f40f58ba94eb8fb6bc158 : SUCCESS in 6m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc4c5b2597214b7e97462fbfe0f30828 : SUCCESS in 1h 05m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8f22bcfcaa694d6daef55a88448dbec3 : SUCCESS in 26m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/99eb9725bf804f75b90bbce00d3a59d7 : SUCCESS in 57m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d36cdcc9b3f148dfb1c221f06b3e1287 : SUCCESS in 24m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/827aba7388b6459aa03f5fb1670055f5 : SUCCESS in 3m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f50d9884d3a84a4089fd93f3f2c15f0f : SUCCESS in 17m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f53ffd6337b434384865ecfdc538a06 : SUCCESS in 57m 10s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":52},{"id":"4eb843de863a022bec57d17c20f17d1ae41b1ea4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-11 17:54:40.000000000","message":"Patch Set 52: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":52},{"id":"07d82293664b02f2e38f2c90cdd23bbad7c0e2f1","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-14 08:25:20.000000000","message":"Patch Set 52:\n\n(2 comments)","accounts_in_message":[],"_revision_number":52},{"id":"7c0da0b0cbdd3c862f2e9ea1a43a639af39fee67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-16 15:00:30.000000000","message":"Uploaded patch set 53.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":53},{"id":"c20de31ff37b96dc3d2e375875cf0d4d0fcffa4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 16:18:16.000000000","message":"Patch Set 53: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df1fe6edf617405cb09c45fb4a48866a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01ed9b8d396a4ca2a78e2bb27252c204 : SUCCESS in 11m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e35164ccfcab42dc9ee7842bca976e55 : SUCCESS in 10m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/3a45646c4708488398475dbcfc96f0a2 : SUCCESS in 31m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/58ca068f892140b18bf0dc35422c1164 : SUCCESS in 1h 05m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/273fbf5a66904b5b915287e53e6ebcdc : SUCCESS in 1h 00m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f838e30b8c7432bb52b2966ded7a17a : SUCCESS in 1h 01m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b23c4bd7b9241d2844dab860141b46f : SUCCESS in 21m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5576822ac96e42f3b5681e3ee9569e95 : SUCCESS in 23m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cce31360ba294eeeb8c860432f4f144c : SUCCESS in 23m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/893ad0c4b7ba46e48ccc9b0e00dedb23 : SUCCESS in 7m 28s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bb5ce7d5bc1147a686054d9bd43145b1 : SUCCESS in 30m 53s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d5c260aa47294cf2b5f35ef6c11e194a : SUCCESS in 22m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/debb07d0f1144be9bdd21879e0ebb154 : SUCCESS in 27m 58s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/0ffe07ea649a461b92fff83d774aa316 : SUCCESS in 20m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/47e1b26adfd04e378f02b51410b7aaba : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a45ff909656f46fcac15ef58b2787a77 : SUCCESS in 16m 00s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8239be365d5e40fabdc739baae62648d : SUCCESS in 17m 53s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/3091c45628864729821c99b8b8aab1a2 : SUCCESS in 20m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a68f3d726b124fd8a80ffab9b8bad76b : SUCCESS in 6m 26s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/86692bd8e3df42389ee99ac94b989cd4 : SUCCESS in 9m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c86529b19134769907c967d35a22db6 : SUCCESS in 58m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/29e8a6f49a864fe48b0d094b0b9ce3f1 : SUCCESS in 52m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/66a7a1d069724d3b812d40a8f58a9a3f : SUCCESS in 51m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78f9938c46394fdbbf7778fc82fcdadf : SUCCESS in 30m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49e4690a6d864b5bb5961b765047c402 : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9c46d4dd653f43a1904d06789018cbcc : SUCCESS in 21m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/caf38412099b4e47a933c180f5989c19 : SUCCESS in 1h 05m 20s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":53},{"id":"182dd768476b36a591fcef2f9bdcd229460b4062","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 19:23:56.000000000","message":"Patch Set 53:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a36a83872d8043319ad482dd32559c21\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d288244a5a614a22a04da060bf97fd03 : SUCCESS in 35m 15s","accounts_in_message":[],"_revision_number":53},{"id":"6a6a210fb33a4f7da2a37090a7e06a50e537d495","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-09-18 20:14:52.000000000","message":"Uploaded patch set 54: Patch Set 53 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":54},{"id":"f46dac68d50c47578e25c3d068bcc4de5e461e65","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 21:12:52.000000000","message":"Patch Set 54:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/494344b754a247299197b2e876a30923\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dc904642828442ac8b217b3066e6d2d8 : SUCCESS in 55m 48s","accounts_in_message":[],"_revision_number":54},{"id":"b1ed478cabbea2232e92b36a1ec127d1b3884650","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 21:36:46.000000000","message":"Patch Set 54: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/801d77fcdd89476fa2c0c15994cf4af2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/685992007e024719b1523b80fc1ab3a7 : SUCCESS in 11m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/398512bb5b4e4c799d2bd8caec46671d : SUCCESS in 9m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/c085d51b162a435abfd1c0f86ba8ac85 : SUCCESS in 45m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/675e213711b147d4896d3285f5d8450a : SUCCESS in 1h 01m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0aae291337b847a096f85ee68bfa088c : SUCCESS in 48m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b1e924a4ce2d46d39353b61c07ecfad4 : SUCCESS in 43m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a6e4982cdd9449b6bd021515e90d47fc : SUCCESS in 20m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/90b17d162163488d8ea868c562601daf : SUCCESS in 12m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0545c36a1939494cb255c6cd02ab0347 : SUCCESS in 21m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9f471bd64d8e47d5aad366e15edad9fd : SUCCESS in 6m 22s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/78b7f84be6954999be58fa5afbe9630a : SUCCESS in 26m 32s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/fcf9b29febf8478da555d8a932e7c79b : SUCCESS in 21m 07s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/fe1ff007531f4162bcfd1b91118c2a0d : SUCCESS in 34m 38s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/82c0ad0d98b540ba950d7dc4264a0d8e : SUCCESS in 19m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1dc998460e9d49778aa4738181056df3 : SUCCESS in 15m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e26958d3413c4248b6db315aabe3025d : SUCCESS in 15m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dc24ddd0aa994e038e9b59cd0b116272 : SUCCESS in 15m 16s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/95ac6c83af86477b9e220acdb505b920 : SUCCESS in 17m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c5175df7a66d48b2b5d8c2db1322a0c9 : SUCCESS in 6m 08s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5ea32d553fe842f69582ddbf9b8f25a2 : SUCCESS in 12m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb49d9919e8145f59aadd3f99896becb : SUCCESS in 1h 03m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ca8b8ef2c8f74abea95facb3df113e9e : SUCCESS in 52m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fa5cefa3f44a4736878c7dad53b8f839 : SUCCESS in 50m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02125a24718146cabc778440c71a2cf2 : SUCCESS in 12m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b0d0c4454344c3397e872c8f254852e : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/17d4a82edbbd48bb9ff62c7c5e1cfefb : SUCCESS in 19m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/164f2f1a268d4d70bfa37079613186d1 : SUCCESS in 56m 55s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":54}],"current_revision_number":54,"current_revision":"a016266c07c7d3554458d2896917bd1dd24d2e99","revisions":{"a5e8e1dc0e5b39f808c48905005504dcd41ec31e":{"kind":"REWORK","_number":1,"created":"2026-07-21 00:00:43.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/1"}}},"commit":{"parents":[{"commit":"c91ce9be05cedafbd3729832079585cc72a38f86","subject":"s3api: User Checksum Persistence for Full Object Uploads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c91ce9be05cedafbd3729832079585cc72a38f86"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:59:49.000000000","tz":-420},"subject":"slo: Add segment headers in slo_manifest_hook callback","message":"slo: Add segment headers in slo_manifest_hook callback\n\nReturn HEADs of segments in SLO manifest callback.\nThis is needed for checksum persistence on MPUs to avoid\nrepeated HEAD-ing of segments to gather system metadata.\n\nChange the SLO manifest callback hook signature to:\nhook(manifest, segment_head_headers)\n\nAssisted-by: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5e8e1dc0e5b39f808c48905005504dcd41ec31e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5e8e1dc0e5b39f808c48905005504dcd41ec31e"}]},"branch":"refs/heads/master"},"a363efdae189c94c6a1d7a58845ee7322997be31":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-07-21 15:58:44.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/2"}}},"commit":{"parents":[{"commit":"a8773ec5f0453328a6b1afa9b1d0bd1a0872ba52","subject":"s3api: User Checksum Persistence for Full Object Uploads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8773ec5f0453328a6b1afa9b1d0bd1a0872ba52"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-21 15:58:44.000000000","tz":0},"subject":"slo: Add segment headers in slo_manifest_hook callback","message":"slo: Add segment headers in slo_manifest_hook callback\n\nReturn HEADs of segments in SLO manifest callback.\nThis is needed for checksum persistence on MPUs to avoid\nrepeated HEAD-ing of segments to gather system metadata.\n\nChange the SLO manifest callback hook signature to:\nhook(manifest, segment_head_headers)\n\nAssisted-by: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a363efdae189c94c6a1d7a58845ee7322997be31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a363efdae189c94c6a1d7a58845ee7322997be31"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"c91ce9be05cedafbd3729832079585cc72a38f86","ours":"a5e8e1dc0e5b39f808c48905005504dcd41ec31e","theirs":"a8773ec5f0453328a6b1afa9b1d0bd1a0872ba52","merge_strategy":"recursive","contains_conflicts":false}},"3b3df33678fe4cc77631ec56735b85c0fe632365":{"kind":"REWORK","_number":3,"created":"2026-07-21 22:17:56.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/3"}}},"commit":{"parents":[{"commit":"a8773ec5f0453328a6b1afa9b1d0bd1a0872ba52","subject":"s3api: User Checksum Persistence for Full Object Uploads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8773ec5f0453328a6b1afa9b1d0bd1a0872ba52"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-21 22:16:17.000000000","tz":-420},"subject":"slo: Add segment headers in slo_manifest_hook callback","message":"slo: Add segment headers in slo_manifest_hook callback\n\nReturn HEADs of segments in SLO manifest callback.\nThis is needed for checksum persistence on MPUs to avoid\nrepeated HEAD-ing of segments to gather system metadata.\n\nChange the SLO manifest callback hook signature to now\ninclude \u0027headers\u0027 key with segment headers for callers to\nuse.\n\nAssisted-by: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3b3df33678fe4cc77631ec56735b85c0fe632365"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3b3df33678fe4cc77631ec56735b85c0fe632365"}]},"branch":"refs/heads/master"},"24dd6be03cb7267e8157367cdd6f61e0ad8f3f91":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-07-22 16:40:54.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/4"}}},"commit":{"parents":[{"commit":"d6bd5b38f367e70690be6178835908b61565a373","subject":"s3api: User Checksum Persistence for Full Object Uploads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6bd5b38f367e70690be6178835908b61565a373"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-22 16:40:54.000000000","tz":0},"subject":"slo: Add segment headers in slo_manifest_hook callback","message":"slo: Add segment headers in slo_manifest_hook callback\n\nReturn HEADs of segments in SLO manifest callback.\nThis is needed for checksum persistence on MPUs to avoid\nrepeated HEAD-ing of segments to gather system metadata.\n\nChange the SLO manifest callback hook signature to now\ninclude \u0027headers\u0027 key with segment headers for callers to\nuse.\n\nAssisted-by: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24dd6be03cb7267e8157367cdd6f61e0ad8f3f91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24dd6be03cb7267e8157367cdd6f61e0ad8f3f91"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"a8773ec5f0453328a6b1afa9b1d0bd1a0872ba52","ours":"3b3df33678fe4cc77631ec56735b85c0fe632365","theirs":"d6bd5b38f367e70690be6178835908b61565a373","merge_strategy":"recursive","contains_conflicts":false}},"09e0f79d526cf422f5122780850c73e982763569":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-07-22 18:26:15.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/5"}}},"commit":{"parents":[{"commit":"ba0979103ff5e43405891f05d200684f3527ab1f","subject":"s3api: User Checksum Persistence for Full Object Uploads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba0979103ff5e43405891f05d200684f3527ab1f"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-22 18:26:15.000000000","tz":0},"subject":"slo: Add segment headers in slo_manifest_hook callback","message":"slo: Add segment headers in slo_manifest_hook callback\n\nReturn HEADs of segments in SLO manifest callback.\nThis is needed for checksum persistence on MPUs to avoid\nrepeated HEAD-ing of segments to gather system metadata.\n\nChange the SLO manifest callback hook signature to now\ninclude \u0027headers\u0027 key with segment headers for callers to\nuse.\n\nAssisted-by: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09e0f79d526cf422f5122780850c73e982763569"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09e0f79d526cf422f5122780850c73e982763569"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"d6bd5b38f367e70690be6178835908b61565a373","ours":"24dd6be03cb7267e8157367cdd6f61e0ad8f3f91","theirs":"ba0979103ff5e43405891f05d200684f3527ab1f","merge_strategy":"recursive","contains_conflicts":false}},"a1b042b6a8ad050a7e3f06f049087572e96c8fe2":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-07-22 22:10:24.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/6"}}},"commit":{"parents":[{"commit":"af40d2205789c4df20dcad5d78b7319d35f9937f","subject":"s3api: User Checksum Persistence for Full Object Uploads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af40d2205789c4df20dcad5d78b7319d35f9937f"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-22 22:10:24.000000000","tz":0},"subject":"slo: Add segment headers in slo_manifest_hook callback","message":"slo: Add segment headers in slo_manifest_hook callback\n\nReturn HEADs of segments in SLO manifest callback.\nThis is needed for checksum persistence on MPUs to avoid\nrepeated HEAD-ing of segments to gather system metadata.\n\nChange the SLO manifest callback hook signature to now\ninclude \u0027headers\u0027 key with segment headers for callers to\nuse.\n\nAssisted-by: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1b042b6a8ad050a7e3f06f049087572e96c8fe2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1b042b6a8ad050a7e3f06f049087572e96c8fe2"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"ba0979103ff5e43405891f05d200684f3527ab1f","ours":"09e0f79d526cf422f5122780850c73e982763569","theirs":"af40d2205789c4df20dcad5d78b7319d35f9937f","merge_strategy":"recursive","contains_conflicts":false}},"52a5a9d3953c80cb45dc606b39e14ffb6c4252cb":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-07-24 19:00:34.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/7"}}},"commit":{"parents":[{"commit":"10b524acd843f58d34d2e39e1a2233a530212ffa","subject":"s3api: User Checksum Persistence for Full Object Uploads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10b524acd843f58d34d2e39e1a2233a530212ffa"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-24 19:00:34.000000000","tz":0},"subject":"slo: Add segment headers in slo_manifest_hook callback","message":"slo: Add segment headers in slo_manifest_hook callback\n\nReturn HEADs of segments in SLO manifest callback.\nThis is needed for checksum persistence on MPUs to avoid\nrepeated HEAD-ing of segments to gather system metadata.\n\nChange the SLO manifest callback hook signature to now\ninclude \u0027headers\u0027 key with segment headers for callers to\nuse.\n\nAssisted-by: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52a5a9d3953c80cb45dc606b39e14ffb6c4252cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52a5a9d3953c80cb45dc606b39e14ffb6c4252cb"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"af40d2205789c4df20dcad5d78b7319d35f9937f","ours":"a1b042b6a8ad050a7e3f06f049087572e96c8fe2","theirs":"10b524acd843f58d34d2e39e1a2233a530212ffa","merge_strategy":"recursive","contains_conflicts":false}},"d8aa4c63338212b591604b70422cb4a1ea6ce4ca":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-07-27 17:04:37.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/8"}}},"commit":{"parents":[{"commit":"56dc229db2fdb567fff1275843c5754dc3f66f4f","subject":"s3api: User Checksum Persistence for Full Object Uploads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/56dc229db2fdb567fff1275843c5754dc3f66f4f"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-27 17:04:37.000000000","tz":0},"subject":"slo: Add segment headers in slo_manifest_hook callback","message":"slo: Add segment headers in slo_manifest_hook callback\n\nReturn HEADs of segments in SLO manifest callback.\nThis is needed for checksum persistence on MPUs to avoid\nrepeated HEAD-ing of segments to gather system metadata.\n\nChange the SLO manifest callback hook signature to now\ninclude \u0027headers\u0027 key with segment headers for callers to\nuse.\n\nAssisted-by: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d8aa4c63338212b591604b70422cb4a1ea6ce4ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d8aa4c63338212b591604b70422cb4a1ea6ce4ca"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"10b524acd843f58d34d2e39e1a2233a530212ffa","ours":"52a5a9d3953c80cb45dc606b39e14ffb6c4252cb","theirs":"56dc229db2fdb567fff1275843c5754dc3f66f4f","merge_strategy":"recursive","contains_conflicts":false}},"dd808b5a3508a22ad48497c1af38f74c487531f1":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-07-27 17:08:09.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/9"}}},"commit":{"parents":[{"commit":"d743a78d5cfdaa40a71977813f584f8da044ebbc","subject":"s3api: User Checksum Persistence for Full Object Uploads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d743a78d5cfdaa40a71977813f584f8da044ebbc"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-27 17:08:09.000000000","tz":0},"subject":"slo: Add segment headers in slo_manifest_hook callback","message":"slo: Add segment headers in slo_manifest_hook callback\n\nReturn HEADs of segments in SLO manifest callback.\nThis is needed for checksum persistence on MPUs to avoid\nrepeated HEAD-ing of segments to gather system metadata.\n\nChange the SLO manifest callback hook signature to now\ninclude \u0027headers\u0027 key with segment headers for callers to\nuse.\n\nAssisted-by: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd808b5a3508a22ad48497c1af38f74c487531f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd808b5a3508a22ad48497c1af38f74c487531f1"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"56dc229db2fdb567fff1275843c5754dc3f66f4f","ours":"d8aa4c63338212b591604b70422cb4a1ea6ce4ca","theirs":"d743a78d5cfdaa40a71977813f584f8da044ebbc","merge_strategy":"recursive","contains_conflicts":false}},"80a52e7c09eef7bebf0fd99d026e011739c3120b":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-07-27 21:16:18.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/10"}}},"commit":{"parents":[{"commit":"bb8e143a96bb9a453ddd3c84d25a77a478f19d68","subject":"s3api: User Checksum Persistence for Full Object Uploads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb8e143a96bb9a453ddd3c84d25a77a478f19d68"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-27 21:16:18.000000000","tz":0},"subject":"slo: Add segment headers in slo_manifest_hook callback","message":"slo: Add segment headers in slo_manifest_hook callback\n\nReturn HEADs of segments in SLO manifest callback.\nThis is needed for checksum persistence on MPUs to avoid\nrepeated HEAD-ing of segments to gather system metadata.\n\nChange the SLO manifest callback hook signature to now\ninclude \u0027headers\u0027 key with segment headers for callers to\nuse.\n\nAssisted-by: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/80a52e7c09eef7bebf0fd99d026e011739c3120b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/80a52e7c09eef7bebf0fd99d026e011739c3120b"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"d743a78d5cfdaa40a71977813f584f8da044ebbc","ours":"dd808b5a3508a22ad48497c1af38f74c487531f1","theirs":"bb8e143a96bb9a453ddd3c84d25a77a478f19d68","merge_strategy":"recursive","contains_conflicts":false}},"291eb176626e9254237a3db1549184fb0c115679":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-07-28 17:29:04.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/11"}}},"commit":{"parents":[{"commit":"fd71876e221355c1a8865a14bde20ac0eb8dbefc","subject":"s3api: User Checksum Persistence for Full Object Uploads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd71876e221355c1a8865a14bde20ac0eb8dbefc"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-28 17:29:04.000000000","tz":0},"subject":"slo: Add segment headers in slo_manifest_hook callback","message":"slo: Add segment headers in slo_manifest_hook callback\n\nReturn HEADs of segments in SLO manifest callback.\nThis is needed for checksum persistence on MPUs to avoid\nrepeated HEAD-ing of segments to gather system metadata.\n\nChange the SLO manifest callback hook signature to now\ninclude \u0027headers\u0027 key with segment headers for callers to\nuse.\n\nAssisted-by: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/291eb176626e9254237a3db1549184fb0c115679"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/291eb176626e9254237a3db1549184fb0c115679"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"bb8e143a96bb9a453ddd3c84d25a77a478f19d68","ours":"80a52e7c09eef7bebf0fd99d026e011739c3120b","theirs":"fd71876e221355c1a8865a14bde20ac0eb8dbefc","merge_strategy":"recursive","contains_conflicts":false}},"b30b99d0bafa80855b944dbbca56c6fe42354eeb":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-07-28 23:23:20.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/12"}}},"commit":{"parents":[{"commit":"e494dac18468cca9dc2f4d5216b0e3ecd6a170f8","subject":"s3api: User Checksum Persistence for Full Object Uploads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e494dac18468cca9dc2f4d5216b0e3ecd6a170f8"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-28 23:23:20.000000000","tz":0},"subject":"slo: Add segment headers in slo_manifest_hook callback","message":"slo: Add segment headers in slo_manifest_hook callback\n\nReturn HEADs of segments in SLO manifest callback.\nThis is needed for checksum persistence on MPUs to avoid\nrepeated HEAD-ing of segments to gather system metadata.\n\nChange the SLO manifest callback hook signature to now\ninclude \u0027headers\u0027 key with segment headers for callers to\nuse.\n\nAssisted-by: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b30b99d0bafa80855b944dbbca56c6fe42354eeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b30b99d0bafa80855b944dbbca56c6fe42354eeb"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"fd71876e221355c1a8865a14bde20ac0eb8dbefc","ours":"291eb176626e9254237a3db1549184fb0c115679","theirs":"e494dac18468cca9dc2f4d5216b0e3ecd6a170f8","merge_strategy":"recursive","contains_conflicts":false}},"50222e47d880de59495c7146dd6982359e51fe60":{"kind":"REWORK","_number":13,"created":"2026-07-29 16:20:11.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/13"}}},"commit":{"parents":[{"commit":"e494dac18468cca9dc2f4d5216b0e3ecd6a170f8","subject":"s3api: User Checksum Persistence for Full Object Uploads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e494dac18468cca9dc2f4d5216b0e3ecd6a170f8"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-29 15:55:56.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a transient headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a\nHeaderKeyDict of the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, copy each entry and remove the transient\nheaders key so backend response metadata does not become part of the SLO\nmanifest.\n\nAssisted-By: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50222e47d880de59495c7146dd6982359e51fe60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50222e47d880de59495c7146dd6982359e51fe60"}]},"branch":"refs/heads/master"},"459b93c3315e96bf4135f3bb84018d34d455fff8":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-07-29 18:20:47.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/14"}}},"commit":{"parents":[{"commit":"05b475672c3576c886939d6426e36416a2f7a392","subject":"s3api: User Checksum Persistence for Full Object Uploads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05b475672c3576c886939d6426e36416a2f7a392"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-29 18:20:47.000000000","tz":0},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a transient headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a\nHeaderKeyDict of the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, copy each entry and remove the transient\nheaders key so backend response metadata does not become part of the SLO\nmanifest.\n\nAssisted-By: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/459b93c3315e96bf4135f3bb84018d34d455fff8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/459b93c3315e96bf4135f3bb84018d34d455fff8"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"e494dac18468cca9dc2f4d5216b0e3ecd6a170f8","ours":"50222e47d880de59495c7146dd6982359e51fe60","theirs":"05b475672c3576c886939d6426e36416a2f7a392","merge_strategy":"recursive","contains_conflicts":false}},"73a05c19e91253fc7df940f683e8d8642d2e4c88":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2026-07-31 19:55:34.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/15"}}},"commit":{"parents":[{"commit":"02ec48df59c892c5bfd0f2e8cd835d6c2821ed8a","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/02ec48df59c892c5bfd0f2e8cd835d6c2821ed8a"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-30 22:06:02.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a transient headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a\nHeaderKeyDict of the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, copy each entry and remove the transient\nheaders key so backend response metadata does not become part of the SLO\nmanifest.\n\nAssisted-By: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73a05c19e91253fc7df940f683e8d8642d2e4c88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73a05c19e91253fc7df940f683e8d8642d2e4c88"}]},"branch":"refs/heads/master"},"ca60252c2c5bddd4e7a35809873bb4262a6ff7a0":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2026-08-03 18:38:22.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/16"}}},"commit":{"parents":[{"commit":"36e0e9eef0b3db84c56164548a8ecd728dc13c48","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/36e0e9eef0b3db84c56164548a8ecd728dc13c48"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-03 18:38:22.000000000","tz":0},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a transient headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a\nHeaderKeyDict of the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, copy each entry and remove the transient\nheaders key so backend response metadata does not become part of the SLO\nmanifest.\n\nAssisted-By: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca60252c2c5bddd4e7a35809873bb4262a6ff7a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca60252c2c5bddd4e7a35809873bb4262a6ff7a0"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"02ec48df59c892c5bfd0f2e8cd835d6c2821ed8a","ours":"73a05c19e91253fc7df940f683e8d8642d2e4c88","theirs":"36e0e9eef0b3db84c56164548a8ecd728dc13c48","merge_strategy":"recursive","contains_conflicts":false}},"ad442c4ba5a3f2a43f7e7711f6cc2724d4453bb0":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2026-08-04 18:18:30.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/17"}}},"commit":{"parents":[{"commit":"a97af4bb27513edee2d57109fed746833202d590","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a97af4bb27513edee2d57109fed746833202d590"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-04 18:18:30.000000000","tz":0},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a transient headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a\nHeaderKeyDict of the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, copy each entry and remove the transient\nheaders key so backend response metadata does not become part of the SLO\nmanifest.\n\nAssisted-By: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad442c4ba5a3f2a43f7e7711f6cc2724d4453bb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad442c4ba5a3f2a43f7e7711f6cc2724d4453bb0"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"36e0e9eef0b3db84c56164548a8ecd728dc13c48","ours":"ca60252c2c5bddd4e7a35809873bb4262a6ff7a0","theirs":"a97af4bb27513edee2d57109fed746833202d590","merge_strategy":"recursive","contains_conflicts":false}},"0836e1674b357ebc26359692776f103213f02e28":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2026-08-04 21:14:23.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/18"}}},"commit":{"parents":[{"commit":"40278adf6879f6f19aa70f311886223b63592afd","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40278adf6879f6f19aa70f311886223b63592afd"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-04 21:14:23.000000000","tz":0},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a transient headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a\nHeaderKeyDict of the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, copy each entry and remove the transient\nheaders key so backend response metadata does not become part of the SLO\nmanifest.\n\nAssisted-By: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0836e1674b357ebc26359692776f103213f02e28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0836e1674b357ebc26359692776f103213f02e28"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"a97af4bb27513edee2d57109fed746833202d590","ours":"ad442c4ba5a3f2a43f7e7711f6cc2724d4453bb0","theirs":"40278adf6879f6f19aa70f311886223b63592afd","merge_strategy":"recursive","contains_conflicts":false}},"63abedeb5af11bb5def77e6e275d3909f65183e9":{"kind":"NO_CHANGE","_number":19,"created":"2026-08-04 21:14:56.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/19"}}},"commit":{"parents":[{"commit":"40278adf6879f6f19aa70f311886223b63592afd","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40278adf6879f6f19aa70f311886223b63592afd"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-04 20:31:04.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a transient headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a\nHeaderKeyDict of the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, copy each entry and remove the transient\nheaders key so backend response metadata does not become part of the SLO\nmanifest.\n\nAssisted-By: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63abedeb5af11bb5def77e6e275d3909f65183e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63abedeb5af11bb5def77e6e275d3909f65183e9"}]},"branch":"refs/heads/master"},"b15aca1d76299e1ac64b1ac51f458ebbc444d8fc":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2026-08-04 23:42:26.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/20"}}},"commit":{"parents":[{"commit":"d08b5c3c3c8b5d837bcf07eb01fe5eecdfb13058","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d08b5c3c3c8b5d837bcf07eb01fe5eecdfb13058"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-04 23:42:26.000000000","tz":0},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a transient headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a\nHeaderKeyDict of the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, copy each entry and remove the transient\nheaders key so backend response metadata does not become part of the SLO\nmanifest.\n\nAssisted-By: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b15aca1d76299e1ac64b1ac51f458ebbc444d8fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b15aca1d76299e1ac64b1ac51f458ebbc444d8fc"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"40278adf6879f6f19aa70f311886223b63592afd","ours":"63abedeb5af11bb5def77e6e275d3909f65183e9","theirs":"d08b5c3c3c8b5d837bcf07eb01fe5eecdfb13058","merge_strategy":"recursive","contains_conflicts":false}},"0ca4a2ba71063bd9986d54fa5b92a5a2d6b5f2ce":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2026-08-06 22:27:20.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/21"}}},"commit":{"parents":[{"commit":"a2ae5d5ac01e9a4666de3ea1b9f52d38773c6e26","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2ae5d5ac01e9a4666de3ea1b9f52d38773c6e26"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-06 22:25:20.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a transient headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a\nHeaderKeyDict of the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, copy each entry and remove the transient\nheaders key so backend response metadata does not become part of the SLO\nmanifest.\n\nAssisted-By: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0ca4a2ba71063bd9986d54fa5b92a5a2d6b5f2ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0ca4a2ba71063bd9986d54fa5b92a5a2d6b5f2ce"}]},"branch":"refs/heads/master"},"3d306cbf558658121dc8c828a3f8101e41658074":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2026-08-06 23:28:35.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/22"}}},"commit":{"parents":[{"commit":"8af38e1088aa48356869acdeb428083915fcd625","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8af38e1088aa48356869acdeb428083915fcd625"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-06 23:28:35.000000000","tz":0},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a transient headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a\nHeaderKeyDict of the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, copy each entry and remove the transient\nheaders key so backend response metadata does not become part of the SLO\nmanifest.\n\nAssisted-By: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d306cbf558658121dc8c828a3f8101e41658074"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d306cbf558658121dc8c828a3f8101e41658074"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"a2ae5d5ac01e9a4666de3ea1b9f52d38773c6e26","ours":"0ca4a2ba71063bd9986d54fa5b92a5a2d6b5f2ce","theirs":"8af38e1088aa48356869acdeb428083915fcd625","merge_strategy":"recursive","contains_conflicts":false}},"37ccf9a8cb5da449e649d65feec6a7a1535356fd":{"kind":"REWORK","_number":23,"created":"2026-08-07 18:32:32.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/23"}}},"commit":{"parents":[{"commit":"8af38e1088aa48356869acdeb428083915fcd625","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8af38e1088aa48356869acdeb428083915fcd625"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-07 18:32:11.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a transient headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a\nHeaderKeyDict of the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, copy each entry and remove the transient\nheaders key so backend response metadata does not become part of the SLO\nmanifest.\n\nAssisted-by: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37ccf9a8cb5da449e649d65feec6a7a1535356fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37ccf9a8cb5da449e649d65feec6a7a1535356fd"}]},"branch":"refs/heads/master"},"d13d083f37d7826ba7af0d230bcc0aa3b6e430bf":{"kind":"REWORK","_number":24,"created":"2026-08-07 20:16:29.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/24"}}},"commit":{"parents":[{"commit":"8af38e1088aa48356869acdeb428083915fcd625","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8af38e1088aa48356869acdeb428083915fcd625"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-07 20:16:14.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a transient headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a\nHeaderKeyDict of the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, copy each entry and remove the transient\nheaders key so backend response metadata does not become part of the SLO\nmanifest.\n\nAssisted-by: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d13d083f37d7826ba7af0d230bcc0aa3b6e430bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d13d083f37d7826ba7af0d230bcc0aa3b6e430bf"}]},"branch":"refs/heads/master"},"d7fba28c12fed7013e3dcbbd7aaaf227061806ea":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2026-08-07 20:16:48.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/25"}}},"commit":{"parents":[{"commit":"2a399a7328c6410517d78baa0662023103d8daa2","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a399a7328c6410517d78baa0662023103d8daa2"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-07 20:16:48.000000000","tz":0},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a transient headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a\nHeaderKeyDict of the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, copy each entry and remove the transient\nheaders key so backend response metadata does not become part of the SLO\nmanifest.\n\nAssisted-by: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7fba28c12fed7013e3dcbbd7aaaf227061806ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7fba28c12fed7013e3dcbbd7aaaf227061806ea"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"8af38e1088aa48356869acdeb428083915fcd625","ours":"d13d083f37d7826ba7af0d230bcc0aa3b6e430bf","theirs":"2a399a7328c6410517d78baa0662023103d8daa2","merge_strategy":"recursive","contains_conflicts":false}},"3af7839c7d78e61f0f9bcafafb7c1e9ae3f8b0f8":{"kind":"REWORK","_number":26,"created":"2026-08-07 20:54:24.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/26"}}},"commit":{"parents":[{"commit":"160dc4df3597d63c10e26e0985fedd5d6e948872","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/160dc4df3597d63c10e26e0985fedd5d6e948872"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-07 20:32:30.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a transient headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a\nHeaderKeyDict of the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, copy each entry and remove the transient\nheaders key so backend response metadata does not become part of the SLO\nmanifest.\n\nAssisted-By: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3af7839c7d78e61f0f9bcafafb7c1e9ae3f8b0f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3af7839c7d78e61f0f9bcafafb7c1e9ae3f8b0f8"}]},"branch":"refs/heads/master"},"82f6c0e2bedbefe170c4736c5795f9a6696baacf":{"kind":"REWORK","_number":27,"created":"2026-08-09 23:52:46.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/27"}}},"commit":{"parents":[{"commit":"5019ee3b720270fec440c05b2171681059197127","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5019ee3b720270fec440c05b2171681059197127"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-09 22:55:13.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a transient headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a\nHeaderKeyDict of the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, copy each entry and remove the transient\nheaders key so backend response metadata does not become part of the SLO\nmanifest.\n\nAssisted-by: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82f6c0e2bedbefe170c4736c5795f9a6696baacf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82f6c0e2bedbefe170c4736c5795f9a6696baacf"}]},"branch":"refs/heads/master"},"583830498dea2fa97395e47adf8ed1b7d6d5a66b":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2026-08-11 00:51:03.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/28"}}},"commit":{"parents":[{"commit":"3839e321ad05aa4c419f7b0e9558e75482f9072e","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3839e321ad05aa4c419f7b0e9558e75482f9072e"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-10 23:41:42.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a transient headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a\nHeaderKeyDict of the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, copy each entry and remove the transient\nheaders key so backend response metadata does not become part of the SLO\nmanifest.\n\nAssisted-by: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/583830498dea2fa97395e47adf8ed1b7d6d5a66b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/583830498dea2fa97395e47adf8ed1b7d6d5a66b"}]},"branch":"refs/heads/master"},"0b95488b21e984bd9b5aed9bc7c18be39648d83f":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":29,"created":"2026-08-11 19:39:36.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/29"}}},"commit":{"parents":[{"commit":"f17e481e43ed67aa12c4a11be3ae2ee5386ff0d9","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f17e481e43ed67aa12c4a11be3ae2ee5386ff0d9"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-11 19:38:59.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a validation-only headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a HeaderKeyDict\nof the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, remove the validation-only HEAD headers\nso backend response metadata does not become part of the SLO manifest.\n\nAssisted-by: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b95488b21e984bd9b5aed9bc7c18be39648d83f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b95488b21e984bd9b5aed9bc7c18be39648d83f"}]},"branch":"refs/heads/master"},"46528d666a326311b1aa54df1279a22a89f94e57":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2026-08-14 23:08:13.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/30"}}},"commit":{"parents":[{"commit":"88291f582c42692be2f5733571ea33e22ec8b706","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88291f582c42692be2f5733571ea33e22ec8b706"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-14 23:07:01.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a validation-only headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a HeaderKeyDict\nof the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, remove the validation-only HEAD headers\nso backend response metadata does not become part of the SLO manifest.\n\nAssisted-by: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46528d666a326311b1aa54df1279a22a89f94e57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46528d666a326311b1aa54df1279a22a89f94e57"}]},"branch":"refs/heads/master"},"e758426e98113d204d8c2bc891f7759ce27492af":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2026-08-17 15:09:49.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/31"}}},"commit":{"parents":[{"commit":"92394d76feeaa7e9ce373f4a10bec83b8540788b","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92394d76feeaa7e9ce373f4a10bec83b8540788b"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-15 04:41:55.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a validation-only headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a HeaderKeyDict\nof the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, remove the validation-only HEAD headers\nso backend response metadata does not become part of the SLO manifest.\n\nAssisted-by: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e758426e98113d204d8c2bc891f7759ce27492af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e758426e98113d204d8c2bc891f7759ce27492af"}]},"branch":"refs/heads/master"},"9d407680236be34b04ac6171e06766621fb6d43d":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2026-08-18 18:51:39.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/32"}}},"commit":{"parents":[{"commit":"1db6f31c25152c662b2f4e07dd48a36c7b0fe536","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1db6f31c25152c662b2f4e07dd48a36c7b0fe536"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-18 17:11:33.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a validation-only headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a HeaderKeyDict\nof the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, remove the validation-only HEAD headers\nso backend response metadata does not become part of the SLO manifest.\n\nAssisted-by: Codex:gpt-5.5\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d407680236be34b04ac6171e06766621fb6d43d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d407680236be34b04ac6171e06766621fb6d43d"}]},"branch":"refs/heads/master"},"0975321af63473ebaf7a27af57ac177cbcab2c9f":{"kind":"NO_CODE_CHANGE","_number":33,"created":"2026-08-18 22:32:30.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/33"}}},"commit":{"parents":[{"commit":"e8acce26182be32b91299e478f5bc3c8f9a7ec15","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e8acce26182be32b91299e478f5bc3c8f9a7ec15"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-18 22:25:15.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a validation-only headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a HeaderKeyDict\nof the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, remove the validation-only HEAD headers\nso backend response metadata does not become part of the SLO manifest.\n\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nAssisted-By: Codex:gpt-5.5\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0975321af63473ebaf7a27af57ac177cbcab2c9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0975321af63473ebaf7a27af57ac177cbcab2c9f"}]},"branch":"refs/heads/master"},"f77b6b3cfda0b221886f77c7ea0bfcb8662158bd":{"kind":"TRIVIAL_REBASE","_number":34,"created":"2026-08-19 22:49:20.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/34"}}},"commit":{"parents":[{"commit":"9775ca346ed14aec2b73086e0bf702ee56d015a2","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9775ca346ed14aec2b73086e0bf702ee56d015a2"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-19 18:42:39.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a validation-only headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a HeaderKeyDict\nof the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, remove the validation-only HEAD headers\nso backend response metadata does not become part of the SLO manifest.\n\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nAssisted-By: Codex:gpt-5.5\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f77b6b3cfda0b221886f77c7ea0bfcb8662158bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f77b6b3cfda0b221886f77c7ea0bfcb8662158bd"}]},"branch":"refs/heads/master"},"fb5fa1b7907f4348beedabfff25bddd70b934010":{"kind":"TRIVIAL_REBASE","_number":35,"created":"2026-08-20 15:48:41.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/35"}}},"commit":{"parents":[{"commit":"9323c90d9b7cde08d79147e3c56aff22bb153b3b","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9323c90d9b7cde08d79147e3c56aff22bb153b3b"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-20 15:43:44.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a validation-only headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a HeaderKeyDict\nof the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, remove the validation-only HEAD headers\nso backend response metadata does not become part of the SLO manifest.\n\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nAssisted-By: Codex:gpt-5.5\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb5fa1b7907f4348beedabfff25bddd70b934010"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb5fa1b7907f4348beedabfff25bddd70b934010"}]},"branch":"refs/heads/master"},"952bf92bc9b6a3d41b310f78be78273bddb0c6a1":{"kind":"TRIVIAL_REBASE","_number":36,"created":"2026-08-21 21:22:06.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/36"}}},"commit":{"parents":[{"commit":"7542bb869409b958ebd3053017d40bc960223ded","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7542bb869409b958ebd3053017d40bc960223ded"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-21 21:15:48.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a validation-only headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a HeaderKeyDict\nof the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, remove the validation-only HEAD headers\nso backend response metadata does not become part of the SLO manifest.\n\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nAssisted-By: Codex:gpt-5.5\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/952bf92bc9b6a3d41b310f78be78273bddb0c6a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/952bf92bc9b6a3d41b310f78be78273bddb0c6a1"}]},"branch":"refs/heads/master"},"c6c7ebe2177963589fe0ff0e4d96c0df33ac4278":{"kind":"TRIVIAL_REBASE","_number":37,"created":"2026-08-24 15:47:51.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/37"}}},"commit":{"parents":[{"commit":"c381f52e5307342218690eed96fe57c799e1306b","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c381f52e5307342218690eed96fe57c799e1306b"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-24 15:26:36.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a validation-only headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a HeaderKeyDict\nof the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, remove the validation-only HEAD headers\nso backend response metadata does not become part of the SLO manifest.\n\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nAssisted-By: Codex:gpt-5.5\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6c7ebe2177963589fe0ff0e4d96c0df33ac4278"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6c7ebe2177963589fe0ff0e4d96c0df33ac4278"}]},"branch":"refs/heads/master"},"4b11615837ea1c0fdbcbcad58405eed0cef49054":{"kind":"TRIVIAL_REBASE","_number":38,"created":"2026-08-24 22:45:24.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/38"}}},"commit":{"parents":[{"commit":"e9f5bdea19664daf9208ae1a1c2851af82a12a0c","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9f5bdea19664daf9208ae1a1c2851af82a12a0c"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-24 21:56:32.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a validation-only headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a HeaderKeyDict\nof the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, remove the validation-only HEAD headers\nso backend response metadata does not become part of the SLO manifest.\n\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nAssisted-By: Codex:gpt-5.5\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b11615837ea1c0fdbcbcad58405eed0cef49054"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b11615837ea1c0fdbcbcad58405eed0cef49054"}]},"branch":"refs/heads/master"},"ac73a9488adf27bf4c3705c0772f67606258a5a1":{"kind":"TRIVIAL_REBASE","_number":39,"created":"2026-08-24 23:45:12.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/39"}}},"commit":{"parents":[{"commit":"01988bafd63c5584de964956f2701ec717c3199b","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/01988bafd63c5584de964956f2701ec717c3199b"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-24 23:41:57.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a validation-only headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a HeaderKeyDict\nof the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, remove the validation-only HEAD headers\nso backend response metadata does not become part of the SLO manifest.\n\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nAssisted-By: Codex:gpt-5.5\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac73a9488adf27bf4c3705c0772f67606258a5a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac73a9488adf27bf4c3705c0772f67606258a5a1"}]},"branch":"refs/heads/master"},"3ad4c5f6c03eb3f5e6b4f30cd2be9475762463c1":{"kind":"TRIVIAL_REBASE","_number":40,"created":"2026-08-25 20:01:23.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/40"}}},"commit":{"parents":[{"commit":"3efbd20e1996455bcbad415a2b7794789cac731d","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3efbd20e1996455bcbad415a2b7794789cac731d"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-25 20:00:17.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a validation-only headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a HeaderKeyDict\nof the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, remove the validation-only HEAD headers\nso backend response metadata does not become part of the SLO manifest.\n\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nAssisted-By: Codex:gpt-5.5\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ad4c5f6c03eb3f5e6b4f30cd2be9475762463c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ad4c5f6c03eb3f5e6b4f30cd2be9475762463c1"}]},"branch":"refs/heads/master"},"1fa76ee74da492a4b5a13fa0c1066f5477091068":{"kind":"TRIVIAL_REBASE","_number":41,"created":"2026-08-25 23:29:14.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/41"}}},"commit":{"parents":[{"commit":"af0e6818c5f54d1ff6e1782e3ae042d64e36709e","subject":"Merge \"Clarify s3api.utils sysmeta helper names\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af0e6818c5f54d1ff6e1782e3ae042d64e36709e"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-25 23:29:09.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a validation-only headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a HeaderKeyDict\nof the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, remove the validation-only HEAD headers\nso backend response metadata does not become part of the SLO manifest.\n\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nAssisted-By: Codex:gpt-5.5\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1fa76ee74da492a4b5a13fa0c1066f5477091068"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1fa76ee74da492a4b5a13fa0c1066f5477091068"}]},"branch":"refs/heads/master"},"1ce802a9f0d6f72e08cc0d1a505d54450aee357d":{"kind":"TRIVIAL_REBASE","_number":42,"created":"2026-08-26 02:01:11.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/42"}}},"commit":{"parents":[{"commit":"efaa4264263ff095b01e0b19d5e571c684b925fe","subject":"s3api: don\u0027t base64-validate checksum values on POST","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/efaa4264263ff095b01e0b19d5e571c684b925fe"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-26 01:59:02.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a validation-only headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a HeaderKeyDict\nof the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, remove the validation-only HEAD headers\nso backend response metadata does not become part of the SLO manifest.\n\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nAssisted-By: Codex:gpt-5.5\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ce802a9f0d6f72e08cc0d1a505d54450aee357d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ce802a9f0d6f72e08cc0d1a505d54450aee357d"}]},"branch":"refs/heads/master"},"d7abe84093c31a630e150928915b6ce72d4e169d":{"kind":"TRIVIAL_REBASE","_number":43,"created":"2026-08-26 03:28:02.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/43"}}},"commit":{"parents":[{"commit":"d603531dd002520d4dd4e3fe641d2fafd8568b3b","subject":"s3api: don\u0027t base64-validate checksum values on POST","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d603531dd002520d4dd4e3fe641d2fafd8568b3b"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-26 03:04:01.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a validation-only headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a HeaderKeyDict\nof the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, remove the validation-only HEAD headers\nso backend response metadata does not become part of the SLO manifest.\n\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nAssisted-By: Codex:gpt-5.5\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7abe84093c31a630e150928915b6ce72d4e169d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7abe84093c31a630e150928915b6ce72d4e169d"}]},"branch":"refs/heads/master"},"d7de6f32cf39a917204f8da2c8b25ed604969d7a":{"kind":"TRIVIAL_REBASE","_number":44,"created":"2026-08-27 05:56:10.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/44"}}},"commit":{"parents":[{"commit":"1b6b56bd697155016ab2fb3ff2e9ed3e84e82223","subject":"s3api: don\u0027t base64-validate checksum values on POST","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b6b56bd697155016ab2fb3ff2e9ed3e84e82223"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-27 05:19:19.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a validation-only headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a HeaderKeyDict\nof the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, remove the validation-only HEAD headers\nso backend response metadata does not become part of the SLO manifest.\n\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nAssisted-By: Codex:gpt-5.5\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7de6f32cf39a917204f8da2c8b25ed604969d7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7de6f32cf39a917204f8da2c8b25ed604969d7a"}]},"branch":"refs/heads/master"},"c7c9d89a68bac5af667516707b0ba61970bca2f1":{"kind":"TRIVIAL_REBASE","_number":45,"created":"2026-08-27 06:36:39.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/45"}}},"commit":{"parents":[{"commit":"3dc364ce62f8b43dde380c9c064ee7098f1b9b60","subject":"s3api: don\u0027t base64-validate checksum values on POST","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3dc364ce62f8b43dde380c9c064ee7098f1b9b60"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-27 06:14:23.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a validation-only headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a HeaderKeyDict\nof the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, remove the validation-only HEAD headers\nso backend response metadata does not become part of the SLO manifest.\n\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nAssisted-By: Codex:gpt-5.5\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7c9d89a68bac5af667516707b0ba61970bca2f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7c9d89a68bac5af667516707b0ba61970bca2f1"}]},"branch":"refs/heads/master"},"09a4bc1dc7b69321bbe59797d3b494717c623838":{"kind":"TRIVIAL_REBASE","_number":46,"created":"2026-08-27 13:56:13.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/46"}}},"commit":{"parents":[{"commit":"62e4431de37cd3f85d71d535f7fdbfb9eb5742d5","subject":"s3api: don\u0027t base64-validate checksum values on POST","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/62e4431de37cd3f85d71d535f7fdbfb9eb5742d5"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-27 13:56:13.000000000","tz":0},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a validation-only headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a HeaderKeyDict\nof the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, remove the validation-only HEAD headers\nso backend response metadata does not become part of the SLO manifest.\n\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nAssisted-By: Codex:gpt-5.5\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09a4bc1dc7b69321bbe59797d3b494717c623838"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09a4bc1dc7b69321bbe59797d3b494717c623838"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"3dc364ce62f8b43dde380c9c064ee7098f1b9b60","ours":"c7c9d89a68bac5af667516707b0ba61970bca2f1","theirs":"62e4431de37cd3f85d71d535f7fdbfb9eb5742d5","merge_strategy":"recursive","contains_conflicts":false}},"77e558675081d8c76b9d07c523d7661a02e6ca6a":{"kind":"REWORK","_number":47,"created":"2026-08-27 21:57:30.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/72/998072/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/47"}}},"commit":{"parents":[{"commit":"7c9065ef5131cb1639b89d6f02086044d292c6ea","subject":"s3api: don\u0027t base64-validate checksum values on POST","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c9065ef5131cb1639b89d6f02086044d292c6ea"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-27 20:38:16.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a validation-only headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a HeaderKeyDict\nof the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, remove the validation-only HEAD headers\nso backend response metadata does not become part of the SLO manifest.\n\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nAssisted-By: Codex:gpt-5.5\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/77e558675081d8c76b9d07c523d7661a02e6ca6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/77e558675081d8c76b9d07c523d7661a02e6ca6a"}]},"branch":"refs/heads/master"},"9990f99d36e05fa2f80b7fe392910839b4277025":{"kind":"TRIVIAL_REBASE","_number":48,"created":"2026-08-28 11:59:22.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/72/998072/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/48"}}},"commit":{"parents":[{"commit":"42cb00e8b8621484240283a6dd908e19ce7c4e45","subject":"utils/checksum: Add combine to CRCHasher","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42cb00e8b8621484240283a6dd908e19ce7c4e45"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-28 11:53:22.000000000","tz":60},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a validation-only headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a HeaderKeyDict\nof the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, remove the validation-only HEAD headers\nso backend response metadata does not become part of the SLO manifest.\n\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nAssisted-By: Codex:gpt-5.5\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9990f99d36e05fa2f80b7fe392910839b4277025"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9990f99d36e05fa2f80b7fe392910839b4277025"}]},"branch":"refs/heads/master"},"82de2ba34e17ef3c276f8fe5b372101887e46845":{"kind":"TRIVIAL_REBASE","_number":49,"created":"2026-09-06 15:46:35.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/72/998072/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/49"}}},"commit":{"parents":[{"commit":"53cb971c4f5313b5c133f8f54371a49bd00734a2","subject":"utils/checksum: Add combine to CRCHasher","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/53cb971c4f5313b5c133f8f54371a49bd00734a2"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-09-06 04:08:56.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a validation-only headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a HeaderKeyDict\nof the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, remove the validation-only HEAD headers\nso backend response metadata does not become part of the SLO manifest.\n\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nAssisted-By: Codex:gpt-5.5\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82de2ba34e17ef3c276f8fe5b372101887e46845"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82de2ba34e17ef3c276f8fe5b372101887e46845"}]},"branch":"refs/heads/master"},"df7612755c5e62d472ef1505fb3d6347fc363605":{"kind":"TRIVIAL_REBASE","_number":50,"created":"2026-09-10 22:53:39.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/72/998072/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/50"}}},"commit":{"parents":[{"commit":"3fafe822ca6a798581a4daec6198af07d92c6788","subject":"utils/checksum: Add combine to CRCHasher","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3fafe822ca6a798581a4daec6198af07d92c6788"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-10 22:25:26.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a validation-only headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a HeaderKeyDict\nof the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, remove the validation-only HEAD headers\nso backend response metadata does not become part of the SLO manifest.\n\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nAssisted-By: Codex:gpt-5.5\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df7612755c5e62d472ef1505fb3d6347fc363605"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df7612755c5e62d472ef1505fb3d6347fc363605"}]},"branch":"refs/heads/master"},"4ccbf064aa1de5e5089e407b0e234a995ba81f05":{"kind":"TRIVIAL_REBASE","_number":51,"created":"2026-09-11 15:21:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/72/998072/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/51"}}},"commit":{"parents":[{"commit":"5435dfdb0531d1d0e1c76851c97b9c980718c101","subject":"utils/checksum: Add combine to CRCHasher","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5435dfdb0531d1d0e1c76851c97b9c980718c101"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-11 15:14:36.000000000","tz":60},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a validation-only headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a HeaderKeyDict\nof the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, remove the validation-only HEAD headers\nso backend response metadata does not become part of the SLO manifest.\n\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nAssisted-By: Codex:gpt-5.5\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ccbf064aa1de5e5089e407b0e234a995ba81f05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ccbf064aa1de5e5089e407b0e234a995ba81f05"}]},"branch":"refs/heads/master"},"c71ada227565ce0457925e8a0b1cc1e4583a8abd":{"kind":"TRIVIAL_REBASE","_number":52,"created":"2026-09-11 15:34:30.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/72/998072/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/52"}}},"commit":{"parents":[{"commit":"a6e91d7d90033bd0bb8dd7e044a197b01ddab606","subject":"utils/checksum: Add combine to CRCHasher","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6e91d7d90033bd0bb8dd7e044a197b01ddab606"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-11 15:33:00.000000000","tz":60},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a validation-only headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a HeaderKeyDict\nof the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, remove the validation-only HEAD headers\nso backend response metadata does not become part of the SLO manifest.\n\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nAssisted-By: Codex:gpt-5.5\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c71ada227565ce0457925e8a0b1cc1e4583a8abd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c71ada227565ce0457925e8a0b1cc1e4583a8abd"}]},"branch":"refs/heads/master"},"b155809c4f8c42573754e25898f30b814e670ff0":{"kind":"REWORK","_number":53,"created":"2026-09-16 15:00:30.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/72/998072/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/53"}}},"commit":{"parents":[{"commit":"d9f6997d8309b8b4a47b27d185bcd56712b17de6","subject":"utils/checksum: Add combine to CRCHasher","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9f6997d8309b8b4a47b27d185bcd56712b17de6"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-16 11:49:15.000000000","tz":60},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a validation-only headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a HeaderKeyDict\nof the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, remove the validation-only HEAD headers\nso backend response metadata does not become part of the SLO manifest.\n\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nAssisted-By: Codex:gpt-5.5\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b155809c4f8c42573754e25898f30b814e670ff0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b155809c4f8c42573754e25898f30b814e670ff0"}]},"branch":"refs/heads/master"},"a016266c07c7d3554458d2896917bd1dd24d2e99":{"kind":"TRIVIAL_REBASE","_number":54,"created":"2026-09-18 20:14:52.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/72/998072/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/998072/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/998072/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/998072/54"}}},"commit":{"parents":[{"commit":"677b7a97925c9e9bc137f749902c3fd7cab1cfbd","subject":"utils/checksum: Add combine to CRCHasher","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/677b7a97925c9e9bc137f749902c3fd7cab1cfbd"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-07-20 23:52:42.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-09-18 20:14:08.000000000","tz":-420},"subject":"slo: Pass segment HEAD headers to manifest hook","message":"slo: Pass segment HEAD headers to manifest hook\n\nSLO already HEADs each object-backed segment while validating a manifest\nPUT. Middleware using swift.callback.slo_manifest_hook may also need\nsystem metadata from those responses; fetching it again adds duplicate\nbackend HEAD requests. MPU checksum persistence is one such consumer.\n\nAdd a validation-only headers key to each successful object-backed manifest\nentry passed to the existing hook. Populate it with a HeaderKeyDict\nof the segment HEAD response.\n\nPreserve None for failed segment validation entries and\nleave inline-data entries unchanged.\nBefore storing the manifest, remove the validation-only HEAD headers\nso backend response metadata does not become part of the SLO manifest.\n\nChange-Id: Ifa38be22aa6b7c0501b0693b36fe7e3c7409e297\nAssisted-By: Codex:gpt-5.5\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a016266c07c7d3554458d2896917bd1dd24d2e99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a016266c07c7d3554458d2896917bd1dd24d2e99"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
