)]}'
{"id":"openstack%2Fswift~998258","triplet_id":"openstack%2Fswift~master~I0104ad018b845e70caf8bc6368aaff8fc6156354","project":"openstack/swift","branch":"master","attention_set":{"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2026-08-26 00:04:10.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-08-28 17:43:23.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I0104ad018b845e70caf8bc6368aaff8fc6156354","subject":"tests: Port boto2 tests to boto3","status":"NEW","created":"2026-07-22 03:36:58.000000000","updated":"2026-08-28 18:51:08.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":31,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"f9081a090dcbe82344f8e25f2d8372546c43235b","_number":998258,"virtual_id_number":998258,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-28 18:51:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-22 04:47:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-11 18:04:33.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2026-08-11 18:04:33.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2026-08-11 18:04:33.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"}],"messages":[{"id":"4f004b322841f8eaf62f5ec758e8f883207a551a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-22 03:36:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36ef3783ed070a80fc9147d68fa406672829a6ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 04:47:17.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1bc8b6a6306f4492b1075f59d39e1c62\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f41259c084a248b59a8eeabb124fc7cc : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca4f9136fddf4840962bfc92eb7c129e : SUCCESS in 5m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1b3ae72031474558a82ce47ee945cd73 : SUCCESS in 1m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b525075353c146bb8002754b1665d487 : SUCCESS in 5m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/498c3c800dd147aca884e6ca7ed16f02 : SUCCESS in 37m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f693961a3919456fb69fa6ba0ccd40e7 : SUCCESS in 43m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c7deeb4a78f940c5be1ad1979d5fade6 : SUCCESS in 32m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c232e84a14464ecaa5d1d10df63d8d30 : SUCCESS in 34m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f5e0f3aaa18649b580066ecdc2bf6a4e : FAILURE in 16m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/961cb9f261894fdfab41fd397397c95b : FAILURE in 16m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/18a9e88cdce44aad847fc2f4d24803e4 : FAILURE in 16m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a71801b71dd54cabb2112d87f46051f1 : SUCCESS in 4m 39s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/98a56afe0aa04d02ace25182efd71beb : SUCCESS in 23m 14s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/831db28e7e2e40f5891f2c7efff66d97 : SUCCESS in 10m 28s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/db40e298d0ee473da7ee56195fc233cd : SUCCESS in 21m 08s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/5a8700f7d57d44948ff58862aa8b453c : SUCCESS in 20m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d711810a863c4b6fbb0eb3da01400453 : FAILURE in 21m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7a6a3b2985eb4d11a97c2a45102d9897 : FAILURE in 27m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/34140974ab454cb38734baa9f64659d8 : FAILURE in 13m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dc69478a7abb4506babfd29dab3a53ed : SUCCESS in 6m 29s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/313dc0d6833f4bbcb48e988db21c031d : FAILURE in 7m 29s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/af2f59499d47420a863e749d3141fa7b : SUCCESS in 8m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/dceb4a11561b455f8c396fb9c7b21318 : SUCCESS in 51m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8b61809741a940b9bb954c9b713227d9 : FAILURE in 18m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/392fb2d89c954cb889939bbd9287ef99 : FAILURE in 17m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b73810579fd04634b03e84a721f2d492 : SUCCESS in 22m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e7d2aaf24344badb598cc62ab540e04 : FAILURE in 2m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a6df1093ed2d48aa91534c9e7029f53c : SUCCESS in 18m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5d433edf1184f15b381b6e68208520d : FAILURE in 56m 12s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"ef5dd278877a487a23bc371ef176fc5ccc984a37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 05:11:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/971b45981e184619a65498959c71f857\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a7e42ac3321d46cfb0d0a129be6e3cb6 : SUCCESS in 58m 40s","accounts_in_message":[],"_revision_number":1},{"id":"0564d7a446fe1d4b7bfb4cf8b9fb58cd00dba8eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-22 19:28:20.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"d8222e8f4af5ce7868936730c3c30b4d0f3bf3eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 20:19:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d65972ff92b403da941a75d0b1fc3e7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/58f2fd4320ed4857a893ebb5c2b9fb1f : SUCCESS in 49m 51s","accounts_in_message":[],"_revision_number":2},{"id":"54202daa677248f6175ebeba4510190235f9ddab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 20:37:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e3812945eeb4e82946f6550313f18d6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/642079dabbc84189b42ea525ec6c7fe0 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a811ccb303c434e89ef04ed02939b3c : SUCCESS in 9m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/00cfd24156ec4a93b9381c97b53a9307 : SUCCESS in 3m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81927283036e492896b49ad57eb4df8d : SUCCESS in 7m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/cd553a9cf39144e4956f6e2316fac8c5 : SUCCESS in 1h 04m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5b4224236a98457e911239e1011315e5 : SUCCESS in 1h 03m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ada409e50cf140878eebc73d90069704 : SUCCESS in 49m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/493b76b4c8ff470cbc1761fb0c6354b0 : SUCCESS in 48m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9614c80ce58d4878a655e60a1959df34 : SUCCESS in 30m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7bcc1a2ce454422bb224ebfc80d1651b : SUCCESS in 23m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b5bb094e9bb4630a36361364a6fe066 : SUCCESS in 24m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5e43ed7d91634e1a85437c9a95850839 : SUCCESS in 6m 43s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ffcd3150485a4a5e96f1f75246adf58c : SUCCESS in 23m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d72d64c2b954462581d752be713d0783 : SUCCESS in 21m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/05fad6e568874036bbb3aa53d36baddb : SUCCESS in 31m 08s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/70593291580b460498657289225461c3 : SUCCESS in 22m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6f6b5512257b469a9de90dd7e6b566e4 : SUCCESS in 19m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0a6a4c374e8645ce80c1a3222b4f8443 : SUCCESS in 20m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5dd232ed98394443b9dbe925302a4387 : SUCCESS in 21m 43s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc4f0a5a94e744fda62016c03dc02b8f : SUCCESS in 23m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/44e25964c9ff43ed8920a662ae96b595 : SUCCESS in 6m 43s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/794adfbe770345ac83a1698024906bf2 : FAILURE in 8m 32s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/cc007de83d9b47da8544878191ccd27b : SUCCESS in 10m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/30a830561ffa45c2a04384dafe0f0edd : SUCCESS in 1h 04m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f5f981bf24d048a1b38356ca140e790a : FAILURE in 31m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/874c22a44d1d459587efe1d63af623e0 : FAILURE in 35m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96904802de32473ca658a4423e6f79e4 : SUCCESS in 31m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69ce7d69e3f643fa8c132d1d9700ba76 : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e6cace5e363041ac9d33e87861c4fd0f : SUCCESS in 20m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9386da703c92494a9b23d8375043074b : SUCCESS in 56m 51s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"b88e1e10ec8ad6df920d2893d6879b81c2dc3df9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-22 21:23:50.000000000","message":"Patch Set 2:\n\n(9 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a8b5dcb67d461bd159e1ae33982c4939d812da5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-22 21:25:24.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"2774d62f3cd132bf073a7a926fbe751078386493","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 22:35:31.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/be6ae0db8c044752bca69c41fb35cf5c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/037805ba1e694a4d82add0235f272117 : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f07fd6ea6df6469eb2444ac0799dc4de : SUCCESS in 6m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8452d304ab824fffb294d210f6976588 : SUCCESS in 2m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87965bc6cfb94ecea3b38cbd1f1af3ac : SUCCESS in 8m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/1711bb2434fd4db68e7b5e020fbbc948 : SUCCESS in 1h 03m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/897d3f4ec6bf43d49809d3adfddbbd99 : SUCCESS in 42m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f0bea585f0934e4c9852a18d657560b0 : SUCCESS in 48m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/629574bf6e6247e1879a8590e0da4252 : SUCCESS in 29m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/87e29bac36214dc8890516c56a437f58 : SUCCESS in 20m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1930d2f393234b9ea79a2617b67d0dc0 : SUCCESS in 22m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5179521725684d0f8bbc966d91227856 : SUCCESS in 24m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/371586223b92492b8ede65365cccce06 : SUCCESS in 4m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e9c527f592624d46a5ab1352bd4951b5 : SUCCESS in 25m 56s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/166e47d7c4e945b9beb7fd2c0dc19d33 : SUCCESS in 15m 31s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b8a485e6f0284122acb45c8e4fefbd9f : SUCCESS in 28m 27s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/206c0194f8eb4e7db4cc2a215170e9d4 : SUCCESS in 10m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8c97fa7bebdc4fc495b08f03488163a3 : SUCCESS in 18m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/969240d3db1246ce90becf871a522802 : SUCCESS in 17m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/16b842335ffc4330bafb3c01579a0273 : SUCCESS in 22m 16s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/b47e1219400e4f80bad06f64aa38dfaf : SUCCESS in 16m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9137efa72c2445df82ef7ffb8909aece : SUCCESS in 5m 39s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/591c02de6cfa45ce9196597176e9c685 : FAILURE in 4m 54s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/792d88c05e084b6d81f95c01a19aeb39 : SUCCESS in 8m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/da76526d84624e40b648e620ce3f6bf7 : SUCCESS in 55m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fc225deee6844d15bed4c72469ad8fe3 : FAILURE in 17m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7cdd68c428cd471db267f834165a3246 : FAILURE in 32m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ef9aea2341b4e58863b63f3e0e30352 : SUCCESS in 28m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b0345ef53b445aca8dc29f540b676de : FAILURE in 3m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2bd2961e85d94c95a79bfd7ca8f3cfe5 : SUCCESS in 18m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a09fe7a27a814e37959d867e4b146d48 : SUCCESS in 42m 19s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"174e83a8cc8e40f8ea3b6afa17ad60ec6b830c33","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 22:40:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd0c43f2fec948c4a433bffa9a42efb6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1349e957bab74293b23154c5ffba51c6 : SUCCESS in 44m 58s","accounts_in_message":[],"_revision_number":3},{"id":"ade9a55e12fbabb2902745cc01edf70bb6540370","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-22 23:23:52.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"bb39f5b9833658a4675bdaccf028a805515286c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 00:17:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5da26b58bd54460cb6c3f02fc7d0a61d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/25592f3479c1460bb14ff62c3c27d7f6 : SUCCESS in 52m 31s","accounts_in_message":[],"_revision_number":4},{"id":"beaf45fe10f2358cda2b8d8a398605a2f0be65d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 00:32:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b18bad92c7346b4b194726059ce27a6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/115754cae9ef4af6a7a4a08b818c00f6 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16b2f46a150a468987dae7303d4f906e : SUCCESS in 11m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1af5b1046204463f98f150623b047a78 : SUCCESS in 2m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd70c69f144a4ea0b9316018f350f5bf : SUCCESS in 9m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/b7c1d259a7ed4f68be1ae388c06b81be : SUCCESS in 59m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f8ed66f89541411582c9af33f6844897 : SUCCESS in 50m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b6e043b4c7984129b94505dd149e76d6 : SUCCESS in 52m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/313d0dcc482746a9966ae9dda815a1cc : SUCCESS in 1h 02m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dd5d09c42cb74973aa1022a09ffac80b : SUCCESS in 20m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4914020109b44b06b33edbec72f3dddc : SUCCESS in 20m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1c99dddc513c431b898237b433dae0fc : SUCCESS in 16m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9573254f9ccc44079438302f445b969a : SUCCESS in 6m 39s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8c61996f00934fd9b2819d2eae419eff : SUCCESS in 21m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7eb5eda2af8e4b4b913ccc93202f3f82 : SUCCESS in 21m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b923987ceed048a4bb1ab55dc7c0296b : SUCCESS in 19m 47s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/bf0ccba280c843bb95f6e2c5ffbd8a6c : SUCCESS in 14m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cf7613df20ea47f6ab756d5dabd8c3c4 : SUCCESS in 18m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7faf572d62e74252bc190ce8ed333e91 : SUCCESS in 19m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/73dbde5a0c4343608265539d6d3c958b : SUCCESS in 14m 07s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/d29abce2a43e422da53b203c5be12ca2 : SUCCESS in 14m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9bcce41adfd343e5b20e004a94ab1f9f : SUCCESS in 3m 32s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f2a2e12125ac4290ba3072d612eb462c : FAILURE in 7m 58s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/bd4b2f52eb2d4b2c9821fe7a9c3e32e3 : SUCCESS in 9m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/674d653a45304827aedee4aa4a0adb1c : SUCCESS in 42m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7b8bd8f92094494b95c7e22ecde81bcb : SUCCESS in 24m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5f9ae559151343e3a800984a55b0d86c : SUCCESS in 58m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/320952d39c814c6e808ea34df7fa2787 : SUCCESS in 21m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff241c3ba2a8414792d5973704eca95b : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e592ae70b5234cf5824d8a34a02f1cd7 : SUCCESS in 9m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d87bdc12ac9243fcafd9a643230106d5 : SUCCESS in 36m 22s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4},{"id":"d9ab4d49517a0c61940d67735238dc8a065e3317","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-23 01:14:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c053e955fa9537227f42941e1fefc6f27b38c940","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 01:57:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8970cc3388ec4acf8a5b8ef316e00c83\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9b7c9988b3634f80a8e0be4094f46334 : SUCCESS in 41m 14s\n- swift-tox-func-encryption-py312-arm64 https://zuul.opendev.org/t/openstack/build/fdf0bcc119c144599121aa401bdc22a8 : SUCCESS in 38m 24s\n- swift-tox-func-py312-arm64 https://zuul.opendev.org/t/openstack/build/1b4b34e21d4141379358899a2d951344 : SUCCESS in 38m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e94bb82fd0f347f7988253a613a638e1 : SUCCESS in 19m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/96c94b05c5534a4da4a33327f41cbbe2 : SUCCESS in 27m 42s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/60cd562177b54d489851ad04692560be : SUCCESS in 9m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/eca385e7a530424384cbb1abf516f1c0 : SUCCESS in 24m 51s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/a681016580d7487ebb32f034c6d84a05 : SUCCESS in 22m 49s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/426b20dfa3e149f1a98fc3500f357f11 : SUCCESS in 22m 47s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/ef9b1387a8a544998fa62a46e5a6af13 : SUCCESS in 21m 00s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/5f0b5e3716744a319207ebbc69641056 : SUCCESS in 15m 59s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/e96301ce75284232b20c83ed81702887 : SUCCESS in 19m 08s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/6a1369be67d9442c8b7914af36bc8c98 : SUCCESS in 9m 52s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/3e86bde1f75647a198a8035631320916 : SUCCESS in 19m 23s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/b8abc33798cb4ef0b4bd5d1a1c4b07c3 : SUCCESS in 8m 58s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/f86f253f0dc2403eaf7731e30750b56c : SUCCESS in 23m 52s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/af8cc54fdac043a9b0a97a6b572a474e : SUCCESS in 24m 41s\n- swift-multinode-rolling-upgrade-dalmatian https://zuul.opendev.org/t/openstack/build/ce06c76155224338b51c5fabb1a66f82 : SUCCESS in 18m 19s\n- swift-multinode-rolling-upgrade-epoxy https://zuul.opendev.org/t/openstack/build/9bc53dec10e34d2d9376cade491cf2fa : SUCCESS in 19m 43s\n- swift-multinode-rolling-upgrade-flamingo https://zuul.opendev.org/t/openstack/build/8e21c6e0a1564f6f862ab113294863f0 : SUCCESS in 9m 08s\n- swift-multinode-rolling-upgrade-gazpacho https://zuul.opendev.org/t/openstack/build/8978e8ebea814f1e9793d4eb9a590549 : SUCCESS in 19m 19s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/f3e977954864475195f183e3ff68a048 : SUCCESS in 25m 05s","accounts_in_message":[],"_revision_number":4},{"id":"780e31ed87dab49baf122dc29554bb0e8bdf0c86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2026-08-11 17:49:00.000000000","message":"Patch Set 5: Patch Set 4 was rebased on behalf of \u003cGERRIT_ACCOUNT_15343\u003e\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"_revision_number":5},{"id":"2e6c4b7add2cd20e586da6e1b83099a94b003501","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-11 17:57:59.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"c5434d2243eedf96fcd2ae9fc7fced2a9e46c7e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 19:03:07.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/471381cca37245829bf4bc9c079d47a8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9970983472b547c7b0304442582c20cc : SUCCESS in 1h 11m 59s","accounts_in_message":[],"_revision_number":5},{"id":"7d5625aacef6aab90eb5fb8fb83065ec6cc96dad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 19:05:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad73a38d241345b09af160a2646f0ebd\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/764bbe83c3eb47e298a6a2ad519ec4bc : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a4559171d954774a169e7908a18110a : SUCCESS in 11m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6cebb2de1f354c58bc933a4cfbecdcb1 : SUCCESS in 2m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90379c1dcae34cf49d2bcf29f1eb7a22 : SUCCESS in 5m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/20d0de37749f41f19d7d69b385c0ccf9 : SUCCESS in 1h 10m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a67f958979e14f5091ac7c1b299f56a6 : SUCCESS in 48m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/644dcc84f8a747088bc09a7ab9538dd8 : SUCCESS in 52m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e70ac58be424202ae38aeb0ddfeb9d9 : SUCCESS in 48m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ecbad8610c794316bb65895a5e78230d : SUCCESS in 19m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/de380aee4a1b4153b4da06a3bb024385 : SUCCESS in 23m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8d14ea2498ef46afb963feb1a2026662 : SUCCESS in 19m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1f064b2ca3a94a0b8e247b2f32424015 : SUCCESS in 6m 38s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/436820ecda7a469c93f043292c638b1b : SUCCESS in 24m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/29f0c46282a8457f977f372368a626dd : SUCCESS in 21m 25s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d0a52fc6292c4398a04998f1ff166773 : SUCCESS in 31m 40s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/72c478dccfa74d7789e5b1bd6595eb91 : SUCCESS in 20m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0af5a32115b046739d6b9010873f83c9 : SUCCESS in 15m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/18a7a9e2ebf141048ff50dc1920bed0a : SUCCESS in 17m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8b39571f66cb427cb1e6d8af12603bd6 : SUCCESS in 22m 35s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/00e5c371e6c94659975f9152db987818 : SUCCESS in 20m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3e2812b3811d4df09cf11c6660263c98 : SUCCESS in 3m 40s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/83683c1ba63d48b19b7c1c6bfa9f4afb : FAILURE in 7m 23s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2826e936c16a4a1498d3efb9e390edd4 : SUCCESS in 7m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5cca3742534f492d935801d712a72a60 : SUCCESS in 35m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/15c757db729d4d75b19eb83c4a068dec : SUCCESS in 36m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0b3016ca0d8742e586a21e4c1ade7b49 : SUCCESS in 55m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e56c22817564fbaa7edd435ae0abea8 : SUCCESS in 29m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9b993ad776c43cb921c773f28451d7e : SUCCESS in 3m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cb3ff54fa6024162b47dc83c76c8ee3c : SUCCESS in 20m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/038c49edc66c4fc790e32a6c94c09d45 : SUCCESS in 49m 39s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"bbe1f25f049b294a822e48c8af2210caeb3f589c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2026-08-18 19:39:54.000000000","message":"Patch Set 6: Patch Set 5 was rebased on behalf of \u003cGERRIT_ACCOUNT_15343\u003e\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"_revision_number":6},{"id":"46404fde29e41737ee8298fa9753e699b8c023a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 19:56:26.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0f21ab891b84fd693435478e94fe169\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/98c0cf50264343fea5aed294c8b0bd46 : FAILURE in 5m 58s","accounts_in_message":[],"_revision_number":6},{"id":"f5462e5d4873cfbdd2337fc4a18a801ce4aa5981","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 20:58:38.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e475fe01bf4a4b798623040599e4605e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e012cbf3066242babe2d34f389b036c8 : SUCCESS in 2m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fa2aae2babd4d939e81f45fbaff135b : SUCCESS in 11m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/63f6e90447cc47b29f2b16cab9af679e : SUCCESS in 1m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9f5d20c23ac4d45a8cc28154b33b92b : SUCCESS in 10m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/59e95fbff4f84aa390320c82221d69fa : SUCCESS in 36m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b411cd6ef2b04a3f835969b8934e98f5 : SUCCESS in 1h 11m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/983530d5ab2c461a868be414baec7133 : SUCCESS in 53m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/737eba5a6db349089882cebc60aa5045 : SUCCESS in 44m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/718e021070e245e09294c686872387d8 : SUCCESS in 15m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cb218848a81d40b8bf02d18e905c01b0 : SUCCESS in 19m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f419617530df4157a132f247b0db3291 : SUCCESS in 27m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4b8d4095d72143798dcd94532eec5aa9 : SUCCESS in 6m 38s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4660133d60e74522a6a3a215a8b41ab1 : SUCCESS in 12m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/036f0bed2d374fd9bf47ca5a52f30353 : SUCCESS in 22m 41s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/78a23216c7df4cf9ba5850aea1860776 : SUCCESS in 26m 02s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/b663ba0ed15743bfba14900ce851a68f : SUCCESS in 20m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/509916d3c40743ea8e018193f32e5e3b : SUCCESS in 24m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/860c6a13bb994044b968965dc71d2430 : SUCCESS in 14m 15s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/01888f1447e04d4a8b43a7d1701d5d09 : SUCCESS in 18m 48s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/b31edb0857ee4914b5626014f1614bff : SUCCESS in 12m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8e4061cf37e545a68f2ab6fc4aaa6114 : SUCCESS in 7m 05s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bced0651e37f4e5b8a13c6cfac7ea894 : FAILURE in 9m 15s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5377c62f3c65499886a3b0b6009873cd : SUCCESS in 5m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/80bc9068fd204f6fb53482235c108e16 : SUCCESS in 47m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e9c95444a0154b26a9066a356c1374c7 : SUCCESS in 39m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9e643bd6832d46689f260b10f78e873e : SUCCESS in 29m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36a048a2f2e24094a31a468aa96ef2db : SUCCESS in 28m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dc84f19187d4349afb6a459169b4c53 : SUCCESS in 2m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6c5b4b6c542e42c9984720c77e8c9ccc : SUCCESS in 9m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8dc0289614f342649bfb4269d2f6d601 : SUCCESS in 1h 06m 52s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":6},{"id":"21f9b8117686ba337bd3df4738841df599519a35","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2026-08-25 21:35:08.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nThe following files contain Git conflicts:\n* test/functional/s3api/__init__.py\n* test/functional/s3api/test_multi_upload.py\n* test/functional/s3api/test_object.py\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"bcb375ac94e39e85b14fb6cc5eff41a8e1ef0c70","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 22:45:42.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ddd036e5c05486b9a494cab997bd446\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6cd58b975db64c2b9f02efeb1ee0a9b8 : SUCCESS in 1h 05m 08s","accounts_in_message":[],"_revision_number":7},{"id":"2f135e5c94953e71d5f5b0b201631cd29e08929b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 22:47:21.000000000","message":"Patch Set 7: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/20f7078a38604479b7b8ed52cd4c48f1\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ef4912d024574b14b77d34d8a763003e : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e69a36a70dbf4bc0afe8ffa3a04220ab : SUCCESS in 6m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/269aa8e84275480f96523d998b988023 : SUCCESS in 2m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/099aadf3b0614bf597e09d08b1b60f83 : SUCCESS in 6m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/a78dadf788d64dacb4aacb7170a991e7 : SUCCESS in 1h 03m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4f075026ff8240ae870d72d9ea9dbf4c : SUCCESS in 33m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9f2d9b29bd7d48a1b01e0b5987f111d3 : SUCCESS in 50m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3fa5132058b54f2994f4e52ba2ec999e : SUCCESS in 1h 06m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e07de6e79f8744e29015f03d8aded191 : FAILURE in 14m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b81d4d97785440be9d49a5db943f9bb6 : FAILURE in 16m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c0e83e318d3b4d2ca36635b482c70a3c : FAILURE in 14m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e9953f8d48db408fbfcc48d81d298652 : SUCCESS in 10m 03s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b0f999a179e3490b9a7b7c132d76dbc5 : SUCCESS in 14m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b027b82f6eec4f9ea0ef85ebe6ae99bd : SUCCESS in 10m 14s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c4be74067afd477f94c2b29cb10edb85 : SUCCESS in 25m 15s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/06cb6346978245bb9bbd8d84c1eb1082 : SUCCESS in 26m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f3c521ada48141d39b4345176f11d69b : FAILURE in 13m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9ca5c502312040149080eb936f35df56 : FAILURE in 10m 15s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c43169f056be4140a592663e6b74401d : FAILURE in 13m 25s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/23be4c2b00684964bad42fc29e841e8b : FAILURE in 13m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/023727c239b148eb84d872ae447089a2 : SUCCESS in 3m 51s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/48f394b1fa51477f98a3a5ffa0b885b5 : FAILURE in 5m 40s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0b9be75c0bbd4ae6b6dd0f49982b34c8 : SUCCESS in 5m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/aff1658a645145f79e7857a96d1b7737 : SUCCESS in 42m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/85babb70c44c47d7ba5716c86f9d0d6c : FAILURE in 30m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2a607025a7f046c381cdb17e7dfe5408 : FAILURE in 33m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d80742220c044f39d5dc5a6d6b17789 : SUCCESS in 15m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5102e4ad65748cca99ed40e8a1f6e9b : FAILURE in 3m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/550f3f7b44004647bbe65a24a974840c : SUCCESS in 26m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/595b4c23ad30444483bf3e9f6817228f : SUCCESS in 53m 13s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":7},{"id":"6e4bae60c15e6a61dbf8412168082c2e083e4d29","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2026-08-25 22:56:37.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"20216df2f3926ebda99e0eee24b85bb862233968","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2026-08-25 22:56:48.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":8},{"id":"da1930f42e0cedad789ee29cf86632671a4a2c8b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 23:51:43.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74b1efe2f8844c2796bccea2ebd4c5c3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8be15f2f57d84e34ba394f41beffc990 : SUCCESS in 43m 25s","accounts_in_message":[],"_revision_number":8},{"id":"97e57d70e45683dd7803c4c0d8d702a8052898db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 00:04:10.000000000","message":"Patch Set 8: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a4b189dc4114261807a317671cfabc7\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9b07d043e99a499281099569116f72e6 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d53cb0eac114e54906f9efa3b24e04d : SUCCESS in 12m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/57ff16de0d6e42cebc7f8118a25060e7 : SUCCESS in 1m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69e3611c90964130b5ae4e9572d2c2f7 : SUCCESS in 10m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/4c541d19985a4d36935a05df67593f45 : SUCCESS in 46m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cfb07830092a4a978d27f7ea12f26246 : SUCCESS in 1h 01m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9a8f6d27e87a401f95070141130b2462 : SUCCESS in 1h 00m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03ceb3719b384a85a55a80870082152c : SUCCESS in 43m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d0e97b9000b4f428e9e16d569fe964f : FAILURE in 22m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/78090e07a4c94085a51f2de6641213dd : FAILURE in 24m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee902238b37248c1a7aaebfa4fb42825 : FAILURE in 22m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/90c02d2316744cd18660c04996f82160 : SUCCESS in 6m 19s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2cb5a428380242a393c85c3d90fed8af : SUCCESS in 17m 57s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/be5ca6ce17384dde9c306468b635bf7a : SUCCESS in 17m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b0209c2ce0b4487dad4bd042d28adbc2 : SUCCESS in 12m 49s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/debc945b467e4e9dacbb0f79ffdc32a0 : SUCCESS in 9m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5ccfe02259454632b43b81985b3511ba : FAILURE in 12m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e2aaa34225f64852abe23a8feeb7f478 : FAILURE in 16m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/86ec984167ca4801b6ff70f400e3e59b : FAILURE in 22m 06s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/cde6b9c8f3024b5e97588764f3d1b878 : FAILURE in 20m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a155799b689f4a0db75a5f15be624199 : SUCCESS in 4m 25s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/59df5220d61f4d33a53b76d1977757a8 : FAILURE in 6m 15s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d43988c1e85544b49a959c7931bef344 : SUCCESS in 8m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/254974cb2a644f8e8afa2895ce5237d8 : SUCCESS in 39m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/26215ea1afa34f2faf8f932519af1835 : FAILURE in 31m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ffa5bbabb7f04bacaa1f35422e8912b8 : FAILURE in 22m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f621923a4dbb408cb67af9b0f2c0701a : SUCCESS in 11m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6f665401d73451d8ce0b782bb67a41d : FAILURE in 2m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7e357d39b36f4cac8459225b1e85956a : SUCCESS in 18m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed4c3dec050c4e13b5e2c99093752710 : SUCCESS in 35m 49s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":8},{"id":"85804c615e40b7df4aa6da4a571e859f8044f688","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2026-08-26 00:19:51.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"6d67f565495b4ff7887ce4ccf26e4a3c9819f097","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 01:26:38.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eedb22d3182f462f8380278864bd9cc1\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9dcdf9ffe8c2401f8c3faae883e26e24 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a8a3236ddbb4af897904fd926c4a1a1 : SUCCESS in 11m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d33b02c520524d208ff817c20b01bfc8 : SUCCESS in 4m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec6869c1a92f4c5db89535659bacde01 : SUCCESS in 9m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/b18bc2333f3c44479edc5a0b1ec35cdb : SUCCESS in 42m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c4e9329c8f6b4277b7c0197dfca8d58a : SUCCESS in 38m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ca8ce135c50c4ef9b62f46c489ca8a4d : SUCCESS in 51m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ab007b235394d579d35133e9162d373 : SUCCESS in 50m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d644f1f33a024b86a9a5de256a02ba06 : FAILURE in 21m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4d34c1f4eb5346cb921a3f22014ef1b6 : FAILURE in 21m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/36fe37dbd1c5468aa95dafd1ea0012e0 : FAILURE in 22m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/858a3f70e7374a9996ebdbb163914b69 : SUCCESS in 6m 25s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c4cac9476ea447fb9202d4b82c7c53b8 : SUCCESS in 27m 14s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a1d2d4e38acb49959bff685aa9d04fee : SUCCESS in 12m 58s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/956bcef586144c76ad18dddd344e6b64 : SUCCESS in 16m 56s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/bd3275d579364002bb1c5fca19857019 : SUCCESS in 21m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e1db64464ad147a7b2e1c30efa046beb : FAILURE in 20m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4ca0a21201fd4a3381f2023cea035d6f : FAILURE in 12m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/bfb6f5579b2a4eb8a85b2b921de04bc9 : FAILURE in 19m 40s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/42d6a800885a456a8e22e3167efb68a6 : FAILURE in 22m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8a3da4a1b7fa47a89f204eb59cedeb2e : SUCCESS in 3m 27s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/05c5f444548447e8b81eaa7da902ebb8 : FAILURE in 7m 42s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/236c68a6fc824ab092d71defd40d0797 : SUCCESS in 11m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f1ff7761d88d44d9a7942920a46ada6b : SUCCESS in 1h 01m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/05f481bcdc484c98b0b2ff58c52d149f : FAILURE in 34m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a57f8a9d148540238112ecf3d7f607a8 : FAILURE in 15m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5e0adaf2e5443c6af01e9ae670c0552 : SUCCESS in 27m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5caa77750814d6a95325d82bb5e94d6 : FAILURE in 4m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4ba7537cf15949ec870513b2a1e577af : SUCCESS in 27m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc9a53cd51514fc488cb0ab8703cc12d : SUCCESS in 48m 24s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":9},{"id":"1b882f4b98e3ee16d522bdd1e0600dadfbe0e564","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 02:18:03.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bad1b697ca4443529f8194604eaa08a3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/593be357e3024d77b7d13a15c63bc3da : SUCCESS in 59m 00s","accounts_in_message":[],"_revision_number":9},{"id":"1f2239a0ec79f3e5614670eb6b122798ee207d35","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-28 17:43:23.000000000","message":"Patch Set 9:\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"249d41a867f8c1c328159963ec3d45cb771e1963","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-28 17:46:10.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"e40583a660d970e19031cce6f2e83aa500be6f37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 18:41:50.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eaa76b3f94d74125a8cb04944de78789\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4a0f90fb61d043d1aaf6146a53287b4f : SUCCESS in 32m 36s","accounts_in_message":[],"_revision_number":10},{"id":"f9081a090dcbe82344f8e25f2d8372546c43235b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 18:51:08.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04c1dd4423dc4287977c240f41310d52\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c50710383259447c8295c2d0618d2dda : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2180c4a6eb384173ab2745a61f589462 : SUCCESS in 9m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/441aa42ab87e4ed5a66793c3777d7736 : SUCCESS in 1m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6413aaa161e446bf95f27b2397f7fd23 : SUCCESS in 8m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/318213c9a6e7428c881ce87ce2ee372a : SUCCESS in 59m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b52254683a2747f2af5249b20cf2be37 : SUCCESS in 35m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cae100d4ddec4bc583727a28f481b9ce : SUCCESS in 33m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d64be0cf75f4150b46bc6b6bc9079ce : SUCCESS in 31m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/502f0cfd0174466ba194709dffe3aa18 : SUCCESS in 15m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/83ba02d80d6f46bdbf3fb07697ed3ee9 : SUCCESS in 15m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/da6a937dfbf5461fbae4b83cfed65864 : SUCCESS in 23m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b21693111ef44bd991e49cb35190341e : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cdd111dc4f3c4cb8a300c085cc5f9e95 : SUCCESS in 12m 12s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ee456ac940314b748cfd0ee13418282f : SUCCESS in 9m 20s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0578abdc6d4343e6ba2e713d209db41c : SUCCESS in 20m 53s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/97e06948dbee4213b5267d3bb3d03f32 : SUCCESS in 19m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/61185d2914bb4c1bad3b3d66a260a655 : SUCCESS in 20m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/55c0b0bdfbb24e42b296cdd88afc66d7 : SUCCESS in 19m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9ac85d0aa6fb4c2b83a0d8057ed304fc : SUCCESS in 22m 40s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/559857bcc67a40b4890e5940c1a3a6e6 : SUCCESS in 20m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5efb620da4c74bf98b0edacf3a68469d : SUCCESS in 8m 04s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/da8e1836f61e4433968bfa2a62657d0e : FAILURE in 7m 51s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/18f82f7ee4a04d718195a0d9f80cebba : SUCCESS in 11m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b99e4dcb6974277b984288f730cf90f : SUCCESS in 39m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/befddb9dbb4f4f93bd6ce4cf120daca3 : SUCCESS in 25m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0db10ed83a974eb195202059328450a3 : SUCCESS in 58m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e9d7335cca94a02bae51689e622f4b6 : SUCCESS in 19m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8202a74a69b54fa98ea97cdfc1214a78 : SUCCESS in 2m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9dc5f43cc4f64cb0ae2d2e6318a28be8 : SUCCESS in 9m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1475d1b7a4fa48f294e51e03241ab223 : SUCCESS in 29m 46s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"78715de9235f45f8ff3f9ac38db23ab2549dfb57","revisions":{"1e0a09461e633b36d81671497feb0290ccb92799":{"kind":"REWORK","_number":1,"created":"2026-07-22 03:36:58.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/58/998258/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/998258/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/998258/1"}}},"commit":{"parents":[{"commit":"dc8a59c6b2f32015aac309a21d64b67d8b1f5a30","subject":"Merge \"Validate object POST Content-Type-Timestamp before diskfile access\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc8a59c6b2f32015aac309a21d64b67d8b1f5a30"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-17 19:22:03.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-22 03:36:47.000000000","tz":-420},"subject":"tests: Port boto2 tests to boto3","message":"tests: Port boto2 tests to boto3\n\nAssisted-By: Claude:claude-4.8-opus claude-code\nCloses-Bug: #1557260\nChange-Id: I0104ad018b845e70caf8bc6368aaff8fc6156354\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e0a09461e633b36d81671497feb0290ccb92799"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e0a09461e633b36d81671497feb0290ccb92799"}]},"branch":"refs/heads/master"},"866489a652c747ccb2602054670d161e8f00f504":{"kind":"REWORK","_number":2,"created":"2026-07-22 19:28:20.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/58/998258/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/998258/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/998258/2"}}},"commit":{"parents":[{"commit":"dc8a59c6b2f32015aac309a21d64b67d8b1f5a30","subject":"Merge \"Validate object POST Content-Type-Timestamp before diskfile access\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc8a59c6b2f32015aac309a21d64b67d8b1f5a30"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-17 19:22:03.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-22 19:28:12.000000000","tz":-420},"subject":"tests: Port boto2 tests to boto3","message":"tests: Port boto2 tests to boto3\n\nAssisted-By: Claude:claude-4.8-opus claude-code\nCloses-Bug: #1557260\nChange-Id: I0104ad018b845e70caf8bc6368aaff8fc6156354\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/866489a652c747ccb2602054670d161e8f00f504"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/866489a652c747ccb2602054670d161e8f00f504"}]},"branch":"refs/heads/master"},"4602e1a3bceffd9013f993e3c4b1ebb9f8528fff":{"kind":"REWORK","_number":3,"created":"2026-07-22 21:25:24.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/58/998258/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/998258/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/998258/3"}}},"commit":{"parents":[{"commit":"dc8a59c6b2f32015aac309a21d64b67d8b1f5a30","subject":"Merge \"Validate object POST Content-Type-Timestamp before diskfile access\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc8a59c6b2f32015aac309a21d64b67d8b1f5a30"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-17 19:22:03.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-22 21:25:01.000000000","tz":-420},"subject":"tests: Port boto2 tests to boto3","message":"tests: Port boto2 tests to boto3\n\nAssisted-By: Claude:claude-4.8-opus claude-code\nCloses-Bug: #1557260\nChange-Id: I0104ad018b845e70caf8bc6368aaff8fc6156354\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4602e1a3bceffd9013f993e3c4b1ebb9f8528fff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4602e1a3bceffd9013f993e3c4b1ebb9f8528fff"}]},"branch":"refs/heads/master"},"ba9111fb7b9097efef45c0d10ae3663be525cc1c":{"kind":"REWORK","_number":4,"created":"2026-07-22 23:23:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/58/998258/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/998258/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/998258/4"}}},"commit":{"parents":[{"commit":"dc8a59c6b2f32015aac309a21d64b67d8b1f5a30","subject":"Merge \"Validate object POST Content-Type-Timestamp before diskfile access\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc8a59c6b2f32015aac309a21d64b67d8b1f5a30"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-17 19:22:03.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-22 23:23:46.000000000","tz":-420},"subject":"tests: Port boto2 tests to boto3","message":"tests: Port boto2 tests to boto3\n\nAssisted-By: Claude:claude-4.8-opus claude-code\nCloses-Bug: #1557260\nChange-Id: I0104ad018b845e70caf8bc6368aaff8fc6156354\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba9111fb7b9097efef45c0d10ae3663be525cc1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba9111fb7b9097efef45c0d10ae3663be525cc1c"}]},"branch":"refs/heads/master"},"f4261c5fadc06d06f05140b61146a412cf15be90":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-08-11 17:49:00.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/58/998258/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/998258/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/998258/5"}}},"commit":{"parents":[{"commit":"482a93529b5822e28e707bccd94037acf4c16d8d","subject":"Merge \"versioning: more tests for OPTIONS requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/482a93529b5822e28e707bccd94037acf4c16d8d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-17 19:22:03.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-11 17:49:00.000000000","tz":0},"subject":"tests: Port boto2 tests to boto3","message":"tests: Port boto2 tests to boto3\n\nAssisted-By: Claude:claude-4.8-opus claude-code\nCloses-Bug: #1557260\nChange-Id: I0104ad018b845e70caf8bc6368aaff8fc6156354\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f4261c5fadc06d06f05140b61146a412cf15be90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f4261c5fadc06d06f05140b61146a412cf15be90"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"dc8a59c6b2f32015aac309a21d64b67d8b1f5a30","ours":"ba9111fb7b9097efef45c0d10ae3663be525cc1c","theirs":"482a93529b5822e28e707bccd94037acf4c16d8d","merge_strategy":"recursive","contains_conflicts":false}},"f8845d985df58f28508d360475447987a439a3dc":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-08-18 19:39:54.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/58/998258/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/998258/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/998258/6"}}},"commit":{"parents":[{"commit":"3064444bf433dfb7ade3d91130b58f7bceae9add","subject":"CHANGELOG for 2.38.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3064444bf433dfb7ade3d91130b58f7bceae9add"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-17 19:22:03.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-18 19:39:54.000000000","tz":0},"subject":"tests: Port boto2 tests to boto3","message":"tests: Port boto2 tests to boto3\n\nAssisted-By: Claude:claude-4.8-opus claude-code\nCloses-Bug: #1557260\nChange-Id: I0104ad018b845e70caf8bc6368aaff8fc6156354\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8845d985df58f28508d360475447987a439a3dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8845d985df58f28508d360475447987a439a3dc"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"482a93529b5822e28e707bccd94037acf4c16d8d","ours":"f4261c5fadc06d06f05140b61146a412cf15be90","theirs":"3064444bf433dfb7ade3d91130b58f7bceae9add","merge_strategy":"recursive","contains_conflicts":false}},"2e6b10cf142a0d3f662961be0738f69c25a9c453":{"kind":"REWORK","_number":7,"created":"2026-08-25 21:35:08.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/58/998258/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/998258/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/998258/7"}}},"commit":{"parents":[{"commit":"a3ca4e2abb3af9d2e40df14c056e3c036088107a","subject":"Merge \"object_versioning: test footer callbacks on overwrite\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a3ca4e2abb3af9d2e40df14c056e3c036088107a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-17 19:22:03.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2026-08-25 21:35:08.000000000","tz":0},"subject":"tests: Port boto2 tests to boto3","message":"tests: Port boto2 tests to boto3\n\nAssisted-By: Claude:claude-4.8-opus claude-code\nCloses-Bug: #1557260\nChange-Id: I0104ad018b845e70caf8bc6368aaff8fc6156354\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e6b10cf142a0d3f662961be0738f69c25a9c453"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e6b10cf142a0d3f662961be0738f69c25a9c453"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"3064444bf433dfb7ade3d91130b58f7bceae9add","ours":"f8845d985df58f28508d360475447987a439a3dc","theirs":"a3ca4e2abb3af9d2e40df14c056e3c036088107a","merge_strategy":"recursive","contains_conflicts":true}},"2814d16664c92a15705dfa66d43ee1db826de0fb":{"kind":"REWORK","_number":8,"created":"2026-08-25 22:56:37.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/58/998258/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/998258/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/998258/8"}}},"commit":{"parents":[{"commit":"a3ca4e2abb3af9d2e40df14c056e3c036088107a","subject":"Merge \"object_versioning: test footer callbacks on overwrite\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a3ca4e2abb3af9d2e40df14c056e3c036088107a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-17 19:22:03.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2026-08-25 22:56:01.000000000","tz":-420},"subject":"tests: Port boto2 tests to boto3","message":"tests: Port boto2 tests to boto3\n\nAssisted-By: Claude:claude-4.8-opus claude-code\nCloses-Bug: #1557260\nChange-Id: I0104ad018b845e70caf8bc6368aaff8fc6156354\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2814d16664c92a15705dfa66d43ee1db826de0fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2814d16664c92a15705dfa66d43ee1db826de0fb"}]},"branch":"refs/heads/master"},"eaf061d95310f692633b87f3486ab06eb913d752":{"kind":"REWORK","_number":9,"created":"2026-08-26 00:19:51.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/58/998258/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/998258/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/998258/9"}}},"commit":{"parents":[{"commit":"a3ca4e2abb3af9d2e40df14c056e3c036088107a","subject":"Merge \"object_versioning: test footer callbacks on overwrite\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a3ca4e2abb3af9d2e40df14c056e3c036088107a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-17 19:22:03.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2026-08-26 00:18:14.000000000","tz":-420},"subject":"tests: Port boto2 tests to boto3","message":"tests: Port boto2 tests to boto3\n\nAssisted-By: Claude:claude-4.8-opus claude-code\nCloses-Bug: #1557260\nChange-Id: I0104ad018b845e70caf8bc6368aaff8fc6156354\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eaf061d95310f692633b87f3486ab06eb913d752"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eaf061d95310f692633b87f3486ab06eb913d752"}]},"branch":"refs/heads/master"},"78715de9235f45f8ff3f9ac38db23ab2549dfb57":{"kind":"REWORK","_number":10,"created":"2026-08-28 17:46:10.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/58/998258/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/998258/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/998258/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/998258/10"}}},"commit":{"parents":[{"commit":"a3ca4e2abb3af9d2e40df14c056e3c036088107a","subject":"Merge \"object_versioning: test footer callbacks on overwrite\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a3ca4e2abb3af9d2e40df14c056e3c036088107a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-17 19:22:03.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-28 17:43:59.000000000","tz":-420},"subject":"tests: Port boto2 tests to boto3","message":"tests: Port boto2 tests to boto3\n\nAssisted-By: Claude:claude-4.8-opus claude-code\nCo-Authored-By: ashnair \u003cashnair@nvidia.com\u003e\nCloses-Bug: #1557260\nChange-Id: I0104ad018b845e70caf8bc6368aaff8fc6156354\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78715de9235f45f8ff3f9ac38db23ab2549dfb57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78715de9235f45f8ff3f9ac38db23ab2549dfb57"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
