)]}'
{"id":"openstack%2Fswift~998284","triplet_id":"openstack%2Fswift~feature%2Fmpu~I1d00ac1a7f63407a6c966accfbb69d8ca754bae6","project":"openstack/swift","branch":"feature/mpu","topic":"merge-master","hashtags":[],"change_id":"I1d00ac1a7f63407a6c966accfbb69d8ca754bae6","subject":"Merge origin/master into feature/mpu","status":"MERGED","created":"2026-07-22 09:01:32.000000000","updated":"2026-07-22 11:57:12.000000000","submitted":"2026-07-22 11:54:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"998284-merge-master","meta_rev_id":"37789d952082207ff26c1259c5262540cf589734","_number":998284,"virtual_id_number":998284,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-22 11:54:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2026-07-22 10:41:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2026-07-22 10:41:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-22 10:07:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"227d3d3c4bf44eca87f851e35abd414853828f65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-07-22 09:01:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b23fa5dd42c04233932eb6e5cb6bbeee80e2792f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-07-22 09:04:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b02db743fc80eb4f3dfdf6288ce884e9787da18d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 10:07:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1de623040684ee0aa556018afa8ca3b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/932a7acba9614ba28c0de4708fdc2836 : SUCCESS in 8m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42e38d510bb9476abf3728029fba9f3b : SUCCESS in 3m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f225772242c24391841a5a050c581ac6 : SUCCESS in 21m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5a42f96156734e76b00ae75f66d51b93 : SUCCESS in 22m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7bee4c54fcf948579aa87e1e3e61f33d : SUCCESS in 23m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1e72a9dc3b4647d2afd515212d95e90c : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0769ead4cc694bbc9f815089ff857a4d : SUCCESS in 31m 13s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f1f1c4ca71714a83809dcaa09e9e6a16 : SUCCESS in 23m 14s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/262d24f6448c49bcae7ee2b92349851a : SUCCESS in 35m 08s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/4cbca60d3ecb44deb772cd1c18069ad9 : SUCCESS in 22m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5f9b9ec0b944431c9f897e63ed26ada7 : SUCCESS in 16m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fbcec13ebd144ce1945381f4619658e3 : SUCCESS in 10m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/aa102feec150433d82dc6f25f62dae0c : SUCCESS in 17m 37s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4e265ecbc39468da6a4112bbea59562 : SUCCESS in 18m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/543ea1e6cd04419ea004c7a927d8755b : SUCCESS in 6m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/788e89d9051c4e3fb3184c97964d498f : FAILURE in 7m 26s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ffb2cc34dbe84dccabf187fa8c900abb : SUCCESS in 9m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f69a98630d94d4aa35cc457624b5943 : SUCCESS in 58m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c439e82f22954fcba57958ff39900de1 : SUCCESS in 30m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74f3ebfe0a734401a153a05ddac368a5 : SUCCESS in 3m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/13240c1d7c164f5087454df9868fd1f7 : SUCCESS in 21m 07s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"4fe65985f5e25f8566a6209d3df4c1a1ddca6420","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 10:24:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/317df428263b4295ac4198db7dc80da5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5d18e4c0623a4c2c952a031fc8e4389c : SUCCESS in 1h 16m 16s","accounts_in_message":[],"_revision_number":2},{"id":"1db4e1b7881f33dd3a396f8d0bf739c59adcdefe","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-07-22 10:41:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d2648b4a97ddeee48cc87632d62b20e82f953d78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 10:42:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5ad6ec1b831ad5f0d619fb7bb3ed4917021d6998","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 11:54:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c36b738c17d4bb4ad7b9957c23e7c19\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fd423265fb848a3a67f496cf3ca2ca1 : SUCCESS in 10m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ef36c5a1d9a457aa34cd8bb39bf67b6 : SUCCESS in 8m 33s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f5b285caf0a04e508c2b7eca67c7b7ea : SUCCESS in 4m 38s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4963b6bd592b418e9dcd2aed45439c54 : SUCCESS in 24m 58s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3e467e4a22c84c018c6e3e38b01d9ad1 : SUCCESS in 9m 36s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e2397265a0884ea58c2adfbbe31213f0 : SUCCESS in 18m 21s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/66063300a6ee4f31bb72ea85ad0947e1 : SUCCESS in 14m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6002f3388f3f4e109c83ba785ddd686b : SUCCESS in 16m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/25caea20685a4cd699544a690d16bca5 : SUCCESS in 9m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/efb501c9558847c1afe2c7b7b53067d5 : SUCCESS in 9m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/970d354be4564bbba07647601d79aa8b : SUCCESS in 3m 59s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ef65f0a6bd6b49f5a03d5896fc838114 : SUCCESS in 7m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/02acf8db3ac4440786791fdb60edac3c : SUCCESS in 1h 07m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4020822646664d18a1e5614c8e628339 : SUCCESS in 30m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae4044d388af448faf74ed502259380f : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5c7537ae420f462c97b78fb5ab96bba5 : SUCCESS in 16m 54s","accounts_in_message":[],"_revision_number":2},{"id":"61211d6db35d6f01ca058603e23f869e1a0fd74b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 11:54:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"37789d952082207ff26c1259c5262540cf589734","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 11:57:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f81e89a71c6844cc8612f5496e5b2335\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f29075f4634a4391bd846ea171812bad : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41083075cd014f0e8eb7283a25ff3d65 : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/8c30fee419a643a5b362c92309746339 : SUCCESS in 1m 01s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d7690327d692a8752181448692f180a7b4469005","revisions":{"28480b04fb6e67393830e3ff3451c93d483e0b8a":{"kind":"REWORK","_number":1,"created":"2026-07-22 09:01:32.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/84/998284/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/998284/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/998284/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/998284/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/998284/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/998284/1"}}},"commit":{"parents":[{"commit":"81b7c3ec870dcc548a51f8c76a9059689da46d0e","subject":"mpu: Fix native mpu acl handler class discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81b7c3ec870dcc548a51f8c76a9059689da46d0e"}]},{"commit":"c79f5d1b7094bbb7020bf004dac27c79e1c303df","subject":"Merge \"Run functional tests on py39 in the check pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c79f5d1b7094bbb7020bf004dac27c79e1c303df"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-07-22 08:01:12.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-07-22 09:01:09.000000000","tz":60},"subject":"Merge origin/master into feature/mpu","message":"Merge origin/master into feature/mpu\n\nChange-Id: I1d00ac1a7f63407a6c966accfbb69d8ca754bae6\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28480b04fb6e67393830e3ff3451c93d483e0b8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28480b04fb6e67393830e3ff3451c93d483e0b8a"}]},"branch":"refs/heads/feature/mpu"},"d7690327d692a8752181448692f180a7b4469005":{"kind":"REWORK","_number":2,"created":"2026-07-22 09:04:47.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/84/998284/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/998284/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/998284/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/998284/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/998284/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/998284/2"}}},"commit":{"parents":[{"commit":"81b7c3ec870dcc548a51f8c76a9059689da46d0e","subject":"mpu: Fix native mpu acl handler class discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81b7c3ec870dcc548a51f8c76a9059689da46d0e"}]},{"commit":"c79f5d1b7094bbb7020bf004dac27c79e1c303df","subject":"Merge \"Run functional tests on py39 in the check pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c79f5d1b7094bbb7020bf004dac27c79e1c303df"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-07-22 08:01:12.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-07-22 09:04:40.000000000","tz":60},"subject":"Merge origin/master into feature/mpu","message":"Merge origin/master into feature/mpu\n\nChange-Id: I1d00ac1a7f63407a6c966accfbb69d8ca754bae6\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7690327d692a8752181448692f180a7b4469005"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7690327d692a8752181448692f180a7b4469005"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
