)]}'
{"id":"openstack%2Fswift~998391","triplet_id":"openstack%2Fswift~feature%2Fmpu~I6152cbc6b5833289074d977b34ad741da5160824","project":"openstack/swift","branch":"feature/mpu","hashtags":[],"change_id":"I6152cbc6b5833289074d977b34ad741da5160824","subject":"container: Avoid rebuilding decoded timestamps","status":"NEW","created":"2026-07-23 00:28:50.000000000","updated":"2026-07-23 01:26:52.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"af69e50a9e400ba17a2369c78e96b2d550a518b6","_number":998391,"virtual_id_number":998391,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-23 01:26:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-23 01:23:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-23 01:26:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8e6d6cbf377d800a6c8228108691b81b8b9d20f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-07-23 00:28:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ac7c334713d37337f7dfda9b3d7294f446d9987","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-07-23 00:32:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7540e74e340b6e85e0121123472bdd48fe24d164","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 01:23:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4ec5bf72a63442d81f82bef430a6d0a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6d23b471a5b94200bd320fc49684e1a7 : SUCCESS in 41m 12s","accounts_in_message":[],"_revision_number":1},{"id":"af69e50a9e400ba17a2369c78e96b2d550a518b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 01:26:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a8951dc9fe64749a42fe9c48e7d70a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ef7db32859f4924b1f50f2510cab73b : SUCCESS in 4m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94712b3d87614a24b4c22f2021f6237c : SUCCESS in 8m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/34416d858a7b43caa0edf956b9916f99 : SUCCESS in 17m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b297f014b8204a82a949c8f53e1d68d6 : SUCCESS in 12m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/16d33ff1617d46b49f20fbc79e77201b : SUCCESS in 19m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6430f04d0bf0441eb1c8bdfd696fce72 : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a6acab16ef4649c8819bd028e1f7dd68 : SUCCESS in 10m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/76e0bc6f89414e51af0bb9ea2d31c5ad : SUCCESS in 17m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/46def545656645839be20b45d2a30e39 : SUCCESS in 14m 40s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/97d6d180a4744ea495e694d7bc2e3c5a : SUCCESS in 23m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ee5aec10f02f4057908fb884789faf3a : SUCCESS in 8m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7a1958e1bba241069575bf6e2a67fdba : SUCCESS in 10m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/569f6ba07a874730a1e737372789322c : SUCCESS in 19m 43s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/462af936d6e84b6fb0a1cfd2540a60c4 : SUCCESS in 13m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b493794cf4c944ffabccb5577110bd9c : SUCCESS in 3m 35s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cac5e714470e4176be14a491d05a4646 : FAILURE in 4m 01s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4dab3c038ed749e68aee47cc0c45e39d : SUCCESS in 9m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b7bc344e9b246e7a3457607ecefa013 : SUCCESS in 56m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b05017cacf44e92b803b8588b63902c : SUCCESS in 11m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e12d418dfcc41ac8f11798035b67de5 : SUCCESS in 3m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ee217324f23f45b5aa801beb13828a91 : SUCCESS in 18m 25s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fd66e5082190715689ffb8d91f16d9916eeffcfc","revisions":{"fd66e5082190715689ffb8d91f16d9916eeffcfc":{"kind":"REWORK","_number":1,"created":"2026-07-23 00:28:50.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/91/998391/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/998391/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/998391/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/998391/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/998391/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/998391/1"}}},"commit":{"parents":[{"commit":"40226ecd741e62d3656771f0ecc842c92f74445a","subject":"mpu: store parts in hidden account","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40226ecd741e62d3656771f0ecc842c92f74445a"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-07-23 00:22:24.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-07-23 00:28:39.000000000","tz":-300},"subject":"container: Avoid rebuilding decoded timestamps","message":"container: Avoid rebuilding decoded timestamps\n\nProfiling 10,000-row merges showed that each existing row rebuilt six\nTimestamp objects, even though expand_item had already decoded those\nfields to Timestamp objects. Timestamp(Timestamp) serializes and\nreparses the value, so redundant construction dominates stale and update\nmerge workloads.\n\nDirectly compare decoded Timestamp objects. Document and test the\nexpanded-item type invariant, keeping the comparison safe and obvious.\n\nChange-Id: I6152cbc6b5833289074d977b34ad741da5160824\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd66e5082190715689ffb8d91f16d9916eeffcfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd66e5082190715689ffb8d91f16d9916eeffcfc"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
