)]}'
{"id":"openstack%2Fswift~999329","triplet_id":"openstack%2Fswift~master~I18a9f2be3042477fa18d827826ba5082ba099944","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-09-20 06:27:02.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-09-20 06:27:02.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-09-20 04:33:29.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I18a9f2be3042477fa18d827826ba5082ba099944","subject":"tests: Cover diskfile quarantine suffix fallback","status":"MERGED","created":"2026-07-30 19:14:22.000000000","updated":"2026-09-20 06:28:23.000000000","submitted":"2026-09-20 06:27:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"999329","meta_rev_id":"fb6d8c7bb3fc75894f8f2cbe0274d35f8b84bc64","_number":999329,"virtual_id_number":999329,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-20 06:27:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"value":0,"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"recommended":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"all":[{"value":1,"date":"2026-09-18 21:46:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":2,"date":"2026-09-18 21:46:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-18 21:47:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"value":1,"date":"2026-09-18 21:46:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-18 21:47:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"value":0,"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-31 21:27:27.000000000","updated_by":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"reviewer":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"state":"CC"},{"updated":"2026-07-31 21:28:44.000000000","updated_by":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"reviewer":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"state":"REVIEWER"},{"updated":"2026-07-31 22:28:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-14 17:01:19.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2026-08-18 13:07:41.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"},{"updated":"2026-08-18 14:57:16.000000000","updated_by":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"reviewer":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"state":"REVIEWER"},{"updated":"2026-09-18 21:47:39.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"99589a8ea8736ad1eacf6355df60d57c92bc273b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-07-30 19:14:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60763d698f19b3b4bec9a1132195c8c91130ba9f","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-07-31 21:27:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fad302ac922d21f7e05520305880540a5027d3b6","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-07-31 21:28:44.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0865de93cca109cd5859ceac97d387f7956041c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 22:28:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d979b53486e34d1e8668c199ff034ffa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57af5fc476aa4e4d915e41bf070d36d3 : SUCCESS in 11m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e028d0ad1d7a4a378a87e26ba7b1f053 : SUCCESS in 8m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3737f23c11fe405bb9e25dac3215ea5e : SUCCESS in 16m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/26b6ad56969541738939b7ebc5565622 : SUCCESS in 16m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8d5a1d78bc4547f4aed39aa1455afd2e : SUCCESS in 20m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/114470f3d9424230967975fa97d4f13c : SUCCESS in 6m 38s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b957cbe88150413c9e5c8eb01cc654c3 : SUCCESS in 26m 29s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6b8cea07c4cc4ed4bc75a0547d8de075 : SUCCESS in 19m 20s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/992bee54d78d4175a48982f43a3ebec6 : SUCCESS in 27m 26s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/99daab88c44f4efc816965d2e0bccd87 : SUCCESS in 19m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3de0d135179349208b89ee84f61ab287 : SUCCESS in 16m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/54df9fa6e07541d6ac380af55153db5d : SUCCESS in 14m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/15d22cfd7c8747c68583dbf09cd8f35d : SUCCESS in 16m 45s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/16c8b6cd4284420691db4770e3e7f2e7 : SUCCESS in 11m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3c8b6900b14f4989bfab51f6784cf664 : SUCCESS in 28m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bea0f3e881e5401d9ef0a2d74a6a27ad : SUCCESS in 55m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61ceca1f0fba4733af3ea87cfe4f27c6 : SUCCESS in 28m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2eb02a683feb4cd3bd1db85db9f7080e : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1a26fb4950f44c2f8043efc30097c2c5 : SUCCESS in 19m 42s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"b5dbceadfe88f3f57985aed20b885516053932b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 22:31:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eea9052d4e1145528c50cff55904afd4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/57eb27539baf4611bf9fa257fe077516 : SUCCESS in 1h 02m 41s","accounts_in_message":[],"_revision_number":1},{"id":"8abba4fffa88db0148c7586d7d2d05527cc1d3ce","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-14 17:01:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aeec5801899d576448455cf597415fd3031439c0","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-18 13:07:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9c394fadab6e961f4d02d6af71d2f7c1a3ded288","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 13:08:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"883e264f0ef2d34cf21712313e49a5708d197545","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 14:20:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0551b7d766e4883b8b8e648defa9d14\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b08225b8b6540bc9e9eac7d99eb1b0c : SUCCESS in 10m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0ecba6b8221432eb04a632a47376b19 : SUCCESS in 6m 44s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d4be8327a9b74613b9076ffe4bd1eb43 : SUCCESS in 7m 37s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b04c9b17b8c444738d0ab11c7a03d3d4 : SUCCESS in 26m 14s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/bb610c70035b419ea10d556d50d54b24 : SUCCESS in 10m 33s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7856e4a5d1d14da1bcc5fdb1f61557e8 : SUCCESS in 32m 19s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/0c4dde32635a49df8b94b129ad323e69 : SUCCESS in 20m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/23d87d96e2fe42ea8235f5bc2ff4ea59 : SUCCESS in 14m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0b9a7221509d4b98bb5a85e9d91cbfe6 : SUCCESS in 8m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ae7f06e638fc442fb39cfdf6ee225fee : SUCCESS in 17m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/22ed30f0ffe944989279e501aca58e91 : SUCCESS in 7m 12s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/746c4e110ecb493fa4b18993e9d60e67 : SUCCESS in 10m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f865793bc6b448e396134e0575538bad : SUCCESS in 55m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cc7e0cc3632c4fcda32edaa610e40f20 : SUCCESS in 52m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e5926f9c5244c7db3c9dd078189eafb : SUCCESS in 11m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7df208c1c1643afb0035f6173950cf6 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bf0374c78e3e4c18a6768a0b1d16a4cd : SUCCESS in 19m 44s","accounts_in_message":[],"_revision_number":1},{"id":"56b7545b4c918a5666ff05bb64bcdf304afba567","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-08-18 14:57:16.000000000","message":"Patch Set 1: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2e77c7c3f2110747d44f0651b88ee19c411db5f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-18 21:46:49.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+2 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"9277c65c2d236f4a62b0e7ce3734454c9723749e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-18 21:47:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"73f0c319f720d01ed3df3d7b1ce137d14fa0fbe8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 22:43:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c5aea4584ee45789a2e0165e4a9f4ca\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f226d244ebc84712b8d50935440c1639 : SUCCESS in 54m 31s","accounts_in_message":[],"_revision_number":2},{"id":"703e65ba9a33da9d429706971b7cf997a74db390","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 22:55:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e41a0795e024681a733d44eb57ed571\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4101a13ca1fe4bd2aee2b437ab96af79 : SUCCESS in 10m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67c37a6d39ae46f388c82e9f7946a138 : SUCCESS in 10m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a6b87f1c3abf44b4bada409385d9b7ae : SUCCESS in 10m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ffcf837e6bcd47c6a4a881574dc72693 : SUCCESS in 11m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7031e4c0703f4912975f9de7fa1fe2fe : SUCCESS in 11m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/285c8d2a585a42d2aecee3f8cd443134 : SUCCESS in 6m 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1a0969ce15654a588c78ca7f1b0f3611 : SUCCESS in 10m 56s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/659f78a8fcfa4f1298efa02826027cd0 : SUCCESS in 21m 42s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5222643be7c74b28bf87d0ca1efdaf91 : SUCCESS in 25m 41s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/a3e06374827e41e7841c08b611f9ae36 : SUCCESS in 21m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2e520dece8984b55b504a9c681edaf42 : SUCCESS in 8m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c2a167ab27a64eee9a5aa52e66c2ac2e : SUCCESS in 8m 37s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5fbae9584b8f4a08bec717f23073de76 : SUCCESS in 16m 44s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/746af91a99be404a8eca1105992980f5 : SUCCESS in 10m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2d490441c86d4013966bcf2c236bc365 : SUCCESS in 27m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/58781cdb85df4be390ba8e3eccd2411c : SUCCESS in 54m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ce2a053ce0e47d3951b7540d46208b3 : SUCCESS in 31m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/156a75d276e6417b837c5bf4126a0adb : SUCCESS in 5m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dbe2fb178ccb415aaa29d32478887f84 : SUCCESS in 19m 25s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"75c9c972a299d970b311369971580e87a40d0952","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 22:56:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e619d65829ffc22a010a8c9a415a389bd91d785b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 00:06:16.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/abf26203266e481682ee92957204b669\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c723045cbc9043eb862be3a7c62719fe : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dce4aa51d6e84fe0a5e9afe62d929e83 : RETRY_LIMIT\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5ec08ea889b248c8bdf63233d3e91a3b : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6c87921287af45f3bb9466e8f8336b5d : SUCCESS in 27m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d3f4a9621bcf4655a849f479a2d964eb : SUCCESS in 21m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/570f00e097834b7a84a8c6dbd8ff38b8 : SUCCESS in 34m 59s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/d9d19477976040a1bd31eb1ff2e68ec9 : SUCCESS in 21m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7b37142b4ca14e87a89251f2d0db2fdc : RETRY_LIMIT\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2e409a79b8264c0a8a10fb94788e0a68 : RETRY_LIMIT\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/14721822f3d8409c8b15c6daa86fbb85 : RETRY_LIMIT\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d0a78027c2a749a78fe81c8a32fbb63f : SUCCESS in 3m 17s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c83faae138b9407e8aa1f5977b9685eb : SUCCESS in 8m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/52a0e4953c4a40128cfe40da0fbc5d2c : SUCCESS in 50m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/153c93f40fe949bb86a6eec3b239309e : SUCCESS in 52m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f24198e5601a45a8bbedab150c582a6d : RETRY_LIMIT\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8fda866d78e4412ba38aca24f337f20 : SUCCESS in 3m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c4989d6ef9d14b0cad4ef29916e88c92 : SUCCESS in 18m 13s","accounts_in_message":[],"_revision_number":2},{"id":"4242a7d39b87c9b0aa67dd66c69e901562f54606","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-20 04:33:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b30a402677448283be18b2fefb813304f7f5d1b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 05:21:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f863b47abfc14cd98b8e3fab24e9c9a6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f6df2ade3e31442a9061ad3e160c6a9c : SUCCESS in 45m 57s","accounts_in_message":[],"_revision_number":2},{"id":"4b8b2a65aa759722717aeaf1e416cf8cfa340c48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 05:28:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72d75124b4c44ce3aed80af4b30294e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c40bcba5cca43fca991996e36cbe335 : SUCCESS in 9m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c79b2275baaf479aa03411866ce5f2e4 : SUCCESS in 9m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e3117688877448c9bbc6f105fb583b8f : SUCCESS in 19m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8a01dd949dc461c9427ff2064d99770 : SUCCESS in 19m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ff70c2d541043f687edbfe934d9157e : SUCCESS in 10m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a93b876a3ddf4066afe7b515ac59a239 : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/77eff5a025fc45c1b2d2ead6593ba739 : SUCCESS in 19m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f4d9f940474e41e9bb50be0fbeab3866 : SUCCESS in 21m 58s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6883662922d3465fb0fb45033271f774 : SUCCESS in 24m 26s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/6f1f4e7183314c43a101baed1cc5233b : SUCCESS in 17m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a48726bd163d47fa8b9c4b691c7ab335 : SUCCESS in 14m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/73e346bdb1794dd38684e84c297b57e9 : SUCCESS in 14m 41s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c75b830cad9f43388639b1676fc4b757 : SUCCESS in 13m 25s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/41ab8b97fe9a4c93bbd36e4850c8690f : SUCCESS in 17m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5a969f5f8c004359bd724e40bb03ba60 : SUCCESS in 28m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/276ee44afe264d789b078b014371fb6d : SUCCESS in 49m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd41fb95bc0047be872cabcbe5362697 : SUCCESS in 21m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12cae23fa1894f8cad5c316ac9450c55 : SUCCESS in 3m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/87cc32448c4f4b2daebc0db579a4582b : SUCCESS in 19m 33s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"9adfd8f502509e95c36684c2609397f1c77b3b87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 05:29:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3eae259f9fb523a36757870db4a0694979b22f9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 06:27:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76c014e677fa4c6682f3931a42365c1c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf6bd0e4d9ac4d3090dba27bbde4eda1 : SUCCESS in 10m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fae6ca472fda4c10a4575e33a21bb73d : SUCCESS in 9m 43s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/1db58b8f4da646f5847a6549e70568ab : SUCCESS in 6m 48s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bb1e645f156d4e65ab44b297deb6f839 : SUCCESS in 17m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0f68f23fb0cd4cb3860bc9435f98a662 : SUCCESS in 21m 49s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/15e753c5afc348518a747e6b9f1d1b92 : SUCCESS in 32m 58s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/7ab74939f5e645ca9b77c65007745059 : SUCCESS in 17m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6ff1db1979d14160b301e881c6bd8c56 : SUCCESS in 8m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9d9293b00fa4465abdd78738d88138ba : SUCCESS in 9m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4f801c570e49425794d78bd519c89068 : SUCCESS in 15m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/383e97978d7948dab47fd32f060fc626 : SUCCESS in 4m 05s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/620a97be633447d182072806435711d6 : SUCCESS in 5m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2ce12e87d94a4a3f86f66c80954be000 : SUCCESS in 32m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0d93a367415b4a0483d008419479bc2d : SUCCESS in 53m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/299ca8c7b4a0400cb8c278184ea9575a : SUCCESS in 20m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88a8045f64a34a04ae7fc0744e714c33 : SUCCESS in 3m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/64272a46183b4e859deded710ce597dd : SUCCESS in 11m 47s","accounts_in_message":[],"_revision_number":2},{"id":"db470cef840f516f96d8bc6b5b2f025da7344c82","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 06:27:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"fb6d8c7bb3fc75894f8f2cbe0274d35f8b84bc64","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 06:28:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8d67ceb2416436f926526ae10ccb4cf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1de5a351599b46469d11a8f62d4b2705 : SUCCESS in 1m 06s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4bcc0dc4344c486db2390ab6a062a4ff : SUCCESS in 50s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3e18a9ffa8f305e6f7a69d1e5eeb7e93e11086e4","revisions":{"462222829c158ca9ca09bc17a14fe9d183c8a695":{"kind":"REWORK","_number":1,"created":"2026-07-30 19:14:22.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/29/999329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/999329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/999329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/999329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/999329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/999329/1"}}},"commit":{"parents":[{"commit":"fafa244a00afe49ed8abc6cb90fcd1bb883c4485","subject":"diskfile: Support direct quarantining of directories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fafa244a00afe49ed8abc6cb90fcd1bb883c4485"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-07-30 18:51:27.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-07-30 19:04:19.000000000","tz":-300},"subject":"tests: Cover diskfile quarantine suffix fallback","message":"tests: Cover diskfile quarantine suffix fallback\n\nCover BaseDiskFile.open() recovery when an unlistable hash directory\ncannot be quarantined directly. The test verifies that ENODATA and\nEUCLEAN both quarantine the containing suffix instead.\n\nAssisted-By: Codex:gpt-5\nChange-Id: I18a9f2be3042477fa18d827826ba5082ba099944\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/462222829c158ca9ca09bc17a14fe9d183c8a695"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/462222829c158ca9ca09bc17a14fe9d183c8a695"}]},"branch":"refs/heads/master"},"3e18a9ffa8f305e6f7a69d1e5eeb7e93e11086e4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-18 21:46:49.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/29/999329/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/999329/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/999329/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/999329/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/999329/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/999329/2"}}},"commit":{"parents":[{"commit":"caf11b1e93047eb09b97dc025a33fe4d5f7f7a8d","subject":"Merge \"relinker: Add audit step to quarantine stale hash directories\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/caf11b1e93047eb09b97dc025a33fe4d5f7f7a8d"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-07-30 18:51:27.000000000","tz":-300},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-18 21:45:45.000000000","tz":-420},"subject":"tests: Cover diskfile quarantine suffix fallback","message":"tests: Cover diskfile quarantine suffix fallback\n\nCover BaseDiskFile.open() recovery when an unlistable hash directory\ncannot be quarantined directly. The test verifies that ENODATA and\nEUCLEAN both quarantine the containing suffix instead.\n\nAssisted-By: Codex:gpt-5\nChange-Id: I18a9f2be3042477fa18d827826ba5082ba099944\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e18a9ffa8f305e6f7a69d1e5eeb7e93e11086e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e18a9ffa8f305e6f7a69d1e5eeb7e93e11086e4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
