)]}'
{"id":"openstack%2Fswift~999740","triplet_id":"openstack%2Fswift~feature%2Fmpu~Ic5f6bb57caf294b10b5424a412e070739c246d3a","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-hidden-rows","attention_set":{},"removed_from_attention_set":{"38496":{"account":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"last_update":"2026-09-23 19:24:33.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-09-23 15:42:44.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"Ic5f6bb57caf294b10b5424a412e070739c246d3a","subject":"mpu: improve ContainerBroker listing interface","status":"MERGED","created":"2026-08-04 15:59:43.000000000","updated":"2026-09-23 19:27:26.000000000","submitted":"2026-09-23 19:24:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"999740-p-mpu-hidden-rows","meta_rev_id":"a3bb628ace89829a130ddddc7b69e1f2695ad8dc","_number":999740,"virtual_id_number":999740,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-23 19:24:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":2,"date":"2026-09-23 17:28:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":1,"date":"2026-09-23 17:28:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-04 16:35:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-04 17:07:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-05 10:48:25.000000000","updated_by":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"reviewer":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"state":"REVIEWER"}],"messages":[{"id":"866650a971cd454f3fe3e6714b3fdfc868b3a2c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-04 15:59:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1fbe25ed2f50cd1bddf8bb1ae656c40606e56c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 16:35:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25b4b40b44c4490b84b5877c57d0f1ee\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/77024de023cd4aab8b01cdedbc27891f : SUCCESS in 34m 28s","accounts_in_message":[],"_revision_number":1},{"id":"761fbd3e0602ebc9de17c01b7332228382e4fb29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 17:07:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/95d31bb2bc5d4529a8b5c5399d4e3ca6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4ab27b25f064a8cb08ab4d35c97c891 : SUCCESS in 9m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7746268d991b4e22aca1f904fb86e4e0 : SUCCESS in 4m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ef30c94bb9f4b97abd5ba4a711db9e9 : SUCCESS in 18m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eeb61a4bad0e43d2a44cadab7292d51c : SUCCESS in 22m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e09bc9b83f7d4676a30b1747d1b6cb75 : SUCCESS in 27m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/26b9bee0f89b4f0c9a38e70a57b79a0b : SUCCESS in 8m 12s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5dc6e07b02914a708230618f913cf3e3 : SUCCESS in 21m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6797929a13504a89a022640d5cbec497 : SUCCESS in 23m 16s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/14f2c30edbad4737b14c900ff0b71099 : SUCCESS in 12m 57s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/e43171a35c6e479eb1736fc82b0325ae : SUCCESS in 9m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/226fcfc749414bbbafb0ec4bfbbebf41 : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1eba9c8b301d43bfafa1479effd9d939 : SUCCESS in 16m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/526543c86f0040718916942872d5e9c7 : SUCCESS in 15m 35s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/97bb579368664db9a3ad2f404b1d955d : SUCCESS in 15m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5a4b0ac046ec4310bcea95fedb245ace : SUCCESS in 8m 30s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1ca4f0ef22984d5f9bde107ddc385bcc : FAILURE in 10m 23s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e27dc0b4dc864b8aaa6eabe2c4436e6f : FAILURE in 13m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/21cc453a1dd344bca8ad1c6cdb463504 : SUCCESS in 1h 02m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6861eeec0c374e808c0348c619a07f1a : SUCCESS in 30m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed466211d1eb4684b06b7e76c3dd6708 : SUCCESS in 4m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/03e7421778e94d3aa64a819fec331b7b : SUCCESS in 20m 40s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"3f40696fbf05e6ca81219f3dac4886917ee0bbb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-04 17:32:00.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7c2615a7e0f99005bf3015e93278967c085c3675","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 18:38:48.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a41135def014c7585b5641086fdac0f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7f9c7eae07ef4d91b5dd8bab3cc76e19 : FAILURE in 56m 41s","accounts_in_message":[],"_revision_number":2},{"id":"e8f058b6ed158efb59ab36424488540c288a8cf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 18:41:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9544776d4c14dfcaa35d835e562cc34\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d5669217ac54292b4bf68802f36c573 : SUCCESS in 10m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34c0ed38adc74c72b5c893ce10e5968e : SUCCESS in 8m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4bdd15206d4a443ea938fca6e060cd4a : SUCCESS in 21m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dc1e4c1a229a47aea3be43d0c03dd97d : SUCCESS in 25m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9b65b71548104557bf774889623bea6e : SUCCESS in 21m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/caab5dbe36ea419aa120fcbc44dc623a : SUCCESS in 10m 39s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0711f8b55ebd46b4a004b66c934d2b93 : FAILURE in 24m 52s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c0eba14bb13a4503ba3118169d66795b : FAILURE in 22m 25s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6f68b6209d7b446c87044aa4aa2c43d4 : FAILURE in 27m 56s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/9ae933a4af434bddbc6a660a078a1fe2 : FAILURE in 23m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/22ec01794fbd4a6d9eb4446cf87bf655 : SUCCESS in 16m 16s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b7fd1bba55f04bf1ac308125535f3ba8 : SUCCESS in 16m 45s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/17135f61406840908209e41f43795878 : SUCCESS in 17m 31s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/07c80100d89c49c98483f81768f043af : SUCCESS in 18m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7685e25b76f7483aa9e62f087981b798 : SUCCESS in 6m 26s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/510e6632b4604339a551730dbfaf1cf1 : FAILURE in 7m 50s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e163742145454bb19b411d953c1a49fe : SUCCESS in 10m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7bc509740a49425189cac467838e65ae : SUCCESS in 1h 04m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c61bab0ec7674399b90edcd76b670c05 : FAILURE in 26m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6d546d12a734bc2aff25cb03b019b4c : SUCCESS in 7m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6761063fbdb5490caa7bbab96c42203a : SUCCESS in 26m 47s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"bf00f00509e9c408f4bbc0e44e48c82048597b78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-05 08:31:39.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"b55d745cc568a401ca3edc551dbcf2c7357dba4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 09:11:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c43fe1a96424b07a8f0afc5626e8e60\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4aa0896f66db4f96b9eada4e3086e529 : SUCCESS in 10m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/476a6078f9504f02839c97b81695396d : SUCCESS in 8m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4e5007496179487caf4e1c140d78e880 : SUCCESS in 16m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0507083f3a05431b9601000411ed2b3b : SUCCESS in 20m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0005b12b93fc4491a2af668ec8c10480 : SUCCESS in 19m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1623d47f48874a4ab58a1696aaf5c3e2 : SUCCESS in 7m 13s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0d7f67f7f2dd4b8b97c03aa866367f1e : SUCCESS in 28m 42s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ef2de85966d847269caf3e89945278a4 : SUCCESS in 24m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b4fb782642c145e1a89770a9ce3f7dcb : SUCCESS in 24m 15s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/50fd92b623544d1cb884ceceebb4372f : SUCCESS in 11m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/21fd944a40414308941d2cc567534024 : SUCCESS in 9m 44s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/962230efaa544a759d7a28893ce1bc85 : SUCCESS in 9m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/60e97491103d4f28b79a0d1c0ca14961 : SUCCESS in 9m 00s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/657c0b0ea49249368f8898361b838413 : SUCCESS in 18m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d40c988225df497387e8555af78a0e15 : SUCCESS in 4m 55s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e875e4b9602f49e0a0c2374808b35e1f : FAILURE in 8m 24s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/caf6d1c00ce74b25890bf6a2b6037088 : SUCCESS in 9m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bdd98bab3c2843b7b27c71419d060ba0 : SUCCESS in 37m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef6a703ed4f4491a98be4d1153e4ef45 : SUCCESS in 14m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddd7779e5d8b4860a79c15a832a8dd81 : SUCCESS in 6m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a1ec32259aed4381ae6ceee39339a6ea : SUCCESS in 22m 12s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"e3a3c7084215f00b28ae02088edce6ffe0e9a1e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 09:12:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef3a83fc0c8347c0b3826af2e6c5ee96\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f2322d0cefe84ccfb98cf7f92910364e : SUCCESS in 38m 30s","accounts_in_message":[],"_revision_number":3},{"id":"dbe3524f2933f3ced6b9c68be308925d3bda6640","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-08-05 10:48:25.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c1971574def04df502811214b1405fcbd7a5a775","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-05 17:32:32.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"4cdcd381ee828a6f7b3168955c8334301e72bf8b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 18:05:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e9ce80a3c194ee29ade236c6bc479ce\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aa509f0004e44fa99136fbefe4497c47 : SUCCESS in 31m 09s","accounts_in_message":[],"_revision_number":4},{"id":"7abf75eddebb285141b5397052565bfa975b4a28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 18:46:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0db39cf7f854db8b3ced48b48efd1f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09d1ec3bc4be4c35b25735770f8f48c2 : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68a2a505199445939331f53c1036bc05 : SUCCESS in 9m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0bcdca96adf6440eb4a1ccb4b9f02fe4 : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/64312954f1a34be3bf05a27ecc7c2ac2 : SUCCESS in 20m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0506698a82714469b671d8daa26adf5f : SUCCESS in 20m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f9e81f0829074f0d80655680882122cb : SUCCESS in 6m 33s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/18d8815c237c4699826711a52eced99a : SUCCESS in 27m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/eff9bf6341bf4778a2dcfa19576db3e2 : SUCCESS in 10m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/02f00e99006b4f6182451c199f931cf1 : SUCCESS in 20m 20s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/461ede96bd8a46fe84498b9c8d460bfc : SUCCESS in 25m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d2e4dd54c5e5477db4daf1ffe9bb2c28 : SUCCESS in 16m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d220cf3a6bb546abb2a340c0ddcfceba : SUCCESS in 8m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/48f47002ee9c4a28b9172922e32e9752 : SUCCESS in 15m 39s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/e7dc50a2999745829661f3ec1517138f : SUCCESS in 16m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/777441404f8f441e8c75069c82873708 : SUCCESS in 5m 52s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ea5b343586414312afd9e20769dc3875 : FAILURE in 7m 05s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e0a921dad0e742b88e6a68f1bc5c3243 : SUCCESS in 8m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1917f2cf014c450a82c02bca8971872e : SUCCESS in 1h 08m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d49ce293699e4b9eaf41e3b4bad10f2a : SUCCESS in 27m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19171b8027984f06abae148a93626548 : SUCCESS in 2m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bf69ede28e7344fea542ee6cb11ed3fc : SUCCESS in 9m 00s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4},{"id":"a12c02337a74bce7b1d694183a96dbbb684d3e5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-23 15:42:24.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"a8e79c92571543bde30707a68c80badbd5e7043c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-23 15:42:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b6f6305fc0e84f4712a96b48f9dd869b3b676d59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 16:24:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28f56ca5b4754a4da24bda05cc37f468\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56509b1d5efa4211975ee41c9fd51569 : SUCCESS in 12m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44354bf3dac140c1a67f92365361e05c : SUCCESS in 8m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b1a57a54fba1409fb059a34785252d73 : SUCCESS in 19m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eda067ef54e04733a7030aae91ae7c0f : SUCCESS in 20m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ad6c07c1ace4a5683cbe3eb75f63d0a : SUCCESS in 12m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6fec17e9f9824a6980780fe63c87e29e : SUCCESS in 6m 49s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/63f9b92104f24b89817343de4b5c46c0 : SUCCESS in 11m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/12dea58b1cc2407c8c38103eaa271646 : SUCCESS in 11m 12s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2eb8b86dbb354c07ac0f6314ff3f340c : SUCCESS in 21m 44s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/592216b90ac04bcd9e39b7af658a67db : SUCCESS in 20m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f8e0cebbc6aa4a4a8cc0473622cd0e32 : SUCCESS in 16m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/54bb43cc7a2a41c19f45e7baafe1fa86 : SUCCESS in 16m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/048f72bc7fcf44bd9d39bf1b12155b5d : SUCCESS in 9m 43s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/083af16f525946f58e369c61ee87ce50 : SUCCESS in 19m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b9e66ba4b3ab4535964a10b55c123ef5 : SUCCESS in 3m 59s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e72caf5854b14a74b5151e92c4faadf1 : SUCCESS in 9m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/540572ff5b8e49599b3d3348a4729c03 : SUCCESS in 40m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee01b630c6294c658ea9193442752297 : SUCCESS in 31m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57973ebe1e0f48f4a0a9b25ac05cb58c : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b5587e7c7117462e9e82ac7602201661 : SUCCESS in 28m 21s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"45d14290e400581a33afc01ce637b9f11031f924","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 17:00:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/518fc4a31dd7463e9b0f5cf529c07b97\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aa45ef7f99b34cc1b8627f3eb03ff8c6 : SUCCESS in 1h 15m 47s","accounts_in_message":[],"_revision_number":5},{"id":"e40300f8bbac469bc4fce42c824968687326ceb0","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-23 17:28:33.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"21b075ffa0d14b6a17a444845122ce348b72d3fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 17:32:08.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8acb8b2a24695e51873f2a40119e880eea7e48a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 19:24:33.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/772ec9246c4144cfa9c537ba01ddc9f9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56789365863440868dae4c204df7a6db : SUCCESS in 11m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ec06dea020446f18fb6c44c216d2cc2 : SUCCESS in 10m 25s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/23c244e5cc7e4ac692c5f5b31d589b7e : SUCCESS in 7m 08s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1b4ac525599c46148b3a8d94273c512a : SUCCESS in 33m 21s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/161b6b9841564fffaf9f7630ec7ecf1e : SUCCESS in 22m 09s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/626286cead2246c1a2034b593dae460e : SUCCESS in 13m 25s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/4007355cbc2c4249b0c3794bb7ef8486 : SUCCESS in 20m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0cbc64e0b88f48449f06fb9b428bb00a : SUCCESS in 15m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/46032aea7f3947188ace079284e3ead2 : SUCCESS in 14m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/06ac1cc3b7ef461cae03e7cb93eb8488 : SUCCESS in 10m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/840b31350eff4858b5a852c242b88ea7 : SUCCESS in 6m 40s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/98dab4d4eb254d96a86e54667d8a8bc5 : SUCCESS in 12m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/21cfd3ed8f9d4625ac80ade3603338df : SUCCESS in 1h 01m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b39c6d0ad4a749d0859864f4184ad18e : SUCCESS in 29m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20186840efcb40c18711c9a3859a4c5b : SUCCESS in 4m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a26193612cd84d39959d71b94af43cd4 : SUCCESS in 19m 12s","accounts_in_message":[],"_revision_number":5},{"id":"ab29e7b10d4a787d41e56401843c286d51ec0375","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 19:24:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"a3bb628ace89829a130ddddc7b69e1f2695ad8dc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 19:27:26.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c11bb45690b4f8aa6cfab067e9748ae\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed88b7495d444b44a1de310ddf56f860 : SUCCESS in 1m 01s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/8547c1b08bce4c8691abe1da12d2f437 : SUCCESS in 1m 46s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d21c8371fb057f45776be2b1a47e293fb1ff54b4","revisions":{"25eedf7d22e119c8e8be19a7739abd14dadc50b1":{"kind":"REWORK","_number":1,"created":"2026-08-04 15:59:43.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/999740/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/999740/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/999740/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/999740/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/999740/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/999740/1"}}},"commit":{"parents":[{"commit":"66962745904c891be5935e02d2d6a6d5f08499d4","subject":"mpu: remove MpuAuditorContext class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66962745904c891be5935e02d2d6a6d5f08499d4"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-03 17:14:31.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-04 15:58:55.000000000","tz":60},"subject":"mpu: improve ContainerBroker listing interface","message":"mpu: improve ContainerBroker listing interface\n\nAdd list_objects which by default returns a list of dicts and provides\ndefault values for more arguments. Decode created_at into the data,\nctype and meta timestamps in the returned dicts. This better\nencapsulates the created_at column overloading in ContainerBroker.\n\nFix bug in list_objects_iter: transform_func was not being called for\ndir_name items.\n\nChange-Id: Ic5f6bb57caf294b10b5424a412e070739c246d3a\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/25eedf7d22e119c8e8be19a7739abd14dadc50b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/25eedf7d22e119c8e8be19a7739abd14dadc50b1"}]},"branch":"refs/heads/feature/mpu"},"9afb8029ae518892a9dbdd5c07e0aaf61edc9db6":{"kind":"REWORK","_number":2,"created":"2026-08-04 17:32:00.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/999740/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/999740/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/999740/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/999740/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/999740/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/999740/2"}}},"commit":{"parents":[{"commit":"9efb66953be7c3668798e6b7995aa27b6e61e25a","subject":"mpu: auditor: fix user account","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9efb66953be7c3668798e6b7995aa27b6e61e25a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-03 17:14:31.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-04 17:31:40.000000000","tz":60},"subject":"mpu: improve ContainerBroker listing interface","message":"mpu: improve ContainerBroker listing interface\n\nAdd list_objects which by default returns a list of dicts and provides\ndefault values for more arguments. Decode created_at into the data,\nctype and meta timestamps in the returned dicts. This better\nencapsulates the created_at column overloading in ContainerBroker.\n\nFix bug in list_objects_iter: transform_func was not being called for\ndir_name items.\n\nChange-Id: Ic5f6bb57caf294b10b5424a412e070739c246d3a\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9afb8029ae518892a9dbdd5c07e0aaf61edc9db6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9afb8029ae518892a9dbdd5c07e0aaf61edc9db6"}]},"branch":"refs/heads/feature/mpu"},"0714581f26b84b8b2631da7341d5b017d946cd1a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-08-05 08:31:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/999740/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/999740/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/999740/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/999740/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/999740/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/999740/3"}}},"commit":{"parents":[{"commit":"2cf8c8e1b6cfca3dd35807edfca81a9a08d72b8d","subject":"mpu: auditor: fix user account","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2cf8c8e1b6cfca3dd35807edfca81a9a08d72b8d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-03 17:14:31.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-05 08:28:34.000000000","tz":60},"subject":"mpu: improve ContainerBroker listing interface","message":"mpu: improve ContainerBroker listing interface\n\nAdd list_objects which by default returns a list of dicts and provides\ndefault values for more arguments. Decode created_at into the data,\nctype and meta timestamps in the returned dicts. This better\nencapsulates the created_at column overloading in ContainerBroker.\n\nFix bug in list_objects_iter: transform_func was not being called for\ndir_name items.\n\nChange-Id: Ic5f6bb57caf294b10b5424a412e070739c246d3a\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0714581f26b84b8b2631da7341d5b017d946cd1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0714581f26b84b8b2631da7341d5b017d946cd1a"}]},"branch":"refs/heads/feature/mpu"},"6a408cc220fbf2d01a75b2a07f3dd6e57b99ecd2":{"kind":"REWORK","_number":4,"created":"2026-08-05 17:32:32.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/999740/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/999740/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/999740/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/999740/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/999740/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/999740/4"}}},"commit":{"parents":[{"commit":"2fd30afbdb773e2269313b8a5c72cab88763e82e","subject":"mpu: clean up more references to history objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2fd30afbdb773e2269313b8a5c72cab88763e82e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-03 17:14:31.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-05 17:28:08.000000000","tz":60},"subject":"mpu: improve ContainerBroker listing interface","message":"mpu: improve ContainerBroker listing interface\n\nAdd list_objects which by default returns a list of dicts and provides\ndefault values for more arguments. Decode created_at into the data,\nctype and meta timestamps in the returned dicts. This better\nencapsulates the created_at column overloading in ContainerBroker.\n\nFix bug in list_objects_iter: transform_func was not being called for\ndir_name items.\n\nChange-Id: Ic5f6bb57caf294b10b5424a412e070739c246d3a\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a408cc220fbf2d01a75b2a07f3dd6e57b99ecd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a408cc220fbf2d01a75b2a07f3dd6e57b99ecd2"}]},"branch":"refs/heads/feature/mpu"},"d21c8371fb057f45776be2b1a47e293fb1ff54b4":{"kind":"REWORK","_number":5,"created":"2026-09-23 15:42:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/999740/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/999740/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/999740/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/999740/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/999740/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/999740/5"}}},"commit":{"parents":[{"commit":"c6c7138a0fbba2bba24f04d8002dacbbbb417a48","subject":"Merge remotes/origin/master into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6c7138a0fbba2bba24f04d8002dacbbbb417a48"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-03 17:14:31.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-23 15:35:29.000000000","tz":60},"subject":"mpu: improve ContainerBroker listing interface","message":"mpu: improve ContainerBroker listing interface\n\nAdd list_objects which by default returns a list of dicts and provides\ndefault values for more arguments. Decode created_at into the data,\nctype and meta timestamps in the returned dicts. This better\nencapsulates the created_at column overloading in ContainerBroker.\n\nFix bug in list_objects_iter: transform_func was not being called for\ndir_name items.\n\nChange-Id: Ic5f6bb57caf294b10b5424a412e070739c246d3a\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d21c8371fb057f45776be2b1a47e293fb1ff54b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d21c8371fb057f45776be2b1a47e293fb1ff54b4"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
