)]}'
{"id":"openstack%2Fswift~999792","triplet_id":"openstack%2Fswift~feature%2Fmpu~I21e515b4b33b8bf57ad1a0603b1da581610c101d","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-systags","attention_set":{},"removed_from_attention_set":{"38496":{"account":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"last_update":"2026-08-05 14:43:48.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-08-05 12:14:43.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I21e515b4b33b8bf57ad1a0603b1da581610c101d","subject":"mpu: Rename MpuAuditor to MpuAuditorDispatcher","status":"MERGED","created":"2026-08-05 08:31:39.000000000","updated":"2026-08-05 14:45:40.000000000","submitted":"2026-08-05 14:43:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"999792-p-mpu-systags","meta_rev_id":"6f1cb7af6ee40eaee8b60ddf526fa0810f3e3e07","_number":999792,"virtual_id_number":999792,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-05 14:43:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"recommended":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"all":[{"value":1,"date":"2026-08-05 10:42:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":2,"date":"2026-08-05 10:10:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":1,"date":"2026-08-05 10:10:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-05 09:11:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-05 10:42:35.000000000","updated_by":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"reviewer":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"state":"REVIEWER"}],"messages":[{"id":"763e845faf5c7417d1a0681da14dcba23b3d144b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-05 08:31:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c447f7410df9ed1d89b23624e33519616905ea69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 09:11:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/720196b981614feda07a74f3ef91e5dd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/358e52ca7d424d8986cb81edb0999aa4 : SUCCESS in 11m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb9e54d61fdf4274bcc794ab74dabcb0 : SUCCESS in 9m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/08de4303100a41a5ba8646231f1b82b0 : SUCCESS in 23m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0026710906674cbc83d501dd737ecccd : SUCCESS in 21m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/47f24c0460a04ccebb1ca6626dc3f7a6 : SUCCESS in 23m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8688c7b8a5d0437fa5d679effb9b398f : SUCCESS in 6m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/38d247996c7044a984b0c9a5f1037176 : SUCCESS in 26m 10s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/39fc5bff798e4105a4516a46beadf78b : SUCCESS in 19m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4ed7a4af487946cdace95fdb06f30e77 : SUCCESS in 23m 44s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/993f2ac8044a40cebb6fb5980b84d238 : SUCCESS in 10m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5bba61370e57455b98b49de832adc039 : SUCCESS in 8m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2c1623cf87f841769b0cb7baf5db1807 : SUCCESS in 15m 13s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ed3eddfe21ac406b9104b327be72f397 : SUCCESS in 8m 56s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9d61325f6404bb2a320077eefb68a1d : SUCCESS in 19m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1eacc28e7da5410289bedb73120f2b9d : SUCCESS in 6m 29s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6764d8ec5d3848e595a6242172301034 : FAILURE in 10m 20s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e7505c16f32c49ca99c8e03b9f768dd6 : SUCCESS in 10m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2415cea702634b89aab2e991b8a3c96e : SUCCESS in 37m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3053dc68352043b0aafbd8e273149bda : SUCCESS in 11m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c14f3b545c9943dc9dabc2c6c0687f96 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a08e3fe9870d4ca3a4bdfb8eb32a1654 : SUCCESS in 20m 19s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"5941eb5ceffb9be988d645cc6b8e718ebd884fac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 09:14:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/866c9b39572548029ec654ab36ec983f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/84d7f5017f1c4a5fae772121c3e01757 : SUCCESS in 40m 59s","accounts_in_message":[],"_revision_number":1},{"id":"2bddb994a393a598c82ba1d85a326d39c4a653b2","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-05 10:10:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6fe2dac9c7b6e02765b4aeb950283b2c4c70db9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 10:11:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"57f1aa3a5482b13161d3dfe7b975f75d1d93d67d","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-08-05 10:42:35.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"866c91a3eeeaf1352aad9011e04105a3d608441b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-05 12:14:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"898e6a72a5032f1e83a653c778b04c9f32082056","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 14:43:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66000714bb8d4df4a1e893f5f65172ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5675f98163dd46e3b7c614d458b4548c : SUCCESS in 8m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8c7b1d66f974c7cb416e6d2dc81d062 : SUCCESS in 8m 04s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0ab6058823414748b6de785efdd79f01 : SUCCESS in 7m 13s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d0d5c82d6d0743a3a86771c9f1011d62 : SUCCESS in 28m 59s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/531261b78858485ebe8ee8d73d581879 : SUCCESS in 27m 08s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/afe09bde0e5f48579df4b15754823560 : SUCCESS in 34m 25s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/21daddf852cf4c10b18630aeda60013c : SUCCESS in 25m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9b8a9fb0232f4767a0abf7b58f7f113a : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/75767b90b0d7408f976bea8ae27a9a77 : SUCCESS in 16m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/123911f0998b487c83554b52c2759432 : SUCCESS in 18m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ef9a06aa4cbf49b7abcf2d39e542cedc : SUCCESS in 6m 20s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ee57438233d1409bb43f328421f9a911 : SUCCESS in 8m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb6e7f61b56547ba8829510fed2b8cfd : SUCCESS in 1h 12m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e35364cf4535496db57074b695059050 : SUCCESS in 32m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b32dbb18a0204ac0aa697dfe35138264 : SUCCESS in 3m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/96d3e3c6d3ac410daff648e3422981fe : SUCCESS in 19m 29s","accounts_in_message":[],"_revision_number":1},{"id":"9f053b2d9a08697ca4ba18ddcf96204816f174c1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 14:43:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6f1cb7af6ee40eaee8b60ddf526fa0810f3e3e07","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 14:45:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e002f0d3cf2f41bfabd4db77c99ca478\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afd86291529d4e6da8d3b1c2844375ae : SUCCESS in 1m 06s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/9ccca6ff0fbc487ea44a112f7f05e156 : SUCCESS in 50s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a9bef880d1c059b4afd87b2b67ad0684be46668b","revisions":{"a9bef880d1c059b4afd87b2b67ad0684be46668b":{"kind":"REWORK","_number":1,"created":"2026-08-05 08:31:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/92/999792/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/999792/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/999792/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/999792/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/999792/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/999792/1"}}},"commit":{"parents":[{"commit":"66962745904c891be5935e02d2d6a6d5f08499d4","subject":"mpu: remove MpuAuditorContext class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66962745904c891be5935e02d2d6a6d5f08499d4"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-05 08:22:14.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-05 08:22:14.000000000","tz":60},"subject":"mpu: Rename MpuAuditor to MpuAuditorDispatcher","message":"mpu: Rename MpuAuditor to MpuAuditorDispatcher\n\n...and clarify other naming.\n\nChange-Id: I21e515b4b33b8bf57ad1a0603b1da581610c101d\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a9bef880d1c059b4afd87b2b67ad0684be46668b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a9bef880d1c059b4afd87b2b67ad0684be46668b"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
