)]}'
{"id":"openstack%2Ftacker-horizon~679933","triplet_id":"openstack%2Ftacker-horizon~master~Ia148f2959f30e098c7693396ba1d6d633e4ef198","project":"openstack/tacker-horizon","branch":"master","topic":"build-pdf-docs","hashtags":[],"change_id":"Ia148f2959f30e098c7693396ba1d6d633e4ef198","subject":"Generate PDF documentation","status":"MERGED","created":"2019-09-04 08:33:40.000000000","updated":"2019-09-13 08:49:27.000000000","submitted":"2019-09-13 08:48:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679933-1568364492015-919711b3","meta_rev_id":"9df1bed6e9e1aded78d2ea5bca95830c0f3227dc","_number":679933,"virtual_id_number":679933,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},{"value":2,"date":"2019-09-13 08:48:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2019-09-12 08:46:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":2,"date":"2019-09-13 08:37:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":1,"date":"2019-09-13 08:37:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-04 16:35:28.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2019-09-06 03:35:59.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-09-12 08:46:22.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-09-13 08:37:32.000000000","updated_by":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"reviewer":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"state":"REVIEWER"},{"updated":"2019-09-13 08:48:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ee83e7ed091e885f21c6c2da1c21acad149bdaca","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-09-04 08:33:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cff1e3cb4fbba4f02c10384df9d7f4ade38b3cfe","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-09-04 08:39:43.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"08b61d90c07b5975e7ed59f2d7d4051b4aef252e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 09:15:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c758c129642848b58dd168332284247f : SUCCESS in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a85754f2a714c1a915e5855ac079061 : SUCCESS in 3m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/172cde17e58f4895ba3c9b57598ba595 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/565b44243db64a3db3b306cabe300ca9 : SUCCESS in 5m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a3f922432fa749d9891716b9a77e9f04 : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a25076c7e6d4b6994bb0800adeb62e5 : FAILURE in 11m 13s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/c40d87e0de2d49739f257a0a96c2652d : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"427167f6dc07f2f9bbc7f0c074847996bd3e6e6d","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-09-04 10:57:45.000000000","message":"Patch Set 1: Code-Review-1\n\nIt seems, we needs to set latex_use_xindy as False to fix gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2996b595f33b63015aa3e71ee8001d9497ee9f3b","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-09-04 11:43:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4d878d8734fb784f405cf28c25f5e44b2c7d7722","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-09-04 11:43:24.000000000","message":"Patch Set 1:\n\n(1 comment)\n\n\u003e It seems, we needs to set latex_use_xindy as False to fix gate\n \u003e jobs.\n\nyeah, agree\nDone.\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"37ea7efefca1dc972cbb93d84a9e78134db69b4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 12:03:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be146dd11c804894b9e29dbfe3b30a35 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98491c21705344f9bb3e4bcf6421ec0e : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fadc34c68b0843c0a3020b12ed2f0aef : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e2a5e32b7064aaf806f875955c59e72 : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a613258b6c51415bab8e8b0920ead900 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26471f7ba1e34e93ada23ce8e35e9f5a : SUCCESS in 4m 49s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/3b86df06536b4b3c87572803e0220c66 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":2},{"id":"354d7eb143dc0c5d2f888b5b43db3b29dfc22f29","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-09-04 16:35:28.000000000","message":"Patch Set 2: Code-Review+1\n\nLooks good","accounts_in_message":[],"_revision_number":2},{"id":"fa74f92e732322416379350c648b07d815f60d47","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-09-05 17:25:55.000000000","message":"Topic changed from 679933 to build-pdf-docs","accounts_in_message":[],"_revision_number":2},{"id":"b8acc4c0b60e260438a7b38f044d175868e60740","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-06 03:35:59.000000000","message":"Patch Set 2: Code-Review-1\n\nAlthough the PDF doc can be built, some contents of the PDF doc looks tricky.\n\n- \"Indices and Tables\" section is meaningless in PDF file\n- Looking at TOC, the chapter \"Contents\" looks unnecessary (as it is the only chapter)\n- The volume of the document is small so \"howto\" in latex_documents make more sense to me rather than \"howto\". You can see 9 page PDF file with 4 blank pages.","accounts_in_message":[],"_revision_number":2},{"id":"3df5facc4e43e360e31ae3db3fd265255713d373","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-09-10 06:02:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d494556b8df84d900656c49d9803a162c5643c37","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-09-10 06:04:42.000000000","message":"Patch Set 3:\n\n\u003e Although the PDF doc can be built, some contents of the PDF doc\n \u003e looks tricky.\n \u003e \n \u003e - \"Indices and Tables\" section is meaningless in PDF file\nCould you suggest how to remove this?\n \u003e - Looking at TOC, the chapter \"Contents\" looks unnecessary (as it\n \u003e is the only chapter)\nCould you suggest how to improve this?\n \u003e - The volume of the document is small so \"howto\" in latex_documents\nDone in updated patchset\n \u003e make more sense to me rather than \"howto\". You can see 9 page PDF\n \u003e file with 4 blank pages.","accounts_in_message":[],"_revision_number":3},{"id":"89fc20be49753e1cb97285dc32ab4959c1413e18","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-09-10 06:04:48.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"f7d7a648d197545c8340e1a565b3293ece03fbd9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 07:20:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f995ab4c8d04189bcd3479bbf664a80 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72b067e09d5f4a2eba6f7fa2620b0529 : SUCCESS in 3m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a6be6f64410d49a996eb87a46ca49e67 : SUCCESS in 2m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04076771327641858d97170079355693 : SUCCESS in 3m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/beaaab5c0a034f099f73628def4888e6 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12c72237c8d64a8bbd3e4968b9beec55 : SUCCESS in 6m 58s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/a3cfbcdc4781416393756ae94364e6d0 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":3},{"id":"90e2e0cadf99207afb22c0d9e2c1ff2fce9d3c34","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-09-10 07:30:14.000000000","message":"Patch Set 3: -Workflow\n\nlooks fine to me.","accounts_in_message":[],"_revision_number":3},{"id":"a8f1930d0867b8fce76b1d53f5a6403010cbb2c5","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-09-10 15:39:49.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"0e29645e754e1376ea23c533bc258267c7900d7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 15:59:26.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cea1a38000b14eef9c021c6a78ec334b : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45298c23415c4635abff2562f83c3252 : FAILURE in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/651cd04e6cf842eebeff5c075942c0d9 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d37217d16c344f8a694e37ced1114c7 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8d522b65757f4dc686f28a7f5ea0178b : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6457b3383cf4cecb23db724a4e13f93 : SUCCESS in 4m 51s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/3a21dbd1eaf747f49d637a89e424984f : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":4},{"id":"d7077138d3176e6c783d7577c147449a7e376663","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-09-10 16:02:33.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"88d0c444b1d021b7b1b2ba74da1424cdcab1c45b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 16:15:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/438b48d8c6bc46d4b4b668ff219480b5 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/408cbb60841e4fe88eeac61c1555798c : SUCCESS in 2m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/763f47fc270d40b9bcb41e620f4b04ce : SUCCESS in 3m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/508096fa49bd477198f6c71ff1249742 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9031da85590a467983ad6346dd5bba01 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2bed773e963429bbf5901aa75afe8e9 : SUCCESS in 4m 42s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/4d7405d8d8d44030996b6d33af026c10 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":5},{"id":"23c9b5988f0d70196023d0a80f7c4d0aa01e374f","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-09-11 07:15:05.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"297f02371419a54572cbfa254e83f556a142c181","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 07:34:08.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19b78f9192a644a099db77c9e2a621f8 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd3c0609ba914c31a6cd376a7c89defa : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b96018ceacba4db487963c010f59c086 : SUCCESS in 3m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc45d045d56543399a4d217fa264f8c2 : SUCCESS in 2m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3075f90789784981a5bccf971ab9f053 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5d2b2fe0e874b58ae5738bf28dc91bb : SUCCESS in 4m 41s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/6de2775297bd4c77b7b7d4a8c745efbe : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":6},{"id":"bac0659e763600bca69d52afe897655d01814b16","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-09-12 08:46:22.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"6ea5c7ef4c8b963094a09e0bfb92f2b3c3e0e38e","author":{"_account_id":26588,"name":"Hiroyuki Jo","email":"joxyuki@gmail.com","username":"joxyuki"},"date":"2019-09-13 08:37:32.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"ef674e62cabdaf87cc0ae540b8868d7538d821ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 08:37:44.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"aa5ef1f1dc8a92889b87c5b3de6fa29703cb5a50","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 08:48:11.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b8e1e8cc86a4b00977db5d21ae845b9 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03fafa77fcef4bf7b82f45eddfa685d3 : SUCCESS in 2m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/be0ab9bba6cc40edb383521536185b9f : SUCCESS in 3m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ddfb9d191af4ceaad4e622d5d36777f : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e256473a4844ffeaa80341a1ebe7341 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd2442b1f4cd4686b2f0bf7632a5a793 : SUCCESS in 7m 19s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/57251cb8948a45de92bbd62b5af5e0f5 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":6},{"id":"ab650ec34eb4023663522f452f02399b6a18ecb2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 08:48:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"9df1bed6e9e1aded78d2ea5bca95830c0f3227dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 08:49:27.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a0565a8b7084e01962c6ff37ec71dae : SUCCESS in 57s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c3806b8b688732472599502c641157f2f207b44b","revisions":{"a29818519667e086fbec196f4087cdcaaa924a3f":{"kind":"REWORK","_number":1,"created":"2019-09-04 08:33:40.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/33/679933/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-horizon","ref":"refs/changes/33/679933/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/33/679933/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/33/679933/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/33/679933/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-horizon refs/changes/33/679933/1"}}},"commit":{"parents":[{"commit":"599152dbf520791d75b2549491b4494911d2b934","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/599152dbf520791d75b2549491b4494911d2b934"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2019-09-04 08:32:33.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2019-09-04 08:33:38.000000000","tz":0},"subject":"Generate PDF documentation","message":"Generate PDF documentation\n\nChange-Id: Ia148f2959f30e098c7693396ba1d6d633e4ef198\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/a29818519667e086fbec196f4087cdcaaa924a3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/a29818519667e086fbec196f4087cdcaaa924a3f"}]},"branch":"refs/heads/master"},"37b5f910cc9201b9326fc38768a9248f9f60def2":{"kind":"REWORK","_number":2,"created":"2019-09-04 11:43:20.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/33/679933/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-horizon","ref":"refs/changes/33/679933/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/33/679933/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/33/679933/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/33/679933/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-horizon refs/changes/33/679933/2"}}},"commit":{"parents":[{"commit":"599152dbf520791d75b2549491b4494911d2b934","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/599152dbf520791d75b2549491b4494911d2b934"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2019-09-04 08:32:33.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2019-09-04 11:43:08.000000000","tz":0},"subject":"Generate PDF documentation","message":"Generate PDF documentation\n\nChange-Id: Ia148f2959f30e098c7693396ba1d6d633e4ef198\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/37b5f910cc9201b9326fc38768a9248f9f60def2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/37b5f910cc9201b9326fc38768a9248f9f60def2"}]},"branch":"refs/heads/master"},"f325a6c14b44a8cf9eeb7ddd9f66351206669694":{"kind":"REWORK","_number":3,"created":"2019-09-10 06:02:35.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/33/679933/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-horizon","ref":"refs/changes/33/679933/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/33/679933/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/33/679933/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/33/679933/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-horizon refs/changes/33/679933/3"}}},"commit":{"parents":[{"commit":"599152dbf520791d75b2549491b4494911d2b934","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/599152dbf520791d75b2549491b4494911d2b934"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2019-09-04 08:32:33.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2019-09-10 06:02:21.000000000","tz":0},"subject":"Generate PDF documentation","message":"Generate PDF documentation\n\nChange-Id: Ia148f2959f30e098c7693396ba1d6d633e4ef198\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/f325a6c14b44a8cf9eeb7ddd9f66351206669694"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/f325a6c14b44a8cf9eeb7ddd9f66351206669694"}]},"branch":"refs/heads/master"},"7c0330e15c98bfa5720b5ac4f323ffa985658c96":{"kind":"REWORK","_number":4,"created":"2019-09-10 15:39:49.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/33/679933/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-horizon","ref":"refs/changes/33/679933/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/33/679933/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/33/679933/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/33/679933/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-horizon refs/changes/33/679933/4"}}},"commit":{"parents":[{"commit":"599152dbf520791d75b2549491b4494911d2b934","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/599152dbf520791d75b2549491b4494911d2b934"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2019-09-04 08:32:33.000000000","tz":0},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2019-09-10 15:39:42.000000000","tz":0},"subject":"Generate PDF documentation","message":"Generate PDF documentation\n\nThis commit adds a new tox target to build PDF documentation.\nIt\u0027s a community goal[1] to have PDF docs available.\n\nUpdate conf.py for latex building.\n\n[1]\nhttps://governance.openstack.org/tc/goals/selected/train/pdf-doc-generation.html\n\nChange-Id: Ia148f2959f30e098c7693396ba1d6d633e4ef198\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/7c0330e15c98bfa5720b5ac4f323ffa985658c96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/7c0330e15c98bfa5720b5ac4f323ffa985658c96"}]},"branch":"refs/heads/master"},"a6dc56db7bc672d5966dc8502191688e5fbe1dc1":{"kind":"REWORK","_number":5,"created":"2019-09-10 16:02:33.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/33/679933/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-horizon","ref":"refs/changes/33/679933/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/33/679933/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/33/679933/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/33/679933/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-horizon refs/changes/33/679933/5"}}},"commit":{"parents":[{"commit":"599152dbf520791d75b2549491b4494911d2b934","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/599152dbf520791d75b2549491b4494911d2b934"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2019-09-04 08:32:33.000000000","tz":0},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2019-09-10 16:02:25.000000000","tz":0},"subject":"Generate PDF documentation","message":"Generate PDF documentation\n\nThis commit adds a new tox target to build PDF documentation.\nIt\u0027s a community goal[1] to have PDF docs available.\n\nUpdate conf.py for latex building.\n\n[1]\nhttps://governance.openstack.org/tc/goals/selected/train/pdf-doc-generation.html\n\nChange-Id: Ia148f2959f30e098c7693396ba1d6d633e4ef198\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/a6dc56db7bc672d5966dc8502191688e5fbe1dc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/a6dc56db7bc672d5966dc8502191688e5fbe1dc1"}]},"branch":"refs/heads/master"},"c3806b8b688732472599502c641157f2f207b44b":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2019-09-11 07:15:05.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/33/679933/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-horizon","ref":"refs/changes/33/679933/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/33/679933/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/33/679933/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/33/679933/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-horizon refs/changes/33/679933/6"}}},"commit":{"parents":[{"commit":"599152dbf520791d75b2549491b4494911d2b934","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/599152dbf520791d75b2549491b4494911d2b934"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2019-09-04 08:32:33.000000000","tz":0},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2019-09-11 07:14:22.000000000","tz":0},"subject":"Generate PDF documentation","message":"Generate PDF documentation\n\nThis commit adds a new tox target to build PDF documentation.\nIt\u0027s a part of community goal, see storyboard for more\ninformation.\n\nStory: 2006123\nTask: 35520\n\nChange-Id: Ia148f2959f30e098c7693396ba1d6d633e4ef198\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/c3806b8b688732472599502c641157f2f207b44b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/c3806b8b688732472599502c641157f2f207b44b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
