)]}'
{"id":"openstack%2Ftacker-horizon~869927","triplet_id":"openstack%2Ftacker-horizon~master~I514b6e6d4adf3ed3b7afd0690d5b5a6c6c4971b2","project":"openstack/tacker-horizon","branch":"master","topic":"tox4","attention_set":{},"removed_from_attention_set":{"31857":{"account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"last_update":"2023-01-13 00:32:50.000000000","reason":"\u003cGERRIT_ACCOUNT_31857\u003e replied on the change","reason_account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"}},"25701":{"account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"last_update":"2023-01-12 20:11:08.000000000","reason":"\u003cGERRIT_ACCOUNT_25701\u003e replied on the change","reason_account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}},"32102":{"account":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"last_update":"2023-01-13 00:47:02.000000000","reason":"Change was submitted"},"17255":{"account":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"last_update":"2023-01-13 00:47:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I514b6e6d4adf3ed3b7afd0690d5b5a6c6c4971b2","subject":"Fix tox4 error","status":"MERGED","created":"2023-01-12 10:31:38.000000000","updated":"2023-01-13 00:47:58.000000000","submitted":"2023-01-13 00:47:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869927-tox4","meta_rev_id":"66d8f4b6a465ac6369cd67c4c6c4147498813e44","_number":869927,"virtual_id_number":869927,"owner":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-13 00:47:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"all":[{"value":2,"date":"2023-01-13 00:32:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":2,"date":"2023-01-12 20:11:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"all":[{"value":1,"date":"2023-01-13 00:32:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"}],"reviewers":{"REVIEWER":[{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-12 12:08:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-12 12:13:37.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":17255,"name":"Toshiaki Takahashi","email":"ts-takahashi@nec.com","username":"takahashi_ts"},"state":"REVIEWER"},{"updated":"2023-01-12 12:13:37.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"state":"REVIEWER"},{"updated":"2023-01-12 12:13:37.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"}],"messages":[{"id":"869d6f1826e78910ec40b3d1beadafe33df8ee3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2023-01-12 10:31:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22e4ac93458d29764eb5674db08d01a06cc7f4ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 12:08:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40264da320ba4b3ab8351d3e2ac493a1\n\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69c01dd808cd4a3d9ecec10d77d592ae : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b893995e21574e5ba772024b29805cba : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5161a45f9444416a916da1037ae8db32 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/efa017ce335447e69e22385aa3199ead : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f6aecd8f66b42cd8770d2c636729b6c : SUCCESS in 6m 51s\n- tacker-horizon-integration-tests https://zuul.opendev.org/t/openstack/build/0cd2cbf93a7d427395931da56e1f7590 : RETRY_LIMIT in 29m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"934324c46b7b7a2687b8216da771ab9a6ad8645c","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2023-01-12 20:11:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cb92cb58e10b6eee934ccc82538f1a6173f3903e","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2023-01-13 00:32:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f0e6e2a1b8cf1a030d850101224a6867605ae6e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 00:33:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a2c033e07ee930fe2f128a052b167990ab41a265","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 00:47:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/677accd638fb4d769f6c768e217678e7\n\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f8d943f37b5479290254030c7d08b95 : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7301ae34966341379b453bbb9d744fd0 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ed586ab74a84026b46c3f1b8258cbee : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/35276050c8af44fea8e968117c419a68 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a37a5e08b27e4455a74d5a0dac53ef1f : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a3a17ea8ed23bcf368d0432cb243cda09da10df1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 00:47:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"66d8f4b6a465ac6369cd67c4c6c4147498813e44","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 00:47:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fbb6ec1b6f04448b8fc3c2204ccbabb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71af3d2e5f1f49c0b2e62c1de406eb24 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e9a7d3a520745fc677f4b7c040b670cf5a2cca5","revisions":{"1e9a7d3a520745fc677f4b7c040b670cf5a2cca5":{"kind":"REWORK","_number":1,"created":"2023-01-12 10:31:38.000000000","uploader":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"ref":"refs/changes/27/869927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tacker-horizon","ref":"refs/changes/27/869927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/27/869927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/27/869927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tacker-horizon refs/changes/27/869927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tacker-horizon refs/changes/27/869927/1"}}},"commit":{"parents":[{"commit":"5a15ddd0ad76bfa49ca47e10b6ee0182944d8be9","subject":"Update python classifier for python 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/5a15ddd0ad76bfa49ca47e10b6ee0182944d8be9"}]}],"author":{"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","date":"2023-01-12 10:27:42.000000000","tz":330},"committer":{"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","date":"2023-01-12 10:31:32.000000000","tz":330},"subject":"Fix tox4 error","message":"Fix tox4 error\n\ntox.ini started failing with Tox4 which had some\nincompatible changes.\nRemove skipsdist option as it is causes breakage\nwhen used with usedevelop.\n\nChange-Id: I514b6e6d4adf3ed3b7afd0690d5b5a6c6c4971b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/1e9a7d3a520745fc677f4b7c040b670cf5a2cca5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tacker-horizon/commit/1e9a7d3a520745fc677f4b7c040b670cf5a2cca5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"}}]}],"submit_requirements":[]}
